)]}'
{"id":"openstack%2Fkayobe~952969","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I825ab7331fbe8f28c17b87ff79a436144426a78c","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2025-06-20 15:14:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I825ab7331fbe8f28c17b87ff79a436144426a78c","subject":"Fix infinite loop in seed container deployment","status":"MERGED","created":"2025-06-20 10:24:22.000000000","updated":"2025-06-20 15:15:13.000000000","submitted":"2025-06-20 15:14:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952969","cherry_pick_of_change":952809,"cherry_pick_of_patch_set":4,"meta_rev_id":"6d9afd2f58f86a1727b96cd7396d20104d5d2a30","_number":952969,"virtual_id_number":952969,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-20 15:14:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-06-20 10:27:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-20 10:36:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-06-20 11:47:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-06-20 11:47:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-20 10:27:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-20 10:36:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-06-20 11:47:17.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-06-20 12:51:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89212b635ef331cf5bac854bbebdff0533969ab5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-06-20 10:24:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7186e6ca526f1f8647cbe6ca0e0789c00c05c605","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-20 10:27:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f2b13eb742cd4d4d3a6ed5b04def3d9ad2d63bb8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-20 10:36:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"61da2736494a8cb49c436d1f175bc18c96e05590","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-20 11:47:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e17c03dcffc8abf4dbd7da5ebce92111a9252952","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 12:51:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7614db10af354ba69895d4264f6a4d1a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65c31e5d1dae410f82affb8ce309b399 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/383d789d54cd42f591e224593b71f543 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d048325d9cf4cf7a65e503d34d27c06 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d2ce1b2652d4aa08d9534a61424af9d : SUCCESS in 8m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fc13956da434047ad77c4741e01875f : SUCCESS in 2m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/65708bb32e5f4d339015a696ae223b1c : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fd1252bb6aae4a5c9b997d5c5a202c80 : SUCCESS in 11m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e96afb56b034494cb497580965fabc79 : SUCCESS in 10m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0dda4a515d2c4db3b95d4adfe359ddb3 : SUCCESS in 59m 28s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/5e6c0a11227542b1866cd5b23b0e0a78 : SUCCESS in 1h 02m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a2163cb04014f289dc06a3e3795a738 : SUCCESS in 51m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f1e2124357e64faabebb414ddc529227 : SUCCESS in 58m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/373195b57df64ee4b9bf8d9fe0c2d2b2 : SUCCESS in 1h 10m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b8ad2a7794954220a2884ac527d94882 : SUCCESS in 17m 54s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffae0d01b09f4bb095ccf7d08babf17c : SUCCESS in 17m 39s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cbbe8763cac341228a8d6c3ad1844da6 : SUCCESS in 1h 59m 27s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f2b58d91f2d48d8a79c52f8959f3866 : SUCCESS in 2h 22m 46s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e06b5a54b35c4caba629ed549a94d0fb : SUCCESS in 2h 03m 22s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c704b58ad8b94ab39e048354c14e00db : SUCCESS in 2h 05m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dc001aad2182447086e7e6e64ecbdf01 : SUCCESS in 15m 12s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/35505de8b94344db80ce1b4e13062847 : SUCCESS in 20m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11eac576bce24dfdab6aeeafb6ab2407 : SUCCESS in 17m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/10fe52014c8f45528e5dea9ced222cdc : SUCCESS in 18m 04s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7f4a508957564164b8013133444b23ed : SUCCESS in 44m 12s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/27ac0dcf4b3448aeab682085901a0e2e : SUCCESS in 40m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bdf6b5cddad43cda6013f73338af781 : SUCCESS in 28m 15s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/df298fa7541c4f86800119d6906ca0fb : SUCCESS in 30m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e2d2590dcf5a4403931619ce6ae6280e : SUCCESS in 14m 48s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/ffafdaa40424412b83e49ffe610d82d8 : SUCCESS in 13m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd6387a28f45411b9f904e2de2d0fad7 : SUCCESS in 11m 07s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/d145305f1d364312b66876865a9a9fe3 : SUCCESS in 13m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c0cd7946a1944a3e9be0aa3ceffcf8cd : SUCCESS in 15m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a7a104468684e7397fc7854293d327a : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":1},{"id":"0c9a429639b75691ddb2c5c852f6e110d4116d61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 12:52:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"53b8a02e83cdf3ec659675da3cc1c6cb5baa53bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 15:14:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5aed1719b7542268345401005f79f6b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4988235d7a324f618b8bcf8cb5177f78 : SUCCESS in 5m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/29beea9c0a3d41be94c2a88b91aedd6c : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c24d0db6a74fb3aaf4f3dffa3e6da6 : SUCCESS in 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9aa9947fb1ff423fb62ce87a11c8317d : SUCCESS in 2m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3b797fbce747464a9a7a98c4e95e5ee8 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b16ba364171e440b9f8b056055582419 : SUCCESS in 16m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7afaf61a233248fdacf4b97d5d2cdbe3 : SUCCESS in 10m 46s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d267d759c1f4447abb0f62c929cd6fc9 : SUCCESS in 1h 11m 49s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/3561a8211c0b41c1adfa0ebe2c38c9be : SUCCESS in 1h 03m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c781e38815b47c88b912ff1ff02db99 : SUCCESS in 55m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/808aa02f3bba4102a0a6d3b69a22209a : SUCCESS in 51m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/12efb7c2252a4b4cb8915d6e6cda73c9 : SUCCESS in 1h 07m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1ac1cb1265b6498d9dcb67afe8d71252 : SUCCESS in 18m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cd2c425275549fe9c646062717bc34a : SUCCESS in 13m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ebb99ff44e4945f0bade0aa0290a9d26 : SUCCESS in 2h 17m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe7b269f5b3748458bc8fa5194436e44 : SUCCESS in 2h 03m 16s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0598c268820d494298c2c7d8a2f1890c : SUCCESS in 2h 11m 02s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3475a75b33444749a8531dc22afff6c : SUCCESS in 2h 10m 42s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8e78e1a8044d4e8188fd290f6fddc7cc : SUCCESS in 13m 22s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/3375bbeb173c4c30b8cddf52a1254c83 : SUCCESS in 15m 01s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fa8107663564c82a18671b95c1aaac5 : SUCCESS in 13m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/da5c4f68812649f29e0fdc41c2737a19 : SUCCESS in 12m 10s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/259a1a55eea4416797314e23f678251b : SUCCESS in 43m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/488223369a184d21812c0ba586330e67 : SUCCESS in 38m 45s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/49e353252f2344f49056823d8e00aa00 : SUCCESS in 35m 42s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/32e28bd3fa17494e8191446745539939 : SUCCESS in 11m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80d3ad5a8ef94f66b25e1eb3fbbd10ab : SUCCESS in 12m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8224e103a72a4391820a0cb127868247 : SUCCESS in 10m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af9b4e24ae5b4c8898ce7421833a03b5 : SUCCESS in 10m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e323dbcd795794ba5a2a62c29df55d11ea1a42e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 15:14:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6d9afd2f58f86a1727b96cd7396d20104d5d2a30","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 15:15:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cfaa42877b0430e94f1694027a8e3b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c4734849f77453dbd13b75f9dee630e : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ca391a7aca742d8baf714cfdcc2c625 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08d10d19d9d0b1d53b461a17020c5f6b427a5663","revisions":{"08d10d19d9d0b1d53b461a17020c5f6b427a5663":{"kind":"REWORK","_number":1,"created":"2025-06-20 10:24:22.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/69/952969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/952969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/952969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/952969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/952969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/952969/1"}}},"commit":{"parents":[{"commit":"46ceb39ad84412ced69f0c66cdf48156d7a13a44","subject":"Use latest CentOS Stream 9 GenericCloud image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46ceb39ad84412ced69f0c66cdf48156d7a13a44"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-18 09:09:56.000000000","tz":60},"committer":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2025-06-20 10:24:22.000000000","tz":0},"subject":"Fix infinite loop in seed container deployment","message":"Fix infinite loop in seed container deployment\n\nThe deploy.yml playbook previously called deploy-container.yml, but this\nwas changed to call itself when podman support was added [1]. It would\nthen recursively loop until failure. This change reverts that, so\ndeploy-container.yml is called instead.\n\nAdd a test to seed jobs, to catch any regression in CI.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/909686\n\nCloses-Bug: #2114845\nChange-Id: I825ab7331fbe8f28c17b87ff79a436144426a78c\n(cherry picked from commit 7bd7929ea77f9843717a424d88422837d19d0d4a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08d10d19d9d0b1d53b461a17020c5f6b427a5663"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08d10d19d9d0b1d53b461a17020c5f6b427a5663"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
