)]}'
{"id":"openstack%2Fkayobe~953074","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~If2a26b42e6fe0152922e10df924c0797259e2e53","project":"openstack/kayobe","branch":"stable/2025.1","topic":"fix-multiple-deprovision-stable/2025.1","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2025-06-23 14:53:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If2a26b42e6fe0152922e10df924c0797259e2e53","subject":"Fix deprovisioning multiple overcloud hosts","status":"MERGED","created":"2025-06-23 07:32:34.000000000","updated":"2025-06-23 14:54:12.000000000","submitted":"2025-06-23 14:53:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953074-fix-multiple-deprovision-stable/2025.1","cherry_pick_of_change":952964,"cherry_pick_of_patch_set":2,"meta_rev_id":"9e67c1e2d2d93cf2ef71aa6c6e1670952eb148b5","_number":953074,"virtual_id_number":953074,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-23 14:53:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-06-23 07:52:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-23 07:54:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-06-23 08:30:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-06-23 08:30:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-23 07:52:10.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-23 07:54:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-06-23 08:30:53.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-06-23 10:07:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"005d995fd66b844942983f16b7cc139d388de364","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-06-23 07:32:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"114e5dfcf99c15aa919f797ac06571c207d05f91","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-23 07:52:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"157ab60cba0ef1b9823453d3debca7311a0203a2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-23 07:54:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8dab3197da014436df978613e0b846b6f3297deb","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-23 08:30:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f331ed2721443facc1561ad90a96967c86c0f88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 10:07:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af4354a376154bb6add3d003f9e4d855\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c41cd004eafe43108006840de0842737 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcf98fb8229e476d8390e54cacbab17f : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e4ae8cdeb7b4cc6928bc094298a73eb : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0169500f06684bd2836a1ba26c7c1e2c : SUCCESS in 9m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/974fdc17af3646429b612b8bdfdd3fc7 : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6e4ad48804774d768b98f6532cfa58d9 : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0e2119e7a4344cc28f8441b7ac4af38c : SUCCESS in 10m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ffd89e7ff3424cea84d9ea52d8b4609c : SUCCESS in 9m 19s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d24385dd4d7d4d30abce9ae59a022ac2 : SUCCESS in 52m 44s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/ac8cd26aa8044edabed4e2d79a86b020 : SUCCESS in 1h 10m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b11db0deec34929b68e16709aae5b1e : SUCCESS in 54m 28s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ec06bc468f984acaa5645fa58d166e94 : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8b4b6af2bd2c4b779738f3f47b2b9560 : SUCCESS in 59m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9e840b1ba6874301aa3e5ffdc92137d4 : SUCCESS in 17m 19s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30c01ab3f09543cbab21e41ef6ecdd28 : SUCCESS in 13m 08s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/44db59a90ac34b1d94e794ffd57a3fcf : SUCCESS in 2h 20m 46s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed8a99cbf9d3465a926386a5d6b85c4d : SUCCESS in 2h 31m 25s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/03d3073b236b4fed98dd3a0c563fac7b : SUCCESS in 2h 28m 48s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0337025225e45cf824e119bf4cfafa4 : SUCCESS in 2h 17m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4a8b726debed401da16960b8caa92a18 : SUCCESS in 17m 24s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/e16d3d82527e49fc9edf40fae91b65a0 : SUCCESS in 22m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71056d46b85a4d22ab20bc4edb1830b9 : SUCCESS in 15m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f9a3179dd8f84caf83495b4e9bf68601 : SUCCESS in 14m 39s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4880437452874ea9a93bbfac7254412f : SUCCESS in 55m 39s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f864e4d3e32241d3a6de723f7132ba26 : SUCCESS in 33m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36231f2d191e4df2acdc6d9448abca76 : SUCCESS in 31m 20s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/d230ec45c28f42dc8c35ca5d7ba8f591 : SUCCESS in 35m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d5bd7a5fc677466c8d42a1c46ceb2a60 : SUCCESS in 17m 16s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/2b0fcc01f6304ff09e4e8e50faad2ba5 : SUCCESS in 12m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/606b50680ba549d0a131c45ace7405bb : SUCCESS in 12m 16s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/6ca2ccc1f4004d1298bbccc77e990ace : SUCCESS in 11m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a96cad8ea2484c348c040758dd6cc531 : SUCCESS in 11m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61a8b11b5487479a8455d0d9be4efe64 : SUCCESS in 10m 05s","accounts_in_message":[],"_revision_number":1},{"id":"db3079fc714a22db0c54ff2efaf769907f7583d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 10:07:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"977a30c0fd42c8d4e92941b6ad10e4c76b670586","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 14:53:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edc9523a10d04ca3af61963138a0164b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd6408144ee44fb8b175a649f44d73a8 : SUCCESS in 5m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7980bc92fde646f9bfb49a1cb0c84cc0 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df086181a62a402c9e15b274832ac372 : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f0a162da38747279e9a703f9efb5803 : SUCCESS in 4m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/37e34e6bb6874fa5a682790f1ab8ef62 : SUCCESS in 10m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/42a9422703e049f8bfa39bf67e45030c : SUCCESS in 20m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/db89a6b6c57449958e0e2230c9c522ee : SUCCESS in 14m 01s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8bb6f30c89244db48efa49dfad5faf4c : SUCCESS in 1h 11m 29s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/8175566d80aa40a78498ff79456b5b65 : SUCCESS in 1h 23m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50ea22e59e3d4bcb8f2f9121e4f5565a : SUCCESS in 1h 08m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/df901f4f38a747eb8c45f42d4925c3c0 : SUCCESS in 1h 11m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/790e990600a843f896fb36019cd40050 : SUCCESS in 1h 21m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/220bc945663a4befb2e6332818429852 : SUCCESS in 30m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7b541275b27456c9457e5b773476c49 : SUCCESS in 24m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/773ced12b8854fd5a3c71c10efa2b8c8 : SUCCESS in 2h 28m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/008a06ee30864de4ad7f61e507c9714e : SUCCESS in 2h 26m 40s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b11814532bb144168f078341422f32ee : SUCCESS in 2h 27m 51s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a1daf63f00e4543ad9f01a09bb1e283 : SUCCESS in 2h 16m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a7be37b6db9644c98c9e99946fc298ac : SUCCESS in 30m 50s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/9af2d591a7ce4065ab8d09dab1edf6eb : SUCCESS in 31m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8fd73f1e570481594be562d1a13a7d5 : SUCCESS in 28m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/50823aa3a4c54239b6a0552fa805b0f0 : SUCCESS in 28m 36s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2ff94f15173647a68646f18a68602e74 : SUCCESS in 42m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc828cb8b90a4adcb36a4e5741e76c38 : SUCCESS in 42m 55s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/eeafd0d6c5874e2da14428808892473b : SUCCESS in 40m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/57b823d5c2cb41e69c62fa8c725f65a2 : SUCCESS in 27m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e206b5daea544be889be38276462b842 : SUCCESS in 30m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ef417550997c47edad5096eee5c36ec5 : SUCCESS in 27m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ed483a61f504f09bb5b0734343b7dd8 : SUCCESS in 24m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b2bc633cf012f9fec8b29a3d47bca977e83e80e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 14:53:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9e67c1e2d2d93cf2ef71aa6c6e1670952eb148b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 14:54:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3d2307f4e0a49f0b615ad5a2b074ef5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ffa553d7d46437886fc3cbff0023ac1 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17500b6bcb574a28b29a9e007d2b88b1 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b2246066c5ec937fbf2571bba987321263661db","revisions":{"0b2246066c5ec937fbf2571bba987321263661db":{"kind":"REWORK","_number":1,"created":"2025-06-23 07:32:34.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/74/953074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/953074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/953074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/953074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/953074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/953074/1"}}},"commit":{"parents":[{"commit":"08d10d19d9d0b1d53b461a17020c5f6b427a5663","subject":"Fix infinite loop in seed container deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08d10d19d9d0b1d53b461a17020c5f6b427a5663"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-20 08:17:32.000000000","tz":60},"committer":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2025-06-23 07:32:34.000000000","tz":0},"subject":"Fix deprovisioning multiple overcloud hosts","message":"Fix deprovisioning multiple overcloud hosts\n\nPreviously, the user prompt ran on one host, but the assertion ran on\nall hosts. This meant the check could only pass on one host, all others\nwould fail every time.\n\nThis change ensures the check and assertion always run once on\nlocalhost.\n\nCloses-Bug: #2115037\nChange-Id: If2a26b42e6fe0152922e10df924c0797259e2e53\n(cherry picked from commit 9a7aaffff723fbae4483f38b55e6ea55d4d908fe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b2246066c5ec937fbf2571bba987321263661db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b2246066c5ec937fbf2571bba987321263661db"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
