)]}'
{"id":"openstack%2Fkayobe~953190","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~Ia368ff68f4c3ac69406040be25fa09711e221fa3","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-06-24 09:20:10.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-24 11:27:59.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-06-24 11:27:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia368ff68f4c3ac69406040be25fa09711e221fa3","subject":"Bump stackhpc.linux to 1.3.4","status":"MERGED","created":"2025-06-24 06:52:40.000000000","updated":"2025-06-24 11:28:46.000000000","submitted":"2025-06-24 11:27:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953190","cherry_pick_of_change":952983,"cherry_pick_of_patch_set":2,"meta_rev_id":"4b34f077677af6184cfdf6d2726de9488a4833ee","_number":953190,"virtual_id_number":953190,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-24 11:27:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-06-24 09:20:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-06-24 06:53:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-06-24 09:20:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-24 06:53:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-24 08:54:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-24 09:11:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-06-24 09:11:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"14dc601ece5548f02cb30808284d2c40d341ae6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-24 06:52:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7c2ed947dc65055075810b6445035a74a734ae71","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-24 06:53:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cfc32503f72c8c27bcff2c79bff80f83599cbf96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 08:54:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b513a0224e79489292fd2a8d9ca794ee\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/540e7f8a3b2340b08bea9d81c3653cad : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a2e7e2d6a8644d3b550aba2bd641b9a : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/900da57fb24d49819fc1560b6f8abf8d : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a607ee79a6404f41b381e8e651db9234 : SUCCESS in 2m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4a4b2ff3a7a54d298853519302aad650 : SUCCESS in 4m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a5a6d9de15f3429d994508e357a34c43 : SUCCESS in 9m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/68fd884a83324524b6b4e0b748cf50c0 : SUCCESS in 1h 06m 38s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/72eac02a96c14d718c3c429b3601af7f : SUCCESS in 1h 02m 48s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e8fb080dd14493ea82520b77119d3f8 : SUCCESS in 41m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f5a091b42654453b93b0ef77626bbbe : SUCCESS in 29m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1be7f0ff764845108cb1d6ada8eb7f7d : SUCCESS in 1h 13m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3a0778b7428d410fa3048a4c3c133ad6 : SUCCESS in 16m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2d89fe1b91a3454180a4c6c32210c2b0 : SUCCESS in 16m 03s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d518484ecaf4bdcb4f40cf8b720d45e : SUCCESS in 8m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f2953d01ae845a69f582b402b136bc6 : SUCCESS in 11m 35s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a52e3b47479b4c4bbba2bc6307eb0d12 : SUCCESS in 1h 56m 18s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c1a7235ad1a419d964f8ce02c9a4b95 : SUCCESS in 2h 00m 01s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/afe7b40717284c97a423949d98ef675c : SUCCESS in 21m 05s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/882f982df0324a5ab36597a43e6abbfd : SUCCESS in 21m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e547eafb95664f9e90bcf36f37b5089f : SUCCESS in 10m 52s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c45993a151f54768b215b874ccb0e6fc : SUCCESS in 41m 26s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7a2bafeca9e342419316ce863670a031 : SUCCESS in 24m 23s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf168fbf8dda4ca29adef9f071a4f3c3 : SUCCESS in 12m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d20f4fb584f14c3ba7d9bc2d2eba9e0a : SUCCESS in 17m 55s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/8cbb725b6c3f4ac6b503e77a71252265 : SUCCESS in 17m 22s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8987b4b10af64be2a40f7e33ff7f68f5 : SUCCESS in 15m 25s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/1826265d414241b3a1ef67b9b7cb4f94 : SUCCESS in 13m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbe2c21f387a4818810741b67794e070 : SUCCESS in 16m 20s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/a3bd88bd1a814bd7a50bd9d042d3c875 : SUCCESS in 13m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a71560917c984f9a8838bf12231b6238 : SUCCESS in 16m 11s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/678d41864c99489fa2bd0cfacc649324 : SUCCESS in 12m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14f884ef465047f9a55f451e981b23cd : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c38038ba5b1c4fb5b9a3681f8718697d : SUCCESS in 2m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/096edda93091462d9227cb13f74461b0 : SUCCESS in 2m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3212db97077f4af88c5e6913644d4ca0 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bddcfbc61df4e86b07a5a2c75871b6c : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"abb95b5fb1ed778d8158afc02a80a130a4c4c3b2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-06-24 09:20:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"14ca30a47bd75af49a2eb9dac28ffc9e87f0fb1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 09:20:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b31caf10a01290aaac1dbee4715fa614697b3464","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 11:27:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46d974e280c64a2199432e1ae304f28c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b3508a78ab646d79c6eea0d0c29e4aa : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da9bc083666847ef91779f50795e4cb7 : SUCCESS in 2m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7702acd719e348668d66a4af07e38193 : SUCCESS in 3m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/718337e561a94ec5a05e8d8895ca6e5f : SUCCESS in 11m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f082b00aaf094a85ab7648267794e652 : SUCCESS in 6m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b57470a638ac4a1eb11e192871c91c20 : SUCCESS in 1h 02m 38s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c790f205702b497a8890d4792611a4c6 : SUCCESS in 38m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f2d0359f1ab40e3a7152ca424b400bd : SUCCESS in 50m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2220bd104f834e898ecca3bdae1c1ac8 : SUCCESS in 1h 07m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/77cc8b58380c445ca678762623a9ee8e : SUCCESS in 15m 43s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05e1e7d42e32447c80141e67bfe1fb75 : SUCCESS in 13m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4e32dbfe70942b0a613bf43b37b7c56 : SUCCESS in 11m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fdf77b5cf54449f0aa778ea8c4aa97d3 : SUCCESS in 19m 49s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2ad456417b542d4912b78112e7db9b6 : SUCCESS in 19m 52s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7927d9d0ef66403488dda8c724590040 : SUCCESS in 16m 01s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f495fcffbcb24eaaa523cf76c9d53764 : SUCCESS in 16m 00s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69a17727cf104a50bd136b46ae545be8 : SUCCESS in 23m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/82693d2a187d441eb41175d1aab9e533 : SUCCESS in 15m 03s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/384164a0d8e0418995b39c794393dcd4 : SUCCESS in 9m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bc1b6b212bf4607b83f63883507d26d : SUCCESS in 14m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c6edb03091b14a86856b1e00bfd3a8a3 : SUCCESS in 11m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c45731f6f2b44dc7ab325f969fa26fb2 : SUCCESS in 13m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/555f2f6d9fc4481fad4fc54539f818b5 : SUCCESS in 10m 38s","accounts_in_message":[],"_revision_number":1},{"id":"eccf29c0002d44e774ad0207c3a2dab7b857ffe8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 11:27:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4b34f077677af6184cfdf6d2726de9488a4833ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 11:28:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5348d40cdae94ee8b97df66759545911\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c4994332a0b48d898b37f64f47108d2 : SUCCESS in 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/626a4da8cf7c4df6ae40aa8eff5a7d7f : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a166ab81e40efdafc299283a64a51afcf806f57","revisions":{"1a166ab81e40efdafc299283a64a51afcf806f57":{"kind":"REWORK","_number":1,"created":"2025-06-24 06:52:40.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/90/953190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/953190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/953190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/953190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/953190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/953190/1"}}},"commit":{"parents":[{"commit":"d9a0af887bb0178036cb1a25e08197bb64372d3c","subject":"Use latest CentOS Stream 9 GenericCloud image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d9a0af887bb0178036cb1a25e08197bb64372d3c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-20 13:41:04.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-24 06:52:40.000000000","tz":0},"subject":"Bump stackhpc.linux to 1.3.4","message":"Bump stackhpc.linux to 1.3.4\n\nThis includes fixes to SR-IOV and VGPU configuration [1].\n\n[1] https://github.com/stackhpc/ansible-collection-linux/compare/v1.3.1...v1.3.4\n\nChange-Id: Ia368ff68f4c3ac69406040be25fa09711e221fa3\n(cherry picked from commit 17254c42fa0539891fa2718145ef0798484db789)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1a166ab81e40efdafc299283a64a51afcf806f57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1a166ab81e40efdafc299283a64a51afcf806f57"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
