)]}'
{"id":"openstack%2Fkayobe~953207","triplet_id":"openstack%2Fkayobe~master~I3937cb29143f3b9965032c186e8086854de7fbef","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-06-24 16:12:19.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-27 16:26:44.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-06-27 14:36:56.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I3937cb29143f3b9965032c186e8086854de7fbef","subject":"Bump dellemc.os10 to 1.2.7","status":"MERGED","created":"2025-06-24 12:30:51.000000000","updated":"2025-06-27 16:27:39.000000000","submitted":"2025-06-27 16:26:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953207","meta_rev_id":"479076d0c215e75f6d0a0acf038bc838e72c6ed9","_number":953207,"virtual_id_number":953207,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-27 16:26:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-06-25 15:05:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-06-24 16:12:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-27 14:36:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-27 14:36:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-24 14:38:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-24 14:45:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-25 15:05:31.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-06-25 17:10:10.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"e33e8515e70a41a9c953854b717201028bc330e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-24 12:30:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88912816394c5a861b1f6d8354eda00eae1a2f7c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-24 14:38:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8c271b9e93e63bda182a9c4e23254b13707c67c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 14:45:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/201b4ad0bcb44a0fbee9aeb3f7bbce89\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9b7eb97c67b48f9b26169576a4fafd4 : SUCCESS in 4m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/efb8876b4707432aa64bf767930911e9 : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a61f661ef57846a8aade0385a6564d02 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1f34c1490564b0db8f8becde14ef318 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de97d6b0129a4bedbccebbaaa0211090 : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8afaeea07ce9489ab5d57d1a86c01218 : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4950f01767c74254be6f7b99a277770e : SUCCESS in 6m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea3338c681464be8b7ff655678c116af : SUCCESS in 2m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b691e0aa1af54d84851c82ca031e4e4f : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/da4924525d044d999ab6171333dcf6c2 : SUCCESS in 11m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6efc8df9a2204da2ae0c2cdcc3b015d2 : SUCCESS in 9m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0bb89f098c84bc49f48c669c7d27aec : SUCCESS in 55m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/28d553e024a942e99202835a4d0da1d5 : SUCCESS in 1h 07m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5378fc2783014f4b9bd78344e56ce881 : SUCCESS in 21m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84e59a6899544430b5ba6e3198bc2c6d : SUCCESS in 16m 07s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f20bb35f2e834490865d28098506c958 : SUCCESS in 2h 07m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2260b7aab7904214b31a772c12d8c4b7 : SUCCESS in 19m 50s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/f2cc289263c14125b07112016da2a092 : SUCCESS in 19m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ce86493df8e4ff49d82dffbe666b801 : SUCCESS in 11m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ada74a47c0644cb1a9a4a6cd16b84c0b : SUCCESS in 16m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8f9df6e931ad4796a507f356084be0f7 : SUCCESS in 32m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/46383229b2c7468f9ccf5c4b471bfe19 : SUCCESS in 33m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52f0399eadd74af894ad54fd0bdb5959 : SUCCESS in 14m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/aa48e12fe30d494299a71fa8b26bfbfb : SUCCESS in 7m 21s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/4a610194004a40ff82d46774b7b101b8 : SUCCESS in 7m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c60ccf88fdf04ed4b66d90467dc309b1 : SUCCESS in 7m 22s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/3ca317d76de84d3a947cd974a7b608d7 : SUCCESS in 6m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/30a1a164bced4c2098ca751c4b78d462 : SUCCESS in 7m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2483fc6939874fc5ad6e0a58917eae05 : SUCCESS in 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"24370d35e901a96d39222c059f88fef4e2bb20f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-24 15:56:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3223a5be4306084444604cffdf48a752b1bc921e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-24 16:12:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ee093b1d1a8d91d6883af25238f8611afc05b080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 17:17:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c790d412da294216853c35d112bae06d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17919b2bea5c47839079c20de488101f : SUCCESS in 10m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5c18ba3e4d5e4e64b5ab60f5018c54c1 : SUCCESS in 11m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6b792f7431b4f58bc067916f9f34ec2 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a7a2ece7a504195970cb7e1522021df : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9fb2bde0d914b2bafd02236442675d4 : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa4741593d684b589923d8d6fdad68e0 : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/583f9580c61f4cacbe9c47239b4a504d : SUCCESS in 8m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/256fd4407ee44ff8bee90735c9900a3b : SUCCESS in 1m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fa8d917039ff450593d470cf336aaa69 : SUCCESS in 12m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3a6031b4b9684354af8a5c2daa3f1b96 : SUCCESS in 26m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bdc3e04362944936946cc58c6db15868 : SUCCESS in 14m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/222810bb06e647678b20e75a94ab29b1 : SUCCESS in 1h 16m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/896e68581257456da76be918bd579a06 : SUCCESS in 39m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1a8588a416964b35a03a59711b3f42ac : SUCCESS in 28m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa43c61652ae4cab9dda13cbdc142fe1 : SUCCESS in 24m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87a8ef34f0ad41aea38f7e5a73f2fb6a : SUCCESS in 1h 17m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6dcb46dbb0354ddaa642171f6edf5dd9 : SUCCESS in 28m 22s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/1d10feca07dc438cbe0b5379cf791b06 : SUCCESS in 31m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f4ebe56bf6745fb92a9645757841e64 : SUCCESS in 22m 36s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9c72cfffc7eb4895b77addf126b942db : SUCCESS in 25m 51s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b4c6eb10aa7449c5b673b7cec74c9c1e : SUCCESS in 52m 37s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/beb5b23db37c4ed3b689c29a211ebc87 : SUCCESS in 51m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea120df3e2a4484ab48025e11fcd12aa : SUCCESS in 44m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/66a1fca8173a4141ba31fde2f9adfbeb : SUCCESS in 25m 03s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/38cc5402824d4dcc97c27e05419353c7 : SUCCESS in 24m 40s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa92904484344bf3bbfcc6214c889458 : SUCCESS in 25m 52s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/bea1f816311342bfa3de07fcb0996e81 : SUCCESS in 24m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/79ad0a983b51493b9de04fc43d220ecf : SUCCESS in 26m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6dfb3640945740c598efd38322aa21e1 : SUCCESS in 22m 40s","accounts_in_message":[],"_revision_number":2},{"id":"9f01468532ae5a7a11dba068269b4af4df19c7b8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-06-25 15:05:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dcc8868ee909473b4a563bbf886f03dafdd3a3cf","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-27 14:36:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7e5ee4131617de16f0fcb8bccc9a965cb21c83e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 14:37:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ccac65408fa3bf3ced261e982ef4c184026ba42a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 16:26:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3b811dd87484cc39b4d7944bd25c599\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67a40a3a41584465815f4602c743d2e2 : SUCCESS in 9m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a45698280a02414e9f3599397c461123 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adeedd8b2302475f912d912c86b4e810 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a3ce9db1dfc4a7599fceb5bae20cee6 : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/585b4a1403fd4ffda41747a58f54041e : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b18d0aa454174d6b9a2575735093cc52 : SUCCESS in 2m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/779ce75535934c9a9b721a5ea263b020 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a98646e1ebd439ab4625d6b2ca98cbb : SUCCESS in 11m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e8b9bd95376a4b9e8e96461d20d62ecf : SUCCESS in 12m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7feed0b7bb71417b85ecb16c4e117954 : SUCCESS in 1h 02m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/78bc48dc89f44701a7558b5c719adfae : SUCCESS in 58m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/34f7a50b48bf43e7bcb8cd87770a0d84 : SUCCESS in 17m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5936b6d93334b2aac8c4f3896694d8e : SUCCESS in 14m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa0c44c3f95744ababfbbe7cf1336932 : SUCCESS in 1h 44m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/887de87c077a463ebd5e8b3657ae698f : SUCCESS in 16m 08s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/b3f6077eb19740d4a437909329e8f61c : SUCCESS in 18m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7d88c2bfa9f4e319494971b203d4b8f : SUCCESS in 19m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a42dec294aa647a1b69082577b7978c2 : SUCCESS in 8m 29s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0dae461d28994d02a7cd6a3c53f48205 : SUCCESS in 31m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c55344f4f4f441b190b5c6809cc19eed : SUCCESS in 31m 07s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d93b9ba277ed4439a376676474339ebe : SUCCESS in 13m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d98da8969a404dacb624da99abbb5b71 : SUCCESS in 11m 38s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/70555c0f00db4bb1a8fb73c45adf3eab : SUCCESS in 13m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc49bc2674a64670b8bb59b2d507d574 : SUCCESS in 12m 32s","accounts_in_message":[],"_revision_number":2},{"id":"f40cbf602d1214ad1750ae26f7ecd40ec4564f04","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 16:26:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"479076d0c215e75f6d0a0acf038bc838e72c6ed9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 16:27:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7d8cea67f5d4227a83f4a59c138cde7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f84cbe7ff5547ecb83eafd4c8c780c6 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/984bc23829514a3aa7af844e5fd741dc : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4007b0ee240904c83b908baa50a0003906f81876","revisions":{"95b10e978fddf8c1e76352853145ad8deeecd3e1":{"kind":"REWORK","_number":1,"created":"2025-06-24 12:30:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/07/953207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/953207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/953207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/953207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/953207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/953207/1"}}},"commit":{"parents":[{"commit":"87459afc2d1ec68de0a94269134c7c6a2fd246f8","subject":"Merge \"Drop some leftovers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87459afc2d1ec68de0a94269134c7c6a2fd246f8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-24 12:30:20.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-24 12:30:20.000000000","tz":120},"subject":"Bump dellemc.os10 to 1.2.7","message":"Bump dellemc.os10 to 1.2.7\n\nThis resolves an issue with applying some configuration blocks.\n\nCloses-Bug: #2115121\nChange-Id: I3937cb29143f3b9965032c186e8086854de7fbef\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95b10e978fddf8c1e76352853145ad8deeecd3e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95b10e978fddf8c1e76352853145ad8deeecd3e1"}]},"branch":"refs/heads/master"},"4007b0ee240904c83b908baa50a0003906f81876":{"kind":"REWORK","_number":2,"created":"2025-06-24 15:56:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/07/953207/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/953207/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/953207/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/953207/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/953207/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/953207/2"}}},"commit":{"parents":[{"commit":"87459afc2d1ec68de0a94269134c7c6a2fd246f8","subject":"Merge \"Drop some leftovers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87459afc2d1ec68de0a94269134c7c6a2fd246f8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-24 15:56:41.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-24 15:56:41.000000000","tz":120},"subject":"Bump dellemc.os10 to 1.2.7","message":"Bump dellemc.os10 to 1.2.7\n\nThis resolves an issue with applying some configuration blocks.\n\nCloses-Bug: #2115121\nChange-Id: I3937cb29143f3b9965032c186e8086854de7fbef\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4007b0ee240904c83b908baa50a0003906f81876"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4007b0ee240904c83b908baa50a0003906f81876"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
