)]}'
{"id":"openstack%2Fkayobe~953274","triplet_id":"openstack%2Fkayobe~master~Ib27428f89b0e3f2865ab1c498f05de741fa49587","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-06-26 08:09:49.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-27 16:11:01.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-06-27 14:26:30.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"Ib27428f89b0e3f2865ab1c498f05de741fa49587","subject":"Support empty strings in dev-tools package lists","status":"MERGED","created":"2025-06-25 06:17:47.000000000","updated":"2025-06-27 16:12:06.000000000","submitted":"2025-06-27 16:11:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953274","meta_rev_id":"8e221c93ae6c5697a459e77968d37316e3402e88","_number":953274,"virtual_id_number":953274,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-27 16:11:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-06-26 08:09:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-06-25 07:57:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-27 14:26:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-27 14:26:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-06-25 07:57:44.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-25 07:57:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-25 08:04:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-26 06:50:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-06-26 06:50:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"f5639107f08e93628ac585f9b3b394f003741ad7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-25 06:17:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7234d9096c436b1e0964e69eed13f9c9794e355c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-25 06:31:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"084c7c8b3a33c23890b8fb4cae53f383513b838b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-25 07:57:44.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e46bcb3a217c1ad43afa8fb079f1a14811a074b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 08:04:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a69f8e00fc09401786d5c29e7cbb04ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f0e995679544f679c57857f58ca38fa : SUCCESS in 7m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dbe5e24bf99b478a960d005bfebbef8f : SUCCESS in 6m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37faadca768049c1a33a4f9ba8382bd1 : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e6e21a90b7f4e9e841b9d3b16065901 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c19a1c4334f4b57a30179f960baae31 : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e3e0ed679cc14966bbc2801de4de0f7d : SUCCESS in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/892611fb53b4449b8c4d81397221eb22 : SUCCESS in 8m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4258224b95464a65971a152778da1e09 : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5a1afd380e2044b6bf6a382b2dac415e : SUCCESS in 5m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5806d2ed945c49cc8eed162354baa797 : SUCCESS in 9m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ef9dd41f916a48669bffff3840928702 : SUCCESS in 5m 58s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89ddf1394b3648059a79d40f607f1d09 : SUCCESS in 1h 01m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4018304d8a23455bacc1bcaf3bd8d7bb : SUCCESS in 1h 06m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6c65e5840a0a446e853af72e1abcb17d : SUCCESS in 17m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d8f30a1f6604cdbbccaf6e439770521 : SUCCESS in 12m 17s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b996bc33d5c04117aebaffaff90e8024 : SUCCESS in 1h 32m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/abcbd4e06f34451ab46b3c7cddd6abfa : SUCCESS in 15m 37s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/ac5367aa19ec444eb486e429d679d445 : SUCCESS in 16m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60982d594466410ab298161dace52238 : SUCCESS in 8m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/45edf1a00810400397569852e752ab6e : SUCCESS in 13m 29s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bbb31f97ad2b47919aada48898722d56 : SUCCESS in 20m 08s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bf5d6f46b3654c4d8a515e7b6c166e6e : SUCCESS in 26m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0dde00616514fe6a6cc6ae9e1bec286 : SUCCESS in 32m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8a79f30d6a8b470ebbd41a357da82fbd : SUCCESS in 13m 36s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/a9dea31848a347a995319db44e0d3ad8 : SUCCESS in 12m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abb128157baf4cefb9c6bb6ff6f9736f : SUCCESS in 12m 38s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/1827f6b22e464a98b7595b3df2368630 : SUCCESS in 11m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/667c831d534243688f010d290eb7d860 : SUCCESS in 15m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76faf91bd35d4c1f970dc31206e230f5 : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":2},{"id":"ceaa0c65c399ad3bd55eebc71ef03153da5bcd98","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-06-26 08:09:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"341ad7b085fec61a7cb8ade3fe2cbe8322377e92","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-27 14:26:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"544195d6ee88abae3715a1437820ed6828f985fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 14:26:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eb736a62065404a405deceb6611977d708a99c89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 16:11:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4eb08019b0e45b4914deebbfa331203\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8807359b9414f4580eed101ade14838 : SUCCESS in 8m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b3bb2b42c6a84231b18a2f829416ce65 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5395884399764f049f4177fc4a5957fa : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/164db0372eff40e2b392be286dc60d70 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e323997ad3b04ed68722f1c06b69724e : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da767a45905d4492a0cbcf42ac7233a7 : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3fd77c690bd5429db11847b705477646 : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7dbd35f19fee409b976372e3d77dcc90 : SUCCESS in 11m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/84eeb287329b496ab6810ca7f300c015 : SUCCESS in 9m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7aa588a401d448cb9aaa0972e1e85dac : SUCCESS in 53m 28s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5407fc34d43f4406945c3e4b674f96ff : SUCCESS in 1h 03m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5c984fcc6eb94e80bbbc47c327321508 : SUCCESS in 21m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e33f93caf284b968f12ba8cd3a9786b : SUCCESS in 6m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c94acdaa8bdb4bf582ccd32643db7b5d : SUCCESS in 1h 38m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/13c7a5591345486b9cd62c62032a0ca2 : SUCCESS in 17m 05s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/c7e7fd7f3a474dbd80e32955efc29be3 : SUCCESS in 16m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44d07469724c46c288019c760fd437f5 : SUCCESS in 15m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7932e270f59a45e9832dd01b81530e93 : SUCCESS in 18m 10s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/21af4b702f2148bc9d88ca6c6287ba74 : SUCCESS in 32m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28d586e42ac04ba8a6786c0a1f3351dd : SUCCESS in 35m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f9137a135a1f4ac6b45f6f40e91eefe4 : SUCCESS in 13m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1ce15074c6e4f34994583758e29cf19 : SUCCESS in 10m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/142e5367255b43e28ea5255b866b6a2c : SUCCESS in 13m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bc44222567642d78da0c5ca8248018b : SUCCESS in 11m 28s","accounts_in_message":[],"_revision_number":2},{"id":"76a13fcf5c45068979f64fbf1149f1d4fc8e8cbe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 16:11:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8e221c93ae6c5697a459e77968d37316e3402e88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 16:12:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/013f9709a5814d5e87aa6d1a68af653d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbaf0cf1ca804e9c8f3d4ddbd033b304 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f70435787db74d708835577d24182658 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"36e7a7b0d0441413913406a68d886248838f794c","revisions":{"99d7543c099141f92ebd5a2cd71405cce96db40b":{"kind":"REWORK","_number":1,"created":"2025-06-25 06:17:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/74/953274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/953274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/953274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/953274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/953274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/953274/1"}}},"commit":{"parents":[{"commit":"87459afc2d1ec68de0a94269134c7c6a2fd246f8","subject":"Merge \"Drop some leftovers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87459afc2d1ec68de0a94269134c7c6a2fd246f8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-25 06:17:30.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-25 06:17:30.000000000","tz":120},"subject":"Support empty strings in dev-tools package lists","message":"Support empty strings in dev-tools package lists\n\nCloses-Bug: #2115000\nChange-Id: Ib27428f89b0e3f2865ab1c498f05de741fa49587\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/99d7543c099141f92ebd5a2cd71405cce96db40b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/99d7543c099141f92ebd5a2cd71405cce96db40b"}]},"branch":"refs/heads/master"},"36e7a7b0d0441413913406a68d886248838f794c":{"kind":"REWORK","_number":2,"created":"2025-06-25 06:31:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/74/953274/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/953274/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/953274/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/953274/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/953274/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/953274/2"}}},"commit":{"parents":[{"commit":"87459afc2d1ec68de0a94269134c7c6a2fd246f8","subject":"Merge \"Drop some leftovers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87459afc2d1ec68de0a94269134c7c6a2fd246f8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-25 06:29:58.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-25 06:29:58.000000000","tz":120},"subject":"Support empty strings in dev-tools package lists","message":"Support empty strings in dev-tools package lists\n\nCloses-Bug: #2115000\nChange-Id: Ib27428f89b0e3f2865ab1c498f05de741fa49587\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36e7a7b0d0441413913406a68d886248838f794c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36e7a7b0d0441413913406a68d886248838f794c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
