)]}'
{"id":"openstack%2Fkayobe~953316","triplet_id":"openstack%2Fkayobe~master~I74c8234f77115d6a478559a53ad86534eb12e5f4","project":"openstack/kayobe","branch":"master","topic":"generic-etc-hosts","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2026-04-09 12:43:57.000000000","reason":"Change was submitted"},"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-07 11:00:53.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-04-09 12:43:57.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-01 12:04:28.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"37855":{"account":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"last_update":"2026-04-09 12:43:57.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-04-01 09:37:59.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I74c8234f77115d6a478559a53ad86534eb12e5f4","subject":"Add custom_etc_hosts_entries variable","status":"MERGED","created":"2025-06-25 14:09:08.000000000","updated":"2026-04-09 12:44:53.000000000","submitted":"2026-04-09 12:43:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953316-generic-etc-hosts","meta_rev_id":"f38bbab189d0ff90ee60b0579da8b3f8c35b5aac","_number":953316,"virtual_id_number":953316,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-09 12:43:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-04-07 11:00:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-04-01 12:04:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-01 09:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-07 11:00:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-07 11:00:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-25 14:12:40.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2025-06-25 14:14:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-25 16:51:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-26 08:07:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-11-17 12:04:12.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-11-20 14:12:34.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-02-19 15:19:50.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"REVIEWER"}],"messages":[{"id":"c60863a358d15e8989ea7316d6587da89f1419fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-06-25 14:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97fafaae0a442e19d6434dd53fa27992b22254e0","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-06-25 14:12:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7434d1a3b94e9162f29c31bd2b318f2dde9e56f0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-25 14:14:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"953371f37842e593853df05e6bbfda79dc247aec","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-06-25 14:26:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c3749c12ca590bbad8381d2f1e2221fd3fed6f4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-25 15:20:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"399ce1479f3314cf9de1fe6f58e3b03160cb440f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 16:51:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/347e5f1f202749298608bf403782e175\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65f64b3807864b9280a18d4737f92059 : SUCCESS in 4m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6187d95902d043a98aeadace9a33bac6 : SUCCESS in 5m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca0cd6c3cc6a4fc7a8326ae02d07bc60 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c67989c095a048cba3c827c908cb81bb : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8c09e5ca6dd47c0baffb8e835d2acfc : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/767d8dd147984b0f8e2151e3860652ee : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10949ad3d11f443db6d69249033eef8c : SUCCESS in 7m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9738d6d8f894b31a70034d766253b48 : SUCCESS in 2m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/82b6fcb9827541b69f8b20de27236dc7 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cac012e2dfa74f3aa15ffa510654f196 : SUCCESS in 9m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9609e18a51e643c983dc91809444c071 : SUCCESS in 12m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a6c96058014483c9785a363f8e3cf0e : SUCCESS in 57m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d0f3922811794b9ba4140d4e53e6162b : SUCCESS in 1h 09m 22s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/83bd628732b2432bbd715573a30ce661 : TIMED_OUT in 2h 02m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d88ad5c367344eab89d5e8c4f2301217 : TIMED_OUT in 2h 01m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa347c5b4c9a4f3096b4f9cf238f2760 : SUCCESS in 2h 12m 13s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7b791bb00f6147e099f4f9f07bdd6cfb : SUCCESS in 17m 12s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/8156cb3c027749baa7ad3465da31d91a : SUCCESS in 9m 38s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e65fc1e293c34d5fbf12a530add46f10 : SUCCESS in 8m 22s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/21ceb3fee42e42beadd86b9a38a0040d : SUCCESS in 7m 43s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/22b988c069104385a36546b6aa6bc54a : SUCCESS in 44m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e318b1ee47c84444ba2d4292e31aa6e4 : SUCCESS in 27m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fadc8deafbe4cf0910164b106bb1dfb : SUCCESS in 17m 18s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/79e7ec508ed649b6b9c1813efbc68bf4 : SUCCESS in 13m 21s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/d208033db35b4fc89788219316b17dd4 : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee3f9c1c95f44499a81889e9342135b2 : SUCCESS in 6m 02s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/6bf1bacb97b24ea4aa04526fd3d0ba97 : SUCCESS in 12m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ad6903cbba444a74b415ea3c0ff51e08 : SUCCESS in 13m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/120c2d3f52ab4779b84674abc33b1b21 : SUCCESS in 14m 37s","accounts_in_message":[],"_revision_number":1},{"id":"cfd350b191336bc76b753a391c3c47fc514ab2d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-06-26 07:57:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3fef7181fee4e923517be5cedfec80d441159114","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-26 08:07:18.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f39b7501e97df4a238577a244339607cb14162d4","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-06-26 08:41:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f75c179e72e71c43f0f8515cca7f8e46e234aeb6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-26 08:59:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ec9b96899267193b348e014385e36e896d716b17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 09:47:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c00cfcb83d4421c8d00e9bce985bb6b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73d5b69240d34595a81bebc7b727be24 : SUCCESS in 9m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d3384a89c7454dc89bcf86ed97163f89 : SUCCESS in 6m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8536402d5b9141e7acf639615d512904 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/788893de12334bb092b669677787ef96 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd7129198dd54147ab483be3ef3a7d14 : SUCCESS in 6m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aff1b5b410e74b33945dd8e28abc4d65 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92bcf7131eee4c24941f1ef64f66c52c : SUCCESS in 7m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/453ee3906f4346e0a6000933e82b9ce1 : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/be79beda5edb4e939d8e4048437de54a : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/35d2876515384695870993c62240b222 : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/640cc9751d2d441ba7464126a07d5e31 : SUCCESS in 6m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fea03511a9164dffa5345ed10eb26e0e : SUCCESS in 56m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7f465b8ea12b4cd7b4a357d72519ca0b : SUCCESS in 58m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f55035fa0efa4600a5c10d063f651c7a : FAILURE in 17m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f6aba8ac86c44eeac8e80a3c875c908 : FAILURE in 8m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a4220c18c91432ab57c738e7e665179 : SUCCESS in 1h 45m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f418248ef573439f8618725b90ea12ff : SUCCESS in 11m 28s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/ab1c376879ac4c28a44f5b3daf98a17b : SUCCESS in 13m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/480160efdf3745198fda96787b83f874 : SUCCESS in 15m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c558b69c77c5421db92bde60c88c53e5 : SUCCESS in 11m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/85a8121e3727405688fd8de7e9345489 : SUCCESS in 26m 00s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ba8c2827c2c24a71a3c0dcb83882e130 : POST_FAILURE in 24m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/239185c1c4194722bbc160f221a82080 : SUCCESS in 20m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9fca4184b2a0414cabe51234284da3b9 : SUCCESS in 13m 37s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/9415dec3e8e44173b802fe913726b260 : SUCCESS in 10m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6fa398d57ed4d7591e86fe1dbcfe1c0 : SUCCESS in 11m 55s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/a662258438ef4048ae71b6b3477e4791 : SUCCESS in 9m 52s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5db9fef6d4e445409c93635f17b2365f : SUCCESS in 13m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa2799a835d5465387da0c734bad8fd9 : SUCCESS in 11m 25s","accounts_in_message":[],"_revision_number":2},{"id":"ea68b25476461e4c4e73b5c624759a9d90517b50","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-06-26 10:39:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0014b697ff1efbb32f98cfb2dccfabc00b48b8f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-06-26 10:40:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6f74e0325d0b5ebc567a29092a8a7870fcf3c00b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 12:23:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/713abb47929e4f32b44812030f2f9a42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c37d8b327db3460eb8d3989c687e6122 : SUCCESS in 9m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ec9b64695b174e9baf9bbb021f60acc0 : SUCCESS in 5m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e69149d956d42e2959c53da65a61a4c : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2076c02e0400495fbab910e01c069c8b : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7033c93069e42f2ad4504160d40f348 : SUCCESS in 2m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4ea900c7aca46c7a39476e49db57c6b : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8aed8db33b25405f9f98189d3a69ba40 : SUCCESS in 6m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc0b4b971e1e454e9ca9f5ea6f8652df : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/63ab197cfe8742988fbbe3e5f7f4049e : SUCCESS in 3m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a7ef0c9b16df4d6b8a1404334e16b9d2 : SUCCESS in 10m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6952d2f3e5c44bd2b9a346a30a2d7e72 : SUCCESS in 7m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e75785e0050f4274ab401a4066c5a3eb : SUCCESS in 35m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6b09391eaff9483c9628e424213db566 : SUCCESS in 56m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/84bd969525b94c6f829ad6644bf46f14 : SUCCESS in 9m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/299e540b6e5047e1a673d01d778d8ee7 : SUCCESS in 8m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/355e8cd5c0bc40adb253184d807a297f : SUCCESS in 1h 42m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/36442869af2c4365be9cf220f23bdc48 : SUCCESS in 21m 55s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/1db5aae777dd4efba97af8a7bad773d1 : SUCCESS in 9m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/295739b167cb4f97b1b581232be0366f : SUCCESS in 14m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/441b277e21044756ac2dbdac85410c44 : SUCCESS in 8m 25s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a5231d4e9d3c4352b3cc5c8549c38297 : SUCCESS in 36m 42s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9b87bd5bf6734e379a24d5a8318c6f74 : SUCCESS in 18m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7535ce9ad63c4a858a0c4f1dda72dcf3 : SUCCESS in 26m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/18b74d832e9640b49f8c926725bf871d : SUCCESS in 16m 46s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/5ea0353dc83b461e98790f78703ee864 : SUCCESS in 9m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0a124eb261e4301b24dfd14cf4060e8 : SUCCESS in 10m 46s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/7ee924ed53d84b4598296cbd4af91ff9 : SUCCESS in 9m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a27b871151f844afb7eeababa36322df : SUCCESS in 9m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb599990de0b46108003dbc1235057f3 : SUCCESS in 8m 38s","accounts_in_message":[],"_revision_number":3},{"id":"3f4f75576c2ff63cd6d8484cdfe99d99c2927a27","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-08-14 16:48:32.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4ba30f54605ed39184088746870a6c93aceb6bed","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-17 12:04:12.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"85a49184782c36a5117d91422a8691368749970e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-17 12:28:56.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"28fffd6ef60fa848f39f48fbefeb7407cd69bfef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-11-17 12:53:44.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"1d35abcc42ce80f89f024b612e0a4717ee6f2054","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-17 13:04:08.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"1b8221c3353a7c13ea3d40d3809a4740ad26967e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 15:06:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00239d9482144b43836bce6ad1972020\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4bbe0d60ba540ef9123b293f0f6ed79 : SUCCESS in 8m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9bc849f5a8784c879bc69188eb068a98 : SUCCESS in 7m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3d06143ef134850b3d540d83c088c7e : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40738fd3793d4864a679fae1828cc353 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/701fe099520347288cde947f1b1af263 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/664e70d9c221452ca165df39dec5e093 : SUCCESS in 2m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fed9c0da8d1a4a21a638c192f9f298fa : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7e8ee663e0724aecb595808c83b48feb : SUCCESS in 11m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8053dbe7e05d4d3ba42dc48f2076a9be : SUCCESS in 11m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f1ca8ece25e047b39b2e0dbe917e5a06 : SUCCESS in 8m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5d65001793e4bbbb4a5bda65b144948 : SUCCESS in 12m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/68f6db42475c4e93a7f952a420782ae0 : SUCCESS in 11m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfb73016d8ab4e95bada64c50093d73b : SUCCESS in 7m 31s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1750851ecd294c59bad6e462e411a36c : SUCCESS in 58m 00s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/df32f5970ae5402ba050ea74a52298c5 : SUCCESS in 36m 10s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2666d44f035040d790da114efcb1bb6a : SUCCESS in 1h 23m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64f139a9164b497b85adca296466ed2f : SUCCESS in 48m 51s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8541f0e627ad447a92fc18f91e996a56 : SUCCESS in 1h 12m 22s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8c74de85b1747abb0310d6d2a63e1d9 : SUCCESS in 1h 44m 11s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/09f4deec9d4d4dacaca4664149cada32 : SUCCESS in 15m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9141554984ca488fa79b01707eb97b35 : SUCCESS in 9m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/215cbc8abf944a038307ddc859caa4fe : SUCCESS in 19m 19s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4f6b4211d35f400c9b4626206a1ebbd3 : SUCCESS in 14m 37s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a7c35d3ddab499b8703adff15747ac4 : SUCCESS in 27m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2e70a8a2575646109d5442021d12c7c0 : SUCCESS in 20m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82e783834ebf4b148a3fbf9fdd656328 : SUCCESS in 14m 01s","accounts_in_message":[],"_revision_number":4},{"id":"ed61753e57216984f7b228085978b5f0bbc99493","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-11-20 14:12:34.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9568a15a45a4ade74ee151608be8f4f70bf0a5c2","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2026-02-19 15:19:50.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0df5dfe1f0c10c125584cd6bf74c66aa5b6c4fbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2026-04-01 09:15:37.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0c2ce3f50be32aea94e64bc390238d62e32ba77f","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-01 09:37:59.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"6da0bf9e8561da6509e15add220f2eb893e90c88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 11:02:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b525c71319774e28b1c444aefc860699\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b441693755954e12956af1026cc84ad6 : SUCCESS in 9m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/70125e35af39433da997505daa3788cd : SUCCESS in 3m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2e7da8314c242558f08db808008c4b8 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85f788a21f8d4276a9deb816f6d3232b : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b2505190f9c643d6809fcba3afcc3eea : SUCCESS in 3m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/226d1762d3e84b8e8002cdd9a30ab386 : SUCCESS in 6m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ba2feec1ae549748f13474806e40570 : SUCCESS in 2m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/89b6e8cd6f97485a9dcae64290323fcc : SUCCESS in 9m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c8b87e89d1a54bb5abfe16aac21aee69 : SUCCESS in 26m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f6f2e0db4773479ea9ce2f6c30af51fc : SUCCESS in 13m 45s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/73e116e23c06453e85ec8377ca39f0a0 : SUCCESS in 19m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/654f266334d0492db39e9b2659318d73 : SUCCESS in 13m 32s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5d3b9a2fa264462089c5b9cb9dde8037 : SUCCESS in 19m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05364cf6060b4592b266df445edc35f5 : SUCCESS in 20m 36s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d4ea4257eab74a57ba27a83ff0f40201 : SUCCESS in 1h 01m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7fb13795bf744dabb8576fcf9c0c446e : SUCCESS in 59m 47s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/80dca5313db947249205c5b5fb385ded : SUCCESS in 1h 10m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4367933afc944d0da34e9fc2e3612854 : SUCCESS in 1h 02m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6ca41f8f1cae4b99bb4178ff1264f6f6 : SUCCESS in 58m 55s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/756f83d827bd4bf5bf660b8e7cf087cd : SUCCESS in 1h 45m 53s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3fcae66bae0541358ffb30d8f258c996 : SUCCESS in 1h 42m 04s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3044af5c66942a0aefdd53556fbe54e : SUCCESS in 1h 37m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fdcbd92fa4f4da5ad818ec8a0f6ddfe : SUCCESS in 1h 43m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4f711d7733014ae08080ea190fc837dd : SUCCESS in 18m 54s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4577b294f0c14207b10d5000a93ad81a : SUCCESS in 15m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa425f8e609c4208847d17a3477f9ff8 : SUCCESS in 17m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f420606357c547ef9c03d3cb438c853b : SUCCESS in 17m 58s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/02a7d7a899aa4affa7eb7aa82f2f46e9 : SUCCESS in 35m 04s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/342336b4ac0b4ed09ba368c6dbc58206 : SUCCESS in 26m 11s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34b9d807e10e4f60925654e556b4a4b1 : SUCCESS in 30m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bdb642cfe2f47d09715f430a2310f8a : SUCCESS in 27m 48s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/be02db10ab6443d7b02b994c6fee6cb6 : SUCCESS in 18m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97a5dacfbd8640f9a97b6a11d48ad457 : SUCCESS in 15m 19s","accounts_in_message":[],"_revision_number":5},{"id":"ae5c058f82748e692928c4bb6e52a6e27b5cdc77","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-01 12:04:28.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"feaabcdc6c9ca0b4788eaa8792d5a038189474c7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-07 11:00:53.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"033e91fd2edc73d7b73b1a15eb7cbb7240f77ace","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:01:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"83a950f9ce5e70e37764aa401aaf90f62e40d0c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:54:02.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4868c4a33dd4189835d1d4e9bfe0514\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05088acc99394ca3af45dca9d7947ad0 : SUCCESS in 5m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/76a93686c1c64e51bcc5b3caa24e61a5 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc37e1c49a204ab597061dab4bcb1729 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/066f161fb18f4bbd97218be13af54ff0 : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98c6f313287e4732a28f72c713e8b6aa : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/023939c4408b4293a2256415cad37c11 : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/368c58f1426e420198f4688325edced6 : SUCCESS in 11m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e9de8bc0d15646a9b69137850282f64e : SUCCESS in 13m 30s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3222bb1c97164d64ae13fa3c2cabdb40 : SUCCESS in 19m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad975e5dada442069d3dba8c06551aa7 : SUCCESS in 14m 18s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ba8d33fdbab4479f8ccaca6d73f8cb23 : SUCCESS in 23m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7680fd4c2da49d4b692166bcf556223 : SUCCESS in 17m 21s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/bd1b62441d5a47969fa3fa91d809a146 : FAILURE in 40m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bd1fa4427ab049e49a07c71d14f2b1c7 : FAILURE in 41m 51s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7b210d116bc74b5e9b028f2ce0042095 : FAILURE in 48m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca0204b5cc5f4616b7296584bc434e2f : FAILURE in 40m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7291c3b992d74cb8b7282a8848a09599 : FAILURE in 41m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/54553e7c329f45c7ba69c339a9b6fd30 : FAILURE in 16m 00s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/724730d3d9324d99851a09ab6d117e51 : FAILURE in 13m 01s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb1ebf9f6de44880a7d1e39e17406d05 : FAILURE in 13m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7fbf90e6611b4f9fae663dbe41f19daf : FAILURE in 14m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/dce4f7f8098b405786b15a7c5d6f5c77 : FAILURE in 13m 39s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8476288a21194a4cb1bab4228f547f47 : FAILURE in 11m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37a07db99781455893593d121818695e : FAILURE in 17m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/180ff93ff0304532ac51003efe1de293 : FAILURE in 15m 46s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7e2fa24125b740d2be5bfe34fe6dc721 : FAILURE in 13m 45s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9f5017f3cd8841cf97248565fc57220b : FAILURE in 15m 57s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fac6e7763884ca59a4977a937d0cfa8 : FAILURE in 12m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc0562696395472bb61ec401c7988eb2 : FAILURE in 28m 30s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/725ac509c2c74325ba2c332498df1d11 : SUCCESS in 9m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09202a6b31be4971885a747f12f19e6d : SUCCESS in 15m 46s","accounts_in_message":[],"_revision_number":5},{"id":"c83733607e0234e189319933697fe53c5bda66f0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-08 08:14:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f6d8deedd86422e81280178f54a714c8dc650138","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 09:57:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c045aa313b9e4f96984254248e7c214b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/177cd456557149f69ea417cd5a8f6dda : SUCCESS in 9m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fcb79b372f564d1db4bf0027eaa8c914 : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13868139e6cd485ba6851f1133735525 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ab51a829e3342659bd02845fcec18c9 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e9f670a24bcf4235abbba80b0b3dcaf9 : SUCCESS in 4m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9a637b3b600b4b90a4045b2635d524f3 : SUCCESS in 5m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b32a8b36c14643eab38b9405d0ca3abf : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5faaa027fbb044c3afa39a16e397b6ff : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/95aae7df40854ad58c0881dd9e821a66 : SUCCESS in 12m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b9a7d261c1f4496c95ff75011fbf2ccf : SUCCESS in 11m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8c1dabd62b0546f4bafb497dc1ebda15 : SUCCESS in 15m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/360b72b8d1eb445894b8629bce64a163 : SUCCESS in 15m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/12dfea477b90411a9b85d53b8ffdf36c : SUCCESS in 17m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d45ae21e29e4ef0a8557bd419dd19e8 : SUCCESS in 19m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3bccf854a95144a390b89b40856a5da9 : SUCCESS in 1h 01m 51s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/090c0dadc67a437390cbb3b20314ac20 : SUCCESS in 1h 02m 30s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/64dafb4cb9c642c3ab4b0a23e206c7bd : SUCCESS in 1h 04m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e3ac13f13574e83a6745e719d423d82 : SUCCESS in 1h 00m 44s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dbf90d06c00e492fbba9c89036015b62 : FAILURE in 12m 31s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/92ffdab9d3494f36b359e5dfbf79facc : SUCCESS in 1h 41m 27s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1fa3853958c548e68048978ce5923a86 : SUCCESS in 1h 37m 45s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/431cdeac86c04a829cdc4ec746e115cf : SUCCESS in 1h 27m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6673518117a43c49eb9edc5fed1a10b : SUCCESS in 1h 42m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e9920f3dc9604296bf6ecda5c3d34354 : SUCCESS in 17m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/fc8e98f35c6643969e5a75498aaf6a95 : SUCCESS in 12m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb3e1c5fdd3849238a88b6feb54e0718 : SUCCESS in 19m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6106af206e2641819ac28325d9675f8c : SUCCESS in 14m 04s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/11cbd54f8cb94fbfb2734a4f3d710f14 : SUCCESS in 37m 08s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1f894335042c41efaa24efb484d6a6dd : SUCCESS in 32m 08s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d938eae45f9b45c298bc78bcf06ae1ea : SUCCESS in 17m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73f46f1b849e4a9baee1bafc41f834b6 : SUCCESS in 28m 15s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4d102141d42844089a19858247b4434e : SUCCESS in 11m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7693012c45747289ad6c6fac5dc9033 : SUCCESS in 16m 25s","accounts_in_message":[],"_revision_number":5},{"id":"e7e280d4ec28ca7c6e6f589ec775bf351d80acf0","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-08 11:47:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0fff43a2424a5f0aed8d9342cf4e6aee962481c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 13:46:35.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a13d53a548f45eea8a29f533db859c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9320bc912fdf46488c2f35335b77da3f : SUCCESS in 6m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0c5bc6fcc6014e4aa699e91ac1851ca4 : SUCCESS in 6m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5a1396f51ee4fcf94381770cca04019 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16fdf88b79e0477b927ac9ec526e0dee : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb195b056ac6462f8551905e2fb9fa7d : SUCCESS in 3m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f5a9be5c24b144079e02e64038f50383 : SUCCESS in 6m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3b9b181a2ab43bbb33ef81b8fdd4b91 : SUCCESS in 2m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/932530a7cf714661a6ee0cb5924c4016 : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e80c20857500410d9e475eb799a8f8a6 : SUCCESS in 7m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/28ffb3104b9649c5a674c0c85270c578 : SUCCESS in 13m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/189439510571411aa43795121e6c3307 : SUCCESS in 11m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b62ed820d9c48eaa3ded502e724327e : SUCCESS in 13m 21s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a162d2ae1a0a4dcead8b7d290759934e : SUCCESS in 9m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/129a16cc227440fcab88206434e8c567 : SUCCESS in 15m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8d9530f5da1c4c749c7d93a7bc1509d3 : SUCCESS in 1h 00m 08s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/30be1cd88623489ab578b67468328333 : FAILURE in 42m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/361e8b16f2fb4c3dac66bcb68ce1033a : FAILURE in 48m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/395a013ac7fd42f6883d7b92844a3738 : SUCCESS in 1h 00m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/19a794db858843cb8afedaee1941b080 : SUCCESS in 1h 00m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f56afb8c7f6f4a8286ed3a8918901f7b : SUCCESS in 1h 57m 12s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e072164696234befae7b6611fc26db66 : SUCCESS in 1h 44m 39s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a32213d87994426a84748802f6ca685b : SUCCESS in 1h 36m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21cce5b241544b628c432b4849e96558 : SUCCESS in 1h 57m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/109b9dc7847f4bf799b887e24d4b97c9 : FAILURE in 17m 07s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/39f8b93b4a0d401ea083b653edfa45f9 : FAILURE in 18m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f27c68eab2f442cbdaa6d072d93d666 : FAILURE in 19m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1c076c4346564143a79fea0bc2556798 : SUCCESS in 18m 46s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b52ace5fba434e8f9a03be3a3671c9c1 : FAILURE in 23m 33s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7f28dcd2be364b018f4930f4b8a692e2 : FAILURE in 29m 23s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/085836ea239744c98ed9b8713f3ac21c : FAILURE in 21m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ca0f4f832004ddabaa955058eceeda1 : FAILURE in 27m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/40e9aa4b151a45df8aff4ba6a6823185 : SUCCESS in 17m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f50bcc1e0a5446980692320296d28d7 : SUCCESS in 9m 22s","accounts_in_message":[],"_revision_number":5},{"id":"c9a8069e8676f53c874467a0b11993ce3828cdaf","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-08 14:23:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ae511901a071ab2bb988806ea7b183ab7c0d3c37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 16:01:31.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5587851d15824492b3a507483eba425f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/879614f3f4324ec8882c9d104b296ce1 : SUCCESS in 8m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7a81304eceed4d2f8a4366542d4b8b10 : SUCCESS in 4m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ea2c532e9f4444f84e0a8a003c0636a : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2269e0c02c204bb38ba051ff4fb26167 : SUCCESS in 6m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/965edca4a24948aca906da23c785aed2 : SUCCESS in 2m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a7f78c2d29bb405c96da08bde63f4eef : SUCCESS in 7m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf2a4c265cfe4136a872825cc83d38d4 : SUCCESS in 2m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/00b1fcf1063a46a68b2ce1a5afac131f : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8be493fdd51b437388f6ec0fc4277bab : SUCCESS in 12m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3a4b5ab357bf4f4591977c590fd74aaa : SUCCESS in 6m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c57ded6dc0d249eeba05a8d0de9808f5 : SUCCESS in 9m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b25669cd4ff4ecb92ea710192afe596 : SUCCESS in 13m 55s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/45aeef146ceb488086a39cc2b20d6610 : SUCCESS in 18m 12s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c89859d24bef473bb4158ba8c179b3a9 : SUCCESS in 9m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6065cfbfcc8b4a139572d8885d417815 : SUCCESS in 1h 00m 42s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ffdb6bff052347519a4b32989ddf18f4 : SUCCESS in 57m 28s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/269356b9812b4227b83d41ca38ae8b35 : SUCCESS in 1h 02m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abaae38ee3f746bab339395c0c8262b1 : SUCCESS in 57m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3196db116b2b410d9bb54e22713ce529 : SUCCESS in 59m 06s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/295b094ab11346608952e038e441b643 : FAILURE in 1h 28m 31s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b171797c3f05458daca112bc665893d5 : SUCCESS in 1h 29m 59s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9373fe0927814e198a446e0e80d73b28 : SUCCESS in 1h 32m 03s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bf06af17a7d445c9bbe0d63782f6287 : SUCCESS in 1h 35m 35s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8a86b5f7aeae4335a0b6f72fd22d8665 : SUCCESS in 18m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/53864d869d354e9c9153f30fbe046402 : SUCCESS in 16m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79bd64fd89ee42eb974abfebb2aaa4e6 : SUCCESS in 10m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fefb9d23b44d43a9aac85dfb52cf9fa7 : SUCCESS in 18m 16s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ee14054951ca475ea47206375103ad95 : FAILURE in 17m 12s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/28661b11309d43eeaf70a62e4c9d043e : FAILURE in 26m 52s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3079b2a3b0494efc8ca703bfa8b4b748 : SUCCESS in 34m 14s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0543fdf4baf457b8074ba19c8e963c8 : SUCCESS in 30m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a56620f640bd42a2857216c4f7e98570 : SUCCESS in 17m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8ac5f0b865d4f4b88cb6ba3a3485b78 : SUCCESS in 14m 52s","accounts_in_message":[],"_revision_number":5},{"id":"37017ec1dace20f5ab1c73327837d79fd760b5fc","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-08 16:46:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"51766cae6732ab48dde9b842b2d81263e704f9f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 18:32:00.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec29753745e149debfdd10c2f38099a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aebd83f50784ed2a4cf2ffbc631e2f3 : SUCCESS in 5m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0d609a5fa9dc4aa9b69f1a3d98f692d9 : SUCCESS in 3m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4027011426e48ebbcf19689ec3f84f9 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ad9bb202da144f0aa00ea1131574279 : SUCCESS in 5m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63cfde87a6e5476991f318149e370afb : SUCCESS in 2m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/35e4ec6e5b1e4e32ba15c137e7f98551 : SUCCESS in 5m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbdc2f5e144b45b486448069f13f393e : SUCCESS in 4m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/19a502422a8541c987b82245ba21b1c8 : SUCCESS in 5m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b55005b4172345888b276524d188c012 : SUCCESS in 12m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/076358fe2af94a87990fb701a92f77ab : SUCCESS in 10m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/43003c51612a46f4947c705312f91ee9 : SUCCESS in 9m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbb36b940f5b4ee99afa5d0c7b6a7002 : SUCCESS in 14m 46s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8540643019a242abae511b3c3c82339a : SUCCESS in 22m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2dff44c01ee64553b2187568d018a0c5 : SUCCESS in 10m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9b17d022888f4893bd00894ae0be03c5 : SUCCESS in 1h 07m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f83c9b88f5eb4848b7019873c37199e2 : SUCCESS in 1h 00m 15s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/18b6da490af24e29ab4d0dc7bfa471b7 : SUCCESS in 1h 09m 58s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6aa0aa79a2ea4d92af6ed14cc82e9e57 : SUCCESS in 1h 07m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5d85112dcfa7444e8f389c85f8824dc2 : SUCCESS in 1h 10m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/84fd9f0aee3144f2abb928f039140fc6 : SUCCESS in 1h 44m 04s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/510c19a9621e479989583b9a84fb6983 : SUCCESS in 1h 35m 21s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a608c44ed0cc4084a0a4585526ee8c8a : SUCCESS in 1h 39m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f826289ee62a4daab6c3452ddd088696 : SUCCESS in 1h 37m 52s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/aa4d82360e624119a7c53e8270c4882b : FAILURE in 13m 11s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e694d2bc98954759ae554a4fe3bdeb75 : FAILURE in 12m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e58d550b55da468eb8d15455f6d6fd91 : FAILURE in 15m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/248ce3f411944f40a92532f77213cfef : SUCCESS in 18m 36s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/92252a34b5d4443493601eaf4e4c0222 : FAILURE in 12m 54s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f800d5f3e2884ec0a74db186144a31a6 : FAILURE in 17m 59s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3677371a7ab9462d91a06fc8595b1392 : FAILURE in 33m 34s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0bceef3106f4834bddf488492588feb : FAILURE in 21m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ac924c18e8cf40b288345456cbbf3d77 : SUCCESS in 17m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f8d56eae66a440585c1b68151437157 : FAILURE in 16m 40s","accounts_in_message":[],"_revision_number":5},{"id":"1ec15a3cc5b9c21df5ebc70ce80bff6c6d8736b0","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-09 08:36:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6fbb18c12a8007aeba61d0d4a13b53d61670cb4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:34:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48797ce9a19b42418e3e01c76e8bea32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5c875c9127940b495ec1d9a3990a722 : SUCCESS in 5m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/45cbbacb6c0e41fe97f68b10a17921cd : SUCCESS in 3m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e133015ea26944fb9f06ed6350ee4d8a : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a48b10e220e84ed4bb595dae5e18ad16 : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c3789c4eac24581963f3cb228d5a63b : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/70ace9ab1e4043faae2a09377aae149d : SUCCESS in 6m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d609c89637514b92a5224dcace654c33 : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/359a89173550461980526dc1c931b5cf : SUCCESS in 3m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/165e7938ddb24da3bab26c32796fb573 : SUCCESS in 9m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/16c43eda37b74edea56fb7b8263530ed : SUCCESS in 7m 03s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/48d46f0b0b1c48b892108c7fc6bb31fd : SUCCESS in 21m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39867dba58da4ffd9da9d6a0c99e19e9 : SUCCESS in 14m 37s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8a28ab1c36784b8f889d86666fcb74fd : SUCCESS in 18m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e2305a33c604c5fab34ed97aaf858b7 : SUCCESS in 17m 15s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1340dd8621674bf7951b7ad1cbf833a2 : SUCCESS in 59m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a9afcceeb58f421390b4440f6fb0ba39 : SUCCESS in 1h 03m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/54563061270f43139cef10fc0602edf2 : SUCCESS in 1h 03m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8a8054c2c704e58970a0471c50db584 : SUCCESS in 1h 00m 40s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/531c9e0a745840dbaaad8c8a6898468b : SUCCESS in 1h 04m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/00e8f0438ec544ad905b75d4b323365b : SUCCESS in 1h 54m 07s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/83791a11b28746c285df8ac4d2fb3ced : SUCCESS in 1h 37m 47s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/834b5eb382cb4c80b0fa74d83e246869 : SUCCESS in 1h 30m 04s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fafbfaa19926415bb4252e2bc9d68bc0 : SUCCESS in 1h 42m 52s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9e96e953367a47f189a5360f6e8c5a08 : SUCCESS in 10m 21s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5b6507b420c64394b67bf0f8d6adc160 : SUCCESS in 16m 34s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee2844afe0d540398cd592c225ac7347 : SUCCESS in 22m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/44c109f09ccf412d9ef2002261da5003 : SUCCESS in 17m 14s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cbddd8df61724b34ab16b8b268f8b75b : SUCCESS in 35m 48s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7bb17f1a3dea40718b91206e180b49b5 : SUCCESS in 25m 18s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98076cb48dfe4765b8d81ed87a70467b : SUCCESS in 29m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3a23dd3ecdc4d1983a3541eb3f29d5e : SUCCESS in 28m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/93a58ae622364e859652e262b90eeee4 : SUCCESS in 20m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9df608613ec47cda81062e9dda5be1a : SUCCESS in 16m 58s","accounts_in_message":[],"_revision_number":5},{"id":"41cc4857cce29833bf3e33270b9b86586e871c6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:34:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"08715dc4a7c2634e8bca48f2158d9356a101277f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 12:43:57.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6279cd29bcc4c8ca203a49e20aaab43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01a9e429ab5e4332b1c606ee4cfa7061 : SUCCESS in 7m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c125232a8f234e7a9d1c5cd7588598b4 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58254b0304e941dc93dca46563f8b4ce : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d274f3ed3bac4f2db21a61d0ba0269d6 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b91b2b43a7948a99590bdab40942213 : SUCCESS in 2m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bb8d43fb1a5049fa9ed791f07694e717 : SUCCESS in 3m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ceb5d191d25e412b889ee161fa0b086e : SUCCESS in 11m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/be779fe96b0b43e2907ecdc4fa972a0f : SUCCESS in 11m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fd9b611dfeb549c0945778d4fa326ad1 : SUCCESS in 18m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5043db97599849b3b845d713408b0843 : SUCCESS in 19m 11s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/83a23284232a4370af0ed0e7decec27c : SUCCESS in 11m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2242c3494914505b6df3875580c0121 : SUCCESS in 12m 37s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e0cae2afc3f445b19d6217834b3e4758 : SUCCESS in 1h 05m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/49da57495a7e4e82897ad58ce29fff00 : SUCCESS in 1h 05m 45s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/23b34f90e5a34bb4900aaaeec6e25c34 : SUCCESS in 1h 14m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e130534efb346a2b3e1d8b6882db291 : SUCCESS in 1h 03m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/41b1c7a02b22490f82699b701627d1d6 : SUCCESS in 1h 11m 19s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b598b0b1f1d34c1ab5d0d9fef6f21125 : SUCCESS in 1h 44m 41s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2754704566954f869279e3833a139f15 : SUCCESS in 1h 47m 37s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b9e2235176d4813b718be69533d0725 : SUCCESS in 1h 31m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06a7fda697a843f5aaeece98e4e31f19 : SUCCESS in 1h 43m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c38dceb60d5a421389b8a10ce11bcc72 : SUCCESS in 20m 40s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/48ec42f701d1407cb2b55f7884dd51fd : SUCCESS in 12m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19a720eefa684e219a73979186374003 : SUCCESS in 20m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f16ab66bafa643119bca4179050fa2c9 : SUCCESS in 22m 12s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bfd4410742ef4380bbea2743ae03f2d0 : SUCCESS in 19m 51s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/69ab655aad8942ceb9d546900c427911 : SUCCESS in 35m 08s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b99481720a864cbca03cc9c81a452f94 : SUCCESS in 29m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d2a23cd027a49cdb9ab692cc71279ba : SUCCESS in 28m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9911e2da031b4bef977c978316b2d61a : SUCCESS in 13m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/556631f843054609a80aec1e1dbf1121 : SUCCESS in 16m 01s","accounts_in_message":[],"_revision_number":5},{"id":"2940bf111380173b1c98e406deeb1cea4512996d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 12:43:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"f38bbab189d0ff90ee60b0579da8b3f8c35b5aac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 12:44:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cf1382c33ab4ecbaddefff936cb05d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea120b46c3c244ca83f7806eb63bf270 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d0ecc65207e4f75bc267f00fe174651 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6fb103fd5dec1a7782af2a440c47df3b3c110fdc","revisions":{"f4bac9320de8036a22e86dba54d5678bddbaa1fe":{"kind":"REWORK","_number":1,"created":"2025-06-25 14:09:08.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/16/953316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/953316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/953316/1"}}},"commit":{"parents":[{"commit":"9fdae13f403955e2ac4c5abdbb9c8d7912fc7124","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fdae13f403955e2ac4c5abdbb9c8d7912fc7124"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-25 13:01:34.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-25 14:09:56.000000000","tz":60},"subject":"Add custom_etc_hosts_entries variable","message":"Add custom_etc_hosts_entries variable\n\n``custom_etc_hosts_entries`` is a new variable which provides a generic\nway to add entries to ``/etc/hosts``. It differs from\n``etc_hosts_hosts`` because it is empty by default and can be used to\nadd hosts that are not in the Kayobe inventory.\n\n``custom_etc_hosts_entries`` is a dictionary where each key is a\nhostname and each value is an IP address.\n\nChange-Id: I74c8234f77115d6a478559a53ad86534eb12e5f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4bac9320de8036a22e86dba54d5678bddbaa1fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4bac9320de8036a22e86dba54d5678bddbaa1fe"}]},"branch":"refs/heads/master"},"a50f467fd3ff68914497f05cf417187747097921":{"kind":"REWORK","_number":2,"created":"2025-06-26 07:57:34.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/16/953316/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/953316/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953316/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953316/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953316/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/953316/2"}}},"commit":{"parents":[{"commit":"9fdae13f403955e2ac4c5abdbb9c8d7912fc7124","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fdae13f403955e2ac4c5abdbb9c8d7912fc7124"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-25 13:01:34.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-26 07:58:43.000000000","tz":60},"subject":"Add custom_etc_hosts_entries variable","message":"Add custom_etc_hosts_entries variable\n\n``custom_etc_hosts_entries`` is a new variable which provides a generic\nway to add entries to ``/etc/hosts``. It differs from\n``etc_hosts_hosts`` because it is empty by default and can be used to\nadd hosts that are not in the Kayobe inventory.\n\n``custom_etc_hosts_entries`` is a dictionary where each key is a\nhostname and each value is an IP address.\n\nChange-Id: I74c8234f77115d6a478559a53ad86534eb12e5f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a50f467fd3ff68914497f05cf417187747097921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a50f467fd3ff68914497f05cf417187747097921"}]},"branch":"refs/heads/master"},"d547b715d2c7ce1f7174c8be9656058f2ad51f2b":{"kind":"REWORK","_number":3,"created":"2025-06-26 10:40:03.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/16/953316/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/953316/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953316/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953316/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953316/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/953316/3"}}},"commit":{"parents":[{"commit":"9fdae13f403955e2ac4c5abdbb9c8d7912fc7124","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fdae13f403955e2ac4c5abdbb9c8d7912fc7124"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-25 13:01:34.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-26 10:41:18.000000000","tz":60},"subject":"Add custom_etc_hosts_entries variable","message":"Add custom_etc_hosts_entries variable\n\n``custom_etc_hosts_entries`` is a new variable which provides a generic\nway to add entries to ``/etc/hosts``. It differs from\n``etc_hosts_hosts`` because it is empty by default and can be used to\nadd hosts that are not in the Kayobe inventory.\n\n``custom_etc_hosts_entries`` is a dictionary where each key is a\nhostname and each value is an IP address.\n\nChange-Id: I74c8234f77115d6a478559a53ad86534eb12e5f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d547b715d2c7ce1f7174c8be9656058f2ad51f2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d547b715d2c7ce1f7174c8be9656058f2ad51f2b"}]},"branch":"refs/heads/master"},"19ce77bddc1a246953aecd8c7ffc339b2b5233cd":{"kind":"REWORK","_number":4,"created":"2025-11-17 12:53:44.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/16/953316/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/953316/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953316/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953316/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953316/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/953316/4"}}},"commit":{"parents":[{"commit":"8bb8a44c989ed6e5d1aa2d7f261adc1646c1dd60","subject":"Merge \"Fix comment in globals.yml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8bb8a44c989ed6e5d1aa2d7f261adc1646c1dd60"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-25 13:01:34.000000000","tz":60},"committer":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2025-11-17 12:53:36.000000000","tz":0},"subject":"Add custom_etc_hosts_entries variable","message":"Add custom_etc_hosts_entries variable\n\n``custom_etc_hosts_entries`` is a new variable which provides a generic\nway to add entries to ``/etc/hosts``. It differs from\n``etc_hosts_hosts`` because it is empty by default and can be used to\nadd hosts that are not in the Kayobe inventory.\n\n``custom_etc_hosts_entries`` is a dictionary where each key is a\nhostname and each value is an IP address.\n\nChange-Id: I74c8234f77115d6a478559a53ad86534eb12e5f4\nSigned-off-by: Alex Welsh \u003calex@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/19ce77bddc1a246953aecd8c7ffc339b2b5233cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/19ce77bddc1a246953aecd8c7ffc339b2b5233cd"}]},"branch":"refs/heads/master"},"6fb103fd5dec1a7782af2a440c47df3b3c110fdc":{"kind":"REWORK","_number":5,"created":"2026-04-01 09:15:37.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/16/953316/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/953316/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953316/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953316/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953316/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/953316/5"}}},"commit":{"parents":[{"commit":"8b8ac1c293fb48b5c01fc294bfb0f6e2b313737c","subject":"Add nmstate engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b8ac1c293fb48b5c01fc294bfb0f6e2b313737c"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-25 13:01:34.000000000","tz":60},"committer":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2026-04-01 09:15:28.000000000","tz":60},"subject":"Add custom_etc_hosts_entries variable","message":"Add custom_etc_hosts_entries variable\n\n``custom_etc_hosts_entries`` is a new variable which provides a generic\nway to add entries to ``/etc/hosts``. It differs from\n``etc_hosts_hosts`` because it is empty by default and can be used to\nadd hosts that are not in the Kayobe inventory.\n\n``custom_etc_hosts_entries`` is a dictionary where each key is a\nhostname and each value is an IP address.\n\nChange-Id: I74c8234f77115d6a478559a53ad86534eb12e5f4\nSigned-off-by: Alex Welsh \u003calex@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6fb103fd5dec1a7782af2a440c47df3b3c110fdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6fb103fd5dec1a7782af2a440c47df3b3c110fdc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
