)]}'
{"id":"openstack%2Fkayobe~953616","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~Ib27428f89b0e3f2865ab1c498f05de741fa49587","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-06-30 08:33:11.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-30 10:44:57.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-06-30 10:44:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib27428f89b0e3f2865ab1c498f05de741fa49587","subject":"Support empty strings in dev-tools package lists","status":"MERGED","created":"2025-06-27 16:23:05.000000000","updated":"2025-06-30 10:45:53.000000000","submitted":"2025-06-30 10:44:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953616","cherry_pick_of_change":953274,"cherry_pick_of_patch_set":2,"meta_rev_id":"6c70e4353c8587730f6c9bc3d470cb22f91a85f4","_number":953616,"virtual_id_number":953616,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-30 10:44:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-06-30 08:33:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-06-27 18:30:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-06-30 08:33:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-27 18:30:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-27 18:49:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-30 07:24:41.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-06-30 07:24:41.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"c4d201b790928093cb4030f6dd2a399dffcfaed9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-27 16:23:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"76a4bc5e6191ae6d749a468671bb7e73ca63a2c2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-27 18:30:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0488c83fdd50663e67a6ee5d63c0035e1b49db2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 18:49:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28e294626d7747fdb230b376db3d3fd1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dae17850130449c7afb94131948cbc92 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d3df6b9c59d45b087615ac17cc5a9ed : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6b96ba4f3e7a4bd992e2a3d99dbb66c3 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa2305a79296415c9c8212acb988d2f5 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5ac5e1bfe07405e97d4a3ddbf3d6ff4 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3f8bcf2f3e7a42d6bb058d56bf31632b : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e227449a66cb4f6c83b6be264e94a745 : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ff85eb387cd44433bc5f8dd40f0dec2c : SUCCESS in 10m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/328f45d4390b42f2b8843b991d318f20 : SUCCESS in 53m 07s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/fdf7840488854887b22541da59ea20e5 : SUCCESS in 1h 03m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f4ddb094b8448f0be54a30f3ea10089 : SUCCESS in 58m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/16dad1451a6f46eab9a501de8d5be5c2 : SUCCESS in 35m 28s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/01ddb3f174774b0dbb8bc7c8b25b175d : SUCCESS in 40m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fcae5ddaf4094827a1ab7b34c277e685 : SUCCESS in 16m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f2f352c098c4404811cdeedc0c37242 : SUCCESS in 8m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a2fe90935c8e4c169a580b4e62311424 : SUCCESS in 1h 28m 17s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4b6f50bc06140879da0df0d651a7abe : SUCCESS in 1h 23m 14s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1f2ef294b72a4017ac5cf65205bdeb3f : SUCCESS in 1h 58m 56s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2a287aee3c248d78b90137638cf7ea5 : SUCCESS in 2h 05m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4a88ea90367b49ec8ba02e020bd1b804 : SUCCESS in 15m 59s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/45a37713a76243bd8bd28c4802249221 : SUCCESS in 16m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b39d6f99c68463d80e895f7240d652d : SUCCESS in 16m 20s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a6784bce3e224bc99d08004804d55eae : SUCCESS in 14m 26s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e43704814a4a48429a76280b83c42c63 : SUCCESS in 36m 04s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/eddbf101b92440128bd390f45ca381a1 : SUCCESS in 31m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71909cc77ac74642ab76843bf7fb73f9 : SUCCESS in 32m 13s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f35aefd83936442094cf90ec33f95d8a : SUCCESS in 33m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2d75dd911cba46a3bcc94414a8d535e7 : SUCCESS in 13m 41s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/35d078476b0f4cc3a99bec370f4ae35c : SUCCESS in 13m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f223eb8dd56d4615bdb37c7b74a5268a : SUCCESS in 11m 53s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/8f6414c181964c63999d310d7cb6d91e : SUCCESS in 12m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dd0afc5a26b4470587ae1465a5a7a981 : SUCCESS in 9m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fedaaaaf650347c5878b4053b3ebb61e : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d94f9db49b3bd3fdccce6f144c07b5c6d61f0533","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-06-30 08:33:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e7b6d67566471e3c4cc936cd66f8ad7df2de5866","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 08:33:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89d37a7689bdc9e6291fea55625b0351f5312af5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 10:44:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94b303c53b0f44c099eb7de5105b74ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c98b58ca89845a9bfa369092901b8d9 : SUCCESS in 2m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e297bed4eefd444c8e49df9d26f6b782 : SUCCESS in 2m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb2cc58f5a514b49ba8f634ec1c2a65d : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ca1ff0714b84f12945757d618158941 : SUCCESS in 2m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1066f8d16e304430a60ea924e8a1d3bc : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/96e21c824e754d1d8a31ad37a7475f2c : SUCCESS in 5m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4994e3bff0f5483d85ed7f4d2de79c77 : SUCCESS in 8m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7e39be21ea864f05be65c42a56835518 : SUCCESS in 38m 04s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/d33ab428b8124103a99bd58f5a56c2e5 : SUCCESS in 1h 03m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d15fd3fa8e547a4893c9aad8a881d00 : SUCCESS in 1h 08m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/81e2e9f959fc42d498526f527fb11916 : SUCCESS in 1h 05m 07s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/105495fe019d4dc8bc58cd32e687e49e : SUCCESS in 1h 04m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b728912ba45a4ba6b6bf0e485211e2ef : SUCCESS in 16m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b20487e9e7e44711b0e5b88bacd1dca4 : SUCCESS in 14m 09s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ddb8d9bbb901451c8f0c17c68f5b3fcb : SUCCESS in 2h 04m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f92846884cb43ba94cb2059297e0f12 : SUCCESS in 1h 52m 30s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/77d9cf383928483d9972769d7b9f6938 : SUCCESS in 1h 51m 42s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/594a33bc5ad5401299d52979a75bdc82 : SUCCESS in 2h 03m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a6ca909574bf46ae95dd41c59e4158c2 : SUCCESS in 16m 49s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/81ef1904d6094ffe9c5e076643459312 : SUCCESS in 17m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c55f9719171448489f3412d89047de0 : SUCCESS in 13m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4e197ce11c5344ca892fe4192e8ffcfe : SUCCESS in 14m 46s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d89efb35b9674b0790c5d60f2ceb8c7d : SUCCESS in 29m 20s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6216d7e314e34e48aab3bcfc0efd9846 : SUCCESS in 23m 37s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/212ae4e42a5644d7a36c85eca5c285df : SUCCESS in 20m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ff3c86915f2d4a638b5777f3b8dce7d7 : SUCCESS in 8m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2deeb02000e24a1383eeed9d64cce929 : SUCCESS in 11m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/aabe1fbf8a40429d95a4a7cc2188b20b : SUCCESS in 11m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3be74d39e614792860c0a5490f669e8 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"4fe82469908c2d7c04a88243fabcf490693021bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 10:44:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c70e4353c8587730f6c9bc3d470cb22f91a85f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 10:45:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88534b445af6499cb6abb0cdd86acd24\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c8fe3c1a3404b748e6ee72d3f5258b5 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b56393c37a2c44adb31df8615c3efd08 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9136fef2e58454a358ec3233e256f0707efde53","revisions":{"a9136fef2e58454a358ec3233e256f0707efde53":{"kind":"REWORK","_number":1,"created":"2025-06-27 16:23:05.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/953616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/953616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/953616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/953616/1"}}},"commit":{"parents":[{"commit":"fac4e78a99a0873a159b37913af2f0834c02798c","subject":"Bump stackhpc.linux to 1.3.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fac4e78a99a0873a159b37913af2f0834c02798c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-25 06:29:58.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-27 16:23:05.000000000","tz":0},"subject":"Support empty strings in dev-tools package lists","message":"Support empty strings in dev-tools package lists\n\nCloses-Bug: #2115000\nChange-Id: Ib27428f89b0e3f2865ab1c498f05de741fa49587\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 36e7a7b0d0441413913406a68d886248838f794c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9136fef2e58454a358ec3233e256f0707efde53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9136fef2e58454a358ec3233e256f0707efde53"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
