)]}'
{"id":"openstack%2Fkayobe~953617","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~Ib27428f89b0e3f2865ab1c498f05de741fa49587","project":"openstack/kayobe","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-06-30 08:33:19.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-30 10:45:00.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-06-30 10:45:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib27428f89b0e3f2865ab1c498f05de741fa49587","subject":"Support empty strings in dev-tools package lists","status":"MERGED","created":"2025-06-27 16:24:17.000000000","updated":"2025-06-30 10:45:56.000000000","submitted":"2025-06-30 10:45:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953617","cherry_pick_of_change":953274,"cherry_pick_of_patch_set":2,"meta_rev_id":"093dfdf6abd4e8acbd53da4457f5027f6295e3f2","_number":953617,"virtual_id_number":953617,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-30 10:45:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-06-30 08:33:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-06-27 21:31:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"},{"value":1,"date":"2025-06-27 18:30:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-06-30 08:33:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-27 18:30:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-27 18:30:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-27 21:31:31.000000000","updated_by":{"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"},"reviewer":{"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"},"state":"REVIEWER"},{"updated":"2025-06-30 07:24:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-06-30 07:24:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"0c278a4d940eae08790df8f2c98c2979af125a56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-27 16:24:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b863953af5f70d53265e6758129efaf2f353a5a1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-27 18:30:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"651d7d681c2c1be29eea334a71991c5b434e08eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 18:30:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/192ee451eab345588f1d58daff2c22ef\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48218a06146043c894f5ebe8eb330432 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9132f8787a5c4e65b3f6034fba7d64e8 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f6c9e7e41564218b57d7691bcb1fb06 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d85fc052464d4c2ab53e3ea3b5a81ff6 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f78b1b1de1b4e8c929324c18de19c36 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdbb8088364d463f8889d14a59daad08 : SUCCESS in 2m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c738699b0dd4019aaff0b96b2c385ec : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2751be6c465f4f3d972be49ec0ed096d : SUCCESS in 5m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5e0292215e744f25aee3abce9f043da8 : SUCCESS in 11m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/92ceb977547d42fb94b7fba08506ec5b : SUCCESS in 1h 09m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/bc665241172344568751287cfec0b359 : SUCCESS in 1h 06m 02s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12e7dd85ec81448e81e9988916db36cc : SUCCESS in 1h 12m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d85d1eeb5e83419190ddfcb56f19e60b : SUCCESS in 1h 08m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4199bbca6c0a4840869bc9a83e5da762 : SUCCESS in 17m 38s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a70b3b6f957743ef8ec26ac68d37e38c : SUCCESS in 18m 08s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/677f5dfaf93c4df2be3769b7b5dd58eb : SUCCESS in 18m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bcadee8934dd4f46aa763c5a760e7849 : SUCCESS in 1h 56m 00s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/874de5d3715e4f179d27e16ac1bc6f5d : SUCCESS in 1h 50m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2b3ac7e32f914a86aedb3c5af736d108 : SUCCESS in 24m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edcf86b41b4b420c8dd22d62f961b7c9 : SUCCESS in 21m 20s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/366d534a8ee84e3fb51b208cb0a400df : SUCCESS in 39m 19s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/19f3e1010555438fa753210eb2f7de17 : SUCCESS in 34m 47s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ce1e3e198a64e99ae15387ae562dac1 : SUCCESS in 34m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2753fc6161b14321921b222f88ab41ab : SUCCESS in 15m 39s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/679f31c2dece4212b7904b99c2cffd78 : SUCCESS in 17m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce619e8dca43465281aac4d6cf2f2fb2 : SUCCESS in 15m 47s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/794d4a715ce7452c9f553ace4106718a : SUCCESS in 12m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/514874745d514667b4d63cffc8f58893 : SUCCESS in 22m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b1a6be563364713a6e7df6172877e12 : SUCCESS in 16m 53s","accounts_in_message":[],"_revision_number":1},{"id":"fb2a3313a6c8d31580e2925012c126d0e05a00ba","author":{"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"},"date":"2025-06-27 21:31:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c5431eba70de01e4ca1911457f2cd354992afe3a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-06-30 08:33:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b19bac227c34a40ef400766e87deaef1387387e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 08:33:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be06d5e400e6872e8d53cd1865e199d775c8dccf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 10:45:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06a6cbb562644161af287ecb0f0b4eb5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d075b63b7b548de89b26f96b888d741 : SUCCESS in 2m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/48d54bcd8961406185c92b3ea25045bb : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df64f30547db4accbc8658ac71a4810b : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c728494534ab40c9a491972ec542fcdb : SUCCESS in 2m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ba091f8de893428abb35cbedcb0425e9 : SUCCESS in 3m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/deeb8ebfa9e04424a3cf900b7a05e41f : SUCCESS in 9m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eff81052ad2449f297d3a51baf6632ea : SUCCESS in 7m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/34d33f9f2da64e6695a122bbfafb1af2 : SUCCESS in 1h 06m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2503547031384a0cad5083eba3d41cf2 : SUCCESS in 36m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7027ee8a16dd455a8ebdf01024b23c94 : SUCCESS in 1h 06m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b3d29c9433744129a43d2b4251aefd74 : SUCCESS in 15m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebcab608b8304cda80939ec9b8ba2b8a : SUCCESS in 11m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/59765106518a454eb5990db1fba3c03b : SUCCESS in 1h 57m 16s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44dbcea055d046058c84eb9393cb417d : SUCCESS in 55m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f8e032e11b304a71bbf780aa69f1946c : SUCCESS in 21m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c18d020ce4c4ead80c13a870777cfc0 : SUCCESS in 16m 39s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bbb3c6592bbb45249206bf115da5a444 : SUCCESS in 27m 11s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c47caade3234f7b844d22e318393fe2 : SUCCESS in 34m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9c2c4c7090aa4f0f841ca18f157e7521 : SUCCESS in 17m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/771b128a8ed149faa2f5700e633a75ac : SUCCESS in 10m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/187b4dd2afb8482a815d34e60fba830b : SUCCESS in 13m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d6e8ceebdbb4be2af01c265189517d7 : SUCCESS in 14m 26s","accounts_in_message":[],"_revision_number":1},{"id":"424bff221a5436db0e1a8302d4191af0d740c402","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 10:45:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"093dfdf6abd4e8acbd53da4457f5027f6295e3f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 10:45:56.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a3fe200f48d4914bb2c7b213046d1bf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31f3b7a149b14adc88658c45ec1102cf : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fffcfdf9febf4c36983ec354c9424dc2 : FAILURE in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"471b2056dc968dbe3ef5f80012a685fa6f447740","revisions":{"471b2056dc968dbe3ef5f80012a685fa6f447740":{"kind":"REWORK","_number":1,"created":"2025-06-27 16:24:17.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/17/953617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/17/953617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/953617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/953617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/953617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/17/953617/1"}}},"commit":{"parents":[{"commit":"73aa2587cb00396bb487aedd49c5943e2be48b42","subject":"Bump stackhpc.linux to 1.3.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/73aa2587cb00396bb487aedd49c5943e2be48b42"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-25 06:29:58.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-27 16:24:17.000000000","tz":0},"subject":"Support empty strings in dev-tools package lists","message":"Support empty strings in dev-tools package lists\n\nCloses-Bug: #2115000\nChange-Id: Ib27428f89b0e3f2865ab1c498f05de741fa49587\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 36e7a7b0d0441413913406a68d886248838f794c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/471b2056dc968dbe3ef5f80012a685fa6f447740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/471b2056dc968dbe3ef5f80012a685fa6f447740"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
