)]}'
{"id":"openstack%2Fkayobe~953660","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I3937cb29143f3b9965032c186e8086854de7fbef","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-08-24 18:53:31.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-08-24 18:53:31.000000000","reason":"Change was submitted"},"37855":{"account":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"last_update":"2026-08-24 18:53:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3937cb29143f3b9965032c186e8086854de7fbef","subject":"Bump dellemc.os10 to 1.2.7","status":"MERGED","created":"2025-06-27 18:10:12.000000000","updated":"2026-08-24 18:54:27.000000000","submitted":"2026-08-24 18:53:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953660","cherry_pick_of_change":953207,"cherry_pick_of_patch_set":2,"meta_rev_id":"100d13fffbbcfa8f728d87023841e465cd7c860b","_number":953660,"virtual_id_number":953660,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-24 18:53:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-08-24 16:28:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-08-24 10:02:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-24 10:02:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":2,"date":"2026-08-24 14:57:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-08-24 16:28:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-27 18:41:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-27 20:34:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-19 15:20:22.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"REVIEWER"},{"updated":"2026-08-24 14:57:50.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-08-24 16:28:21.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"5c9b35df9a9f64193144c94b858815ca606ad407","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-27 18:10:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7878630b22a7babc08f7a144393898fa43ef5705","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-27 18:41:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"55323c5bdc982b920dcf7127fc6a4b860e6a1a9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 20:34:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b551d831c3ac468db87a59701f5d5175\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a6cd18fae734852b99791cc655d391b : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02fc0ed42be3465daf30e61992ff784c : SUCCESS in 5m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3f8b4fb2d2e94d8e95c630799d686d83 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ceaa7b5135347da9e754bbe435bf34a : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e3624de7f494ce5a58de011e9db639a : SUCCESS in 2m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9e5dc15081b14334a3204f7a516fd08b : SUCCESS in 7m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/07f125f3035a4f83a374aeaa1b1d65f9 : SUCCESS in 9m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/86d8c326ae8d47c1ab26a9857186acad : SUCCESS in 9m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5de64bf715df4b6788fe7597a4174144 : SUCCESS in 1h 12m 09s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/a7b5fe47214c4a5baeca3e65dc376fb8 : SUCCESS in 59m 08s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29039d8d78a54dab8e0c230c4f502b37 : SUCCESS in 53m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cc3073fc41ae46f08aad0fd191206a03 : SUCCESS in 1h 14m 48s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9c42cc825442411c92056e44ad1b4378 : SUCCESS in 59m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1a707f884d6e4ddcbb600987302a4572 : SUCCESS in 17m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a66ca7a2f7e94d968f9c207e8340f0a5 : SUCCESS in 15m 19s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e8e07753130746a49291799b479fdc83 : SUCCESS in 2h 10m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e2a959f333646b195cbc729af14527f : SUCCESS in 2h 24m 04s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3e61ef79b8d6423c988d3f13f0635d20 : SUCCESS in 2h 08m 20s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d21c6bbec2543168a8129386a3a92d3 : SUCCESS in 2h 01m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ff022ef4906d47fba307f6d3e7dc86a4 : SUCCESS in 10m 57s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/3c2f915d0a414d3a83cb8ed774f20796 : SUCCESS in 10m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1279ee556014c8bb3c1a54151ccc29f : SUCCESS in 9m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c04054fd2be54593a295c06a1074739a : SUCCESS in 9m 30s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/76441b5364e247b7ab4647eb9444a4bd : SUCCESS in 20m 11s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9af78b1c1f21404488f1e96c223e5bf9 : SUCCESS in 33m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e92b1aa08e341f984a0d9721b2a5450 : SUCCESS in 21m 19s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3f76f1bd74f74979bc07f512dc254601 : SUCCESS in 35m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/af178be6901b4ab2a78411fbd14d2c78 : SUCCESS in 9m 37s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/ba49975ed1c3407bad3f308c03ea201d : SUCCESS in 14m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e7beeb6d3044a59b225dd49c6464634 : SUCCESS in 13m 39s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/a90e239daa3e4d0ab065a1684dfacd41 : SUCCESS in 12m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e107113171a346029af2613c030bf408 : SUCCESS in 14m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68cb699168234f6f92df01b0b8ae4dd9 : SUCCESS in 14m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ed0de8bcb2a95b406227bca2dfc04974953c0bba","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-30 11:35:05.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"786d43b75cdae801a8703eee39244773041dc579","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2026-02-19 15:20:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6e9b439e62d8e36b841e70cd658844190d58a89a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-30 09:16:09.000000000","message":"Abandoned\n\nNeeds more work","accounts_in_message":[],"_revision_number":1},{"id":"255cb94ebe17def858ded974a42dbdbb015cd43c","tag":"autogenerated:gerrit:restore","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-24 10:02:06.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"e4ff525e222cdb9a581d42ae4f87f79455830fa9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-24 10:02:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c64caac8ff06da52009be05869de996779c968b5","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-24 10:02:42.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"b848d485ee7977ef55ffe25fba5796f959fe226d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 12:41:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f98c0384b5494058a2478f37e6fc2348\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bc40343fe89470596ddc8b5df1b00ad : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f8f3118f705482a83f9ef43eddcc275 : SUCCESS in 6m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/acb81a71615a469493b3a366dc924726 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4462b2217fb440acba754d2438239b65 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b58c38947b434b53954b29164afca16e : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1cdf5c35b8fb49fe92184b989addfeb5 : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f42612b0b40b4907b2f21cbff35e2015 : SUCCESS in 12m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cc26c3f80ec144308c83117d80644aff : SUCCESS in 11m 18s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e50beb9d086148f88b3e2c71e58e5395 : SUCCESS in 10m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2477c5116c3f433981f356aa7e2a16e3 : SUCCESS in 10m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b1757f857e84853a98f12750bbd2683 : SUCCESS in 13m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8c61e34e6a724924a63c33ac560aae89 : SUCCESS in 20m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/da5b60ac7adb4b9b8a09fd175b2728dd : SUCCESS in 14m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9f579c3a28a4e2ba8b52d48bfc3048d : SUCCESS in 16m 35s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/abf21be4e5364bf1a47b5204506fcfb1 : SUCCESS in 38m 02s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c7fba2a8e5b9456980854f334ef0948d : SUCCESS in 1h 11m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/778e510f6c5c481e9627b851bbfd575e : SUCCESS in 1h 08m 26s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/327ac67619cb479c9cfa7170311b773f : SUCCESS in 1h 04m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/01146b6e9f194d189eac517944f55a56 : SUCCESS in 38m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/27f49afcad7d4565bed5e9b7b0cafedb : SUCCESS in 37m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23aa3ef492a6445d8c5b5bba70298221 : SUCCESS in 1h 07m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/65f28845a0d3498e9ce2b515587c5b0a : SUCCESS in 1h 00m 19s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e4e4f566c1334c48af1e0f7b68b5cdcf : SUCCESS in 2h 32m 31s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ec587a4c867469698f311293c8b9504 : SUCCESS in 1h 08m 55s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/19deb8f52b184c72bd9759da091851d8 : SUCCESS in 12m 07s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/134cc01a076e48d29e037d2203546e51 : SUCCESS in 11m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f54eeb6422ed4ba490307f0ea99ffebd : SUCCESS in 21m 28s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/7d5bcb09ee1f4eec94d513744f3b7d8d : SUCCESS in 19m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba39787ec76a4adcb45cb82d8cc074e9 : SUCCESS in 11m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/177095b6da9443c0a2a96aa7c5a4bc11 : SUCCESS in 16m 00s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/9bf995f5daf84bf6995ed0121e323307 : SUCCESS in 37m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9c9beb94b7b84339b4bc177d0bf08bbb : SUCCESS in 10m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0b259c72de8e4a60befe6fc879d71925 : SUCCESS in 16m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b01a14817d4e41aa9007b2081ef2f105 : SUCCESS in 13m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1a385903c28448bfc3621b464e75cc914024891e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-08-24 14:57:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7519ff98561300495dac3d56a6b5a1db9602d66d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-08-24 16:28:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"329d75d272fca7c931db28ae63e31918ce22a944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 16:28:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1e4cb4779eb9f12769a300796eac39fea4e8dd53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 18:53:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16936116ea2f4b0185c3d56e553ddd74\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0c336ee928c4025a82fac8c3f278415 : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30e71493ae5d4cfbaafab8bec9c63607 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aba78115942b4f89bcdd8ca4ce0beefb : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de598baed26e48da8e0835c92446ea92 : SUCCESS in 4m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4f47d77ecea64a799e7e319db4bf8824 : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8f4491a2a0bb4d0ca0544be1e43d3e5b : SUCCESS in 7m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dccdb51b782f44a29de1d677a55d4d7a : SUCCESS in 11m 11s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/41e283e447ec42ec9c75b91858be2d96 : SUCCESS in 10m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/803902926d504de994fa9e4249234e2b : SUCCESS in 10m 22s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9807f3a23f5d4789a868db08bd53f639 : SUCCESS in 7m 41s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f4c095b25b96498889dcb1b3a706547e : SUCCESS in 16m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/14a912fd6d6340e3a21ad0d9cc6b5731 : SUCCESS in 18m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f91c4380cee14e17b8366ef91be7c5b5 : SUCCESS in 8m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3af6a36daa3d40dea9641288f7df6b9a : SUCCESS in 41m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a04f4c393464486aa6802fe29e0cf29a : SUCCESS in 41m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cb114839bd304753933ea4c5e7d065a6 : SUCCESS in 1h 16m 24s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/14ab1a02921f48d7a5c25056e4df76e3 : SUCCESS in 1h 05m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/220cfbd0c54a4940aeee5a67afa6f8dc : SUCCESS in 46m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cdd467643b474230930aba6b780b60d3 : SUCCESS in 1h 14m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17a1ec92f57e4292a47bb4cd3618ec52 : SUCCESS in 1h 09m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/454c7e09e5f34fd4a5d5dbe7b536f96c : SUCCESS in 1h 03m 20s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/418e3b8c118e41f3837f066f0e3b56d2 : SUCCESS in 1h 22m 07s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a692a9936f0f44c6afa7c81460545c80 : SUCCESS in 2h 17m 42s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b9ac88c3b43f40b9ab36902d5e20cc00 : SUCCESS in 23m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cb4b9797f71641c086c32548ca1ae0a3 : SUCCESS in 15m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/352d2228e04d4923b36475ec333d6ba6 : SUCCESS in 20m 12s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/f209205fa30f4c3c904a92ed893d71fa : SUCCESS in 10m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70fd5e8b11374334aa7e933dc5d2b16c : SUCCESS in 18m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9576062fa776492b991c7feb5abeae52 : SUCCESS in 9m 36s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/28d3ac6877a24f80b5121a48b09e15ae : SUCCESS in 24m 49s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a8616978624e42cdb103c7169d9d24e2 : SUCCESS in 12m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/794b91f88cdc4951834f95a23ef8aafe : SUCCESS in 15m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/065d5da0775a4c5e8646a9c9b702c855 : SUCCESS in 15m 43s","accounts_in_message":[],"_revision_number":2},{"id":"506350e6dfdc83cb02341aa203ad3d184cd4d0a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 18:53:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"100d13fffbbcfa8f728d87023841e465cd7c860b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 18:54:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/876b3735fd1241fba62a6aa6f2036f87\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ee8769969e54738b2fae91d54e7a26a : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4937739615804b0fb511e315be7630f6 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"62c51cdc683a1be627bf3137c1e7263566210ea8","revisions":{"d71c62b637f5fe2e1ef29ce9a87ae80e841e836d":{"kind":"REWORK","_number":1,"created":"2025-06-27 18:10:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/60/953660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/953660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/953660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/953660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/953660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/953660/1"}}},"commit":{"parents":[{"commit":"fac4e78a99a0873a159b37913af2f0834c02798c","subject":"Bump stackhpc.linux to 1.3.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fac4e78a99a0873a159b37913af2f0834c02798c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-24 15:56:41.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-27 18:10:12.000000000","tz":0},"subject":"Bump dellemc.os10 to 1.2.7","message":"Bump dellemc.os10 to 1.2.7\n\nThis resolves an issue with applying some configuration blocks.\n\nCloses-Bug: #2115121\nChange-Id: I3937cb29143f3b9965032c186e8086854de7fbef\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 4007b0ee240904c83b908baa50a0003906f81876)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d71c62b637f5fe2e1ef29ce9a87ae80e841e836d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d71c62b637f5fe2e1ef29ce9a87ae80e841e836d"}]},"branch":"refs/heads/stable/2025.1"},"62c51cdc683a1be627bf3137c1e7263566210ea8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-24 10:02:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/60/953660/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/953660/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/953660/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/953660/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/953660/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/953660/2"}}},"commit":{"parents":[{"commit":"4e33454a7b70371870db9e18fbf002f42da642f9","subject":"Merge \"Set MTU on nmstate bridge ports\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e33454a7b70371870db9e18fbf002f42da642f9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-24 15:56:41.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-08-24 10:02:18.000000000","tz":120},"subject":"Bump dellemc.os10 to 1.2.7","message":"Bump dellemc.os10 to 1.2.7\n\nThis resolves an issue with applying some configuration blocks.\n\nCloses-Bug: #2115121\nChange-Id: I3937cb29143f3b9965032c186e8086854de7fbef\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 4007b0ee240904c83b908baa50a0003906f81876)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62c51cdc683a1be627bf3137c1e7263566210ea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62c51cdc683a1be627bf3137c1e7263566210ea8"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
