)]}'
{"id":"openstack%2Fkayobe~954352","triplet_id":"openstack%2Fkayobe~master~Iebd91d92c603efeaac94cebf009036873dd97093","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2025-07-10 15:42:49.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-07-09 14:56:35.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-07-09 08:02:20.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Iebd91d92c603efeaac94cebf009036873dd97093","subject":"Enable building of neutron-bgp-dragent container","status":"MERGED","created":"2025-07-08 15:08:53.000000000","updated":"2025-07-10 15:44:04.000000000","submitted":"2025-07-10 15:42:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954352","meta_rev_id":"ae82f0393e52b831df1fd7246db89eebebc6f177","_number":954352,"virtual_id_number":954352,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-10 15:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-07-09 14:56:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-09 09:58:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-07-09 14:56:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-10 15:05:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-08 15:21:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-07-08 15:22:39.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-07-08 17:00:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-08 18:35:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-07-10 11:49:23.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"6a4b326327793b353b4f35c0eaeb20036f3c5a78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-07-08 15:08:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1c8092cc32605c04ea3156a965c95ebe251780f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-08 15:21:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"734a29f293d47d04611e75e7f0f32485fdcd8a5c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-08 15:22:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ebbda024de0744931eae865ee48f20f1d62f6b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 17:00:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0efa96401ae4b6ebf5caae2781ab5f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4bd41dafee4694a6d5bf6366eb446b : SUCCESS in 8m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d680143af87b42af95f9e59f9e9a6a9a : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ceef70d0d684e6ca78b86fb10359679 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c86c7698dcc46cc9e48bb795ef5d234 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ba64b7d78eb4f0585fd18ee785cd5d5 : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2374872d0fb544e7b7dbcc06be3e9bc1 : SUCCESS in 2m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f86f9bf71314cc09d6f47f3dc9a767a : SUCCESS in 7m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6a4d6c7095c468e93ed697d046b952c : SUCCESS in 2m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b8f0d0ce026e428fa29525c33e4bd236 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bf74620cb5db447d8cfa14540719785e : SUCCESS in 5m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d049b43e052241faa6b6465fd22b5272 : SUCCESS in 10m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8de2d889dc047609a107088c075e448 : SUCCESS in 28m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/10ead77c484f4e55ae446cd4e6765731 : SUCCESS in 30m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e234469c3dd64f7a9df613343679760d : SUCCESS in 17m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4c06778463843da98400f8d4ffbac0d : SUCCESS in 13m 35s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82a2cb5ef46c43359a88c2bfe74ed8e3 : SUCCESS in 1h 46m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ba7ee955ff8e407bafc62d7e4b68ee3c : SUCCESS in 8m 56s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/1c15a267871348cc97ab91c125e0f604 : FAILURE in 24m 38s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e4f483d0f5c4ac197cf52dc1aa27f28 : SUCCESS in 16m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/71f54cae40f24c2587de6298a9f31650 : SUCCESS in 13m 25s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/83984c95403d4741bc6e45aeba999068 : SUCCESS in 46m 11s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/598ff23acdb24db780423ffc1087958e : SUCCESS in 30m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa0ecf4a775946fd9d9dfce09280dacb : SUCCESS in 32m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fd2ad8dfcd4b458f9e70bc619a8cfb6f : FAILURE in 14m 36s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/d411317fe5ca4b9699ee08035e66bd31 : FAILURE in 13m 48s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68b216bb701442c48e7f7941ce80ce63 : SUCCESS in 14m 36s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/4c45619b6a454f84b4d57c7ae6a3267a : SUCCESS in 11m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5538eacf022749718e5e88b351a40c8a : FAILURE in 11m 11s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cf869c5470d48a192b8fab23ecfeb75 : SUCCESS in 13m 10s","accounts_in_message":[],"_revision_number":1},{"id":"0505fe24c84e5cf44f192f54ef1296a4059bf086","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-08 17:16:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54f61c4b5042542cbd4a5921babbd9b7a4ab7f6c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-08 17:24:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"496e19b849f175309c8d80ab1774d4b1d4d9adda","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 18:17:02.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7ad75cc0c6a4fc990069508f913402b\n\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7badc9ba374c4808bd4c97954df79121 : FAILURE in 34m 53s (non-voting)\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6ebcf06544a04b698af21d26cf15f017 : SUCCESS in 12m 02s (non-voting)\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6897ce53339b4c6384468da2a0e966cf : FAILURE in 16m 51s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/899bed83b4dc46ac83143e9e49c8bb8c : FAILURE in 20m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/77a392f5286042de927f253ca85a3b1d : FAILURE in 33m 31s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/64e484612cf543348a64eafc0f292aeb : FAILURE in 49m 56s\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/52786cd26c70401ea8a3f75c50434b00 : FAILURE in 23m 17s (non-voting)\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/fe016792082140728484fe8a36444721 : FAILURE in 55m 04s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ec859bbca6b4098bdb9b6bc5fb26dff : FAILURE in 31m 43s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/98992ecf661e494ea9af0e6003a333b6 : FAILURE in 41m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d9e22614933b313995a612d334e2b8791972156b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-08 18:35:00.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"59b366d82d86b7c7d36585235d46a6d81b3e3620","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 19:30:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/98783a4e60714db59b9eddc499b669d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/733a28ef320945a78bec5347d453e7f9 : SUCCESS in 9m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b62a508d65694cb28477c46489a1be40 : SUCCESS in 7m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba228a25dbfc44f3b9e2c8703aa71316 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2c016dd26b44905be34ebef03514734 : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ae894b23b0545c9a843ace8e7fd188d : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80242c4f190e42548d80c61e4abcb96a : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b3b379f9163467d8b1ce4ff277e6b86 : SUCCESS in 7m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c39ed1a693a48e8aac3fea5df281a11 : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/de19f97168594d2bbc9569f70be31437 : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/77284e4c9e7c4753883870c9ed238dc1 : SUCCESS in 10m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5f242cfffbec469e9324cdb4db4e3232 : FAILURE in 7m 18s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b080789a278447c2b0de300df497e8ab : SUCCESS in 1h 09m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5d6627ddfd554d73b4fd89b989fdadde : SUCCESS in 58m 24s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/66c81e76c7e3459ab828af8658007424 : SUCCESS in 21m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9e9b7d3f8d244a593558137f58c6f70 : SUCCESS in 16m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd6bb06f8cf8404189d0a83b854cba6f : SUCCESS in 1h 46m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ff80f8bfc55748d6b00fa9b75c8a220d : SUCCESS in 19m 47s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/8e6b566df85d480891f9ba63e7e8b2ee : SUCCESS in 19m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3632d265ecee46898f3d2f0125a333dd : SUCCESS in 17m 02s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6e73c75fd3e34b65ac9e47a2f9855536 : SUCCESS in 16m 57s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/de6006517fcb48feb36d9f4d8b5e27e3 : SUCCESS in 40m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cff9a51848c74e3085268049fca23ed1 : SUCCESS in 27m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14b4f84d786545d29c8b351bafaf46e8 : SUCCESS in 31m 42s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c829d4d9c0c54201acacc1c7dab29403 : SUCCESS in 13m 50s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/840b6a3f60cb44838b2cf8699f598715 : SUCCESS in 14m 01s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac40e1ab375d4eef808c1140ca37e6cc : SUCCESS in 11m 38s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/0b2bee3f431746e88ea8329b3d1c18e2 : SUCCESS in 14m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/07e8df9347fc41e4a618141cf5d6c068 : SUCCESS in 13m 44s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5621e824cf224842a9427252af56611e : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1ba705ba021956066ce441215a75222eb23b7c9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-07-09 07:58:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e3c3229edfa68fbd1b5898ce6faf80d60321ee88","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-07-09 07:59:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d03d6c7f954515861a9af7e96bddb9db4f64dfd9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-09 08:02:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7690c8fb93c96097d08a02e222045d43f2eccc49","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 09:37:11.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c092667f62f84bda83cb68d00efce4de\n\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/989de0baca5845c990fbd34e88f80a99 : FAILURE in 30m 10s (non-voting)\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7543301ca21c41a3a52be2948f8d69d5 : SUCCESS in 17m 37s (non-voting)\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dd0e6e487e5e4504b7cff78479a7afff : FAILURE in 23m 18s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/9e30f49bf13f4529b345aeab34cb297b : FAILURE in 25m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3eef5b8880a34c489f765e8f4ea4318a : FAILURE in 30m 50s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5dace686cc4d48568344287680ee7a18 : FAILURE in 1h 29m 58s\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/daa8767243904f16a255cde6801a7870 : SUCCESS in 1h 00m 47s (non-voting)\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/dd51b142209f41c7bec9726b29c81dd2 : SUCCESS in 1h 15m 29s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adc819ea871145ca8ab47897183c3351 : SUCCESS in 49m 45s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/34a727c952a54d508f2f8d59181caf26 : SUCCESS in 56m 40s","accounts_in_message":[],"_revision_number":2},{"id":"61e369ee4a33bf6b281e7e61c6d45e22d6b6235c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 09:58:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/445ac5af96f64de2a109038a437a2407\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1db7c55f03741f8bb7dec13ca6f1ecd : SUCCESS in 8m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d823d8d8199d4cce8d654bc42da7ac09 : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bbc3c82554c440ebce846738e5fa1f9 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1071d3788f18446ab7fec3518c6eb378 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a63aed8e940430c849cca5b4e9bf324 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d6f43bc9f65d443589771b458db15492 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa8eabdc4e084eacbd1c0004c2d187e7 : SUCCESS in 4m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a2fcf47e8144c0a88d8a46deeaca331 : SUCCESS in 4m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/02e2f29d628747c1a53a6e71d008379f : SUCCESS in 6m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b225789e85624673bad1844dca36e653 : SUCCESS in 7m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7f0a8b52bb3b4a57b7b1341faec5ff1c : SUCCESS in 8m 43s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/941ba98964184b248c326831cb62cf0c : SUCCESS in 1h 06m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5b828945915f413eb7e467b545cf2947 : SUCCESS in 34m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6b7cf22303104f589f4fdb611f6f4f50 : SUCCESS in 18m 10s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5e33415de8c4e1f9a488e22f82fbeaa : SUCCESS in 16m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cc33905bc31441ca89a029630029a7b : SUCCESS in 1h 57m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/789ea2f5eb814f22aef63edb27f114bf : SUCCESS in 20m 45s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/42aba0eb77ed4402b4ec0cde0ca83f6e : SUCCESS in 10m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9daee83b2fe14f86b431c1c842b17af7 : SUCCESS in 15m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b3d4933bc1f547cabaac1f757d615dbb : SUCCESS in 15m 56s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e77301462a9c4b40b443207191913e99 : SUCCESS in 46m 16s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/22db0bc6029245d6825047e3aeae020b : SUCCESS in 31m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c00c63476f842109b62e116b3ce5fa1 : SUCCESS in 27m 59s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/896a1d0555c54a128919d04497b8ae7d : SUCCESS in 14m 37s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/dc9d127a56f744e089e114eb54397857 : SUCCESS in 13m 47s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b1dfdbd74fa43ebbc8915d0aed0bf0d : SUCCESS in 12m 49s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/4619a29973564083ac850eb6d8148fbd : SUCCESS in 12m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/72bd2f15b7ee4df1beb1c10895728bff : SUCCESS in 13m 57s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/978347e93a34418fbc170869c01e4adb : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":2},{"id":"3fe86c95db4821007caaa0cf13013607025f3a47","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-09 09:58:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"69bccc020300ef4eb085a18821f8acdabf3e3a6b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-09 14:56:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"53dae2db40eff90219557f903320333a61a8033b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 14:56:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d67452016366ca51825a27a02c928bc9928e2292","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 16:39:48.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/67a2e51ec1b44671a097dfe097c0ef02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d3cece3a8f24b2eb2d044d1623baa5d : SUCCESS in 4m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/04ca443c2c8444779847f0ec5a5fe82f : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8cb6d4d3d5446ff9fa1e2295b38740e : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53753ed6d42b4ebdb22170f0f63607a8 : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2304f93c6a2942c1bd710c4d5a7c0381 : SUCCESS in 3m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60a9387a4e314612a6913139cf0d4540 : SUCCESS in 3m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0579ba7d9296419cad4c4832350acccf : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f45e19f1923e49029509a86ca340543d : SUCCESS in 13m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9b60bf76b5a842f19b5d4245ba53df24 : FAILURE in 7m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78dbf87b795144919bb830107f98e109 : SUCCESS in 54m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f894aede76c04366800101141352be44 : SUCCESS in 56m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/da4ded25ad414c4f8164b557020f857f : SUCCESS in 16m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25aaf61c158c44a1887c98a1f58d2ebc : SUCCESS in 17m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a33db7536554db3bb1ffce9f83d5ec9 : SUCCESS in 1h 42m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/18bd1f3d90974881b3cb4090e10e6e26 : SUCCESS in 18m 43s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/37c3e219639a4f02a5ee73a1e2edbbb3 : SUCCESS in 21m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9f8322c05564261bf4b956dc9f78c33 : SUCCESS in 12m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/22429f7553904351b0948087ba4c5279 : SUCCESS in 17m 43s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1eaa6f0141cb4239936386fd8b27fc25 : SUCCESS in 18m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6569ae836ee9443fa5fa518d0a4ee51d : FAILURE in 15m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c59d12929ae14f62b9fabe14ace74e19 : SUCCESS in 11m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9f1d604529b4713be3e16c7cfdb2caa : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":2},{"id":"9dc561e231a5c28d3f40bf7aaf99121fbbb7ee9a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-10 10:10:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5772cbb9f3a6cc0a963a73a8b184116d0738dd95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 12:05:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc43c03b04614cc7bdf21eb698856025\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c22b5f1168d446a8aaa12fd3b0274dac : SUCCESS in 14m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d8abef090fb8463598103e5b3dbcfd43 : SUCCESS in 5m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a862141388343ad929ccef8f7574946 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81458a2ffa7d4281921275378aceb759 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56b00a3f729f4ccab748fdcf6cd7f521 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4887960578849ce99f0c62ea8921abb : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d67e1cb292eb41c8b9eee9e225b82eb8 : SUCCESS in 7m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddce731b4f6941b6817d0edeacd6b6de : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c5f4e14a5b7e4457b8a3801f1c49cf68 : SUCCESS in 5m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7ddab06c35a24fe3bd765c4e6e3a308f : SUCCESS in 11m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f4a2cc9a9cc2495cb7e7162a5eae9c3c : SUCCESS in 9m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/860467bb746247aca95fc0d3163b4465 : SUCCESS in 1h 02m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a2cb8bb990a1401da055e0f0afacfb71 : SUCCESS in 55m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0cc274c2bfbd4b6eae658b4b9bff7147 : SUCCESS in 18m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c4be2a6c6214807a8f5ac2367f35b81 : SUCCESS in 13m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c4c3b60696f449ca7ceea28c9c360ec : SUCCESS in 1h 42m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d03570aef5c241d58df7d26fe3b5f7b9 : SUCCESS in 18m 18s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/6e2dcd9a26c642348c548da8bba771f1 : SUCCESS in 17m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9851515b15be4684b37bb77b840a8531 : FAILURE in 9m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/24b6b8b96321425e9156451cf894bcc9 : SUCCESS in 12m 41s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e83a6d28d24e4ab1a860a8f2f764c9cb : SUCCESS in 39m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/088e30ce7a4d4d38abcdc49046813930 : SUCCESS in 30m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c479378451304d3184b80a8ef71995ec : SUCCESS in 26m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5aed6f39d79743db9144c8fee27c0247 : SUCCESS in 11m 58s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/63488c8ca22e48d2ac13ffdd194e6af9 : SUCCESS in 13m 25s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6339d87a97b84e0d905e67b6ed253d56 : SUCCESS in 11m 52s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/9742b724afca4cd4b4650a49c1dfb7bb : SUCCESS in 14m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2ee9b0efb9004fcd9e96402d0d7291d8 : SUCCESS in 12m 56s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f2291a5baaa4620ae240e11bba69df5 : SUCCESS in 10m 43s","accounts_in_message":[],"_revision_number":2},{"id":"dea6d11ed81ebac46deee018d6b3ac50e8a721e1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-10 12:12:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"130082b668aa6c5e50b11dd7d658157d2026db6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:03:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9211773305bd432593fed675cb78351d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af6c690a6de4ca2b19baec30a7d2535 : SUCCESS in 4m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6cfe1f93e3fb431d9bfe8eb283dc0a5c : SUCCESS in 7m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85e81213455b434e8c9225a60af9714c : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1e088d5620542bfb68103ed6f31d526 : SUCCESS in 2m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb5fadfd9b8f4bc5b5d2876bc0aece5c : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4eac029b7fda4b618254ac86cd9b97df : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4ccb1b38b4542b3b799e20457559d77 : SUCCESS in 6m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18a294f686f04ec5b97cd866d83e3f26 : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/101abd714990400fac01e425f4624631 : SUCCESS in 2m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/352eb028b61d4edba0e556f3f5800fd5 : SUCCESS in 13m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b826e33a7efa43fc84e7affd7d87e2de : SUCCESS in 9m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2db2158558fb4af287c002dc4e795028 : SUCCESS in 36m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0547816211e64c96afa80cd1b90031cb : SUCCESS in 30m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5f47faf4c35447e395acdb3ffba97b13 : SUCCESS in 10m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bcac4cacf5f48a0855121aeb27821d4 : SUCCESS in 16m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5ebb82276c94da0a3f4088a662e1124 : SUCCESS in 1h 43m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/818d91de7d5740429f711b373682eabb : SUCCESS in 21m 50s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/82995015aca8466493f257ea573a7119 : SUCCESS in 16m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9d30272fc8f4d71ad7d3588f9d3b226 : SUCCESS in 15m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9289a5a1448c4cebb0801ba3e80cc25b : SUCCESS in 15m 20s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3da6d286a07a44bc8c0dabbbebdb0251 : SUCCESS in 49m 00s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/641780ccef9b43b5b817cb0d161aa594 : SUCCESS in 29m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/141ecbf7de004c4dbf487d7099e67113 : SUCCESS in 25m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7ae29280323f4b479093440fa497c823 : SUCCESS in 13m 05s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/233afbfefd91435198385cf327cab8ed : SUCCESS in 9m 33s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ce5604ef58341cbb9058ca08bda67a7 : SUCCESS in 11m 49s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/0cd578dcc3db4e5392fa57f25c193946 : SUCCESS in 11m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e35babccb3804e838e51073e36e3971e : SUCCESS in 12m 59s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d9792653195427fb52b28f736fef5cb : SUCCESS in 12m 51s","accounts_in_message":[],"_revision_number":2},{"id":"afb81176133fee8c46918ae32beb8def0a0f99b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:04:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8b81cd6bfb00b8d6427dc34823db789d1a678b42","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-10 15:05:20.000000000","message":"Patch Set 2: Backport-Candidate+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"50c37f9212dcdcd8361c2f3fa3a974b72faa2f7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:42:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38a8aadbe084436488291a7c5e7ba0fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41c8d17da6084a47b1b47ac55eb38813 : SUCCESS in 8m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/65635074372949648615c5fcb0891c18 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0688215c16fb47398e655d792b44a328 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba73456dbd234544a07f21f12fe77b55 : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b375c7570ff4d27b60f929d037062c7 : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a0bc005dc9549a5846ee5d697a47242 : SUCCESS in 2m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/51a7fe8056774669ba0e6c51faf35534 : SUCCESS in 4m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5fe5a734c05547ad833b7aa61fb40fbb : SUCCESS in 11m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5097e84e6f334f26a0050dfb52962078 : SUCCESS in 10m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c450336c9c04242a218d0ac6c6a881e : SUCCESS in 58m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3b463a60f5044a6aad0b03da2edaa962 : SUCCESS in 54m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/edb24ed94683476bbec8739bed4e844d : SUCCESS in 16m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b53fd9f066843d4879d9468fe4a8450 : SUCCESS in 12m 28s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/276501ab5aea4a88960bdf6fbb6297bd : SUCCESS in 1h 38m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fb10375e829245478332b80b72dd716e : SUCCESS in 18m 21s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/5e38f37f29284567a71f413a3561845b : SUCCESS in 15m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/112d9f52f09d4d6f8c156e6defda1000 : SUCCESS in 15m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6e4c4ffa2620499b8ef56f6d6228fc30 : SUCCESS in 12m 57s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1b5810f2a9f84ba6b9a7db75dc24c7ca : SUCCESS in 29m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47d0ab5602ab44f5b5264f8331a19159 : SUCCESS in 25m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78f19bcc6b7e43ae9feac0a239295f3e : SUCCESS in 10m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd6513d2968148b480000f682eea2629 : SUCCESS in 9m 59s","accounts_in_message":[],"_revision_number":2},{"id":"2a1e4f0b6e731748f8993b5e504571a5e745278a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:42:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ae82f0393e52b831df1fd7246db89eebebc6f177","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:44:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a31406ef19b4cf284470570e80d9286\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85aa3d6b0a4045608b21ab83d09cdbca : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/108ca4d5a32644d0ad9df8bb69a0f2b8 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5e9b7738fcc00c1c4750b88b7b5a39c1f4437959","revisions":{"6eedf4641322a31a7f8e8f675bcc3347fc73bb6a":{"kind":"REWORK","_number":1,"created":"2025-07-08 15:08:53.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/52/954352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/954352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/954352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/954352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/954352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/954352/1"}}},"commit":{"parents":[{"commit":"4051b0cac06e5910ee88e782e70fed19b90b5cd7","subject":"CI: Remove installation of catatonit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4051b0cac06e5910ee88e782e70fed19b90b5cd7"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-07-08 12:58:19.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-07-08 15:10:08.000000000","tz":60},"subject":"Enable building of neutron-bgp-dragent container","message":"Enable building of neutron-bgp-dragent container\n\nChange-Id: Iebd91d92c603efeaac94cebf009036873dd97093\nSigned-off-by: Alex Welsh \u003calex@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6eedf4641322a31a7f8e8f675bcc3347fc73bb6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6eedf4641322a31a7f8e8f675bcc3347fc73bb6a"}]},"branch":"refs/heads/master"},"5e9b7738fcc00c1c4750b88b7b5a39c1f4437959":{"kind":"REWORK","_number":2,"created":"2025-07-09 07:58:26.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/52/954352/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/954352/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/954352/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/954352/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/954352/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/954352/2"}}},"commit":{"parents":[{"commit":"4051b0cac06e5910ee88e782e70fed19b90b5cd7","subject":"CI: Remove installation of catatonit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4051b0cac06e5910ee88e782e70fed19b90b5cd7"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-07-08 12:58:19.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-07-09 07:58:00.000000000","tz":60},"subject":"Enable building of neutron-bgp-dragent container","message":"Enable building of neutron-bgp-dragent container\n\nChange-Id: Iebd91d92c603efeaac94cebf009036873dd97093\nSigned-off-by: Alex Welsh \u003calex@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e9b7738fcc00c1c4750b88b7b5a39c1f4437959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e9b7738fcc00c1c4750b88b7b5a39c1f4437959"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
