)]}'
{"id":"openstack%2Fkayobe~954630","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~Iebd91d92c603efeaac94cebf009036873dd97093","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-07-10 17:21:08.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Iebd91d92c603efeaac94cebf009036873dd97093","subject":"Enable building of neutron-bgp-dragent container","status":"MERGED","created":"2025-07-10 15:44:21.000000000","updated":"2025-07-10 20:31:13.000000000","submitted":"2025-07-10 20:30:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954630","cherry_pick_of_change":954352,"cherry_pick_of_patch_set":2,"meta_rev_id":"67eb56b0163b5c9e4d0c6f16fc31cbdbeeab8bd6","_number":954630,"virtual_id_number":954630,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-10 20:30:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":2,"date":"2025-07-10 15:46:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-10 17:21:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-10 17:21:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-10 15:44:21.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"CC"},{"updated":"2025-07-10 15:46:03.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-07-10 17:59:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c5622ce8c37d41cd7fa87839de3a5a8a1c6ae06d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-10 15:44:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7182eac10db2a9c83c78490d2a1db4a47de64b87","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-10 15:46:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c3e9882c31f617fdf4ea1dbd5d5e30182882bf1b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-10 17:21:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cc3ccdf8e679e9f87d584b6ddaece67ad60c0bab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 17:59:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1565cc2b073640669080b3924ffb4545\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85fdd5118abd493f9dcd453d526fe84f : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11e2f9d437b14c70b2c078d12c4efcc0 : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5be9b776bbc8427ba43cd5705ec3ac58 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ee38378a6c44a479bff54aa8ada1137 : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e8886bc09c043d2b9ef18ad77eb7000 : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8af45eeecc964613a23c2d1807979d93 : SUCCESS in 5m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7bb7a9f9291842a7a69b7418a8a7cd90 : SUCCESS in 10m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a692d30cff524b62bb2d5e933d97b2a7 : SUCCESS in 9m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/53427178ec114aebb9e4bbfd95a28a10 : SUCCESS in 1h 00m 38s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/cedbc8ebaada427690c8f0523559db8d : SUCCESS in 59m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4f8d89b1e3c45f49d9185f9d1ab3d75 : SUCCESS in 52m 57s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8290d6fd5c73410088e4967224c1aac3 : SUCCESS in 54m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/59ea15ad3fce4544ae7f18d7005df60c : SUCCESS in 1h 02m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f23f68fa288649eba44731c23c38ed1f : SUCCESS in 16m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad4cb4d3dfea4186abe05a4f37d6e9b9 : SUCCESS in 12m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/21550c137f344a48ad9ef897b806cc5f : SUCCESS in 2h 05m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d34bc5ee29fd4fdaa40f51680fb27d03 : SUCCESS in 2h 10m 25s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/9043441ab6364b0ca8bf3412afa8b235 : SUCCESS in 2h 11m 00s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6273c8b789e645df88b83b54e6fdd5cf : SUCCESS in 2h 08m 13s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f80b85190a194f4fb8d338ea78e416bf : SUCCESS in 16m 01s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/9433a05b0da8497fad9a11766d0cf4cf : SUCCESS in 18m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21d70db0341e40abbf3ea5ecbe32e860 : SUCCESS in 13m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e5833b9c8291465e9aac1cb2f3e95608 : SUCCESS in 12m 56s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/368082bb6ad549f5a4dcf2311d600a2c : SUCCESS in 44m 42s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b61a788f4d1d49b29031e1bc1c57361e : SUCCESS in 45m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9920baea0dad4f3483511ff200d56ca7 : SUCCESS in 20m 04s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3453edb341444b4d8db32cad4fee0fea : SUCCESS in 31m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b5fb0a572ebb478d9eaca00e4b60ee8b : SUCCESS in 7m 27s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b08830830d174588a70e5498f2d2cde9 : SUCCESS in 15m 24s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/789fbc40d91c43efb496595f57d6becc : SUCCESS in 14m 31s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/95748d9cd9544ffeade5220849295bd7 : SUCCESS in 10m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a2ed2d5ea19c4b5792abe36f60df0abd : SUCCESS in 14m 44s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bbb5a64ff4141a78726c2261e180ebd : SUCCESS in 9m 09s","accounts_in_message":[],"_revision_number":1},{"id":"55d0e73e4dc01d76dde86a6adbb0e3e2f2022e87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 18:00:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d1d6902b3f603382122715dbd7adf6f4fbc803d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:30:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bcd22e5433645d683e101a60bb19454\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f6d1a71fdea41daa6234b142ebf8c11 : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0dd6a11efdb046c9ae39eba3fe60886b : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e11b276f16af49fbaa30af12b1da0e3a : SUCCESS in 11m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ffdc6cf015b4510a6b8af109881501f : SUCCESS in 2m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4731a876e7ef477f9b58abe37cd9e181 : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a3bc02b404c545da841d210cc11e4b1e : SUCCESS in 8m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/28f24655c3544bfba89a058ba4347138 : SUCCESS in 10m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7d077dedf1a74bb2b0d7036e074ac0b5 : SUCCESS in 1h 01m 08s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/aa1bcdd08f6c435ab5636bb355f4afc8 : SUCCESS in 43m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17620771e1cb46f08ea0069d2ff8f7bd : SUCCESS in 1h 08m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9fcf2459100e4f66b8994693b7f121fb : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2e3d6f325b5e4512b46e7620d3fe38d1 : SUCCESS in 58m 22s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/44803cdb53274adc8588addf17fc82c1 : SUCCESS in 17m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6441248dc4db48158444762780a86cc1 : SUCCESS in 6m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/af5cf68975ec400a8f699ab9bb20273d : SUCCESS in 2h 18m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22a7f217073d4418bef09b08b39dec90 : SUCCESS in 2h 28m 41s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f3dd89a4f8784d2baf56de8dbedf2cf7 : SUCCESS in 1h 15m 59s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3d8fed161074a599267e2fd66fc586e : SUCCESS in 1h 08m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7770f396f9fd4769a81a58f5dd99e343 : SUCCESS in 19m 59s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/63d9d34b11ef49b99bef5d67c644a8cf : SUCCESS in 8m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf0e05775b3c43018975367de4e29deb : SUCCESS in 17m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/521ebd9e868a4ff892f48091cb4563e1 : SUCCESS in 17m 56s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/52b476c2bf9a414580e18841a7eeb8c7 : SUCCESS in 20m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81e0b21c15b74046ad64cd2b585c8ad8 : SUCCESS in 31m 52s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/88ef2fa2955a47c59fba984b769335aa : SUCCESS in 34m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a235491c78c34922a921c15f61e23a81 : SUCCESS in 11m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9133965755e4e7e96fabe1655d308d1 : SUCCESS in 14m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a4ca9e4d3077abfc9394befa234edef6bb62865b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:30:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"67eb56b0163b5c9e4d0c6f16fc31cbdbeeab8bd6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:31:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fc3e9b3b2334bdb9f67457225b4d28b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68407e6692e34628b080c3c4e9e018c7 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f8ede7348e0470ca885212117439ae2 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4fc2ac3ad6ab9ca6a986428c421d2b0248a62dd","revisions":{"c4fc2ac3ad6ab9ca6a986428c421d2b0248a62dd":{"kind":"REWORK","_number":1,"created":"2025-07-10 15:44:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/30/954630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/954630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/954630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/954630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/954630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/954630/1"}}},"commit":{"parents":[{"commit":"3b4fa3d9529a0fba1c7106e46b6e2cac3186025f","subject":"[2025.1 only] Fix kayobe target venv backport","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b4fa3d9529a0fba1c7106e46b6e2cac3186025f"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-07-08 12:58:19.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-10 15:44:21.000000000","tz":0},"subject":"Enable building of neutron-bgp-dragent container","message":"Enable building of neutron-bgp-dragent container\n\nChange-Id: Iebd91d92c603efeaac94cebf009036873dd97093\nSigned-off-by: Alex Welsh \u003calex@stackhpc.com\u003e\n(cherry picked from commit 5e9b7738fcc00c1c4750b88b7b5a39c1f4437959)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4fc2ac3ad6ab9ca6a986428c421d2b0248a62dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4fc2ac3ad6ab9ca6a986428c421d2b0248a62dd"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
