)]}'
{"id":"openstack%2Fkayobe~954632","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~Iebd91d92c603efeaac94cebf009036873dd97093","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-07-10 17:21:13.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Iebd91d92c603efeaac94cebf009036873dd97093","subject":"Enable building of neutron-bgp-dragent container","status":"MERGED","created":"2025-07-10 15:45:46.000000000","updated":"2025-07-10 19:13:33.000000000","submitted":"2025-07-10 19:12:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954632","cherry_pick_of_change":954352,"cherry_pick_of_patch_set":2,"meta_rev_id":"c0e7c924abfa8bbb6e9bf5b623f768078b9e00c5","_number":954632,"virtual_id_number":954632,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-10 19:12:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":2,"date":"2025-07-10 15:46:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-10 17:21:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-10 17:21:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-10 15:45:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"CC"},{"updated":"2025-07-10 15:46:08.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-07-10 17:54:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"421604b10c9fb02ad01303ea98319985348e32e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-10 15:45:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"40ec8d7614a9dfb8f86ee8318c067c09181214ef","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-10 15:46:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c7c7e7aca2460e411fd9bafe16e8be52f39e6079","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-10 17:21:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e01c9779eb6ebcbb12fe636e6692d8c2268ac8ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 17:54:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/443bb18901e4472e9232ebba1587a621\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a71109287a35427da3d7d7b76b8b3cb5 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e83344821214995a9ad78472a73f73e : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d276dd852bd142359eee6033148d6c90 : SUCCESS in 2m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d0146df6e8c4413193aa9b020fc0c8d3 : SUCCESS in 3m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7b4849ccb184b30b987f94c6f94495f : SUCCESS in 7m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/97699d311a3847d3a4e1293fa41086c6 : SUCCESS in 9m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/43eefdc36594456dae852d405fcfde50 : SUCCESS in 37m 57s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/749f8c4438314fe48d8ed4c3abea120b : SUCCESS in 44m 53s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e357f3db0e1840c683578caba2e53b02 : SUCCESS in 40m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/243edfdd0e8e46049c6512b6e87acb67 : SUCCESS in 45m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4f5dc27448a44e25af7b3d8cc97c6ac5 : SUCCESS in 1h 08m 49s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3026c1c08fdb48679091e0b4fd359bfb : SUCCESS in 12m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7f0fd2a2cee943919dbd58a419d36c91 : SUCCESS in 17m 42s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a90595535e940fbb7816a05d61cd1df : SUCCESS in 11m 02s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b63ecb894db4aa8ab4a981e64711278 : SUCCESS in 11m 43s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/cd71deeadee74194b552b0240794cbd8 : SUCCESS in 1h 18m 23s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce003ce59a4f48989144c084f74b62b4 : SUCCESS in 2h 01m 55s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/578026c290864a20859dc9eba3b7b9a5 : SUCCESS in 13m 46s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72fe97621a7e4d56b18887c092601081 : SUCCESS in 16m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0047be40d4644c56a76c9dc959e5a9bd : SUCCESS in 14m 33s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ce7a39ed2cf845b08d81810b84c0d1a7 : SUCCESS in 25m 41s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8692ef45a4124752af83c9f128fc0457 : SUCCESS in 27m 58s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e84e4043de64b9e87b4d23b7ad13f76 : SUCCESS in 28m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2827f751851340e594f1c6a870d0a35a : SUCCESS in 11m 32s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/606d455b1b6b43749322bf27fd97992d : SUCCESS in 17m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa82342246a14ce8b823fa2d4a38b5f3 : SUCCESS in 15m 51s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/0fa33adc61ac4c31935cd440a9a3edfc : SUCCESS in 13m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3f92b57f5274f1a85b708db1718eadc : SUCCESS in 10m 14s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/dddc572616b148f0ac9f120347a05aa7 : SUCCESS in 10m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/763456ca0f9c48019bca8befee1344e7 : SUCCESS in 12m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7591eb5c25574c83b83fd5670a0fd540 : SUCCESS in 17m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/853a972b88ee4180abbd0d609ac8e744 : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b56da598c6644e548759e2782a70cfa0 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2642defc18b43748af4db16123f10bf : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b3f4a4af843475f83c71d0d791488d2 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f18009b483645618bd8f611c1a5f35b : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a1544843942c6a6a04930cb4eeb3e9177efa3daf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 17:54:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6861ea93dcbc3d654166fdda45c08c9054ad2bc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 19:12:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de8dcef03439453d997a958958930dab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b71d9a57f8047d79e8329816c111b3a : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aaf39240ed3c4a27b8cde1bc91ba6b62 : SUCCESS in 2m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d9f76d3cb36040c3bfed54fbed4392f3 : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d6d307269a824c41a9b515e1c1df99ab : SUCCESS in 11m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1e71d820a5d145adb3331905841423bf : SUCCESS in 9m 02s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b548f981526e4c088c4e9e816a5992e9 : SUCCESS in 54m 42s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c12f8e1c81ab4473872da99b85fba48d : SUCCESS in 46m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7346a598bf874e0d8ff26644f41be4d8 : SUCCESS in 53m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/39e41520127c4362a58082d181f15e57 : SUCCESS in 1h 07m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/261392783f1742f39afc929a3e95f1b9 : SUCCESS in 15m 36s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b357cc0ff89849c68cfc99290de12980 : SUCCESS in 16m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70f3940001284a5ca62c1543b1e9bcee : SUCCESS in 18m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4a95a088b9534710b4776e0cdc97c1d0 : SUCCESS in 14m 42s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db23ad224d794e4a9396d00cac4ef9d1 : SUCCESS in 17m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7eef87bf697443b8a31114beae118a32 : SUCCESS in 15m 05s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2bda6bdca657404f9c500718bb249dce : SUCCESS in 27m 46s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4cef9be654ac472f8e064e07bcba71f0 : SUCCESS in 25m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/07e53b8b2a304b4e82c80f2602797fe3 : SUCCESS in 15m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca431801f4974b248e239d6c45ee28c5 : SUCCESS in 14m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1365cfc847884a25851ecd4718933ea2 : SUCCESS in 15m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/978ca7a34c6c48d486f9e292d294b88e : SUCCESS in 18m 39s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8b6ee1679344b6887e1a1ff046e00f7 : SUCCESS in 13m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67aec7aca3a54bff8233d6980d1fcdee : SUCCESS in 17m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3ca76703a72f084fa108f57199f1eb6bc1186242","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 19:12:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c0e7c924abfa8bbb6e9bf5b623f768078b9e00c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 19:13:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e062b8401be4ccbb83fe80cb9d713c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca1fc7e6208944929146956fe5f19c55 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d60167c881b46e8b4d2a785b0f15d3a : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a67a98f72ce967b13f85c597e9428d047cff505","revisions":{"2a67a98f72ce967b13f85c597e9428d047cff505":{"kind":"REWORK","_number":1,"created":"2025-07-10 15:45:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/32/954632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/954632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/954632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/954632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/954632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/954632/1"}}},"commit":{"parents":[{"commit":"263e14d94ed96c7e8f20f1567b97dcff45c5106f","subject":"Support empty strings in dev-tools package lists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/263e14d94ed96c7e8f20f1567b97dcff45c5106f"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-07-08 12:58:19.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-10 15:45:46.000000000","tz":0},"subject":"Enable building of neutron-bgp-dragent container","message":"Enable building of neutron-bgp-dragent container\n\nChange-Id: Iebd91d92c603efeaac94cebf009036873dd97093\nSigned-off-by: Alex Welsh \u003calex@stackhpc.com\u003e\n(cherry picked from commit 5e9b7738fcc00c1c4750b88b7b5a39c1f4437959)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2a67a98f72ce967b13f85c597e9428d047cff505"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2a67a98f72ce967b13f85c597e9428d047cff505"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
