)]}'
{"id":"openstack%2Fkayobe~954652","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I531862fa9bebb8ef8ee900457278e0959dbe66cc","project":"openstack/kayobe","branch":"stable/2025.1","topic":"junos-fix-stable/2025.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-07-11 10:10:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I531862fa9bebb8ef8ee900457278e0959dbe66cc","subject":"junos-switch: Fix junos_config call","status":"MERGED","created":"2025-07-10 19:16:42.000000000","updated":"2025-07-11 10:11:29.000000000","submitted":"2025-07-11 10:10:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954652-junos-fix-stable/2025.1","cherry_pick_of_change":950339,"cherry_pick_of_patch_set":5,"meta_rev_id":"679b7f745fdf65a89474959bedc31c3d430b09a5","_number":954652,"virtual_id_number":954652,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-11 10:10:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-07-11 08:24:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-11 08:04:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-07-11 08:24:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-10 22:01:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-11 08:04:25.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-07-11 08:24:05.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"d8073a0f0f1ec941eb394882d69fff76d388ade7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-10 19:16:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a7364cb3182db3202ac2debd866094cda3066ec3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 22:01:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a7cab3269424bc897efddde684a58f0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e6f295915d940c5891a313124d31d92 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5751bf27a794863b7074367e67a5b96 : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3db5286212349bd87084662d102463d : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d471981f8616460ab667e9a7e15b414c : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee9b7442d22d4b03b5d47eca38886a5e : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c2f3701e9ca4b46b765bcd35db1fc1a : SUCCESS in 2m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ff75d4c16d7b488689629e8b3278e71d : SUCCESS in 5m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b2bff30e469945148bf670a0d109f330 : SUCCESS in 12m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/86f2fc2ed1874e3793c9df1bb38c9bd5 : SUCCESS in 30m 53s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/4adba324367544c89f817e70ca3c1f9e : SUCCESS in 1h 11m 44s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6a43791097d470fa1426f0a75fa68cd : SUCCESS in 49m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b97faab8b9f6485394df4b72fbec9f56 : SUCCESS in 1h 01m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7567c697d2b0410daf042f9e1793426f : SUCCESS in 1h 02m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f139ecfafc07427a8f2cd181df156e7f : SUCCESS in 17m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97d57390cbc94d7a8414327178903712 : SUCCESS in 13m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dccef8b7df5f48208e5b559dadfa5b9d : SUCCESS in 2h 18m 35s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/214754bec00a4a06b39525a8276266e9 : SUCCESS in 1h 11m 03s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/38eabb3cc08b455cac7454b535b1b198 : SUCCESS in 2h 21m 30s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5c8fdf2892947329dbbf9e5491fb583 : SUCCESS in 2h 11m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8f866f2957e3476ab6404af594a4fcdc : SUCCESS in 18m 25s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/ff0b6b2321e64ee2adc59dede04d2f03 : SUCCESS in 18m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9963f950d804f53aa2aebeed6d5550b : SUCCESS in 15m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a1110d81b09742708f60cc0dc0e59c09 : SUCCESS in 17m 44s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/cbe81829752548ae96d77fbba8877b22 : SUCCESS in 18m 38s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e4033ff0089b4dd182e85b686c87c33c : SUCCESS in 38m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4fd4bd9713f477a92f78edf5e78e312 : SUCCESS in 32m 57s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/431887732bd9406886eacfd1864e4cd2 : SUCCESS in 35m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d24f51ae45cc42dabaf20ed8b3cc4be9 : SUCCESS in 14m 23s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/4e024503f3f64b7c9356fee7d40d1ead : SUCCESS in 14m 02s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/476a0b20b550402b869e6e4343da4d3d : SUCCESS in 11m 12s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/e5cbc300c9924d3e8532835ce7fbd4d7 : SUCCESS in 14m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/71f51b4d8650403bb2dc7d22727bbbc9 : SUCCESS in 15m 36s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d165df8e80784204ac981b8a56ecf0ba : SUCCESS in 14m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ecaad7e1add73693a397004060a1fc098d6440d3","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-07-11 08:04:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7bda98f2ed07f6e61f860cbb89d59b809a300985","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-07-11 08:24:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80078da2c266aa38720d792420f468dcd7dee1cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 08:24:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81fdf2d75a2f0840fed67a65c1798371b4d25a99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 10:10:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8278f61b360045278a614def499d36bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b4e0f623259487e8e78b0613f4e64ff : SUCCESS in 5m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/72da728ec28b423fb6ff93b81721dfdc : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a6acf8a087f4d4295d2511d70d3e882 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66f4daa882e64f52b4369655151d3a2a : SUCCESS in 4m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fa8841dd92844d7296984e8022056564 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/adf573dcebf14106b77dba082f10ccd4 : SUCCESS in 12m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e2cdd5cf112a46b9bccdb0be280b2e3d : SUCCESS in 12m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/59639a6bb2a64ed3acdfd6fb1c33f679 : SUCCESS in 1h 10m 28s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/837a464556fa4abf81fca8c2600ac36c : SUCCESS in 1h 14m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fcb17c157c14c8ea7557ae9c1be9ee9 : SUCCESS in 1h 07m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b604ef9a8ccf468696c7cddccb46244c : SUCCESS in 56m 58s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e22a51bdec704360882e7897764017de : SUCCESS in 1h 16m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4396e208fc9c44369da6cd0fc4f78cae : SUCCESS in 21m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/610ee6aedea346eb96e611e6e1e1c129 : SUCCESS in 10m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/08bc09c74a894ea6af70569c8193638f : SUCCESS in 1h 45m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5377be09ef624c7bb82923c9efe4d7be : SUCCESS in 1h 36m 20s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/830b3dcae1f4429e8e456a01db777821 : SUCCESS in 1h 45m 12s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e71f12d58268429fa6c5606b516a4cd9 : SUCCESS in 1h 36m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/493ee5e9b5714ca0b078e182d27e7c0f : SUCCESS in 12m 41s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/04b9529194194dfd94bf457d94a23d15 : SUCCESS in 12m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1dfcebd6ffaa41a1a3a14146b67c43d6 : SUCCESS in 11m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/29e05100fd4e4aea842a2543fc85ae1b : SUCCESS in 8m 34s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3df8f36797884d6d91469641ceb6103b : SUCCESS in 26m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb5b80b32c0f4f9bb7bf0f560a793d02 : SUCCESS in 24m 09s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a99fc2a95bda48b7a252eabf8e15ac43 : SUCCESS in 25m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/648836babda8433cad5be704f08c407b : SUCCESS in 8m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56c0b07a44714d8ebdf7682bde0830c1 : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c16c6e72f8e9493a088b0ffbb788cc1c411b8382","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 10:10:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"679b7f745fdf65a89474959bedc31c3d430b09a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 10:11:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1448fce4de554a90a7338f09ac5c79c5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75d3c8859fae4cdc8a86e0b9d7aee85a : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24bfed7edf5c4a97a54a522b0f0fdc20 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba81b6fc6be6403c087deb1641c0cee32322e6c7","revisions":{"ba81b6fc6be6403c087deb1641c0cee32322e6c7":{"kind":"REWORK","_number":1,"created":"2025-07-10 19:16:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/954652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/954652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/954652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/954652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/954652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/954652/1"}}},"commit":{"parents":[{"commit":"3b4fa3d9529a0fba1c7106e46b6e2cac3186025f","subject":"[2025.1 only] Fix kayobe target venv backport","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b4fa3d9529a0fba1c7106e46b6e2cac3186025f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-03 10:25:42.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-10 19:16:42.000000000","tz":0},"subject":"junos-switch: Fix junos_config call","message":"junos-switch: Fix junos_config call\n\nThe junos_config module dropped support for the provider parameter and\nfor connection: local [1]. Apply a similar to fix to the Arista one [2].\n\n[1] https://github.com/ansible-collections/junipernetworks.junos/pull/333\n[2] https://review.opendev.org/c/openstack/kayobe/+/922631\n\nCloses-Bug: #2111341\nChange-Id: I531862fa9bebb8ef8ee900457278e0959dbe66cc\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 1372a7b01e1b4022f486aba01f4010efc724ada8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba81b6fc6be6403c087deb1641c0cee32322e6c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba81b6fc6be6403c087deb1641c0cee32322e6c7"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
