)]}'
{"id":"openstack%2Fkayobe~954655","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~I531862fa9bebb8ef8ee900457278e0959dbe66cc","project":"openstack/kayobe","branch":"stable/2024.1","topic":"junos-fix-stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-07-11 10:35:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I531862fa9bebb8ef8ee900457278e0959dbe66cc","subject":"junos-switch: Fix junos_config call","status":"MERGED","created":"2025-07-10 19:17:34.000000000","updated":"2025-07-11 10:36:28.000000000","submitted":"2025-07-11 10:35:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954655-junos-fix-stable/2024.1","cherry_pick_of_change":950339,"cherry_pick_of_patch_set":5,"meta_rev_id":"9682b7277165f13947dee0a4b0fa83bdbbdc2d9a","_number":954655,"virtual_id_number":954655,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-11 10:35:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-07-11 08:24:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-11 08:05:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-07-11 08:24:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-10 21:38:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-11 08:05:02.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-07-11 08:24:21.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"a2acfe542b33d80d01b14c86fe820b2551b1a3ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-10 19:17:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3f0ea3b60ae6f5eeb74630c24f949c1fd1096e66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 21:38:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59bdb3630bb44431a9652f3110fc26f1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36834db8e415439fbaf6d837b8183cb8 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1859ed618ae748ea81a5100ba8061df2 : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13076ae922cd49c58df12fb6579af4eb : SUCCESS in 3m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2cb41fbbb69a4152a95a41527e003fad : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e8344330bc054f3c9a8fca4a03ae338c : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e04e26fea0b94b66a3c2bc1cc6a0bc9a : SUCCESS in 9m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/63d421a03fe848bda34ddce21d376a1e : SUCCESS in 1h 04m 31s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b68b1212ea5a415f832c5e6187b1a267 : SUCCESS in 33m 06s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/041116c6ee9848ca836505381b9b8726 : SUCCESS in 40m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77775ad99fa5445ba9e8a9e5e1c2627b : SUCCESS in 1h 04m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/129cf7b307014fdca67110aa8d6d79cb : SUCCESS in 1h 07m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/32ad3f0cdd1546488f6f1528d2e1bf0a : SUCCESS in 15m 26s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a567062d11ff4769b5190bd74a377fb5 : SUCCESS in 20m 51s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1bba2613cb84ddeb88219260b41c666 : SUCCESS in 12m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb3334e949a24e15b46b3e3a19e5a9c3 : SUCCESS in 7m 17s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a7ad537def6c4fa19ee21ebe7afa4e5a : SUCCESS in 2h 15m 09s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6974b4c95a0a4bd88eafc7ff3c2d0f1e : SUCCESS in 55m 55s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2f09b4d239e9489ca3c5d39c474af996 : SUCCESS in 17m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9599582fa3544880bab5c7d574080625 : SUCCESS in 19m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4428d02f5a94fe99e6c5af812dc3430 : SUCCESS in 14m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c72051c2e7ce41c6a914fa3e5d8da379 : SUCCESS in 44m 01s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/366102e3d2e044c58d3ee9ef8b5d4c5c : SUCCESS in 23m 10s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84c0c707f9e643d9b944c024ff14e1ba : SUCCESS in 24m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/257952581e3d44078a3740c9dfab5faf : SUCCESS in 18m 05s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b4321e7edee1404f902d4e6f061bcb23 : SUCCESS in 17m 55s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50424bebdbfc42baa0a7da81c2f930d4 : SUCCESS in 11m 07s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/1d51646aabd94637b60254a0417be803 : SUCCESS in 12m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43aa7802b14d43069825f63d99b413e7 : SUCCESS in 14m 10s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/011da375d21640b9a5e912edd3a1aa8b : SUCCESS in 14m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/85184d6e77a6484eb5e5bde75f890343 : SUCCESS in 13m 25s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ac22aa0d9aa4b5a8449dffbc983fe62 : SUCCESS in 10m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb6557082be948c0b3b4491215872b19 : SUCCESS in 13m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b0c79675e4f47a4a55410bd64d90cbd : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b5d95a84a4b4b5b884a2bc4948f5325 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84c44a8b9c994281819f334bfd856103 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/42ececc65b3a4b79aaf141b07130fa2d : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2cc0eb39ae8d9361e18a8a81ebfb808b503510d3","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-07-11 08:05:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"445855764e9367913ccad9799ad9105fc3ea7fe8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-07-11 08:24:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac62ec6b5acd8334b7161c0e1fed64b91bf5ca96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 08:24:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3475426a40646c16eb8e41eeab8a0bb398cbf3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 10:35:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bb197657f204110bc8b8202cc2f6513\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47cf784ed2ac4cb58cde5308b2927fff : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/455039ec8fbe4a279876ff2515564900 : SUCCESS in 2m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/60e4a4562535437690eb0a7085ee79c5 : SUCCESS in 3m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/629c2dc87bdf4d5c8f4c8296eacb2fb7 : SUCCESS in 12m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/92c00e056344434cac0d8d61eb33ddb3 : SUCCESS in 8m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b95134ea0215475b987b850116d5cf07 : SUCCESS in 1h 03m 36s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31bc7702bcae4f5189643c1fe0b98619 : SUCCESS in 48m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7644b23bbad4e1bb4b66d4fcfc4bd9a : SUCCESS in 1h 04m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/699521bcbfd842a79724a945a6ad833b : SUCCESS in 1h 02m 26s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bf84c7f598bf470588d62da9c90caade : SUCCESS in 16m 13s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17480cb3f2fb47caa0229e942570b5d0 : SUCCESS in 12m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b68ed5b1dbdf4deeba38b196629929ca : SUCCESS in 7m 21s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/86bf1345f17a420ba1300163f7f3c32c : SUCCESS in 11m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efc885f9c03544d08a1b30070b30fd27 : SUCCESS in 10m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf3140e50540406ead5fda76308504f2 : SUCCESS in 11m 06s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/eb3e122ed8f84aab83dc69741b2d75f3 : SUCCESS in 32m 12s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0d3f38e731a4cd091b1ea383043054a : SUCCESS in 22m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ec516475f21a4130a192bcab45349657 : SUCCESS in 10m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6bdc2151bb54b08af08602fbe5d51fd : SUCCESS in 9m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32a1c5e049c34a6e835f739df1eb1378 : SUCCESS in 18m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c6b9db74232b4e2dae8647b019ff5bc7 : SUCCESS in 17m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62f8f668678545d397bd314805de6133 : SUCCESS in 12m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b42263f87cf4e9cbafcfb76ff00e2dc : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":1},{"id":"21d3a52ad09a7d88b91337e1475391ab90de049f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 10:35:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9682b7277165f13947dee0a4b0fa83bdbbdc2d9a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 10:36:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15e08bb12426445d915508c56db86c5b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df4bfd1fad64bc38c12dd85883baaf1 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dad301a1b5de42a098dcb6cbc0a1a951 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0251eda6eea9a61f07608f6febebe6e1cbba73c4","revisions":{"0251eda6eea9a61f07608f6febebe6e1cbba73c4":{"kind":"REWORK","_number":1,"created":"2025-07-10 19:17:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/55/954655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/954655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/954655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/954655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/954655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/954655/1"}}},"commit":{"parents":[{"commit":"2a67a98f72ce967b13f85c597e9428d047cff505","subject":"Enable building of neutron-bgp-dragent container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2a67a98f72ce967b13f85c597e9428d047cff505"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-03 10:25:42.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-10 19:17:34.000000000","tz":0},"subject":"junos-switch: Fix junos_config call","message":"junos-switch: Fix junos_config call\n\nThe junos_config module dropped support for the provider parameter and\nfor connection: local [1]. Apply a similar to fix to the Arista one [2].\n\n[1] https://github.com/ansible-collections/junipernetworks.junos/pull/333\n[2] https://review.opendev.org/c/openstack/kayobe/+/922631\n\nCloses-Bug: #2111341\nChange-Id: I531862fa9bebb8ef8ee900457278e0959dbe66cc\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 1372a7b01e1b4022f486aba01f4010efc724ada8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0251eda6eea9a61f07608f6febebe6e1cbba73c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0251eda6eea9a61f07608f6febebe6e1cbba73c4"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
