)]}'
{"id":"openstack%2Fkayobe~954712","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~Ic2fef4c6925d2d799f096de6b6fa519d77130c3e","project":"openstack/kayobe","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-07-14 12:23:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic2fef4c6925d2d799f096de6b6fa519d77130c3e","subject":"Fix documentation of seed_containers usage","status":"MERGED","created":"2025-07-11 10:38:46.000000000","updated":"2025-07-14 12:24:31.000000000","submitted":"2025-07-14 12:23:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954712","meta_rev_id":"365bc6fb67290206cd45d579123a176c036d8403","_number":954712,"virtual_id_number":954712,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-14 12:23:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-07-14 08:34:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-14 08:27:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-07-14 08:34:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-11 11:45:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-14 08:27:48.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-07-14 08:34:46.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"d843297320c33249d9b14872bcf0e0fa150e7d93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-11 10:38:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"392035f0318ee7960a5dcda5c20adf44b765b404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 11:45:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0a14c8388d54dbfbd4272d4530931e9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbbb8ebc9cd34169bce5456733492555 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ee36af6dbdc4bf8bc6334ca7b7f990d : SUCCESS in 2m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b7261c3491540c085fc8448e8e01af2 : SUCCESS in 2m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd98fcd21813499783bfce8324be01ea : SUCCESS in 2m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85764e43a74746768f60fc744a634858 : SUCCESS in 7m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f0744b50915f4fe2b5b8df00361a6edd : SUCCESS in 2m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/179fcce1d86b4f31bb0c795d5e3357aa : SUCCESS in 7m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/39a4f75438214a43a8ebc7971a47add0 : SUCCESS in 5m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b3f387c189fa43709165085d9b0cb78c : SUCCESS in 34m 07s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/382153c001fd4cea977a880ed37a8c1e : SUCCESS in 35m 04s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c223e4016164b14a247c7b951b2ba4d : SUCCESS in 52m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e4d7bdddeeaa4ac5981ad1dcdfc38924 : SUCCESS in 52m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/29f3aad93ef24e96997e8f940724d525 : SUCCESS in 10m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7fce25beb22a4fb48e4e5e934505535a : SUCCESS in 10m 44s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9e408364ad547258cdd19d2cccc9a02 : SUCCESS in 7m 50s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4193eb68ca1f44859d420ec6498d9758 : SUCCESS in 1h 02m 36s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81a60bf56c7646f4b1f766c17a13d439 : SUCCESS in 1h 05m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/22e19f46001d4f0c84166e50cf6c810e : SUCCESS in 20m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b092fd7df98e46a8b00fc30dfb84be2c : SUCCESS in 11m 52s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/fac3b8cde04544bd946b3a06b9934a9e : SUCCESS in 33m 42s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6fd330f45a9c41d3906e746c5b216533 : SUCCESS in 30m 47s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34c8b5c3e9fa4a8ea5c0cdb8b296c781 : SUCCESS in 18m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/225ea7eb022e4e43b9e95d2f5b5d7e13 : SUCCESS in 11m 19s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/58c1e884ce1f4458a6d9b92f62e58d2b : SUCCESS in 17m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd21056674224ad5adae2b15740be9a5 : SUCCESS in 10m 54s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/f2980f8766774b498bd50ef28786d21b : SUCCESS in 14m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8d702efdd34742d08a80c936067471a4 : SUCCESS in 14m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b32c44cee1834baa863b2deab0c22cba : SUCCESS in 14m 44s","accounts_in_message":[],"_revision_number":1},{"id":"61dc840ac137da4228ea0fffad3c3715288794e9","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-07-14 08:27:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"572ca566d191a212d910a4b3747a104e019bc96d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-07-14 08:34:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12c26d93f2da4bc2cfbe62360bc04be665ccc705","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 08:35:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8fa928ce15ced0fe5926e5d8d67543bc9cb72d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 12:23:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edbb9cd55b69498bb846226d0124ffdd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21e3168e5bf84da89c77efc5754755d4 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a01b2befcd94604863d17aadbd94218 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba6a8e130b794dcd80870c782bbbe46b : SUCCESS in 5m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fc7c5d8b65e2498a855e6ccc96cc7841 : SUCCESS in 2m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/813b053a1e2343fe92c43b8a97953d7a : SUCCESS in 7m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1f47183e87704ab6abd9bbe491395339 : SUCCESS in 7m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7c683b5911814d39b4686f5bc1612b73 : SUCCESS in 48m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b23fb4e5d8d469abb0b15ebb8414352 : SUCCESS in 48m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/789b78a943e34487904fc65603bfd159 : SUCCESS in 41m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e6193534e8c1424288e1a9e5722f5c8f : SUCCESS in 14m 05s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cef0a597a3734e4d9057c4c07863b64b : SUCCESS in 11m 47s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/660e38a8b41c474f9308e8550517f568 : SUCCESS in 1h 56m 36s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/faf920f0376a4e249fbb41e844a60789 : SUCCESS in 1h 48m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a61f67014fae47cd8eacf44689e566f1 : SUCCESS in 13m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/158eef047d214ab1bf4e4354b46b94dd : SUCCESS in 19m 55s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/991382121d0848f0889ce629708f0283 : SUCCESS in 26m 08s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c04670e22b93475c974ce8dfecbfac93 : SUCCESS in 33m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/07ea664291bc4716ae1b0d42aaca2df4 : SUCCESS in 17m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3390306450394040b811cc1630f18b54 : SUCCESS in 10m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/344a46e2b482488ebdb53f5d8ae93094 : SUCCESS in 12m 22s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e31f40fe355746b58987a651e215a789 : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5f5899ec955768762ed51bb1922da944652a98fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 12:23:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"365bc6fb67290206cd45d579123a176c036d8403","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 12:24:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef8a1ab3d4c5468a82bdb5e5ea2d6012\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a1283b872ba4c8ca7ffdfc17245a7c1 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7dd59e2982649007a686f424cd156e410e8878c","revisions":{"b7dd59e2982649007a686f424cd156e410e8878c":{"kind":"REWORK","_number":1,"created":"2025-07-11 10:38:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/12/954712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/12/954712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/954712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/954712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/954712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/12/954712/1"}}},"commit":{"parents":[{"commit":"ca93abef9af875f610ad408f17e242bce4fea4ba","subject":"Merge \"junos-switch: Fix junos_config call\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca93abef9af875f610ad408f17e242bce4fea4ba"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 17:25:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-11 10:38:35.000000000","tz":120},"subject":"Fix documentation of seed_containers usage","message":"Fix documentation of seed_containers usage\n\nPulling a specific image version requires using the tag parameter.\n\nInclude docker.io in container image name for compatibility with Podman.\n\nChange-Id: Ic2fef4c6925d2d799f096de6b6fa519d77130c3e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 65c129143623995d16ec4701afcc5c23717323b8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7dd59e2982649007a686f424cd156e410e8878c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7dd59e2982649007a686f424cd156e410e8878c"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
