)]}'
{"id":"openstack%2Fkayobe~955772","triplet_id":"openstack%2Fkayobe~master~Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-09-29 12:59:01.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-09-29 16:13:48.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-09-29 16:13:48.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-09-29 12:58:12.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7","subject":"CI: Add CentOS Stream/Rocky Linux 10 host configure jobs","status":"MERGED","created":"2025-07-24 11:05:41.000000000","updated":"2025-09-29 16:15:29.000000000","submitted":"2025-09-29 16:13:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955772","meta_rev_id":"7ff4e35a24abcaede6049bee99d0d9f55b941e82","_number":955772,"virtual_id_number":955772,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-29 16:13:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-09-29 12:59:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-09-29 12:58:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-09-29 14:27:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-29 12:49:46.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-24 11:11:59.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-07-24 11:24:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-14 16:01:48.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-08-27 08:29:32.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2025-09-29 10:38:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"59736dddfb17f78076f5893874f0887af32c8bd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-24 11:05:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"662326d2677e00976b3c2adeccdc99ce62b6e2fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 11:24:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30960b18d7124bd88cbf2d37d8208eb8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c27ede14f784edc9c89c6c2831677bb : SUCCESS in 7m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6b6f0cab97264f8da375f7e028cc03a1 : SUCCESS in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12cc17551e4e4276aed273af0575106a : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af71dcb284de42d2b63351e50197ce67 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8509c54f98b438ca9d139df8fb665ef : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ef0e5ac38ba4ddf9167e7b8f8446aab : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3336bc3503694be0af15173610d84800 : SUCCESS in 6m 18s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/2b63d3fb9b1e4e12817d50d8b97eed70 : FAILURE in 7m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"47156c61b18139627223a59fe71e1416451684bd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-24 11:57:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41b6c92b17021afe8efe84d7f5ed11bf7d89f4e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 12:13:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6af87bc55cf4efab851ff7225ddc982\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0fd3ff1e79f41aca25ea7b369e863bf : SUCCESS in 7m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a7ee773beeac4111a77dce39c5b96b80 : SUCCESS in 4m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02950a704d1f48809438afe1a55d11d2 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/177f67dbb4bf428bab2e63f422a88ab2 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42f05e8db05e4538900a187ff8961465 : SUCCESS in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/913546934a704dca8271aa604a540950 : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc037faba65647bab67c20034851b427 : SUCCESS in 6m 35s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/9898f4d7812343dcab7d20e52959e4a9 : FAILURE in 9m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed95fd40c988daa537043f83878e4236386fae7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-24 12:32:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a7d4fdae974f036ce582a3e292f3cd1044999c54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-24 12:42:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"41dcc0f492a34db7ac7600a627daf92f6c9afc22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 12:57:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a465990fea7409ba35c3a8b3a002b84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a030ef1269347049cf418cadb19bd7d : SUCCESS in 6m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e86086a60e5e48798ea8aa1b80c5baa0 : SUCCESS in 4m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad96707e6c9d40c3bd2b2027afe3c3f5 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ca1fa542efc41e4bd6716090dcd9b3b : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6bab94a867eb494f98f7b58094237cd9 : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a88048d7266f498da58b7c98063a7829 : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cafcfd9adba5463997d0531672c1271c : SUCCESS in 6m 22s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ab8ca103c1a74c7685c5860bf739ad4d : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/66330004836047ffa95f1f247d1a40d2 : SUCCESS in 10m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/00cc1e3b664f4705ac16be1865285ad4 : SUCCESS in 10m 07s\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/673d839b0ccb46c4bc9da63ca93e3162 : FAILURE in 5m 48s (non-voting)\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/1f81b366f7414cfbaa6f3dbb92f8bf5c : SUCCESS in 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"87bfbb6c4999086dc4dfd76161ab3ae9f7930140","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-24 13:01:49.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f0d4f4057cf05f9492bdf8f1788bff2a47f7abe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 13:20:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f3534682c0d4c78af9625566df20c39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f34f580ebad43c5920b92972f562cea : SUCCESS in 10m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/35f34bedc9b64f088500de1a1a0d0b0a : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8602535a6e1547fd9fcb5ff03a69724d : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5805d4d4f454983b8feba6c255e44f0 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41002f95f78d42b1805a8feadb38431e : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e5b9215060244c2bc54e578305c0326 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c7c16d9e22d42aa88c08380bc3ab56c : SUCCESS in 7m 52s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/024a087213ca4442b5cca3cde68d08a1 : SUCCESS in 6m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b5f2f649b3064667a80f7760543553e9 : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d0b665839a8e4dcdbfb322bae840d888 : SUCCESS in 5m 24s\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/1e715825840b45338e0f03c51579a818 : FAILURE in 8m 10s (non-voting)\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/fa29bde860364b25ae7ad7cd2705fdb2 : SUCCESS in 7m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"75aae5945991b0790c5644d331af186c7fa7e083","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-24 14:28:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"711d8054398542af89e3655f6390adff9a362e65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 14:59:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d70dec25c8b044069bb43ebdddac946f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e9eabce4db43e288923494843eaf9a : SUCCESS in 9m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/864953b2956b4094a44cef3db8c9a128 : SUCCESS in 6m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbcbc5d8fb0a457f9fa7350f0bba19b1 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27bc3a0637534604a5cbb6ff98930400 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/350ebb2a433a41a48068ff1c518d264c : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e486c2a0a1374c84b00e9dd77f130626 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c51c84abeafb4da487cd521fac3d44de : SUCCESS in 4m 23s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3aef30fd34874f0991c748c59eccca8c : SUCCESS in 3m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d1091b3b073241dfac0ace57e0df6cd1 : SUCCESS in 11m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/367277bc84fc43669ee82757a7d65ce7 : SUCCESS in 12m 20s\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/afc7f392857d42329c428621aa313467 : FAILURE in 20m 13s (non-voting)\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/465eedfa8e884b32a186bf1fad89f38a : SUCCESS in 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e7f3d3e965e67ba875809942ea4fd26ab06d6aa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-24 15:25:28.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"0cd29b33ee25777d01736055d303f68270c171f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 15:48:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a18183cc7cbc46df9b76e801219a92aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9f0e16f41ec478383a37324eda6795b : SUCCESS in 9m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5296ddb2f1414a23886292f9213c6178 : SUCCESS in 5m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/434d3d909c254272b29c0f458da7c662 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a8b837233f4c8b861728cf823f0f0b : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/883475374d6b49dd9b5090324daaf0ea : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13e4b4b16bf7427788a9f217103f7245 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8edadee94a7f4c4ea2ec6ade9b7e04ec : SUCCESS in 7m 54s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/61e76061a87c4f5e888317452c2dc5b8 : SUCCESS in 3m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/717d96cf5e804635a14daa8b2a96092c : SUCCESS in 8m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b8c7eba07e264e80a93ea34a96bbd6ab : SUCCESS in 9m 55s\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/58b847280c0048e19d18b032e1f5c3b2 : SUCCESS in 16m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e1ff9e809639ea557d4cbdd4923ac95731ee078d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-25 16:39:10.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"b8a41a983a30f0c0273a9350e391c6beda28691e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 17:02:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6bf386d2e9f49e581ec1dddc0998dac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea5fe0e9442a433a9e5f82e4d21fc4c2 : SUCCESS in 9m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a674c5970d5549c7999e3021749c941c : SUCCESS in 5m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d840353673d04e52a36a63fcdeef8aac : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b910f50eb9f4f8e942fe526755be46f : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a7c7568d23f45858b2349ac2e42376f : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/855f1d794a4e48868b0473ec17e92ca4 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a28695ce9e6e4ed6a9265409bb825b3d : SUCCESS in 6m 51s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4781a94f46a34e61a7011ec7eac3ee51 : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5916008b46884216980c24c48752ddfd : SUCCESS in 13m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/abc7f11739444f20b63e7869abb2b076 : SUCCESS in 8m 04s\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/8d85af5e073d4b64998bca5dab55e754 : FAILURE in 18m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7cb7a1342ac6e9761207b0ae7d18472a89c2aa5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-07 11:03:08.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"c0422ba985d58a7be7ea1cd7a8f3ed3ce27f2be2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 11:29:54.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/297ec3cbf804454883180f745e3e974d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/055883be99cd4a7280715f92ad5d573a : SUCCESS in 9m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/51fb4f183b0043839c1d678e6883656c : SUCCESS in 5m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/338832366a1942d497b72f7a7ba3123b : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb6681e33e15491981f33b3c2dd0a14a : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d76fe66cf12f42dcbdd39db2eea7b7aa : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55c12c42d7104628b04ae5190b9a3bb3 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d90283424fa547eabaa04df3ad039da8 : SUCCESS in 5m 53s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f084ef4c57e247538b3655dc7c4a6a81 : SUCCESS in 5m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/92953addc042443a839ad6b631e5abc2 : SUCCESS in 13m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3edf992a15104112a90cdfef1de81750 : SUCCESS in 9m 20s\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/69e13b902c144f01ba9124565044e61a : FAILURE in 19m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bb60ade37181e2f265d92df4e05de312c862dba7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-08-14 16:01:48.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"b5492927359a92ea24cd22e9c29498de64356ee5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-15 10:07:17.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"70168b35ac51304388469a8dda5c3858112bc57c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-18 08:49:47.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0f967fd725784941e9c4bfe339332c64e7e98c17","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-18 08:53:46.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d434a21e017c977b6694d2ca77733b4e11f2eb87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-26 16:54:16.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"dd5e37a6cf0fe8a6fe05b4a233b56b47935ce19d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 17:36:21.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d01ee1f8417e4d888542cea61c3a681e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e290bcf75aaa4baaa4ff6b532783a9d5 : SUCCESS in 4m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/526ec24df4424f9c95296512ffc8981a : SUCCESS in 6m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/392b8a43bfcf4f1186d85366c418e689 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c493c79339048aab4443e0cef5e6f2b : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fee9e2009cc40428d485d3c63652f23 : SUCCESS in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8dd64b1872f1496a8378d4c918389c4c : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9b01d1e8fad4a429cc55a2dc0c9d4f1 : SUCCESS in 7m 11s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ab1446b763544406800f0c6b6c87266a : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/238540a5707544be8f5775bfd04d2afb : SUCCESS in 13m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9166a317d38e4520bd497ac9d4932846 : SUCCESS in 9m 32s\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/2802777daef04010ac886413b63403d6 : FAILURE in 39m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1fb690a95d5af688affcf881005d85ebd8d72d8a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-26 18:58:13.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"3ca5225f375a90b795eb513b6138512133408043","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-08-27 08:29:32.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4fea408241d603f695cd3a74daafbd8a062d2ad7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-26 12:37:07.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"7818cdfdffcf5707deefd8ee46ff4c025467e29c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 12:37:25.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  The nodeset \"kayobe-centos10s\" was not found.\n\nThe problem appears in the the \"kayobe-overcloud-host-configure-centos10s\" job stanza:\n\n  job:\n      name: kayobe-overcloud-host-configure-centos10s\n      parent: kayobe-overcloud-host-configure-base\n      nodeset: kayobe-centos10s\n      voting: false\n\n  in \"openstack/kayobe/zuul.d/jobs.yaml@master\", line 317","accounts_in_message":[],"_revision_number":10},{"id":"0292934e7fb7e2d736b7dbdd8e914721624ae2e0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-26 13:03:22.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"9b6c2d48bd9ac8ab4f713f86e92f81ce974700ae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-26 13:21:32.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6546fb5fe8568e857eb9d70ebef2ff8e370ad706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 13:37:05.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c665e0d2b05b4bd6ac789ef809b4dffe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72dce054122c42a8a6cac0b9b230e901 : SUCCESS in 8m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bc399d76b0544fd6b2d31e4ced7935a5 : SUCCESS in 6m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83e7dcb6a7074267b8efc8c0b75b0325 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab1c9cf05c2945aaa4e3e7467284646f : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97518dab75184ed5ac8b0134d5c16193 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d70fe77937884ddcab2ced2fcc65d467 : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/629274fda30d4c08866e0bb5e8957383 : SUCCESS in 7m 17s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c51a5768c6dd432c8d34a5607b096765 : SUCCESS in 2m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e090d4f1c15340aeb06c1b7bf92dba18 : SUCCESS in 11m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0e425476c3f44e1fbbd82ccfa69f05bd : SUCCESS in 12m 53s\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/2906194d97684a51b2833e3a6e132b97 : SUCCESS in 13m 26s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"094a07680acffc035647d8eebfac331d1aa9fa5f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-29 09:55:08.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"49b6f65b11dcc4fa61dd031abf10c3eeb1acc81b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 10:17:52.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be0ddbae73b34b7293b708dc2dc61a27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/985ebea3865646d2ae55b67e16d13eb1 : SUCCESS in 9m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1166eaa5b9a24f0782d483d5e8849730 : SUCCESS in 6m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b3001797b56469ca4cb1e000afbe63f : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd2b74f162e34f33970f7fd9b8fa7f13 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/732d820b46a2433282f43e8e6ca1aedf : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e773824a714c4cf3858522146e41cd9a : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9aeb4cf40df84fb68f42d96ad1e29fda : SUCCESS in 7m 06s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/79d3e8601ddf47c6a45c79f0a257d06b : SUCCESS in 3m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/395eaff8a024476599ae651538fce2a2 : SUCCESS in 6m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/342b15ae072a4be2a7ffad665ed841e8 : SUCCESS in 6m 23s\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/70b8100d2cd14d849c14ae86d891f8bd : SUCCESS in 15m 06s (non-voting)\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/08b62f2f39144325ab14fbf4fb3d6b5c : SUCCESS in 17m 34s","accounts_in_message":[],"_revision_number":13},{"id":"52afdc600218fa1c0cf8b2ae71b22a9d10d491c0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-29 10:34:29.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"7cecfb311ba7548c68ddfa19d1be3a292c7b1010","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-29 10:35:24.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":14},{"id":"af9c655e1ee0f292c7f304338f63f73b926646ee","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-29 10:36:57.000000000","message":"Patch Set 14: Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"9ab71055ffbeaab4e0baa708682cc3a0bcd6f231","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-29 10:38:29.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d280805ad68a10a203f164eab9a3fc0503127a13","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 11:33:30.000000000","message":"Patch Set 14:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6f81d1f14c944ec909a09b93ead62cf\n\n- kayobe-infra-vm-centos9s-cloud-image https://zuul.opendev.org/t/openstack/build/d89f20d7c67848008fecd258169e1815 : SUCCESS in 16m 23s (non-voting)\n- kayobe-infra-vm-rocky9-cloud-image https://zuul.opendev.org/t/openstack/build/bc30a127d8f0401d9fe1e0bdfedd63c7 : SUCCESS in 14m 31s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/8c1f246624df4719806f9a45b01e2c72 : FAILURE in 22m 24s (non-voting)\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e8c2fa10ed534c19b7c4c5871a943f19 : FAILURE in 17m 46s (non-voting)\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ef7e00e82abd44e28c1ba8dc8646c41c : SUCCESS in 9m 26s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/c3bc7ec941ab4e0b850c062b3a53faed : SUCCESS in 10m 30s (non-voting)\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2a211811527345e1883ec4d6847051e7 : FAILURE in 25m 00s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/762847c982f043ffa308bb8cb8287302 : FAILURE in 27m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/491f90d9b55a4a6bb509565f1ec0605e : FAILURE in 28m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5f656e442e5a422db8dbfbd1e0601916 : FAILURE in 22m 36s\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/1d036e1a48b14632867e712c06252101 : FAILURE in 16m 20s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c8d14fbeb7a7403db4c5b7d5b25cce8c : FAILURE in 17m 10s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/746754049559415a9aaf32eb6719961d : FAILURE in 14m 38s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4502ed0b49a54239845b720fef788295 : SUCCESS in 43m 26s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/926e8cb9c7924694a382ae6942d42666 : SUCCESS in 49m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e361af78546f4a22bfbe22b6f3099cd4 : SUCCESS in 19m 19s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/bcae36d4aa4940c69e49a3c6422ecafb : SUCCESS in 18m 05s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/36b87617818f4b1681719cb3b86ff9f9 : SUCCESS in 32m 52s\n- kayobe-seed-vm-centos9s-cloud-image https://zuul.opendev.org/t/openstack/build/a510a11b444b431189bfe1fbd64db4a2 : SUCCESS in 20m 56s (non-voting)\n- kayobe-seed-vm-centos9s-q35 https://zuul.opendev.org/t/openstack/build/6121651d4b0f425393186b57e7a80749 : SUCCESS in 14m 19s (non-voting)\n- kayobe-seed-vm-rocky9-cloud-image https://zuul.opendev.org/t/openstack/build/e2bcd379eb2044a8ac9aab5d29e8e612 : SUCCESS in 12m 47s\n- kayobe-seed-vm-rocky9-q35 https://zuul.opendev.org/t/openstack/build/d1b34e264d13423aa0b0a6ee9ead5201 : SUCCESS in 6m 55s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/5d308c1eed4a49f49b206578a566cf55 : SUCCESS in 20m 14s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/e0c7dc45fb954a2a83ed5c5498622d3d : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":14},{"id":"ad3545896e094a8fc616b954b796e0fa527c6ca0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 12:07:47.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e0a250887784624b3bd35facab0f84a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8451b4bbc274407eb0f6e27bff944d84 : SUCCESS in 9m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/91fef2528a874deb94b05b51b1e6b136 : SUCCESS in 5m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b284718f4a84af38cad781f92bfbd74 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/201ab87763444428bf68e4b32aeafd1b : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/550e8bdaa96a4d148fb36dfafa97da80 : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3437a56b27b4113bea90e733b75507b : SUCCESS in 5m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3803180791c4967b6f8be87cf8a6497 : SUCCESS in 8m 11s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d6ab55394bfc4db6bbfa262b3de289b3 : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0f23265d1dea4577b1f90dec8348b8be : SUCCESS in 9m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6758257669774fd1ac153b46267a91c0 : SUCCESS in 7m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ff5e7550e284157ad05b0f501f7a988 : SUCCESS in 22m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f3461e84e4d74b1d852a249421b1ecbf : SUCCESS in 52m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1ed81c1b0fd0463598ba81805a5d6734 : SUCCESS in 13m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ec4fbf25ccbb494b86cedd89fc32c0d3 : SUCCESS in 9m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/104d3fec70964a87aa0de3c98eaa420d : SUCCESS in 12m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df6e934ba2cb430cacfe511b5771aee2 : SUCCESS in 1h 28m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/741b20a1249d4f8b911a79bd5435b23e : SUCCESS in 11m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/426b90b378f0445a80d0c4d1c9182665 : SUCCESS in 12m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ab7a3e100ae488c9fa4314a1a413528 : SUCCESS in 30m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b59bfc1639444052831f21e4b375e19f : SUCCESS in 15m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bdd6c4a66f149d5b0229e0bd80021d5 : SUCCESS in 7m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/867cf387074043b4905d40ab28603480 : SUCCESS in 8m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d277ba7e62db44669f1bc8ed6a44d9ed : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":14},{"id":"0a8a6c915ee13960656c671d65b13993c6591055","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-09-29 12:31:35.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"3019136aada96a1bb682211f171ab875279fa956","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-29 12:45:43.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"1dcbfcba09a69e67553c423b828f0e4773636db8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 12:46:09.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"d4f33005840f092348a8f272623ae58014cb2243","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-29 12:49:46.000000000","message":"Uploaded patch set 15.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":15},{"id":"0d719d0f84e2a9d99aa2d30cb32e6f9eb5d6ec56","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-09-29 12:58:12.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"526fc8e445c8479eb952074272db17fcdb72f94e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-29 12:59:01.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"91af0684c73d4d74f6ad5f2e138deff6ae459d37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 14:24:09.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47abe4e4e49c4e33821c70f35a28407a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2abe0206a0284fc6938d1a28d9c8b7da : SUCCESS in 8m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/31189b4e4e4e4240b6074110938bfdc6 : SUCCESS in 6m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5df01ed63250463294d93edcb0cbe6d8 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b20b2f2ffdba4f90acbea95543c34954 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73bbf179b646463ea6604cf954aa40c1 : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e100442996e54775bff3cc6ea98aae84 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd9f68a1273f46d38e60a6bae5f4ab42 : SUCCESS in 4m 18s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bab9f53e6f90413fbbb1afcca68de471 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ebd4dd4d1db84537a3a79d19fa80cbfa : SUCCESS in 11m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/26d5bb4ba00444b4b0f3b724a183c5f1 : SUCCESS in 10m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/399724585917431587786bb7bcc13465 : SUCCESS in 45m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8e142c195b474116b6bec725bea1cef5 : SUCCESS in 46m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4a71c11d50e341b291f10a237e859c8b : SUCCESS in 11m 57s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5c71b97cc7a7490f8a04236bcc62fe18 : SUCCESS in 12m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d584a82e6ed43fd9061225437fbeda0 : SUCCESS in 13m 19s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf6dbc69561e43e584e3e951e9c7658b : SUCCESS in 1h 28m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b15804ae848049978b156ba5fe98c8b2 : SUCCESS in 14m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4f062358a8c14a90925ef0e5ddbfbd60 : SUCCESS in 12m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a4b7c668d134ef4b979ac6c545d6343 : SUCCESS in 26m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9a17198b350a4960ba0392ce8caa068e : SUCCESS in 15m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5331d1e5828f4a519542e8b1de9078cd : SUCCESS in 13m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/800b62bc36e14ca8bb1616063f852bbe : SUCCESS in 13m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/833a4a85c4354239b4840cd5a8c6a44a : SUCCESS in 11m 11s","accounts_in_message":[],"_revision_number":15},{"id":"d82bbfa0644dd9e9ebc6f81969b104ea87e604b7","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-09-29 14:27:32.000000000","message":"Patch Set 15: Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"692520506af1035a07ee0dd4ae7c06ab6138a0ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 14:28:30.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"a4ccad5050b9d982aae82a4d7fbd3770010c78b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 16:13:48.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6211e5e42617447bb6bdcf164ea72883\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a1867e8354d435a90a59690a7632e34 : SUCCESS in 6m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/45746fa2588246e4af65b88d1b5225e2 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55a37af3cf834949b98f28fdbc15705f : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbcaf89b5ee04eae8f8b9b629ae2d734 : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64370403b7394f3da8547e34699f74e0 : SUCCESS in 4m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cbd8aeb07e5242eeb0e3c692acda3af2 : SUCCESS in 3m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a5db998826bc46b3941f1ff984ecf04b : SUCCESS in 6m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa908ddda96344c0a3bddbbfb6ee5d37 : SUCCESS in 5m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b73e6f3524041c09a948516e7cce402 : SUCCESS in 23m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/33fb7e01ffa84af88cda126d421367dd : SUCCESS in 25m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/397ffb720647451c96e4f93da64f7187 : SUCCESS in 9m 55s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/673da3c9b8e04d9880057644fb7f095f : SUCCESS in 11m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72adf78dba34483380c84061670df03d : SUCCESS in 9m 03s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfba21489a384423b532238daf67e3f8 : SUCCESS in 1h 40m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6b83b042a2045c495a007be51507201 : SUCCESS in 14m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/77fc40b3f3a04fa5b2ec77cb8044ba05 : SUCCESS in 14m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d7a0ef74d744cac8f97be66e8dff49e : SUCCESS in 14m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1bcfc98c0d5e4c98a58abe2fd6b3a8eb : SUCCESS in 9m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee3902d162704927af6a3bfb6ed70d92 : SUCCESS in 8m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/66b8312fe900498c9c01ee6013cdec98 : SUCCESS in 12m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d9a3361079b461c81ceff7cc18871e6 : SUCCESS in 13m 26s","accounts_in_message":[],"_revision_number":15},{"id":"0f0351d02ca4088f87c5f942da438e76df5b79af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 16:13:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"7ff4e35a24abcaede6049bee99d0d9f55b941e82","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 16:15:29.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db5ba3be599044399f7bbd9db55d86a7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ce74770a1184a409cc86dc030c171ac : SUCCESS in 47s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"80782a43e0fecd74b4f5061b6f95a800f78949da","revisions":{"becc915cae3fbf5fa39915e358a159e80759b626":{"kind":"REWORK","_number":1,"created":"2025-07-24 11:05:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/955772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/955772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/955772/1"}}},"commit":{"parents":[{"commit":"3d20c5fdf941a7f759f2b7f2f183036c5b485fb5","subject":"[WIP] Bump Ansible roles to support centos/rocky 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d20c5fdf941a7f759f2b7f2f183036c5b485fb5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-24 11:05:34.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-24 11:05:34.000000000","tz":120},"subject":"CI: Add kayobe-overcloud-host-configure-centos10s job","message":"CI: Add kayobe-overcloud-host-configure-centos10s job\n\nChange-Id: Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/becc915cae3fbf5fa39915e358a159e80759b626"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/becc915cae3fbf5fa39915e358a159e80759b626"}]},"branch":"refs/heads/master"},"f4b9472531fbd214bfc9dc4b689d3c9cab6b9bde":{"kind":"REWORK","_number":2,"created":"2025-07-24 12:32:58.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/955772/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/955772/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/955772/2"}}},"commit":{"parents":[{"commit":"3d20c5fdf941a7f759f2b7f2f183036c5b485fb5","subject":"[WIP] Bump Ansible roles to support centos/rocky 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d20c5fdf941a7f759f2b7f2f183036c5b485fb5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-24 12:32:39.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-24 12:32:39.000000000","tz":120},"subject":"CI: Add kayobe-overcloud-host-configure-centos10s job","message":"CI: Add kayobe-overcloud-host-configure-centos10s job\n\nChange-Id: Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4b9472531fbd214bfc9dc4b689d3c9cab6b9bde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4b9472531fbd214bfc9dc4b689d3c9cab6b9bde"}]},"branch":"refs/heads/master"},"f9196f527afe8d30fb2f8cb6890ce7593c1e78e8":{"kind":"REWORK","_number":3,"created":"2025-07-24 12:42:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/955772/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/955772/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/955772/3"}}},"commit":{"parents":[{"commit":"3d20c5fdf941a7f759f2b7f2f183036c5b485fb5","subject":"[WIP] Bump Ansible roles to support centos/rocky 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d20c5fdf941a7f759f2b7f2f183036c5b485fb5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-24 12:41:57.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-24 12:41:57.000000000","tz":120},"subject":"CI: Add kayobe-overcloud-host-configure-centos10s job","message":"CI: Add kayobe-overcloud-host-configure-centos10s job\n\nChange-Id: Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9196f527afe8d30fb2f8cb6890ce7593c1e78e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9196f527afe8d30fb2f8cb6890ce7593c1e78e8"}]},"branch":"refs/heads/master"},"5c04c253fa4aea0efbfea5a8210290dfbfa61acd":{"kind":"REWORK","_number":4,"created":"2025-07-24 13:01:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/955772/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/955772/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/955772/4"}}},"commit":{"parents":[{"commit":"3d20c5fdf941a7f759f2b7f2f183036c5b485fb5","subject":"[WIP] Bump Ansible roles to support centos/rocky 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d20c5fdf941a7f759f2b7f2f183036c5b485fb5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-24 13:01:42.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-24 13:01:42.000000000","tz":120},"subject":"CI: Add kayobe-overcloud-host-configure-centos10s job","message":"CI: Add kayobe-overcloud-host-configure-centos10s job\n\nChange-Id: Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c04c253fa4aea0efbfea5a8210290dfbfa61acd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c04c253fa4aea0efbfea5a8210290dfbfa61acd"}]},"branch":"refs/heads/master"},"34ca22466da0551a7d9e663984e38f9148943ad0":{"kind":"REWORK","_number":5,"created":"2025-07-24 14:28:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/955772/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/955772/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/955772/5"}}},"commit":{"parents":[{"commit":"3d20c5fdf941a7f759f2b7f2f183036c5b485fb5","subject":"[WIP] Bump Ansible roles to support centos/rocky 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d20c5fdf941a7f759f2b7f2f183036c5b485fb5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-24 14:27:49.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-24 14:27:49.000000000","tz":120},"subject":"CI: Add kayobe-overcloud-host-configure-centos10s job","message":"CI: Add kayobe-overcloud-host-configure-centos10s job\n\nChange-Id: Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34ca22466da0551a7d9e663984e38f9148943ad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34ca22466da0551a7d9e663984e38f9148943ad0"}]},"branch":"refs/heads/master"},"54ae34bc6dd48ab261c71229a2db1902ba02fb7e":{"kind":"REWORK","_number":6,"created":"2025-07-24 15:25:28.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/955772/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/955772/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/955772/6"}}},"commit":{"parents":[{"commit":"3d20c5fdf941a7f759f2b7f2f183036c5b485fb5","subject":"[WIP] Bump Ansible roles to support centos/rocky 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d20c5fdf941a7f759f2b7f2f183036c5b485fb5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-24 15:25:21.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-24 15:25:21.000000000","tz":120},"subject":"CI: Add kayobe-overcloud-host-configure-centos10s job","message":"CI: Add kayobe-overcloud-host-configure-centos10s job\n\nChange-Id: Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/54ae34bc6dd48ab261c71229a2db1902ba02fb7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/54ae34bc6dd48ab261c71229a2db1902ba02fb7e"}]},"branch":"refs/heads/master"},"e240d6d3e52e127e26138fe0d2b4bbcb9be37fa1":{"kind":"REWORK","_number":7,"created":"2025-07-25 16:39:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/955772/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/955772/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/955772/7"}}},"commit":{"parents":[{"commit":"3d20c5fdf941a7f759f2b7f2f183036c5b485fb5","subject":"[WIP] Bump Ansible roles to support centos/rocky 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d20c5fdf941a7f759f2b7f2f183036c5b485fb5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-25 16:38:45.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-25 16:38:45.000000000","tz":120},"subject":"CI: Add more centos10s jobs","message":"CI: Add more centos10s jobs\n\nChange-Id: Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e240d6d3e52e127e26138fe0d2b4bbcb9be37fa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e240d6d3e52e127e26138fe0d2b4bbcb9be37fa1"}]},"branch":"refs/heads/master"},"366f81bc59895cd095783a93f87fab2326993c58":{"kind":"REWORK","_number":8,"created":"2025-08-07 11:03:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/955772/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/955772/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/955772/8"}}},"commit":{"parents":[{"commit":"79a2114ece9355e01ca2ae1e97ac9e453ad7528b","subject":"[WIP] Bump Ansible roles to support centos/rocky 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79a2114ece9355e01ca2ae1e97ac9e453ad7528b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-07 11:02:56.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-07 11:02:56.000000000","tz":120},"subject":"CI: Add more centos10s jobs","message":"CI: Add more centos10s jobs\n\nAdapt epel-release task to support CentOS Stream 10.\n\nChange-Id: Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/366f81bc59895cd095783a93f87fab2326993c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/366f81bc59895cd095783a93f87fab2326993c58"}]},"branch":"refs/heads/master"},"b1b7253aa85c18e7c907c6752f57e343f4f8a877":{"kind":"REWORK","_number":9,"created":"2025-08-26 16:54:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/955772/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/955772/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/955772/9"}}},"commit":{"parents":[{"commit":"de1eb9146f640c4331e1898ab7d0d91c5777f1b9","subject":"[WIP] Bump Ansible roles to support centos/rocky 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de1eb9146f640c4331e1898ab7d0d91c5777f1b9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-26 16:52:53.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-26 16:52:53.000000000","tz":120},"subject":"CI: Add more centos10s jobs","message":"CI: Add more centos10s jobs\n\nChange-Id: Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1b7253aa85c18e7c907c6752f57e343f4f8a877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1b7253aa85c18e7c907c6752f57e343f4f8a877"}]},"branch":"refs/heads/master"},"f802d62f4490764a8d6d7ef8e068c0d5a5eb8bc3":{"kind":"REWORK","_number":10,"created":"2025-09-26 12:37:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/955772/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/955772/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/955772/10"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-26 12:35:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-26 12:35:47.000000000","tz":120},"subject":"Add CentOS Stream 10 host configure job","message":"Add CentOS Stream 10 host configure job\n\nChange-Id: Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f802d62f4490764a8d6d7ef8e068c0d5a5eb8bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f802d62f4490764a8d6d7ef8e068c0d5a5eb8bc3"}]},"branch":"refs/heads/master"},"5866f0019552dbcbf20977b390c58d1b8250ab9b":{"kind":"REWORK","_number":11,"created":"2025-09-26 13:03:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/955772/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/955772/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/955772/11"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-26 13:02:53.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-26 13:02:53.000000000","tz":120},"subject":"Add CentOS Stream 10 host configure job","message":"Add CentOS Stream 10 host configure job\n\nChange-Id: Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5866f0019552dbcbf20977b390c58d1b8250ab9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5866f0019552dbcbf20977b390c58d1b8250ab9b"}]},"branch":"refs/heads/master"},"10946bc47a2dd4be8c672f7110c75b54ecff9841":{"kind":"REWORK","_number":12,"created":"2025-09-26 13:21:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/955772/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/955772/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/955772/12"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-26 13:21:15.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-26 13:21:15.000000000","tz":120},"subject":"Add CentOS Stream 10 host configure job","message":"Add CentOS Stream 10 host configure job\n\nChange-Id: Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/10946bc47a2dd4be8c672f7110c75b54ecff9841"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/10946bc47a2dd4be8c672f7110c75b54ecff9841"}]},"branch":"refs/heads/master"},"7fcc1bdb8862d43ac473203cf0a0adbc7d101b1a":{"kind":"REWORK","_number":13,"created":"2025-09-29 09:55:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/955772/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/955772/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/955772/13"}}},"commit":{"parents":[{"commit":"b7a911966749e8fc3bd3cb53e99666c284ae8577","subject":"Merge \"CI: Fix Ubuntu cloud image tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7a911966749e8fc3bd3cb53e99666c284ae8577"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-29 09:53:17.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-29 09:53:17.000000000","tz":120},"subject":"CI: Add CentOS Stream/Rocky Linux 10 host configure jobs","message":"CI: Add CentOS Stream/Rocky Linux 10 host configure jobs\n\nChange-Id: Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7fcc1bdb8862d43ac473203cf0a0adbc7d101b1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7fcc1bdb8862d43ac473203cf0a0adbc7d101b1a"}]},"branch":"refs/heads/master"},"1713df3d425e48ad36fa06eff738b8cb2371e371":{"kind":"REWORK","_number":14,"created":"2025-09-29 10:34:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/955772/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/955772/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/955772/14"}}},"commit":{"parents":[{"commit":"b7a911966749e8fc3bd3cb53e99666c284ae8577","subject":"Merge \"CI: Fix Ubuntu cloud image tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7a911966749e8fc3bd3cb53e99666c284ae8577"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-29 10:33:36.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-29 10:33:36.000000000","tz":120},"subject":"CI: Add CentOS Stream/Rocky Linux 10 host configure jobs","message":"CI: Add CentOS Stream/Rocky Linux 10 host configure jobs\n\nChange-Id: Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1713df3d425e48ad36fa06eff738b8cb2371e371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1713df3d425e48ad36fa06eff738b8cb2371e371"}]},"branch":"refs/heads/master"},"80782a43e0fecd74b4f5061b6f95a800f78949da":{"kind":"REWORK","_number":15,"created":"2025-09-29 12:49:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/955772/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/955772/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/955772/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/955772/15"}}},"commit":{"parents":[{"commit":"8c2382006a590f41a4ade20a49c364274819ee50","subject":"Merge \"Configure tuned on Ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c2382006a590f41a4ade20a49c364274819ee50"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-29 12:49:37.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-29 12:49:37.000000000","tz":120},"subject":"CI: Add CentOS Stream/Rocky Linux 10 host configure jobs","message":"CI: Add CentOS Stream/Rocky Linux 10 host configure jobs\n\nChange-Id: Ide4cfa11f4867cd9883e7f994443eb4c4849d1d7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80782a43e0fecd74b4f5061b6f95a800f78949da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80782a43e0fecd74b4f5061b6f95a800f78949da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
