)]}'
{"id":"openstack%2Fkayobe~956133","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~I8698e0b0fa01c84318e16a5e61f698ad36996a7f","project":"openstack/kayobe","branch":"stable/2024.2","topic":"bump/config-drive-stable/2024.2","attention_set":{},"removed_from_attention_set":{"34212":{"account":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"last_update":"2025-07-30 11:52:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8698e0b0fa01c84318e16a5e61f698ad36996a7f","subject":"Bump jriguera.configdrive to latest master commit","status":"MERGED","created":"2025-07-30 08:24:34.000000000","updated":"2025-07-30 11:53:10.000000000","submitted":"2025-07-30 11:52:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956133-bump/config-drive-stable/2024.2","cherry_pick_of_change":955777,"cherry_pick_of_patch_set":4,"meta_rev_id":"b61364fae2c277ea7e59872c06f3a2f228ce680f","_number":956133,"virtual_id_number":956133,"owner":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-30 11:52:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-07-30 09:23:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-07-30 09:36:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-30 08:30:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-07-30 09:36:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-30 08:30:59.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-07-30 09:23:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-07-30 09:36:36.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-07-30 09:51:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b46daa1f723ede01ef478d9221407f072f715dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-07-30 08:24:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ec317550c56e27ba808d027e2f2f057d0cdfb9c1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-30 08:30:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1c6035b0cb2c8dce2bcba0f14ff9a3f744ec2c59","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-30 09:23:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"52de38640ee4d9ff6810d592bd0cd545bbd22f36","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-30 09:36:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"01784fed25c3fc6d71a24a77f86737589408c995","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 09:51:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/878fdc65f18040cebcf3474d50a7b7b8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26d8ab09d9af424db97fd355246e4a67 : SUCCESS in 2m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80627e0365d7481891e04d0da5f4af5a : SUCCESS in 2m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b62982563ac41e4ade349fd49aec348 : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48ae573a3d26441aadac166adc328937 : SUCCESS in 2m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f71c560cc2154c3cbec16fcbdc3e8fb0 : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e81bd14d41645b3a0be0afefa454053 : SUCCESS in 2m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6073e9b28bb546ea912b4ed9bb645a38 : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/203c0e9bce814648acd1b359a617cae7 : SUCCESS in 5m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1285891c4e43438ab23299ae7b7a9712 : SUCCESS in 4m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e76ce4e27cdb4f8183b2d764fb988206 : SUCCESS in 36m 02s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e301631919fc4724bd3dd002b54c4e5f : SUCCESS in 35m 58s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a6bcc59a8ad4841a4d0b7a7b7f97942 : SUCCESS in 1h 01m 58s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e01587884b3e4af0af998c240ca4ca3e : SUCCESS in 1h 13m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c1f1aae1f2a34db5827312ac3db59140 : SUCCESS in 22m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/200bf82f3c5d42dd8799965ed0f0d89d : SUCCESS in 10m 28s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96b2e6a035ca428daf47f80c5a358a80 : SUCCESS in 17m 52s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6fd00d0de2634f9f93655b04963c8f86 : SUCCESS in 1h 25m 32s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d88c3a760247436eae1e423ae1fa1d77 : SUCCESS in 57m 07s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4b689c4da4cd43e39facca6e3cedd664 : SUCCESS in 17m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a9fdbca47ab4c77935d4f4cfbb65c02 : SUCCESS in 15m 26s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/423fe33a3a4647d0a7f32a8f400d53c4 : SUCCESS in 34m 57s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/472bc76641134f32b2196dcfde8f0d5d : SUCCESS in 20m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3f7a4deb57b4be0b2c550c7861ac093 : SUCCESS in 28m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9d31565085f34366942c2813d59eea48 : SUCCESS in 14m 45s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b7119441978a44faa68e5dd8a3aa76af : SUCCESS in 11m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1fe428c2af64f928b6080d6fcf6e77c : SUCCESS in 10m 52s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/d259e2750a1b4feab733c507ea81ce6f : SUCCESS in 9m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/aeca00a3ec004cb58a29b67b830984a9 : SUCCESS in 16m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/900631730c6548a9a6674eabb8ae0adf : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ed843bd8aa5631b2b06dff6689febc22e44389c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 09:51:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d4951f214c0877d335b7d6e9c92aca76ce13eb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 11:52:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3a8af5e8e2a4d3a99c659431ea71f72\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4bb3199e2c24a379b6a6afa9c8c6337 : SUCCESS in 2m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27ac7cf902744371a057c01ff947bbcb : SUCCESS in 2m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eed025507cb41b1a6ce44ef2832f42f : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d9dd4af17864211845fb5b80a483b83 : SUCCESS in 1m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/777e33b581ef43ea8a01f2d1910b9934 : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a661fe324eb64c7fbfd6efd6cb7ca246 : SUCCESS in 9m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ceb84d3795bd49a6a077e05154cef1b7 : SUCCESS in 5m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e5ac4c5a93a446dfb271931c66f0af48 : SUCCESS in 39m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0db311fbfb0c4aca97b8b8f1cc6e975a : SUCCESS in 41m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/54d8b30e90f34788b869e7b0d7a2214f : SUCCESS in 1h 01m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b2dc14d9d4aa4d0e8aaa5757c21f61e4 : SUCCESS in 13m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa3510c554354f609f1e1b8889cb2c71 : SUCCESS in 14m 35s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5ee6eff67dd743fa88825ff4d6830bcd : SUCCESS in 1h 42m 14s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8364e69a75de495d8fa692b2249a8835 : SUCCESS in 1h 31m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c80bd72b6c67451c8ca0d7b5317adccf : SUCCESS in 19m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/880222d737054f01a62f6d59a15353cf : SUCCESS in 17m 34s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/557c0c13b1bd47298a69e771485b5e15 : SUCCESS in 35m 13s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/304d2fc887ee4260a781647d8ddc7b7d : SUCCESS in 18m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f621556a02f649e080ea761ee9121a76 : SUCCESS in 17m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/054dc804d14d4fc59be72cadb40600b2 : SUCCESS in 12m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/656d28a267d94330a3e26a2f19ba3211 : SUCCESS in 13m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f81402de76d94d2caeeb05de0203d78e : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a5fe1f91bd33314b1f57518a4a482b1c6893282a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 11:52:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b61364fae2c277ea7e59872c06f3a2f228ce680f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 11:53:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49e6f71211464f3581da9b66c94e7bed\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc48d3d9452a4d9e865dd4717621250a : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c1580ab7b5d4200810db45df415ec57 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d792e3ae7725e65419e7d8764ce6004431db3b6c","revisions":{"d792e3ae7725e65419e7d8764ce6004431db3b6c":{"kind":"REWORK","_number":1,"created":"2025-07-30 08:24:34.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/33/956133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/956133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/956133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/956133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/956133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/956133/1"}}},"commit":{"parents":[{"commit":"b7dd59e2982649007a686f424cd156e410e8878c","subject":"Fix documentation of seed_containers usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7dd59e2982649007a686f424cd156e410e8878c"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-07-24 11:36:23.000000000","tz":60},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-07-30 08:24:34.000000000","tz":0},"subject":"Bump jriguera.configdrive to latest master commit","message":"Bump jriguera.configdrive to latest master commit\n\nThis resolves an issue where incorrect network-data.json\nwould be generated when interfaces without IP addresses\nwere attached to infrastructure VMs.\n\nCloses-Bug: #2118403\n\nChange-Id: I8698e0b0fa01c84318e16a5e61f698ad36996a7f\nSigned-off-by: Matt Anson \u003cmatta@stackhpc.com\u003e\n(cherry picked from commit 28dadfbce180192fc197caa65c943140190e485e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d792e3ae7725e65419e7d8764ce6004431db3b6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d792e3ae7725e65419e7d8764ce6004431db3b6c"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
