)]}'
{"id":"openstack%2Fkayobe~956440","triplet_id":"openstack%2Fkayobe~master~I9c3b78cfba1676d75f829cd5bd5b5dc05fd52dbc","project":"openstack/kayobe","branch":"master","topic":"add-password-view-cmd","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-11-21 18:36:37.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"35010":{"account":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"last_update":"2025-11-28 13:22:07.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-11-28 13:10:20.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I9c3b78cfba1676d75f829cd5bd5b5dc05fd52dbc","subject":"feat: add command to view passwords","status":"MERGED","created":"2025-08-03 11:49:23.000000000","updated":"2025-11-28 13:23:00.000000000","submitted":"2025-11-28 13:22:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"956440-add-password-view-cmd","meta_rev_id":"2e638aa583d34a5a023d8d167d675ad3f9fa9b5f","_number":956440,"virtual_id_number":956440,"owner":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-28 13:22:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-11-21 18:36:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-11-28 09:52:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":1,"date":"2025-11-21 18:59:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-26 12:00:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-11-28 11:05:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":-1,"date":"2025-11-28 11:05:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-11-21 18:36:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-03 13:37:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-14 16:22:34.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2025-11-21 13:40:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2025-11-21 18:36:37.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-11-21 18:59:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-26 12:00:36.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-11-28 12:18:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"ade184148532e93fc61d18ef2281d3e54655d4d4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-08-03 11:49:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a315d982b57f2816d9f5e263a747ab27e9570ff4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-08-03 11:52:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d5d514221cae0abd157c21110f1a15cd63b795a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-03 13:37:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/395642af22e74e0d92f70168ae7a9ea3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/767331bae4f74ac9a9d5c1e659ad5128 : SUCCESS in 2m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1244db180164133a61293f89d4c2693 : SUCCESS in 9m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c84c40d545e1427e812fd2cf08cc8a85 : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/434f6b38796a4352be1229b119cb23b6 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/134d2218d8dd460b8fb4d8cb3edd9be2 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a18e46d4190b4500aa3581ddc580afaf : SUCCESS in 2m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c97258573c646f586cfd592de5207d8 : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6dbdeb64848c4a86a558c8bc868ddb15 : SUCCESS in 3m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e14b02da01474f958ed54290199eaec8 : SUCCESS in 2m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/05e357bd846c4defb53db16b54d43bde : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a42ada45dc42434583d1e31300bc15bc : SUCCESS in 5m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dd8de6a544784d788257f8c8ef02b037 : SUCCESS in 5m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2eda3c745ae84cc29e73e52c80f68082 : SUCCESS in 28m 29s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9615cf23b6314717829fe576a6d81be4 : SUCCESS in 1h 08m 59s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d61dd86f21ae41419f353b5482fb1274 : SUCCESS in 20m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be7b0266b0cf4398aa0dbf65a60c500c : SUCCESS in 6m 36s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33c32428d0a8493a88590c8b8a01c980 : SUCCESS in 1h 40m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/137db3a86cd94038af13544211c90143 : SUCCESS in 17m 58s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/14ae10f8da864cf89645f46a4c368b08 : SUCCESS in 17m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0b1d04affad41759c4fa75727945c6b : SUCCESS in 7m 19s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7807fd1e6d804cd1a8644f07dcf240b5 : SUCCESS in 13m 25s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/283ca68f5c9e42e7a43956363e2152a1 : SUCCESS in 40m 26s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d2a4d7708ed34a7dbb60516a07e72417 : SUCCESS in 16m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1c5ba0753cc4059b47c9d576d9889f8 : SUCCESS in 27m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/00a7f7d590f74b948cae056963db8369 : SUCCESS in 15m 10s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e41636c6c09d486eb74352ca6a3a17ff : SUCCESS in 13m 42s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb5ae909c1674dbeb713549d572b3752 : SUCCESS in 12m 41s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/dddd8ee7ced742a4bf12ba169bb7670a : SUCCESS in 13m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f069e714e491442db275e3c19044f9a7 : SUCCESS in 8m 02s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e438ffc539e14ca9b80bc52e032c17fa : SUCCESS in 11m 49s","accounts_in_message":[],"_revision_number":2},{"id":"2ec2cfe8c4e133cd0aca9994392f54bab43da61c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-08-14 16:22:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3672bd4fad96e460ddb61d878c90217ce435c976","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-11-19 15:12:58.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c80e4e3bda6837489d8888b7dc8d9b0318268a6b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-11-19 15:26:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"37ce229959b4de56c9274d756e534aac81c417b2","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-11-19 15:27:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6c79fb1a4d1da12ace5aa419c6a103ce9c62cb8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 17:16:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b87c2fcb248f47e7b3a5df99a1d13b16\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/033a8bc28e4545c382c5719ebdef4aa2 : SUCCESS in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/833c5f21ba9243b98769771e07e1c14a : SUCCESS in 4m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/093dad9fa4fe4fc58f67970cc28d7545 : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31cbbc7f89d94107a414351fc19d9932 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/355ea06d83024b15ade88b774ea8e0e5 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9eaef7c1988c4075abc0fb985a87ab7c : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8d4688c34824e3ebe2b7642d0ab8b7e : SUCCESS in 2m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9f0d0b68a9fa494b817d070d3eadaa65 : SUCCESS in 3m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/faae95fb68bb440893acbf31192413f6 : SUCCESS in 12m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b14631782e3a44599aea8727b9e6809a : SUCCESS in 7m 58s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/45c830a5b09f4aeda5d1f4572122781c : SUCCESS in 13m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a80f54f928543c58bcbfaef0904e005 : SUCCESS in 13m 33s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/560345b9e4ac4fe99fcd79524e6ff20e : SUCCESS in 13m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/602d2be2129b4c83b657501c1496ab9b : SUCCESS in 14m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b384afb5e60b466f96895466da270b97 : SUCCESS in 53m 25s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f51b123934064b499e4f64c99b4eeecc : SUCCESS in 57m 25s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4cb2d75eb0fa4cd588456415ff906928 : SUCCESS in 1h 13m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b33227fd3494818b9dc86d6d8f3f2f8 : SUCCESS in 1h 02m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3f41a2f774ba4123a536d3a0b3bfd044 : SUCCESS in 1h 15m 27s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4dc18cfaa9154a6a87a74445a299b534 : SUCCESS in 1h 45m 10s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/bb7f0a61424b46f691d8384670584057 : SUCCESS in 19m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d68d9cd2a85f44468f277c102c01ac89 : SUCCESS in 11m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7098594dabe042bc99b1d8086b6d3091 : SUCCESS in 17m 22s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4239f4365589451da6194468f0f7f499 : SUCCESS in 16m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c88aedf72aea4c888b2c7aae0f77db8f : SUCCESS in 17m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2eb6e34a299d4083b086e4054980982e : SUCCESS in 14m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/026f1150bb8a40ab9ba7fc86a2e74911 : SUCCESS in 13m 38s","accounts_in_message":[],"_revision_number":4},{"id":"e26e6aa9fc96314240819d666ddf4bbd00d1e178","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-11-21 09:03:33.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"3cc59bc44afab30fe25979003a1e8bb8e9406283","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-21 18:36:37.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"38a0c089d35e2c548b310db3446e0204306b59b6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-21 18:59:35.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a26a2c13c193f7badea99dec0009c3e9c2765c1d","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-11-26 12:00:36.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d235fcac482e2df99debd7fe81f3d13ed005a4d7","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-11-28 09:52:39.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f04275bd77390cd542b629c0977b4eb17bcd067c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-28 11:05:56.000000000","message":"Patch Set 4: Backport-Candidate-1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8912c2071ba67d7754f58d0b0ceb061048663f8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 11:06:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"dc875b8ec47ada29739dc15a931f04003b272e5e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-28 12:18:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6abd08e94c39042ab23c9e37db9df2aef3b88528","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-11-28 12:30:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"471ce7ce9113be461e532bc4a00e81deadca36db","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-28 13:10:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c03bc168f8f0112c88ea5a3c2e52fa3a42653826","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 13:22:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/174ae248d5d34c02b2e2014668a2751d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5631a0e1c15543758f29c4c93b84c4e8 : SUCCESS in 1m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eca0dca19fa3403b957662310fef916c : SUCCESS in 7m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/663c939c7ef447d799dc86058a51a07d : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a4ce5a832834e1cb6273caa004b8443 : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/57d0562ecccd4de0bbff7c160f39b5d4 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc2bbea3cedd4ac98c86a3e9deebbf72 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e645ee4653364d208b837a1c9c842530 : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/496efdcff78944628884c8f88a8457b6 : SUCCESS in 13m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a4efa14a1df84f2d9cd5925c01898b15 : SUCCESS in 9m 26s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1a2b5f038e3f402ca8e63cd10bfe88bf : SUCCESS in 7m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe423e9b0ce64800b08b98b277e8ab9b : SUCCESS in 11m 47s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/17e704de29be4a4b89c18671da86115e : SUCCESS in 11m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b5ac4e55a2e43adbecb37b6d5a5c8ba : SUCCESS in 13m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/47e41627ab574ef5ab91fc1466068923 : SUCCESS in 1h 04m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7256133cd2cc4c049ba77e32105db670 : SUCCESS in 44m 25s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ec45cf9fc1424efea70519f65198e7c3 : SUCCESS in 42m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cbde322044e4837ad08deed63c400da : SUCCESS in 33m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5e148724e17742e99e1554396b0e4715 : SUCCESS in 57m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2f06aa1ebe24853b2eb4c7bd2757667 : SUCCESS in 1h 20m 30s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a53727335db94bfcaf7a416246185f03 : SUCCESS in 9m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/97b75e7f04e144299f0afccfe0515497 : SUCCESS in 16m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31af8ab816674de090d363975274d84d : SUCCESS in 16m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2d6ba831db7b4a24830e48c2eef18a5a : SUCCESS in 14m 25s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03148edae32c46e79c6bf1295b04e1e9 : SUCCESS in 27m 21s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8e247e90c3fe430a8a8cb2b42ad44547 : SUCCESS in 12m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d74e7cecf53b46bbac153101bda63434 : SUCCESS in 15m 39s","accounts_in_message":[],"_revision_number":4},{"id":"d174fca041db9fa6d97b1f255c6c699441aa4e73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 13:22:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"2e638aa583d34a5a023d8d167d675ad3f9fa9b5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 13:23:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbfca1f4326541ee96ce873ee0c1c4c5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d1dde0de2dd4d079e0c8591ad93a464 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bc321cdf86744a0b9bc639b31eb8e62 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ddccfb1679bc612434d86c72f09a74dd9e42df29","revisions":{"3c25f8f3ba5f871467e8a39afc2a2f0f0222a9de":{"kind":"REWORK","_number":1,"created":"2025-08-03 11:49:23.000000000","uploader":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"ref":"refs/changes/40/956440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/956440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/956440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/956440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/956440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/956440/1"}}},"commit":{"parents":[{"commit":"28dadfbce180192fc197caa65c943140190e485e","subject":"Bump jriguera.configdrive to latest master commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28dadfbce180192fc197caa65c943140190e485e"}]}],"author":{"name":"Jack Hodgkiss","email":"jack@stackhpc.com","date":"2025-08-03 11:46:40.000000000","tz":60},"committer":{"name":"Jack Hodgkiss","email":"jack@stackhpc.com","date":"2025-08-03 11:46:40.000000000","tz":60},"subject":"feat: add command to view passwords","message":"feat: add command to view passwords\n\nQuality of life improvement to make viewing the contents of\n`kolla/passwords.yml` with the introduction of a kayobe command that\nremoves the need to provide an absolute or relative path to the\n`ansible-vault` encrypted passwords file.\n\nChange-Id: I9c3b78cfba1676d75f829cd5bd5b5dc05fd52dbc\nSigned-off-by: Jack Hodgkiss \u003cjack@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c25f8f3ba5f871467e8a39afc2a2f0f0222a9de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c25f8f3ba5f871467e8a39afc2a2f0f0222a9de"}]},"branch":"refs/heads/master"},"d1c178e3e2cd8a0909e15c5290256e4ab6f549c0":{"kind":"REWORK","_number":2,"created":"2025-08-03 11:52:33.000000000","uploader":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"ref":"refs/changes/40/956440/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/956440/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/956440/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/956440/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/956440/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/956440/2"}}},"commit":{"parents":[{"commit":"28dadfbce180192fc197caa65c943140190e485e","subject":"Bump jriguera.configdrive to latest master commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28dadfbce180192fc197caa65c943140190e485e"}]}],"author":{"name":"Jack Hodgkiss","email":"jack@stackhpc.com","date":"2025-08-03 11:46:40.000000000","tz":60},"committer":{"name":"Jack Hodgkiss","email":"jack@stackhpc.com","date":"2025-08-03 11:52:22.000000000","tz":60},"subject":"feat: add command to view passwords","message":"feat: add command to view passwords\n\nQuality of life improvement to make viewing the contents of\n`kolla/passwords.yml` with the introduction of a kayobe command that\nremoves the need to provide an absolute or relative path to the\n`ansible-vault` encrypted passwords file.\n\nChange-Id: I9c3b78cfba1676d75f829cd5bd5b5dc05fd52dbc\nSigned-off-by: Jack Hodgkiss \u003cjack@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d1c178e3e2cd8a0909e15c5290256e4ab6f549c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d1c178e3e2cd8a0909e15c5290256e4ab6f549c0"}]},"branch":"refs/heads/master"},"51bf6e6cdd02da8f75ed5a93f0d8f1ad0c2541a8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-11-19 15:12:58.000000000","uploader":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"ref":"refs/changes/40/956440/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/956440/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/956440/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/956440/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/956440/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/956440/3"}}},"commit":{"parents":[{"commit":"dbde402b6356683f7c232950e91d80f15ac1aa21","subject":"Merge \"Move nvidia.nvue to 1.2.9 for cumulus 5.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbde402b6356683f7c232950e91d80f15ac1aa21"}]}],"author":{"name":"Jack Hodgkiss","email":"jack@stackhpc.com","date":"2025-08-03 11:46:40.000000000","tz":60},"committer":{"name":"Jack Hodgkiss","email":"jack@stackhpc.com","date":"2025-11-19 15:12:58.000000000","tz":0},"subject":"feat: add command to view passwords","message":"feat: add command to view passwords\n\nQuality of life improvement to make viewing the contents of\n`kolla/passwords.yml` with the introduction of a kayobe command that\nremoves the need to provide an absolute or relative path to the\n`ansible-vault` encrypted passwords file.\n\nChange-Id: I9c3b78cfba1676d75f829cd5bd5b5dc05fd52dbc\nSigned-off-by: Jack Hodgkiss \u003cjack@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51bf6e6cdd02da8f75ed5a93f0d8f1ad0c2541a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51bf6e6cdd02da8f75ed5a93f0d8f1ad0c2541a8"}]},"branch":"refs/heads/master","description":"Rebase"},"ddccfb1679bc612434d86c72f09a74dd9e42df29":{"kind":"REWORK","_number":4,"created":"2025-11-19 15:26:28.000000000","uploader":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"ref":"refs/changes/40/956440/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/956440/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/956440/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/956440/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/956440/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/956440/4"}}},"commit":{"parents":[{"commit":"dbde402b6356683f7c232950e91d80f15ac1aa21","subject":"Merge \"Move nvidia.nvue to 1.2.9 for cumulus 5.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbde402b6356683f7c232950e91d80f15ac1aa21"}]}],"author":{"name":"Jack Hodgkiss","email":"jack@stackhpc.com","date":"2025-08-03 11:46:40.000000000","tz":60},"committer":{"name":"Jack Hodgkiss","email":"jack@stackhpc.com","date":"2025-11-19 15:25:35.000000000","tz":0},"subject":"feat: add command to view passwords","message":"feat: add command to view passwords\n\nQuality of life improvement to make viewing the contents of\n`kolla/passwords.yml` with the introduction of a kayobe command that\nremoves the need to provide an absolute or relative path to the\n`ansible-vault` encrypted passwords file.\n\nChange-Id: I9c3b78cfba1676d75f829cd5bd5b5dc05fd52dbc\nSigned-off-by: Jack Hodgkiss \u003cjack@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ddccfb1679bc612434d86c72f09a74dd9e42df29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ddccfb1679bc612434d86c72f09a74dd9e42df29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
