)]}'
{"id":"openstack%2Fkayobe~956702","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I29c0c6bf77e919940c3452b1f5d219462552cdff","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-08-07 09:33:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I29c0c6bf77e919940c3452b1f5d219462552cdff","subject":"public-openrc: avoid duplicate OS_CACERT","status":"MERGED","created":"2025-08-06 15:51:42.000000000","updated":"2025-08-07 09:34:15.000000000","submitted":"2025-08-07 09:33:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956702","cherry_pick_of_change":954593,"cherry_pick_of_patch_set":5,"meta_rev_id":"1d1904996132c6024ba92803ef4b887ab03b189e","_number":956702,"virtual_id_number":956702,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-07 09:33:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-08-06 19:51:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-08-07 07:09:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-06 15:52:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-08-07 07:09:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-06 15:51:42.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-08-06 18:52:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-06 19:51:37.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-07 07:09:30.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"1cae8fa0cfeb5b1833f7db09b9ce5386f973aed8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-06 15:51:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"647bc61641da12b206e02ec7d7fa65c96b8a1940","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-06 15:52:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"37a45c03117bf6b5b33494e69576dbfcd7da54b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 18:52:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f573cd263a35445a975d652f7978d10a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96fdee71470b481f9552cc209274c475 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/452653b62b3245428c848de9dc1d0fcd : SUCCESS in 2m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/536d04e9d402444f8e3f02918f593c43 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d5b883712c440fb503f6d8416c53b4 : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4204665a0fc24d53b9d2296b6335ef84 : SUCCESS in 2m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8105830088fb4e299467ec7e584aa233 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/df4430afc8c2430f853c46a52e7b3d3a : SUCCESS in 12m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5168e214300c474dbe1328681e993773 : SUCCESS in 9m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d8706232d89d488db2f5fdaded9e01ea : SUCCESS in 1h 08m 30s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/e68160de78e64234b5e843a122221f23 : SUCCESS in 1h 02m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e708a3f9292479ca6a3e5783f948554 : SUCCESS in 54m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/99a7640606704f6c920198fb87a647c2 : SUCCESS in 1h 01m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/614078d731c841ffb60237a7ca527871 : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6ba7c83dbbe8436f89f7353c1c8b36fb : SUCCESS in 23m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/201f09b465c3460cb27d0e06f62792e7 : SUCCESS in 15m 43s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6a680f4dc44d4691bfb335283ec04388 : SUCCESS in 2h 58m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f86e40705fd4e9dbead8b14b95ef452 : SUCCESS in 2h 15m 47s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/fd7928cff4cb42c9922a1164168802b9 : SUCCESS in 2h 31m 59s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b65c0ce30bf14d6b81caa3696d280d2f : SUCCESS in 2h 10m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/61ca2497192e48a99a099361224e1ca5 : SUCCESS in 19m 40s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/16620ab010094fd7938b6647e349c362 : SUCCESS in 17m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4cc4bfa11a34a9a9a3dd8b44004ee35 : SUCCESS in 16m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6c6e1e275a2f4176b995c677c63a8137 : SUCCESS in 15m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ad21e1496a0541dba3a1338a20172f5b : SUCCESS in 41m 06s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cbd2b06930b747bca678660364ca1143 : SUCCESS in 35m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/302f1f8bad8b4f32884376494c020470 : SUCCESS in 19m 52s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a2a17e5f2de648fa8c7661e02dc1900a : SUCCESS in 30m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/578ab69c81024eacaf021e8b4fdbbd0c : SUCCESS in 15m 12s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/26bfdb12ee2546e882180bab31b89ba2 : SUCCESS in 16m 27s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a85ec25a5af43fe9d18e7389093fef8 : SUCCESS in 10m 35s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/a9eeccac8aa84a0198d33cc815e98eb8 : SUCCESS in 11m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ae06f816d94240d5beffc1deca2041fa : SUCCESS in 15m 55s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ade8b4b3f324687a92b8c45a331f096 : SUCCESS in 10m 52s","accounts_in_message":[],"_revision_number":1},{"id":"050005a2cdc6636903a66004a6cec9416fa5665a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-06 19:51:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"281e82e8b93d1cbd2c22bf5098684e18d6936add","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-07 07:09:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d0c20af1bdd41a8552b9dbe20a112a3b65f0557","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 07:09:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"38fcc997c607437a5c481c8e975c7ebee9337397","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 09:33:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bf20375f2944d6cb72d08be7b56c3e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c398451ebfaa454d849572b85bcc9560 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c8139ab898e5450dbc51f42c07f2f8c9 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4a4e1cb7af4291aad0312efdb987a7 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65afc14bb8234abe8888aaebfe001d89 : SUCCESS in 2m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/70a65131f712491a8d777dab6f68ac60 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ea61a1b780d5403cb6183b7f916b4cbf : SUCCESS in 10m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/847fbafdaf3e4ee2963d62cd1c7e1b1d : SUCCESS in 5m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/acdfc547f51548d8bfe41f8d0448b846 : SUCCESS in 1h 08m 38s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/149f4efc41c647ecaf35832ca5691263 : SUCCESS in 1h 07m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a879c58b2744891852d642c9ef9e6a1 : SUCCESS in 57m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2202647b02e249829be033289529b2ad : SUCCESS in 1h 08m 26s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/365e28ecd5904e1aa28f6d7b0d048234 : SUCCESS in 1h 08m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a028341a71e74388afe115c3a2cfe0bc : SUCCESS in 17m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0546a89f61c7480db14783c6ab70d442 : SUCCESS in 13m 31s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/393aabce69f34bf2b2aa2affee7f98a6 : SUCCESS in 1h 13m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7baa15dea4b4c259422f8cfc3bd98b6 : SUCCESS in 2h 15m 00s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2a82e9501b454bc588ec55c3b018e7bf : SUCCESS in 2h 18m 16s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffdcbf651c4443d19726eb8bdeae38e8 : SUCCESS in 1h 06m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/35a7f12d785248dc8994ac4b5524c572 : SUCCESS in 20m 49s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/b361a6084981447ebd5cb41b308dfe23 : SUCCESS in 16m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe51f2154710456b854e2234f99cb815 : SUCCESS in 19m 32s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/edba0fab27194dcab2d4e9def3840e13 : SUCCESS in 17m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/39c8e23dc0144a778120e30871d8e152 : SUCCESS in 39m 03s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98006d0c579a4386a99e0839b4e40563 : SUCCESS in 38m 23s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a2172d2c29a5403190676adfdb7553d6 : SUCCESS in 35m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06d9f42c56c541ce95b7d2b8b0fd6248 : SUCCESS in 11m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e427f0f111c4b66a39bb6d09a72b6b2 : SUCCESS in 14m 48s","accounts_in_message":[],"_revision_number":1},{"id":"fecccdec9a54e08cce4be057790a6dda64d2680b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 09:33:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1d1904996132c6024ba92803ef4b887ab03b189e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 09:34:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f9cccfaf98543388c7caee4890d543f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d19f7d431cad4eb183939b4051163207 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20d52724083444f5b564a6d90704699e : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f83a17e5870718b62047af4c3a30b15cd56ca0d","revisions":{"6f83a17e5870718b62047af4c3a30b15cd56ca0d":{"kind":"REWORK","_number":1,"created":"2025-08-06 15:51:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/02/956702/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/956702/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/956702/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/956702/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/956702/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/956702/1"}}},"commit":{"parents":[{"commit":"af5b2ca5166ce7894b6b6685d4b9288a312f784b","subject":"Merge \"Add network group to tls-backend\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/af5b2ca5166ce7894b6b6685d4b9288a312f784b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-10 10:33:41.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-06 15:51:42.000000000","tz":0},"subject":"public-openrc: avoid duplicate OS_CACERT","message":"public-openrc: avoid duplicate OS_CACERT\n\nSkip the admin OS_CACERT line when rendering public-openrc.\nRecent change [1] introduced that.\nIt occurs only when both admin and public cacert variables are set.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/949624\n\nCloses-Bug: #2116318\n\nChange-Id: I29c0c6bf77e919940c3452b1f5d219462552cdff\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit 24b975fe028deb3161463032ff6d4e13f740d3e8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f83a17e5870718b62047af4c3a30b15cd56ca0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f83a17e5870718b62047af4c3a30b15cd56ca0d"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
