)]}'
{"id":"openstack%2Fkayobe~956784","triplet_id":"openstack%2Fkayobe~master~I55e703dcf67e556dd97289a58b8720511b60355b","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-08-14 15:16:53.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-08-18 07:39:11.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-08-11 15:20:17.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I55e703dcf67e556dd97289a58b8720511b60355b","subject":"Fix version of EPEL GPG key","status":"MERGED","created":"2025-08-07 12:53:50.000000000","updated":"2025-08-18 11:42:46.000000000","submitted":"2025-08-18 11:41:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956784","meta_rev_id":"41a06e9c86e5ee0397b043a2ad4a7df6c9b42609","_number":956784,"virtual_id_number":956784,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-18 11:41:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-08-14 15:16:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-08-07 13:48:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-11 15:20:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-08-14 15:16:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-07 13:48:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-07 14:45:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-08 08:50:54.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-08-08 08:50:54.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"7ab6e8bf83bb887f7f5858f2b7999580ff3b5238","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-07 12:53:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3602ce7149c7b749c0f1c2720fe953312470aab0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-07 13:48:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"267c4d42e781dc971e4a393173dd97f315f310b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 14:45:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e61a762c6e1c4a2b8cc2431f52b18175\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f01801d997be4b058de3046332321f69 : SUCCESS in 10m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/080c28c0519a45aa9a245376ec02eb9d : SUCCESS in 8m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/deee22d131cc4761b1baf737734cf515 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44b7f51651c048778e28e37a3414646d : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddc7a27881aa4daf84a2db203938526c : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/23d3fb91cb6445948ab74e2d521a89b5 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7db118327da4912a4d57fb3ca78856a : SUCCESS in 5m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b86331622fe94dacb4910f1d53664525 : SUCCESS in 4m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/124e4412dd594ff1803aa1679d3f062a : SUCCESS in 2m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9162aa32fe4b48e2bcff4672d5836dbc : SUCCESS in 11m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/18b57ec13709433bbda690b0ee65f01c : SUCCESS in 9m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b414c42000a14fd9af45646e84a7e8cd : FAILURE in 1h 00m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5eda618c1f0b492da0eed35775fa8780 : FAILURE in 28m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9c9d4f9d405f4788a00362c7d7810077 : SUCCESS in 23m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7ad5fd3836b4b29bab9657d809f7f4c : SUCCESS in 17m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd31c4125c00484e93357b5e37f84be7 : SUCCESS in 1h 46m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/62f0178eb106409287148a1627070f93 : SUCCESS in 9m 07s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/fc76e52b78324112bac645cf0c54e10b : SUCCESS in 17m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88c30d77de22482a8181c648e613f5df : SUCCESS in 14m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cde983719f0e45bf955ab09b6b5cfa0c : SUCCESS in 15m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3ce9fa5766c842438c9515eaafa4f0ba : SUCCESS in 18m 51s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fecde571c2e7416bbf9a662ba3c9b7f4 : SUCCESS in 30m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af4df6f7ef0f4b3c84a3e45bf21e8d7f : SUCCESS in 32m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/65269e6092fc4f05bfec2e1920d172f6 : SUCCESS in 16m 12s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/a2e55bdcba854ea9976095a7d36bb736 : SUCCESS in 14m 18s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e69a2a708fe441a9a403df050fbd3d05 : SUCCESS in 11m 51s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/86e99c9652dd4efa9191d5b6b844c77b : SUCCESS in 17m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0a49180901424dfc82442beca9e9abfe : SUCCESS in 13m 12s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c655907ad8c4d9aa96fff322bf5e3fb : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"24678aa122abcae6fca3cdae17fadedfa21b69ee","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-08 07:24:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fde3db4a8978f552efed8859552c51029f3e09e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 08:32:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6d4e4ea7fd94d4aa2988e107dc22173\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f8693a6729741dea22b0202c9f1a65c : SUCCESS in 4m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fd63752c65d64539b0cc6c032ddd4d62 : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43dcc33d995649218a63610a221777c7 : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/825abd0f56f74a01b883ba4db38caaaa : SUCCESS in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/969890dbd3ef41b6a1132e9afbed83ff : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/257f14bfebc244d998fc0673b4956ab1 : SUCCESS in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1efa114aca34b348397fd3c7187f769 : SUCCESS in 6m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc62308299cb454ea6c185da0d368dbb : SUCCESS in 2m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/213a9a3a91f44a42bb37e3a9c9f7e190 : SUCCESS in 2m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/89245960987549dfb56c4be37e1dcae1 : SUCCESS in 5m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/559240e1a5c14fb196287e4f73829c06 : SUCCESS in 9m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/576c788eb705403eb80ffa9c1c79dbf5 : SUCCESS in 27m 29s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3ad3db9f4d0e4765a7ca2af722775cf3 : SUCCESS in 28m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4e279083f9494bfeac596befbb1eda5f : SUCCESS in 18m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fa83d1eadff47a78f26be03a69ab4e2 : SUCCESS in 7m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39217fab1a3e4e9e99370755ef94d135 : SUCCESS in 54m 56s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/eda65040197e483fbdd7203c287a67a8 : SUCCESS in 15m 54s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/d2ae5526dd534142b26800cea1bd5362 : SUCCESS in 8m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff3fb95630f04ecc8a338fcbb02a41c0 : SUCCESS in 17m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/641a4068b3ab4c5983e7447b4fad87ed : SUCCESS in 7m 14s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/29142bcfa18a43f8b0e4556cdc57f5dd : SUCCESS in 37m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5439e1d7261f4a81ac16f2f27a471407 : SUCCESS in 25m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c963060e97b4c8aaca3ed32c50dcf9d : SUCCESS in 27m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/05bbc72254314841bd269df370687785 : SUCCESS in 13m 26s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/d1f996de070b4e288d7c688c0ae7f7b2 : SUCCESS in 7m 02s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7130ae80afff47d18dc6e1e155fe9155 : SUCCESS in 6m 38s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/3c7c3c2ff43941808f061fda41cd9e3a : SUCCESS in 11m 38s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8eb7d903cd944f1fb68afcc44c3f173a : SUCCESS in 6m 54s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cef504cfe15a4b15a2eb70eba9e334d1 : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"dd2f0e8f22b198b7d4052d0338e982543d3de013","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-08-11 15:20:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"642a245b4a3a47922e8f45906c856da20c216ba6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-08-14 15:16:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e58ac9a6a84e39789c5e6dace8069dd11a11d18b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 15:17:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3329e6bb1863b1bf2f6f7a71d60b950e6d778967","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 04:14:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd119e1bc23c49419a7d129704876b0b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaa2895872c64faf82af559b6eecc2f6 : SUCCESS in 6m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fb3cc2905d024027b11914f82a7c71ac : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b31934cd45c4e57bbc01fe9cb27abe0 : SUCCESS in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/026eef2168ca4deaaee12ee079b9cc3d : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6677044933043ef872c0b76e61378b7 : SUCCESS in 2m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46731a69ddc24aef99a9d2c9fa6cd408 : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f4bce39603164ec0ab938402674d4233 : SUCCESS in 6m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e2674d459719436db895635b117f060e : SUCCESS in 7m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e5021d281dd44f4b843fe5266bb2fd5f : SUCCESS in 11m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/228412e76f04412d99dc8eac32b88f09 : SUCCESS in 1h 06m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/37921ce772d74d4fa29a5cfc56fc05be : SUCCESS in 1h 07m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d73a765ae124435abfaaebf04ed741e8 : SUCCESS in 16m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4d928e244e24ad780c66ceea956ea7e : SUCCESS in 6m 58s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72c5a83101934c459a31c795a81c1921 : FAILURE in 11m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/24f768d21c1443998ef93dd5133565dc : SUCCESS in 9m 57s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/37231b3ac2124a90a1bdd77b8d815261 : SUCCESS in 9m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe20571be5f74bd7ab9231bd4ba7a1e4 : SUCCESS in 8m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a82ae71bce904545961d3f5b6e032c8e : SUCCESS in 10m 19s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a11e30bcf30d4ecdb1f31bbedceead8b : SUCCESS in 28m 17s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd6b048a46be43ce931b38b57944b0c3 : SUCCESS in 31m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f166bdb549124c139e66ed5a967901e8 : SUCCESS in 13m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a02c718699a44cfe805ea22fb8492ac4 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":1},{"id":"eb6ff4c9c88f63b628066dabe77334272d5e0ccc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-18 07:39:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc01fb51fe68e97639aada22a3dfe0a11598db56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 09:39:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57573b14d0724986b5d3f389bcdc636c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e04a3cea1a094002abc099e6d19771bc : SUCCESS in 9m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/abf0746c36754f8c83d6d91e940d2451 : SUCCESS in 5m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b9a3add759e40dca6f89c60c3e51177 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfa913ce5a2347d6a825fcdb4190b114 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c66b1cecd45243d78358ebc29c664496 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44e3ab077cdf41bf8026adeebf0fdae8 : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/260ec8f66a1a454fb28fc22850215d2b : SUCCESS in 4m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62d62df9ca224cdca4025ec11f91f6f0 : SUCCESS in 2m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/be1d6f905070428d87de6c73f51e57f3 : SUCCESS in 3m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a5dba4610a4a4fe29680fd3f723e2672 : SUCCESS in 10m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/69ac4bb13b6e4e70a13accc16891ab7c : SUCCESS in 6m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edbc874631cb4db1ad11511987a50364 : SUCCESS in 34m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c33bdce37c464d4fb6ff4aa7385903c2 : SUCCESS in 33m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5b7508c3daa14d7f92a821ccfa3a091a : SUCCESS in 21m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba5e2d532a0d4eecbcfbf4e49403ee21 : SUCCESS in 16m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fd02003d7cd462ca236c3acc42be253 : SUCCESS in 1h 46m 25s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/68e492053eb64c49928fb0ecfbe84923 : SUCCESS in 20m 08s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/1355058a84084f76a33b34dcc29bfe35 : SUCCESS in 19m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6e224a7894d4f5c8ce91c772fdc9fe2 : SUCCESS in 9m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/37aa8edae2474e9da587ccce39bbdd77 : SUCCESS in 8m 12s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/42d02e56742d40c39237f3c129b3fffb : SUCCESS in 20m 31s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f71a2ca998124c58a56d31c75ebf813e : SUCCESS in 34m 49s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ead4ac09e52496e9b1e331a3b24e92c : SUCCESS in 28m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c7b4968e25eb487e92a78000d6927100 : SUCCESS in 9m 11s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/f4dfc39fe740452a8bb622e3a2ef8fc8 : SUCCESS in 8m 59s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32a48d76866f44eeae50121ec7016b9a : SUCCESS in 8m 32s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/de7ce2fb5a034bb6907ba95776e3f36b : SUCCESS in 7m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b283b45fd56b484fa3811959deac3e67 : SUCCESS in 7m 47s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c2ca18462f843a6ba6bd9a845ddb1d3 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":1},{"id":"dc6bdeb1947bf124486704585e1a6af8eeded99d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 09:40:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1aeca6fd5011bd05549c1e4cfa940d8722baa757","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 11:41:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c4db7c2047c4acca181deee6dbe1cd5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/665d74f552014f028aba4ad0dc500861 : SUCCESS in 9m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0abcae9c63f44d098944e8b69b6432ff : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e62addf77a8a473b971fcb639e7f8464 : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee4c5e8ca4f247c58456317c5bf9c375 : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9af7a5703cb45d581e2cb276a85a173 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/349b54f0bb444c4d8b152621b0ba6cea : SUCCESS in 3m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/76b0ac13021548289bd8f368b5c73fa6 : SUCCESS in 7m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a62461395f24526894beaf48fde66c5 : SUCCESS in 14m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/12c90df944fd434ba786789b681bed31 : SUCCESS in 6m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cff2052009134c2a8fb8fd0b19074b3a : SUCCESS in 50m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/01df532281f74b17853d9dc314cdb75b : SUCCESS in 1h 04m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/63f5b8bb99c14f14949cb21fc5ae0d4e : SUCCESS in 17m 19s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/667512190dc2405088d273838552f0b7 : SUCCESS in 18m 12s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/018baee7af34426f9dc49ff9bf1a0e1c : SUCCESS in 1h 58m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d2b33ab9369e4706a538dea331308ddd : SUCCESS in 15m 16s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/b378bdcd79cd4dbebf25390a34e4c746 : SUCCESS in 20m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc2ae54d0130430abf7400b32b25fd47 : SUCCESS in 9m 32s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3c38af14535844bca21adc31b112eaf6 : SUCCESS in 13m 20s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e6ba98027ea1481bb3f841d127750fb9 : SUCCESS in 36m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23891f0965e144798b05691bb608bb70 : SUCCESS in 23m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8c014313dfd4851b3187e5f65a6e55e : SUCCESS in 13m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60a90eece5024ad0826db8c9b74ed351 : SUCCESS in 10m 14s","accounts_in_message":[],"_revision_number":1},{"id":"941fb2ccff29685286f7c0b56a80b416cdc8bd5d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 11:41:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"41a06e9c86e5ee0397b043a2ad4a7df6c9b42609","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 11:42:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2cff23dc8134da3a04c82aa97fdf9a0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e58c2574a87a4da08949e55c56c438df : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b71da020701f4be5b0ea1fd522168d1b : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d4109924c4115c462bfab0b82fb03d0266f5e272","revisions":{"d4109924c4115c462bfab0b82fb03d0266f5e272":{"kind":"REWORK","_number":1,"created":"2025-08-07 12:53:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/956784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/956784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/956784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/956784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/956784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/956784/1"}}},"commit":{"parents":[{"commit":"adab3467b45102c851754405b1da7d50d42135a7","subject":"Merge \"use current debian version supported by kolla\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adab3467b45102c851754405b1da7d50d42135a7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-07 12:53:40.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-07 12:53:40.000000000","tz":120},"subject":"Fix version of EPEL GPG key","message":"Fix version of EPEL GPG key\n\nThis is only an issue when dnf_use_local_mirror is true.\n\nCloses-Bug: #2119921\n\nChange-Id: I55e703dcf67e556dd97289a58b8720511b60355b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d4109924c4115c462bfab0b82fb03d0266f5e272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d4109924c4115c462bfab0b82fb03d0266f5e272"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
