)]}'
{"id":"openstack%2Fkayobe~956871","triplet_id":"openstack%2Fkayobe~master~Ib6f815c319a7e92e675382cfe9d4011598e72aba","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-03-18 11:25:04.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-03-18 14:49:57.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-16 12:31:00.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"36238":{"account":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"last_update":"2026-03-18 14:49:57.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-03-18 14:49:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib6f815c319a7e92e675382cfe9d4011598e72aba","subject":"Skip external connectivity check when behind a proxy","status":"MERGED","created":"2025-08-08 10:46:12.000000000","updated":"2026-03-18 14:51:11.000000000","submitted":"2026-03-18 14:49:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"956871","meta_rev_id":"62cf24901f0046afaef81ad23a195c4f20c747d6","_number":956871,"virtual_id_number":956871,"owner":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-18 14:49:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2026-03-16 17:40:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-03-16 12:31:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-03-16 17:40:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-08 10:52:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-08 13:00:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-14 15:58:40.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2025-08-19 09:35:24.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2025-11-28 13:00:21.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-03-13 17:25:58.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"CC"},{"updated":"2026-03-13 17:30:29.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-03-13 17:47:27.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"d10074836d3ed98bcb88d1f7e0e1258370b4a727","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-08-08 10:46:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89ed0967fcc6384b183011295c1a8e9a4abe4c63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-08-08 10:47:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7ce9fdb070d147dac3f67f9128c322deeb63f129","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 13:00:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5534ffa6884740e69cac125de9f20464\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49e8d10e8e5742b283b48401fb524f34 : SUCCESS in 8m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3f48bfe2bbb444619e58e60eddedc007 : FAILURE in 5m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/573aa612248b4f43916eeb8a16a70a8e : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fbd58b251a84252bbcbcf24b4a95da9 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a93c547e6a24306948dd2f2c39f024c : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0275815830834db49277c6fc7cdcd992 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0cdd6d12ef1644c7ae001c5da87c45ae : SUCCESS in 6m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cb2d2ea88254fb1b2f97c04bd98f9e9 : SUCCESS in 2m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5b1b58ee0a474627a803469b2a9438f7 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3d421fe60963442da37661e6e24d6878 : SUCCESS in 12m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3f11565e067e444481bfdfae80d93643 : SUCCESS in 12m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8412fe269d284f868400deedbec9ea59 : SUCCESS in 56m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d0c389abfadb46d3832db5b41279818e : SUCCESS in 1h 00m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/efe0e18304b744cb8b6fb1f880db357e : SUCCESS in 21m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f38f97593c74f1b9eef994e47bd35d2 : SUCCESS in 14m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82bbf81a9b6b445fb377a8fb5ef71df8 : SUCCESS in 1h 52m 24s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9fe0d5b3820d47109a3c5568bd38eb86 : SUCCESS in 17m 11s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/c378238393cd4ea4b9593a9322e295c4 : SUCCESS in 17m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d676d2d529445e5888593d00c910b2f : SUCCESS in 16m 36s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9e81577a377a4d2a8d9058982c044a76 : SUCCESS in 15m 23s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c89c36cc031b4da28efe315017ca9376 : SUCCESS in 44m 16s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5e762d9f9f0b4fbcb54b928d242c66c9 : SUCCESS in 27m 25s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e22a631135724655aa4e5dfc48d85a44 : SUCCESS in 27m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a29f642ecbb44ffb993d1774caecb1b8 : SUCCESS in 13m 28s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b1cd2e26237c46c09b0c6969df39bf7a : SUCCESS in 13m 18s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c59b4f32956b495cb4e6dd55358e2077 : SUCCESS in 14m 45s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/bb6e9a590ceb48768954937776972c8d : SUCCESS in 6m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d755b0a83bb94153979683015e0430b0 : SUCCESS in 7m 02s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99b2fcc0adc84fe79cf3cddda5c53358 : SUCCESS in 13m 21s","accounts_in_message":[],"_revision_number":2},{"id":"b94c2a724bf9d99d069e9d8fdbf21503a65dedc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-08-08 15:27:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"bddd39bebdd0837b8cdb287a736169c7616d56b0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-08 17:13:54.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d689d5089a28085aca4bd4b11a852646ceadd71d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 17:18:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d66055aaf256446a9a270a21b74bea67\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03302fc1247b404fb78a97dab4288c02 : SUCCESS in 8m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2bfffe36df8843f98ce8ad8ba8e8342d : SUCCESS in 5m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bd803a7e4f548cf8087c87d5f219d9a : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c93bee72c0dc4ad786802d83041c7571 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f62a55bbd1114742a1e20c817d1841f1 : SUCCESS in 2m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2be7e7499dc6400eac65387d2025cac4 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68aab7d983b84722846da9c252880a76 : SUCCESS in 6m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/595d77cd500148d7a04ec36ee38437ef : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e238e19769bc4988ab9bf235e497377e : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/313a83b2535f4673a28eec2b6f649286 : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7ef6e169e08a491ab978a456b828ee11 : SUCCESS in 10m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4845ee602ef432392499e98cb1b48ab : SUCCESS in 1h 01m 42s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4fac7173f37c45ceb8e4d4f901e5a882 : SUCCESS in 1h 09m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a2f626ac5fa74a61a2cc6707150a1193 : SUCCESS in 18m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c56109167c9143c9ab55fe1ded87ccb7 : SUCCESS in 14m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81662158def949c58cbda9b190a8e735 : SUCCESS in 1h 45m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fb8862a745664d48b6c58bc438096445 : SUCCESS in 20m 00s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/05274130787d46f68cdd3ebf2831679d : SUCCESS in 17m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2050d8e4e9a14cfcacc0309a0b8e9696 : SUCCESS in 16m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b506e0972cee4598aba0e3dafeef3068 : SUCCESS in 14m 41s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/53c37672a3ac4356954a163b041f4216 : SUCCESS in 36m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a417c1522f06433e99873185b2925a5a : SUCCESS in 28m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58d2a7a6cd0549caafec3179395f47be : SUCCESS in 29m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0b98ea962e5e42bc817df08abf4f8592 : SUCCESS in 13m 03s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/2c0051a6054643b58ae8b3fba7c5bef6 : SUCCESS in 12m 44s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61ca8191fce44f9b9534889fb837bf75 : SUCCESS in 12m 53s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/b7ec171c73b842b8a3c55cda765639cb : SUCCESS in 14m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ef8b7c6f671843238da1681480286bc0 : SUCCESS in 12m 56s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4671ed68b72c43a199b4723707fae604 : SUCCESS in 11m 41s","accounts_in_message":[],"_revision_number":3},{"id":"7b254ae41edad8523e54a6537295f7b2ee806eb6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-08-14 15:58:40.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b4c9ca25c775597e053e7f9251b13c3da241b1b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-08-18 18:51:49.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"6204fe40c7ac2eee6d635cb887c5d6322306f069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 20:12:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3c74b45b666457a8ac15d8fdd48ea05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dfb14b3b0d847689243c8e4047c61dd : SUCCESS in 10m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c5d4d313480a4009a948da6e3c3fb550 : SUCCESS in 5m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3bb07c69edd48fba070aed80fdf464a : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d198954c711420a94580c41e35206bd : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a84a1d32a2fa476aa875eeb2e3fa7d35 : SUCCESS in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/054192b22a314af4a8d97da30e64dd19 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a40922f0a59642f789fcee2703668157 : SUCCESS in 8m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef2c14ab0a93489993d8d3b725c48c82 : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a540ad254da42edab79201facd659dd : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b8564f810cf490188c6ff4f09a0bca3 : SUCCESS in 10m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/08aedc9dd0d94ddc91d37af6e74ab646 : SUCCESS in 12m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/131a424166534e9580bfb19212431f5c : FAILURE in 16m 51s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/28a9729ee9be4a1a832d1be0a2b952c3 : FAILURE in 14m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0c2d3fb16c374450916e1e23f6c87841 : SUCCESS in 18m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6ee9681e32d4238b63fd1e5fdb7ab13 : SUCCESS in 15m 46s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad285f0abac74b2bb3d3935b3a35d8c6 : FAILURE in 1h 19m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c222e54f9703496ea25c939f015e57be : SUCCESS in 18m 41s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/96c3ec3f33c544dba9d5914321f9c065 : SUCCESS in 12m 52s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f8ac240c9f049628a0f8370ba3652e8 : SUCCESS in 19m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/77f2e41f60374677861b60e4223fe17b : SUCCESS in 11m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/cbaa281a46bb475cbd61bbabf0a4a1f4 : SUCCESS in 43m 55s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bd8db5ed8aab471799c333c4a794d77c : SUCCESS in 34m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c14b11fc91be420b9e3acf7b1691c174 : SUCCESS in 19m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/91ba3043f69644cf9a2d132410ddb705 : SUCCESS in 14m 39s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/a8f5aa5d05614fcf8ef0ffbd71c66663 : SUCCESS in 16m 24s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5df9b6b5b1b4a7abba21de3f77caee5 : SUCCESS in 12m 43s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/995617477f1d45c485712b8bf6fb1e18 : SUCCESS in 10m 38s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/67f6b1c16a424478b2bdb919ea2d8693 : SUCCESS in 13m 10s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f80c0d62e0b42c3b54d8ce421f27fdf : SUCCESS in 14m 13s","accounts_in_message":[],"_revision_number":4},{"id":"24265ce0130b3e93045aa246e3ffd3da26aa1018","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-08-19 08:42:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"98e38df6044e355cbc3d5d95ba66c43c9677c562","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-08-19 09:35:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"522f0aa5bc3c62255e7d5f6d17f34587b8c11dc5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-19 11:30:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4832ec24c34ce712b60d60d77dedbd286d83fcf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-27 13:22:31.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_36238\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}],"_revision_number":5},{"id":"14253d7850d26f36b9821043d2f35c9b3f965b9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 15:21:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cecad7c7807f4d668293582e5716dced\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e6ff91c2244d0080885188e44fc836 : SUCCESS in 7m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/25fac230e76d4d21bcaf3923cac0d1d7 : SUCCESS in 3m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48f66943998e48ec9f590898be565944 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8ee2e4161aa4d1dad414afd12c87498 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2448e25258e4421ca6f42b9c657e2cd6 : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdbe229062a7468292bf1df9b2555e56 : SUCCESS in 2m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c0405db9b8ea4ce49023a4e558ded076 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd608c1bef5742a294d12168c9ce313a : SUCCESS in 10m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/77f2c35a3109412bb570927c57572f33 : SUCCESS in 11m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0dca1a85217a4c54a63ffcbffbee407a : SUCCESS in 8m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e35ad7557233402c9686e8d42942e8f5 : SUCCESS in 6m 53s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3e34c19b483d4d3394268bf23ce4f216 : SUCCESS in 12m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cef87047bad14130a12687ecfc3ecdc8 : SUCCESS in 8m 39s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f1b2b5009dcc4026949e2b40a572cf0a : SUCCESS in 38m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/883f0dd4a1ee44dcb2f9063de857fc11 : SUCCESS in 38m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a5d6f6b131f64c9385b1279fb8d5e82c : SUCCESS in 39m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90673c277cea4cf0bd1eb73ca0647582 : SUCCESS in 33m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e0b6b5291fa144f9b65f4186dd73f6e5 : SUCCESS in 1h 17m 36s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74845d06ecee4be68f3d0d0e011aba15 : SUCCESS in 1h 46m 27s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7e4ea9f4e2b5460a90a3f311a375b659 : SUCCESS in 18m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2b9d939c34524c61b12561aa176f6b95 : SUCCESS in 11m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51d5a6b47acc49d3bdab1f444e97c470 : SUCCESS in 17m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6cff54508c874375b91f2b7993ddb690 : SUCCESS in 8m 17s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1dcc75f3c4974e528e66971c69c66db3 : SUCCESS in 23m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a703cb017f1145a5b34a189ff3fde1cd : SUCCESS in 16m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b550400cff04040baa59413eda69a27 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":5},{"id":"f2b017e3bd868550b27bfb5a88197abfc6c588da","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-27 15:55:07.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"63b2038837d4a9652fc50cbcccb217fc57381f26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-11-28 11:37:27.000000000","message":"Uploaded patch set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"2bdeb0496b18eb2f9297420554833ef9d3be06de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 12:45:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eedd4110786e4cbab56da012350edc86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89a303ea4a7847a89725496900aeaa64 : SUCCESS in 10m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a85e046408104bdf9d0e924173f771fa : SUCCESS in 6m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/463f0d46f6af413f9f4c75f113a60c06 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05acf0c6320e40bf9f953ae9c5ac210d : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/921c818731eb409f8a52c3f2aa9fae0e : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc119550af8e4c57a489f3d5776ae6c9 : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/74bba2615188463a86ac0beb02f64f5c : SUCCESS in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e9f1890677bb4aca9e0a932c67de7f2d : SUCCESS in 6m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/062f805220934cfa8f6b4cef67f527de : SUCCESS in 8m 54s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/777b5d6f24664215b3f33172834739fa : SUCCESS in 12m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de92c51e89f14ced8f3ce26388399606 : SUCCESS in 6m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/09a19d4eb313422fa495159dfeada1ad : SUCCESS in 18m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2e52900d54b427998ec7d527fccf821 : SUCCESS in 15m 37s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3ab85aa316104cc7962f45a16d8dcfe7 : SUCCESS in 36m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/da14b5431fca4ab6b24e0d4bf1165a2c : SUCCESS in 1h 05m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/74f5c7a77e37483c8ad7ffb30c157c65 : SUCCESS in 36m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93cd52238c7e4f7e9c21b00900040508 : SUCCESS in 32m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/63325dc849cd40faafd1d43f5f03b645 : SUCCESS in 1h 04m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebb26e8a2f694127aa4f4d21c3d33cf5 : SUCCESS in 52m 07s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/122cbd26c5ed494ba9be454963396d16 : SUCCESS in 11m 44s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c27dfd5f42084331986acf8a26338245 : SUCCESS in 14m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4528bd0edff342d69af5af7979462548 : SUCCESS in 8m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1787b63a69e544249cd3749115b5929b : SUCCESS in 14m 49s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/831b0dfb2c9449c48575dc51732dec26 : SUCCESS in 25m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3f97d1d0be92493196e95794c3c556ed : SUCCESS in 13m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0c644de8fca41228324f0eea794e7b4 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":6},{"id":"21a5be8441601682099af9327f40c621f4104993","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-28 14:52:56.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"77820b4fd1153bde4c460c618dcc391a445d44f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-13 17:18:45.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"51a886f32e7f3b97789d12308a4697a26440c150","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-13 17:25:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c2c8a9fed83caf7bdeaa6a56f36bbb1159a0544e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-13 17:25:58.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"e738b23d55ffdcedd169b72518ff22ce8f894bda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-13 17:28:04.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"9512d8f1a598ada92ec65911ab79190e0876d7e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-13 17:29:58.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"1c2388102a2834f4102fd40321bb5056c49ce8ab","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-13 17:30:29.000000000","message":"Patch Set 10: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"61b37a7a455ea9442db864cf3c3bcdba77adeaff","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-13 17:47:27.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"09e6b7365e9af9203dbdbfe9d117396078eecb6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 19:05:17.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbf1dd81723b403091efdaafa3404e1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6beb3dc532594400b4b9a5ff53d95772 : SUCCESS in 9m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0494541da0194fb58341d8e573c10b6d : FAILURE in 6m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e44f73422704e4096fd682ebee6c2bd : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9fe7a5bdfe246e480d6fb1fd5e12a50 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/361518e81a3c48b8808e9e2e747233ba : SUCCESS in 3m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a7ca5b7ae3624945a0b4617f61c8d198 : SUCCESS in 6m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fab78de21da48d5987daf6ce26909f2 : SUCCESS in 4m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a1a9908989b24acea5efef92161aae7e : FAILURE in 3m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/92e49a48238a4f808c91e07ad2550e5d : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/19a16c028a8c4f3e97c23fe746a6d7d4 : SUCCESS in 10m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f0dd2db89fb2483cbac9ebfff64f5852 : SUCCESS in 13m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb103089a41a46659656bc22f11e7862 : SUCCESS in 12m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/39bc7157ad8e4f1792d4c792a658a9f4 : SUCCESS in 16m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79c0db408b9c4f049d4e47f32b5ceab6 : SUCCESS in 15m 10s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/45cb4789020f4119a225ead020a4938b : SUCCESS in 56m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/930027d0dd9a4de7b79393667819f95e : SUCCESS in 56m 41s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/76fa7bca5cd7427c925a145b40e8ab32 : SUCCESS in 58m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2a4d57536694a42adfe5d2444f43c9b : SUCCESS in 55m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6a274164de5d437782cb74a9fd264540 : SUCCESS in 55m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ccccca250c10441ba995d44874c432f5 : SUCCESS in 1h 34m 19s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d085cdddea8a4ba793efb18c1829c345 : FAILURE in 3m 45s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/863a8bf446fe4f08bc5187803b53602e : SUCCESS in 1h 27m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d74360852ce34b11884f8dbd8e79124b : SUCCESS in 1h 32m 07s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d348a7465a3f40af86ce1fac3ae899c9 : SUCCESS in 16m 04s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b9a2819bc0be42aca16f9f7985330535 : SUCCESS in 21m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2ab5fa3c7844001b2c041fa98c8a004 : SUCCESS in 17m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/79b573776de54c6db2395b800e4c2ee4 : SUCCESS in 15m 24s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/164ab2295ee84092927dd281bd098d51 : SUCCESS in 22m 28s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/15f969930c6a4a1fb84c6f13745c85aa : SUCCESS in 35m 19s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0093a778cb5a455eaac7d5bd4f0dff4f : SUCCESS in 31m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd65d99f64454edb9c8c6d4e50edfe47 : SUCCESS in 14m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2957e23a5ddd4419aafc53882d589ea5 : SUCCESS in 9m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1522bedca2a148728869965a9f96da4f : SUCCESS in 12m 09s","accounts_in_message":[],"_revision_number":10},{"id":"65de7b09d4383b37a8a5d7b017cfb9c0b1c279ec","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-14 20:57:43.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d7c608dfcb8499551494acb2c70402daeb7a9cf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 22:35:00.000000000","message":"Patch Set 10:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c90decfaffe45729556fcc67e312f81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a06a542760348fcb3ce545e8f617272 : SUCCESS in 9m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/849b29bddea34cca83fe7d67c6c3f7f8 : FAILURE in 7m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fee1276467a459bb35738d78f24e4fc : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ea204a6413e4bbe8d49c495d3d29c02 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f673a42df062425195055ad1b77fe8ea : SUCCESS in 6m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2a2d5254d5b84acfa0c4cc2e3ce57dc9 : SUCCESS in 3m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd079a30852c447ea02d491386fd350c : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/be3183dde4104dae944bad52c239e976 : FAILURE in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0873f82e9df84597804bc2c359582447 : SUCCESS in 12m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f1f118962b04eb981f41c9a5eec4e47 : SUCCESS in 6m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1c3092cf36634c5b9e80ed08ef8f7904 : SUCCESS in 9m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19c348b0853e4e5cbd56e14793114b83 : SUCCESS in 12m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/62e4adb30ea24a398df9d77381fb9181 : SUCCESS in 17m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0ed44a83dbb4ca7b1e1c01cab75c9a0 : SUCCESS in 18m 32s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/68d759ccd77844c2a0af2021b339aaab : SUCCESS in 55m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1e9b4c3e3d57431e83d4ffdc73a8ec61 : SUCCESS in 58m 34s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/89d47a6b3ab4447c934ee5d49f701275 : SUCCESS in 57m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f00e07494dd49df99619070799bb96e : SUCCESS in 58m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ed23b036a6634866b62332ae7350e67f : SUCCESS in 1h 00m 04s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b82ad86250ae46a28d940e28b4d65770 : FAILURE in 3m 30s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/cbda7ff20c3f45eb92597558a143b824 : FAILURE in 4m 08s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bd62960e89744e682b7ca0f8688291f : SUCCESS in 1h 27m 22s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa61935cc9b448d9a3208d734ac56789 : SUCCESS in 1h 36m 33s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8f38bb3b63b24c378e7b42acc02c6a9b : SUCCESS in 14m 57s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/036afe9aae884273a94185fb81988cfa : SUCCESS in 16m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ac5dccc1bdd4201be426467186482fa : SUCCESS in 10m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c2158184947d4a14b7608cf48a93428f : SUCCESS in 15m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fa52c6f3ab3e4878bc731239ed1b16c7 : SUCCESS in 15m 38s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/cd471b3269404f3b987a506d1d54eec2 : SUCCESS in 34m 30s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebe936b8675f41ceab54729fd2de1361 : SUCCESS in 15m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18e7ba5e50434d06849aa9df324d1d72 : SUCCESS in 29m 35s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/82d30aac6f38412998c83e8ba02cb3fd : SUCCESS in 12m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7907fbc81e4f4d0eb33d9b6da4472d9e : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":10},{"id":"29205fd94feb3dcb6ebcbff2bc280eacbe7928c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-16 11:05:39.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":11},{"id":"a52914ae022e066b5c2484f95ecb9b3d32b9bab9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-16 12:31:00.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"e91f03552997a2a01aea48045dea2ca6040c4dd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 12:42:11.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eb056b78f3642c8897b13c9b503ecf6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6467f61aefc4280a8c9135426e89842 : SUCCESS in 9m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fbb348b57636471286a14c7c2e8aa20c : SUCCESS in 6m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12fd80f6de834257a427efb3bd57ad23 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e852baf00dbe489a8133a05db891afed : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48dd31f29e934382b3fd0de1b35a30b5 : SUCCESS in 4m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/69d63be40722469cbcbcc5201667a84e : SUCCESS in 6m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dbe145fdbff45dcac999378b441b4e9 : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3d21ee3d08704a3ba234fedeab3dbf5c : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc89e1e17ece4f44bd6e0db922fea42a : SUCCESS in 12m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9d78702e590540a887a40ef8b2399782 : SUCCESS in 6m 19s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2f9f3405ac204bdc8d032d3c9e848927 : SUCCESS in 9m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebb8e55624dc4fd8b654f253de8814c8 : SUCCESS in 13m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/85d0ab0fcc354e74b306d5065a740fba : SUCCESS in 11m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02d3dfbe74e545ab940fd8e5a99dd558 : SUCCESS in 16m 48s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fad8264915404011b83d13639f6c1002 : SUCCESS in 55m 22s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/aca759db6b574e51a3551e7799402ca6 : SUCCESS in 56m 28s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b78850bcea424c89b2e9b2a34fc0b0af : SUCCESS in 57m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc1ccb05317045b89b3df8120c11c5f1 : SUCCESS in 55m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6f4d65308d0746968142ca6eda7d4a25 : SUCCESS in 57m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fca4f72253134a138851259cca3a112d : SUCCESS in 1h 34m 48s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2d328df6a624421b958fda358c156550 : SUCCESS in 1h 30m 04s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ef8c8925b244824bf85ed26ab4a950b : SUCCESS in 1h 30m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d1c75a56d45447c9e0e4115d42e7604 : SUCCESS in 1h 35m 07s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/abda64e2a25046a3aee1a115792c2880 : SUCCESS in 9m 28s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b4128019ac0b45958d463203a175ddeb : SUCCESS in 13m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5644be710e214e2ea8491e91f3028a81 : SUCCESS in 17m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9ed530ba513147668fbcfee122d1ec2c : SUCCESS in 14m 56s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/083ba14d5255435f976e2cea0a744108 : SUCCESS in 16m 14s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/25426397dfb04ea6a1c46f44c69d5f57 : SUCCESS in 25m 49s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44090a0c3b3e4dea8bedc2c491d7307a : SUCCESS in 37m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/335c2ccd735444e3bb6b204a875258b7 : SUCCESS in 26m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d171dd21c8db46f294e4aa5d741e9ed1 : SUCCESS in 15m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9684bc744fe456e8fe5cdee6d8e51c3 : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":11},{"id":"cec832d2fc6e0d59880bf4287a1d6814305089ff","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-16 17:40:11.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"dca6926b9d8679b2a070890f6493568b5e694177","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 17:40:29.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"5b7b6988cb5c972840ce22c2c33f5099a2d6415d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 19:18:33.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fce00ede29ea4e258142d22b1d7250fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29fe89b8473443d7a11edf3e88c4f23b : SUCCESS in 9m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e8112ab56f1d4340a7b0a7665fe2e6cf : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26435380f4e04c3eae400abffedd069d : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3304cb42036945b8b87db82f02128f8e : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0933a2e1e4b4fb7aef83278ea762721 : SUCCESS in 6m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/83067821c6794939a8bbec611e46415a : SUCCESS in 5m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9ffcf2a790af4976a70cc2d0a748be77 : SUCCESS in 11m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9768f6e4156a41599223f2bf71aee5b0 : SUCCESS in 11m 48s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e2f28e88e027442a957a1b244479ef81 : SUCCESS in 13m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f659a41c0de94f16a02497b8bfcd8cfd : SUCCESS in 12m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a6e466934ae34bfca20db3dec539df99 : SUCCESS in 11m 10s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a92e0646cc9d4ff5b4de26b97f959e31 : SUCCESS in 16m 25s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/38a3c610b4f34b79940fde57c765ec51 : SUCCESS in 1h 03m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a090ca32536648d5b47cba355e6343d5 : SUCCESS in 1h 01m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/622a9abc8d3f4b0ca75b5bce60d12935 : SUCCESS in 1h 02m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5165a122c11c41b8bd01aa0a2f43ad9e : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6b69611f26b14c3da72f645b72d8a527 : SUCCESS in 59m 21s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8dc9e616581149f1b54f5a0f109a0f87 : SUCCESS in 1h 36m 04s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/91d0d488114f4beb97e73223d470a7c8 : FAILURE in 59m 53s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cf09a4afd154a88a9b6190e8bcb675f : FAILURE in 1h 01m 09s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa6e6f5a2075424a995167cc58a4b976 : SUCCESS in 1h 37m 29s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2d3f9a3271b745729f206bf4e7e868ef : SUCCESS in 22m 18s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a2bb444961a54ec394d54fd17acbb52c : SUCCESS in 18m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1789235958fe4d969cad353551842fac : SUCCESS in 19m 41s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/423c565aa6994047b5c414a830a822d9 : SUCCESS in 8m 30s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/739f7647d01348c7bbc82b6df11b880a : SUCCESS in 28m 32s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/47818e793a6f4dc5a3aaac6d146b6030 : SUCCESS in 36m 18s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3280c32c1c5419d9653eed9c3bb4743 : SUCCESS in 38m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/283a617677a94854bac7672b239795ca : SUCCESS in 27m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/220953d4f3fa42c38c2f64f4fcc41a9f : SUCCESS in 15m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38ff551c527e43819cad886d209f9d20 : SUCCESS in 14m 38s","accounts_in_message":[],"_revision_number":11},{"id":"e0e61216baab3c18964a0f3ecaf3daf4b87fa2f5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-16 19:56:00.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"1f41a79a30e1527c340c137b0106d82394e172c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 20:59:43.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/abe0174e875944a191565ea3c2c0b1ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/817001a9622546c88b3c6901cb93b54a : SUCCESS in 8m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6aa59b3869fc40ecbec12c4f16e59f1f : SUCCESS in 7m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee50aba18ba04475a97fb02aa9c960b8 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37089b22853d45cab7e9ee450e2c7107 : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/88a01adf9bd6407e91b7c3b6038c540f : SUCCESS in 3m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/27026f45b2b44134819488d5f8ba83bd : SUCCESS in 7m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acba2abff4c04319b996ab1ac0084876 : SUCCESS in 2m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/662f8ebfc38d4e41a815019646298439 : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/48c15474b7284c0d8bb13d90bb67f488 : SUCCESS in 7m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e297cd3b65f145f3a934e303f84fc3d2 : SUCCESS in 10m 43s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/764be977726a4269b73cf2852d556f6d : SUCCESS in 16m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2aed7aacbde94b6583f31d31cba81d6e : SUCCESS in 8m 21s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/745234af730a49629fdb552f3fbbc071 : SUCCESS in 13m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d209fae75a2491e9f7dbc810735bbd2 : SUCCESS in 15m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e68f584cae6d4c72b3e67c310676f639 : FAILURE in 21m 02s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1a84981dc8e14e788461e67419a55f3f : FAILURE in 19m 06s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8c34ffee006f44b2999e14b81cb47b75 : FAILURE in 23m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fea82bcd46c44e08da00d0bebcfdfd2 : FAILURE in 18m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7ec67bd8f3484556b3b064d9e88ef0cc : FAILURE in 17m 47s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e54f3dad948944ee82ca69896fcbc710 : FAILURE in 20m 56s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b7f8d4f853554f979b1011fa838e28c7 : FAILURE in 19m 29s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/307df39a3d3c499da6559b2ebfbda4d7 : FAILURE in 18m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/addbad29250f48c6bf1fb39a1964b806 : FAILURE in 23m 35s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d7451c10e7454c10b990af3df578882a : SUCCESS in 16m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/de655230aa4a45c481daa48edb780295 : SUCCESS in 9m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/519dc804bcaf44a3b11eeda8ce8f1343 : SUCCESS in 18m 22s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6ff40daa782044b797b17e4488351553 : SUCCESS in 16m 23s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3e73c399b7d34b5ba5eff5ef926f8723 : SUCCESS in 57m 05s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ffd93105459142ac92cd6f72b951c740 : SUCCESS in 16m 50s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/856470f325f9465ba98ffea7da7219b0 : SUCCESS in 29m 34s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9f2a7ddeea14588b2986a4e6438cb60 : SUCCESS in 35m 18s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b61fa27602ab44e797c0e20d345f955c : SUCCESS in 16m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e40b7536bdd24d38992a64e8568e475c : SUCCESS in 16m 38s","accounts_in_message":[],"_revision_number":11},{"id":"a6e1925c417512e71345d84627434e11fc618c39","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 11:25:04.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"da5cdb4558b2c079302cf0b64fef3684168300d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 13:02:02.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb0afc3184f94ed5b3c1edf968ee8ed1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38e7e62036b84f67802f95df3f8ef0fa : SUCCESS in 9m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e8dba7a93b9047dfba631a9a5995f88b : SUCCESS in 7m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f95f4bb91d1e45cbba43ff8f35119304 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fad5526bb27349388b6f3b10a3627cd5 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ba603980bb2497bb872747d37eb0689 : SUCCESS in 5m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1998d2ea2fd340b18aa531272054c8c8 : SUCCESS in 6m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14a9c08e1aa84272a48f8cce172badad : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9dae60e3be2e4d41bb5106f3203c7a83 : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b9d07938db440558ce91ef61c791361 : SUCCESS in 6m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7f6333a8739a40c2b3a848ae318d7084 : SUCCESS in 12m 26s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/50dea23e6be24227a027fbf08cdbdc29 : SUCCESS in 19m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f364d97f70cd4e42ba25604f871745ec : SUCCESS in 15m 05s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/07a291e294dd43cdbdc108905dc5ee5b : SUCCESS in 21m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60801f1422fc4be49925f76cd1cdbeb7 : SUCCESS in 19m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e4391b2ebb544974a88051b67a0b5400 : SUCCESS in 56m 29s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/266cc922cb4f4ac3ad6d0db94546e76f : SUCCESS in 56m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/85429af1033b469d9602536bd74fb581 : SUCCESS in 1h 00m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/616d1f6b6d2b4d80a233f1f7970e69ec : SUCCESS in 59m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bc6558ca264043fdb4f0d80ac87f5283 : SUCCESS in 55m 51s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5a35c7a8a2de47c2a1ee334fc3e3be1f : SUCCESS in 1h 35m 42s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/dbe82420f1474392a7f1aa4ecd727eda : SUCCESS in 1h 31m 37s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91951ac65dcf432ab45b7a29638dfabd : SUCCESS in 1h 26m 46s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/998a17d24a8741b98d8e6af8f55d808d : SUCCESS in 1h 35m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/bf78f2918c6f4c429671c599f471b8c5 : SUCCESS in 22m 18s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/327c52d4d7324d28b6a7508dabb40a0c : SUCCESS in 17m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1093471c6def459dbb1a85e8c3814fb7 : SUCCESS in 18m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3a6a75e804d3447f978fc885c42566b0 : SUCCESS in 16m 26s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1bdf667114d54bd6b2c4ec3f960384d8 : SUCCESS in 15m 45s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5f05b918bd5d43b0a4b358fa40c12972 : SUCCESS in 35m 22s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f9aa0f3ef8a4601ae5efc8974c734f8 : SUCCESS in 33m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/277dd7a44d2649ec823549d661a8f35d : SUCCESS in 27m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7fc07df9868f44abad520e16575004b0 : SUCCESS in 18m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4661ba0bb14146d195e93c56b059fb7e : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":11},{"id":"3bb3e61cadb0b876f22e0eb509cbb46e5983c800","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 13:02:27.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"b448653856c8e58c9a3798d3a1a8493e1635c820","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 14:49:57.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19809fc8ec864ef48bc4326e80167278\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29adaf58ba034e2d9b7026dcf41a0137 : SUCCESS in 8m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9aa37826a2de49abafd016141d4e9d8b : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc1c54e56c9e4f97ad748b43db820b71 : SUCCESS in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3239e9d3db684991bf0059f8d52d92e9 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/845595abe998472ba961cebdbc84bded : SUCCESS in 2m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/28c7d166a01d4c75bdbea28e9ab025ac : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/65eae66dd92c4ef0be0109aec50829ea : SUCCESS in 9m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c1f1124ab17243c6b166305e5d2a10bf : SUCCESS in 9m 40s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a0f811fb4f5c4eb0857035a605306920 : SUCCESS in 13m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5c198b3e4da43d2b28e4cf9e8bcf144 : SUCCESS in 12m 20s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2d69efd3d9e74f399bafe1bd5cab69ea : SUCCESS in 11m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/702114accbd84f55ab535efb5e675b4d : SUCCESS in 9m 35s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9f491ca80d5f4cf2835ff4e824f0c18f : SUCCESS in 1h 00m 46s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a966d71af6784fa1aad5cc8a3166242d : SUCCESS in 59m 10s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3787f17cb24c4faeaf8d73b53ac946cb : SUCCESS in 1h 05m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bb7bb50915c41c9bac568d9a507bde1 : SUCCESS in 59m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/62edbf8c35b84755924b7edd747850b3 : SUCCESS in 57m 21s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9dd8792384044ea89bd752c331a83a85 : SUCCESS in 1h 36m 48s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3f06c0dbd2f048a5a3810317a5157326 : SUCCESS in 1h 30m 19s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/039d93ccf8a6409ea57e5e445dbe072e : SUCCESS in 1h 46m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4511ca0dc45d43a0a54ecc52456fac2a : SUCCESS in 1h 39m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a6c74eb39eb04bedaa226f8f937efaf4 : SUCCESS in 14m 29s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/05d43f279301452a884f6f46dbb8b8d3 : SUCCESS in 9m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c11f8904449749c88b39bfebf998766b : SUCCESS in 8m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bb5f7d66cd38435ba5add15c2cbbc5ae : SUCCESS in 16m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3bee7657bce748e590ee7c3dc907cf5e : SUCCESS in 15m 16s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b6e692c19cb6448c8c84c9caea8b3bf0 : SUCCESS in 30m 16s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e72411b01d5645eea6a5ac16a13d22e7 : SUCCESS in 15m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06b3c21568154bdaa595e6a08d8070f0 : SUCCESS in 28m 05s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5f94b43b5c4c48f3acc2c7a75670a805 : SUCCESS in 20m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6c2472bcd8143ffa0012bb4f4352f54 : SUCCESS in 12m 59s","accounts_in_message":[],"_revision_number":11},{"id":"88f6cfb480bece7fc4f6fbe7a51fa578d6d33873","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 14:49:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"62cf24901f0046afaef81ad23a195c4f20c747d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 14:51:11.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f3655642fae4817b36d6a94ba34f4b4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/477a61cfed3d4f7596786b11600768e8 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f45e5f5a7b984314891a766ffd57c9b4 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"8767fcc69281277139f7132f36a2a65f8a06f5e8","revisions":{"96ebf4f0d69f408c7b873ea69c53db4538b49428":{"kind":"REWORK","_number":1,"created":"2025-08-08 10:46:12.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/71/956871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/956871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/956871/1"}}},"commit":{"parents":[{"commit":"adab3467b45102c851754405b1da7d50d42135a7","subject":"Merge \"use current debian version supported by kolla\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adab3467b45102c851754405b1da7d50d42135a7"}]}],"author":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-08-08 09:54:52.000000000","tz":0},"committer":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-08-08 10:33:25.000000000","tz":60},"subject":"Skip external connectivity check for walled garden hosts","message":"Skip external connectivity check for walled garden hosts\n\nNetwork connectivity check fails on hosts which have no\nexternal networking, this patch changes the external\nconnectivity check to only run on hosts which don\u0027t\nhave `http_proxy` defined.\n\nChange-Id: Ib6f815c319a7e92e675382cfe9d4011598e72aba\nSigned-off-by: Jake Hutchinson \u003cjake@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96ebf4f0d69f408c7b873ea69c53db4538b49428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96ebf4f0d69f408c7b873ea69c53db4538b49428"}]},"branch":"refs/heads/master"},"a026fedfb5d733c8e3c77a5e2b1c27d8b40a4ba2":{"kind":"REWORK","_number":2,"created":"2025-08-08 10:47:02.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/71/956871/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/956871/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/956871/2"}}},"commit":{"parents":[{"commit":"adab3467b45102c851754405b1da7d50d42135a7","subject":"Merge \"use current debian version supported by kolla\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adab3467b45102c851754405b1da7d50d42135a7"}]}],"author":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-08-08 09:54:52.000000000","tz":0},"committer":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-08-08 10:46:53.000000000","tz":60},"subject":"Skip external connectivity check for walled garden hosts","message":"Skip external connectivity check for walled garden hosts\n\nNetwork connectivity check fails on hosts which have no\nexternal networking, this patch changes the external\nconnectivity check to only run on hosts which don\u0027t\nhave `http_proxy` defined.\n\nChange-Id: Ib6f815c319a7e92e675382cfe9d4011598e72aba\nSigned-off-by: Jake Hutchinson \u003cjake@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a026fedfb5d733c8e3c77a5e2b1c27d8b40a4ba2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a026fedfb5d733c8e3c77a5e2b1c27d8b40a4ba2"}]},"branch":"refs/heads/master"},"12eeb3fba73d01b057f43b3e18ccaa13dba0cb4d":{"kind":"REWORK","_number":3,"created":"2025-08-08 15:27:18.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/71/956871/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/956871/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/956871/3"}}},"commit":{"parents":[{"commit":"adab3467b45102c851754405b1da7d50d42135a7","subject":"Merge \"use current debian version supported by kolla\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adab3467b45102c851754405b1da7d50d42135a7"}]}],"author":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-08-08 09:54:52.000000000","tz":0},"committer":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-08-08 15:27:09.000000000","tz":60},"subject":"Skip external connectivity check for walled garden hosts","message":"Skip external connectivity check for walled garden hosts\n\nNetwork connectivity check fails on hosts which have no\nexternal networking, this patch changes the external\nconnectivity check to only run on hosts which don\u0027t\nhave `http_proxy` defined.\n\nChange-Id: Ib6f815c319a7e92e675382cfe9d4011598e72aba\nSigned-off-by: Jake Hutchinson \u003cjake@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12eeb3fba73d01b057f43b3e18ccaa13dba0cb4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12eeb3fba73d01b057f43b3e18ccaa13dba0cb4d"}]},"branch":"refs/heads/master"},"0bf8c7f255a082d57da1ad8fce5bc3f87b99c51f":{"kind":"REWORK","_number":4,"created":"2025-08-18 18:51:49.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/71/956871/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/956871/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/956871/4"}}},"commit":{"parents":[{"commit":"adab3467b45102c851754405b1da7d50d42135a7","subject":"Merge \"use current debian version supported by kolla\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adab3467b45102c851754405b1da7d50d42135a7"}]}],"author":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-08-08 09:54:52.000000000","tz":0},"committer":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-08-18 18:51:47.000000000","tz":60},"subject":"Skip external connectivity check for walled garden hosts","message":"Skip external connectivity check for walled garden hosts\n\nNetwork connectivity check fails on hosts which have no\nexternal networking, this patch changes the external\nconnectivity check to only run on hosts which don\u0027t\nhave `http_proxy` defined.\n\nChange-Id: Ib6f815c319a7e92e675382cfe9d4011598e72aba\nSigned-off-by: Jake Hutchinson \u003cjake@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0bf8c7f255a082d57da1ad8fce5bc3f87b99c51f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0bf8c7f255a082d57da1ad8fce5bc3f87b99c51f"}]},"branch":"refs/heads/master"},"bbc54c5d3fcbf8f376e4184f7d08a053ccdf71d9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-11-27 13:22:31.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/956871/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/956871/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/956871/5"}}},"commit":{"parents":[{"commit":"03c849c6cab0045a78257eb37db95c75cc1c42c2","subject":"Test bare metal node inspection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03c849c6cab0045a78257eb37db95c75cc1c42c2"}]}],"author":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-08-08 09:54:52.000000000","tz":0},"committer":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-27 13:22:31.000000000","tz":0},"subject":"Skip external connectivity check for walled garden hosts","message":"Skip external connectivity check for walled garden hosts\n\nNetwork connectivity check fails on hosts which have no\nexternal networking, this patch changes the external\nconnectivity check to only run on hosts which don\u0027t\nhave `http_proxy` defined.\n\nChange-Id: Ib6f815c319a7e92e675382cfe9d4011598e72aba\nSigned-off-by: Jake Hutchinson \u003cjake@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bbc54c5d3fcbf8f376e4184f7d08a053ccdf71d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bbc54c5d3fcbf8f376e4184f7d08a053ccdf71d9"}]},"branch":"refs/heads/master","description":"Rebase"},"ce52975288cbd9d619da11a4a8ef07bb779e2f6d":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-11-28 11:37:27.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/71/956871/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/956871/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/956871/6"}}},"commit":{"parents":[{"commit":"03c849c6cab0045a78257eb37db95c75cc1c42c2","subject":"Test bare metal node inspection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03c849c6cab0045a78257eb37db95c75cc1c42c2"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-28 11:36:11.000000000","tz":0},"committer":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-28 11:36:11.000000000","tz":0},"subject":"Skip external connectivity check for walled garden hosts","message":"Skip external connectivity check for walled garden hosts\n\nNetwork connectivity check fails on hosts which have no\nexternal networking, this patch changes the external\nconnectivity check to only run on hosts which don\u0027t\nhave `http_proxy` defined.\n\nChange-Id: Ib6f815c319a7e92e675382cfe9d4011598e72aba\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce52975288cbd9d619da11a4a8ef07bb779e2f6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce52975288cbd9d619da11a4a8ef07bb779e2f6d"}]},"branch":"refs/heads/master"},"144b61eff59a72202bf49e98a9401e4402902aaa":{"kind":"REWORK","_number":7,"created":"2026-03-13 17:18:45.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/71/956871/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/956871/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/956871/7"}}},"commit":{"parents":[{"commit":"03c849c6cab0045a78257eb37db95c75cc1c42c2","subject":"Test bare metal node inspection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03c849c6cab0045a78257eb37db95c75cc1c42c2"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-28 11:36:11.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2026-03-13 17:18:11.000000000","tz":0},"subject":"Skip external connectivity check for walled garden hosts","message":"Skip external connectivity check for walled garden hosts\n\nNetwork connectivity check fails on hosts which have no\nexternal networking, this patch changes the external\nconnectivity check to only run on hosts which don\u0027t\nhave `http_proxy` defined.\n\nChange-Id: Ib6f815c319a7e92e675382cfe9d4011598e72aba\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/144b61eff59a72202bf49e98a9401e4402902aaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/144b61eff59a72202bf49e98a9401e4402902aaa"}]},"branch":"refs/heads/master"},"cb7ebba4e4059a9a0ef85b45364a4b1031acad49":{"kind":"REWORK","_number":8,"created":"2026-03-13 17:25:02.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/71/956871/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/956871/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/956871/8"}}},"commit":{"parents":[{"commit":"8b36acddab1fc16c35ef86852c6a816da3617546","subject":"Support Python installation through Apt proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b36acddab1fc16c35ef86852c6a816da3617546"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-28 11:36:11.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2026-03-13 17:24:39.000000000","tz":0},"subject":"Skip external connectivity check when behind a proxy","message":"Skip external connectivity check when behind a proxy\n\nNetwork connectivity check fails on hosts which have no\nexternal networking, this patch changes the external\nconnectivity check to only run on hosts which don\u0027t\nhave `http_proxy` defined.\n\nChange-Id: Ib6f815c319a7e92e675382cfe9d4011598e72aba\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cb7ebba4e4059a9a0ef85b45364a4b1031acad49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cb7ebba4e4059a9a0ef85b45364a4b1031acad49"}]},"branch":"refs/heads/master"},"c1eb60261858318d8ebb88175ebecf06be294790":{"kind":"REWORK","_number":9,"created":"2026-03-13 17:28:04.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/71/956871/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/956871/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/956871/9"}}},"commit":{"parents":[{"commit":"8b36acddab1fc16c35ef86852c6a816da3617546","subject":"Support Python installation through Apt proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b36acddab1fc16c35ef86852c6a816da3617546"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-28 11:36:11.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2026-03-13 17:27:58.000000000","tz":0},"subject":"Skip external connectivity check when behind a proxy","message":"Skip external connectivity check when behind a proxy\n\nNetwork connectivity check fails on hosts which have no\nexternal networking, this patch changes the external\nconnectivity check to only run on hosts which don\u0027t\nhave `http_proxy` defined.\n\nChange-Id: Ib6f815c319a7e92e675382cfe9d4011598e72aba\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1eb60261858318d8ebb88175ebecf06be294790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1eb60261858318d8ebb88175ebecf06be294790"}]},"branch":"refs/heads/master"},"f4c4ebef059797f733a19b27740aa05c565fc0e3":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2026-03-13 17:29:58.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/71/956871/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/956871/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/956871/10"}}},"commit":{"parents":[{"commit":"8b36acddab1fc16c35ef86852c6a816da3617546","subject":"Support Python installation through Apt proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b36acddab1fc16c35ef86852c6a816da3617546"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-28 11:36:11.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2026-03-13 17:29:36.000000000","tz":0},"subject":"Skip external connectivity check when behind a proxy","message":"Skip external connectivity check when behind a proxy\n\nNetwork connectivity check fails for hosts that have\nno external network, so this check is now skipped if\n``http_proxy`` is defined.\n\nChange-Id: Ib6f815c319a7e92e675382cfe9d4011598e72aba\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4c4ebef059797f733a19b27740aa05c565fc0e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4c4ebef059797f733a19b27740aa05c565fc0e3"}]},"branch":"refs/heads/master"},"8767fcc69281277139f7132f36a2a65f8a06f5e8":{"kind":"REWORK","_number":11,"created":"2026-03-16 11:05:39.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/71/956871/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/956871/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/956871/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/956871/11"}}},"commit":{"parents":[{"commit":"8b36acddab1fc16c35ef86852c6a816da3617546","subject":"Support Python installation through Apt proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b36acddab1fc16c35ef86852c6a816da3617546"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-28 11:36:11.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2026-03-16 11:02:59.000000000","tz":0},"subject":"Skip external connectivity check when behind a proxy","message":"Skip external connectivity check when behind a proxy\n\nNetwork connectivity check fails for hosts that have\nno external network, so this check is now skipped if\n``http_proxy`` is defined.\n\nChange-Id: Ib6f815c319a7e92e675382cfe9d4011598e72aba\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8767fcc69281277139f7132f36a2a65f8a06f5e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8767fcc69281277139f7132f36a2a65f8a06f5e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
