)]}'
{"id":"openstack%2Fkayobe~957685","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~I55e703dcf67e556dd97289a58b8720511b60355b","project":"openstack/kayobe","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-08-19 11:39:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I55e703dcf67e556dd97289a58b8720511b60355b","subject":"Fix version of EPEL GPG key","status":"MERGED","created":"2025-08-18 12:15:38.000000000","updated":"2025-08-19 11:40:42.000000000","submitted":"2025-08-19 11:39:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957685","cherry_pick_of_change":956784,"cherry_pick_of_patch_set":1,"meta_rev_id":"2028663c857e4e74180fa17885f88a4f5ef0931b","_number":957685,"virtual_id_number":957685,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-19 11:39:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-08-19 09:05:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-08-18 13:24:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-08-19 09:05:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-18 13:24:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-18 13:59:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-19 09:05:48.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"e564431875faedc329f7dbd1f33180e6068979ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-18 12:15:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"873cec6f1ff08b53be3ea4d558382b848ef138c1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-18 13:24:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"485601f07847f8d08bb2bc20b4d383cd83e37259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 13:59:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a7311130dd344ab851d561a47d0b7e8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c6b91a95d074b67926ad91e1eae4a11 : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23455e07b3114439942484471d98163b : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/979fa5c10f66409a82776669233ed6eb : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1d7444a6eb214282a6c93d4be3816859 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a11b768f0e474499a2cea3dd765f9e70 : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79c324d3d9e54390b3a40ebc46aecb6c : SUCCESS in 1m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d2efe99827ce4024a549742d416ab4fa : SUCCESS in 3m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/52f3e4441339475eb8861937d6556b59 : SUCCESS in 9m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/52809d10ce6046d89724597bb3a3dc69 : SUCCESS in 8m 14s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e5a3a999cec8445091fbe354c5bd024f : SUCCESS in 59m 43s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ed6f4a8d97cf40daa079ad6aadae9874 : SUCCESS in 35m 51s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e5f32d65ea74ba68aa6db2441dfd51c : SUCCESS in 37m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a5984a62b44649fa84c049d2e7a56342 : SUCCESS in 1h 15m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1cee1053e9be460db06025367a84a762 : SUCCESS in 12m 32s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/37f1def8ef954feda399332d6b577b90 : SUCCESS in 19m 23s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9bc6dd7117d4d9e9e92ed609119fe3a : SUCCESS in 15m 48s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a8330212baea44899cfe85dbfdb822c4 : SUCCESS in 1h 02m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2711363ce494b1aa3f59c1577356a9e : SUCCESS in 1h 33m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bc462deb0468429089b32d05c689a510 : SUCCESS in 14m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20aaf2171fe045d8857894e0593efe68 : SUCCESS in 17m 12s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/de72251df0ba4d8b93c34355d112ccc2 : SUCCESS in 29m 18s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/40d9a6931fae4f55b9a18c79eb7b7079 : SUCCESS in 22m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b68e8b7909f64696a6eb0eb2d83b45d3 : SUCCESS in 27m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b0d6b08030484628afeb0e82fe014353 : SUCCESS in 20m 09s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/a0d07ae59b1e4a7192f9139ca9e39cf8 : SUCCESS in 12m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f4b8f99c9554258b9479acdd803da2f : SUCCESS in 11m 59s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/b5b3ed10174f43e193ad2952011999a3 : SUCCESS in 14m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/00856f89f8b54970be76d50b1336b4a0 : SUCCESS in 12m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d790f0f4953141d5b8c8fced2aa31e65 : SUCCESS in 18m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1e521c6e4552da5a3a79f7b6ef31437ccf1d6d7d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-08-19 09:05:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2528944f5b7b6167840b7dc7e22f6fd6c4d3596d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 09:06:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50fda5f8904637de65c5c9710168457b6700fc5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 11:39:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04f64fe6e38944ed9a846215d4ec02c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a057b6e2fbf546afafd993f3c1ac5479 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34a218fa5b444184a3013d1fcaa86b87 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb2bf82fe60443bac76a29ea40d56a5 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ea8d9e309a1491d8d0fe0a21532228a : SUCCESS in 1m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dd686da40a4f485fa6235f384606a042 : SUCCESS in 2m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4d073ae5fb284e52b226738d51d72c1e : SUCCESS in 12m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2b86624b6db14620b244bd8a2835bccf : SUCCESS in 11m 55s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bf1094ffaa864d3788a387e76865c4dd : SUCCESS in 43m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a744ed486bf412380c1e95f72a7b611 : SUCCESS in 1h 11m 56s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/dfc90f35ef5242f1bcbbc513c501544f : SUCCESS in 1h 18m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/eb9f9c468ccd469cb9e68bd5ecb4fb92 : SUCCESS in 19m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e730f7adde8e419f9295d1acbb555334 : SUCCESS in 17m 53s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6e5e65d0bde4463797c6c46b41229867 : SUCCESS in 1h 42m 26s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14c8a68ca8964da2bb64465f35c558f0 : SUCCESS in 57m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/969e6968b8944f9ca125632e1442619e : SUCCESS in 20m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4ec76d77d6f48b9a0e7588d6fdd3a8c : SUCCESS in 18m 26s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ae2133d113f74f8c8cb5f9aaa93ae56c : SUCCESS in 38m 27s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4f0bb794e77441da1b6fec7ec04738a : SUCCESS in 26m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3d25205e5fc046f082dec70c7c8cf4ed : SUCCESS in 16m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23634e5969a74d049d39df27ff2a2f9f : SUCCESS in 18m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0d956a262aa54d96b7735e4f79c18096 : SUCCESS in 19m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96385314dff748c1893f9bd063c252b9 : SUCCESS in 19m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7d157ad00a3ebc93eec423dca394433e4f46d5cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 11:39:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2028663c857e4e74180fa17885f88a4f5ef0931b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 11:40:42.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/64788475dd1341668b9e140fe6429acd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14fe028d9e7041fdac5d6a7f5b7b1d02 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/433b3bacba2745da829e8c4575d75875 : FAILURE in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88431b454602e3ea21d21b9557a61f5a2b33d4bc","revisions":{"88431b454602e3ea21d21b9557a61f5a2b33d4bc":{"kind":"REWORK","_number":1,"created":"2025-08-18 12:15:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/85/957685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/957685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/957685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/957685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/957685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/957685/1"}}},"commit":{"parents":[{"commit":"263f9f47ac99fad4a0709b8d913162f9eedc281d","subject":"public-openrc: avoid duplicate OS_CACERT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/263f9f47ac99fad4a0709b8d913162f9eedc281d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-07 12:53:40.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-18 12:15:38.000000000","tz":0},"subject":"Fix version of EPEL GPG key","message":"Fix version of EPEL GPG key\n\nThis is only an issue when dnf_use_local_mirror is true.\n\nCloses-Bug: #2119921\n\nChange-Id: I55e703dcf67e556dd97289a58b8720511b60355b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit d4109924c4115c462bfab0b82fb03d0266f5e272)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88431b454602e3ea21d21b9557a61f5a2b33d4bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88431b454602e3ea21d21b9557a61f5a2b33d4bc"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
