)]}'
{"id":"openstack%2Fkayobe~958026","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I29e7e8a7480009fd359b8aa8b867b11900109f00","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-09-05 12:28:43.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-09-05 15:38:09.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-08-20 14:52:56.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-09-05 15:38:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I29e7e8a7480009fd359b8aa8b867b11900109f00","subject":"Fix known_hosts module issue on centos/rocky 10","status":"MERGED","created":"2025-08-20 08:11:47.000000000","updated":"2025-09-05 15:39:44.000000000","submitted":"2025-09-05 15:38:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958026","cherry_pick_of_change":954671,"cherry_pick_of_patch_set":5,"meta_rev_id":"03aab2108bcdb7ac2c8bf61a61a805933c7fe924","_number":958026,"virtual_id_number":958026,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-05 15:38:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-09-05 12:28:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-08-29 19:48:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-29 19:48:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-09-05 12:28:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-20 10:56:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-20 13:11:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-20 15:02:12.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-08-20 17:29:42.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-09-03 09:34:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"c5e627bd857c008b875346ae0e453cdd9c018924","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-20 08:11:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b519a12ddd585a79027eb17feabd2e6d1bf39d33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 10:56:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2895668db9c041f6840505ef9029b5b4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04425c3ab7b14d4a841be847221d0610 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6402ccbd0ef349f388732083f385a0e3 : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/14d9fba7b526453d909d57630030faa5 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400710fe63ba4f5ab0e5069674f27123 : SUCCESS in 9m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/76957e3c7c9048a88c9aa60ad7c7f061 : SUCCESS in 3m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4f59a727ad8e44a59e196fc85838f299 : SUCCESS in 6m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a1965853e07b45ee95f91762a34f3994 : SUCCESS in 8m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/02f1d34dc2114398870a1e40c97d9b1f : SUCCESS in 1h 05m 06s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/be5b40ec8fe24c719b0e348e61cba9be : SUCCESS in 1h 11m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed7c682a66fa43ac9235776d9751fa90 : SUCCESS in 55m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fa12552c8b604f27bca7c9f36647b2ca : SUCCESS in 1h 10m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/14f0e15d1fe64232b98ebe384ad1ee4a : SUCCESS in 1h 09m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c8298e936cf3497ea6bfead0d72d4b0c : SUCCESS in 17m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9140658e80540cab36292cdb7b15bf4 : SUCCESS in 18m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/08d494cb355142b19455929ba7abf302 : SUCCESS in 2h 17m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/175b069ad39442c8a00b4f728df17c76 : SUCCESS in 1h 15m 18s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/bc75d2e9a778437ea0b74cf36ea923eb : SUCCESS in 2h 37m 39s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9be291182bd4db5b54af0b58fed4cd5 : SUCCESS in 2h 23m 56s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/de1f0c2076e74f82a5ad12abe52c88c0 : SUCCESS in 11m 03s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/5d79f5e7e36f4a43bcf02789da881e03 : SUCCESS in 16m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4df1d3abcceb427fb525e6ad842f9364 : SUCCESS in 14m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/157b2cec29c34b048c8bc5e2af1594e5 : SUCCESS in 9m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6b9be4bf11d44fada3d929f071df1624 : FAILURE in 18m 25s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/eac0d4d2f4bd4e5181535ee080004561 : SUCCESS in 33m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5b99e1a6c744bee81652dd47e99d30b : SUCCESS in 37m 59s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/33127efc73c24225ba6648a4244ad619 : SUCCESS in 34m 20s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/78e9b083efdc4725abb0e042e321d75f : SUCCESS in 12m 47s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/de5ada39cf474952893404816d56c672 : SUCCESS in 13m 39s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2289a79a430e4efd8eee56a64c48ceef : SUCCESS in 12m 18s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/b04222b26de247debf7e32ab52c0521c : SUCCESS in 7m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c73a774827dd438a9aebd57ea4fa158b : SUCCESS in 13m 35s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e10df686cf224f48ba50349f3615b624 : SUCCESS in 11m 20s","accounts_in_message":[],"_revision_number":1},{"id":"807cdbf1e002be99f7e5e6bb39663ceac928d9ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-20 12:35:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5d86fcc01380522184c6be681b152d4a0a1d44a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 14:51:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01cd97bb8f3847faa11ede551f613f46\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9d95f6b6c1048f6be8115beb8c15e10 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c30154ad3e343b9a724bb1435317057 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/33f48a1bbfaf43a280547ed48b8f299e : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd8b22bc3ca3479fba079be3e7aa1b0b : SUCCESS in 7m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/948ce7e0f3fd4f57befa762dec9a0727 : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2b192a68904d475ca8799c734d9b41e9 : SUCCESS in 8m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c651f140fa214ca289f6ba19c7c39be9 : SUCCESS in 10m 26s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f124e975e13246e8ae5e15684443ed91 : SUCCESS in 1h 11m 19s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/723baf02c0d447eba4ed31c5a736e119 : SUCCESS in 56m 22s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47af137667374de6be32400c1b6442ad : SUCCESS in 58m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9b8f8535adfc4b288e8abe58d7333040 : SUCCESS in 1h 16m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8266137fdbeb42a398ed2570536d87e8 : SUCCESS in 1h 06m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a4098c5dd78045858b18c48afbd5fa0f : SUCCESS in 16m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3b8f0122eb04b39b19698510e614e9d : SUCCESS in 7m 11s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7ee2de5f6b654323936db54141816722 : SUCCESS in 2h 11m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/233552b0685a4b01902f010f7fb45097 : SUCCESS in 1h 13m 38s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6d87e2476d5349428bc1cfa1163d1d7a : SUCCESS in 1h 18m 08s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ba45feef2b34e4a80e4734c9952e3e6 : SUCCESS in 1h 10m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6a8a29271ed5461c8c926d4248535d0b : SUCCESS in 15m 29s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/3051f2cdfa5b493c8806002bd22aadb3 : SUCCESS in 17m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fed03f6cae34d618af80d2d1f744e13 : SUCCESS in 19m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2eda6e5579e048c5bf572a283f91c9d2 : SUCCESS in 12m 49s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9ed0cee0f1e1406f95c16f5e08306625 : SUCCESS in 26m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9d44e1a88854c39bfed642491fff048 : SUCCESS in 34m 39s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6d802b697f1448c19a3e1efd90466c95 : SUCCESS in 40m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e67d8d7d2ec94f96b9dd758a4521ab0e : SUCCESS in 14m 13s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b0f89610402d4cdbb5584fd8955e2eec : SUCCESS in 21m 19s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce717d76ba9342d4ba2cd256736a3157 : SUCCESS in 12m 03s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/90f20efcd8e94d7aa869f012a5543929 : SUCCESS in 10m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/069773dd243e4a65897e02a9b102cc7e : SUCCESS in 7m 47s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80a6ac6d42ce4e5786581bd383c479c0 : SUCCESS in 9m 49s","accounts_in_message":[],"_revision_number":2},{"id":"834cf444a7174269c513b126c98624b0eff7c514","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-20 14:52:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"47f3298491b9f5c19405e6c64d6a27455a2b75b3","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-08-20 15:02:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d7be01fb7b5195ccefe9e0871243be03792c1d5e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-28 15:48:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"31fad78ff1a9bc3f8dd028db78976615057eb614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 17:55:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b81bee65315f4452a738e26ff39196b2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e35184de5ebd4f24add689756396ddc8 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e59fcef9f1c349a08fd2ec11c0b9e25d : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e750cabebcc4491fa702a4f92eb4a2f4 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a45a50755234c08aa7977cb274a12b8 : SUCCESS in 6m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e85ad5d808ea44e8ac1f561893d1b0c0 : SUCCESS in 6m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/96fcc9c44c3348dda1281343c3335555 : SUCCESS in 12m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c81344b24d584d8dbc1d285aee046c24 : SUCCESS in 9m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/99a8b8d0ceed460a861998b3ad5e0a16 : FAILURE in 49m 42s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/e0aaa434d5ce413c9563c0a601f6c331 : FAILURE in 31m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/959273c3c6ce411b9e12c45140b712e6 : SUCCESS in 56m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4349dd5b425845f7b3834e5cb56bc801 : SUCCESS in 1h 34m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/03a36eb96d554291b0430178c286643c : FAILURE in 53m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/686b55f1b61848af82f251920188d2b3 : SUCCESS in 17m 29s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5390a9a06e9454fa608692cc5d09608 : SUCCESS in 18m 32s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2e509dcabb9b41b888a07afb80680bc1 : FAILURE in 47m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33879b7c3a4149228309fca794123491 : SUCCESS in 2h 05m 07s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/405a00a0be9247c28af5bb7041cda495 : FAILURE in 17m 34s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c4239273c2c49dd96292b8230839493 : SUCCESS in 1h 55m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e6f36ab5d09147a1b6690f29d218c218 : SUCCESS in 9m 37s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/22ea03386b134cf7b13e6947cba11f53 : SUCCESS in 17m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f01e9a4daa5e49319e341420f69e2c92 : SUCCESS in 16m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8f6236346f3c43dc917fa0166fd9bcbe : SUCCESS in 10m 12s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1ea93fa6c3474becafbae376ee3a0d4c : FAILURE in 22m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42d9371827e64699bc927cfd5002da69 : SUCCESS in 22m 13s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/acc179e0804f4b1ea5d2882e0a1dd7ee : FAILURE in 18m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1d84749210f24a488248fecb8ec7fffe : SUCCESS in 8m 45s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/763155e7b6214fb0975a6e39821bb665 : SUCCESS in 15m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99905133720f4b6395532377a37d2a67 : SUCCESS in 15m 47s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/c7530e4818bb4246bdf7514736a8e39f : SUCCESS in 8m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/596bb0749d444ffdb78101e8766dd0b0 : SUCCESS in 11m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10310f93803a494c9b4f071e629cfaed : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":2},{"id":"234677da9511e60f0050ec17d15c1f3908450449","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-29 19:48:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"cb2232dbd4dbf923f901732ae4b81864ea9b52c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 22:05:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c6828be47804eb3bc579edde0fad2b2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a67420fc0c9844df96399aefd61c8a7c : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8344406909844f9d98da37e824df0476 : SUCCESS in 6m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b92bbad74d84dfb944c7b4d624d7d6e : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1100dee011a1486ebeaa797458a0eb5c : SUCCESS in 4m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0998e1d529ff465f980c3a0de7d59e6e : SUCCESS in 3m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0320cd09ef164c08bb5087d06b712cc7 : SUCCESS in 14m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/15c44fbdb8574a1294e575adb061017c : SUCCESS in 6m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8594bf6014c64f5aa1f8a2be281ccc0a : SUCCESS in 1h 22m 53s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/1793b800da76462dbf77759da8fe0279 : SUCCESS in 32m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a008a50d1c924922a5bb2e2ab48d8d05 : SUCCESS in 34m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7b25fe4fb5054d419bb36198062c2f07 : SUCCESS in 58m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/473f3cce35294f14b558a73d04b54ee7 : SUCCESS in 33m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/837038c6f31c402b83cb4e073f3beca6 : SUCCESS in 16m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a9330116a3f481ea9b25d147ca10b24 : SUCCESS in 17m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6d9e646ccef84a68893cc51d34a81c04 : SUCCESS in 1h 21m 19s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b917ac91f3e34cd199dd4e33f069df39 : SUCCESS in 2h 15m 12s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/19fb47830ce14810a16ee7790cadc1f2 : SUCCESS in 1h 21m 35s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a660d4d27554f32ab47af232a00827b : SUCCESS in 2h 11m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/70074a4320de4ceabd47e844f3a6b69e : SUCCESS in 15m 55s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/0ed1223b2b984098bb81f7d2219f9811 : SUCCESS in 8m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55fe757312a140cea68c21ab36ec3bf5 : SUCCESS in 7m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4b9d64587548431f93c35f933d43b445 : SUCCESS in 7m 29s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/73cf84910086483d9ea8a644a4abdb42 : SUCCESS in 37m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70c68642ed1547968d2e7d072b4d94ef : SUCCESS in 26m 36s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/26a80e7ad0d24043be58e53864b7c962 : SUCCESS in 34m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bde364ea034d4125a72fd06c37f86142 : SUCCESS in 7m 39s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b537945df36b4929a0c4e4570fbd06c1 : SUCCESS in 57m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/793a9d650f274a88804fc1a3057d4ba0 : SUCCESS in 9m 22s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/caae5335eb3543e2bf88255a58cb7517 : SUCCESS in 12m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/af1fab83642748a5b4286465df27d949 : SUCCESS in 10m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3ec0ed7b48c4f129994b1bf43c022ae : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":3},{"id":"3646a2cec3b340e7bfd9e43a653a7757e8b665b9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-05 12:28:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"965a8bb2dc217304eee2dee9374dd4adfa1e1ce4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 12:33:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5580eed237820f2b9ed39d406e3e8b0fc1e483e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 15:38:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de39c874accc41fa9a26f0990917e326\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57a23b9f013244f2946ff3d3312b1579 : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25c60e2768214b9d95bd33dc5c422468 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98d855112fde48d2a8ff848380005c7e : SUCCESS in 9m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/830cbb4896f249fc8949e7e99aac98e6 : SUCCESS in 5m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c9422d57e2fb43f49d5f42f70f0917bc : SUCCESS in 13m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b8d0868552874f6380fd231f9247d35e : SUCCESS in 7m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dc1e669a11a7402299f76aa4ed82fcdf : SUCCESS in 31m 25s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/a458ac3dd705459db7fef6593fcb229d : SUCCESS in 1h 06m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce3b67bf661f4b83b71c1d65b30f8734 : SUCCESS in 28m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8829cae08984470f9c9c95ebb63c0e6f : SUCCESS in 59m 42s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cff4b184e215450ba5b580acaaed8850 : SUCCESS in 1h 04m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9dfc195094c44cceaade65510181e90a : SUCCESS in 18m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65dafdf456114eb0823d22aab84f8a2e : SUCCESS in 8m 24s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/206280f8e37d4f2e89e81b515234127e : SUCCESS in 1h 27m 22s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50140f523d2f4253a8f9164ec1a5907c : SUCCESS in 2h 37m 02s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/672f305b3dc940159d6b44c79dcd5d44 : SUCCESS in 2h 35m 43s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/159894153a934adbb801f7a4680562bd : SUCCESS in 1h 16m 24s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8762821a47404b7c91e362f495865dae : SUCCESS in 18m 05s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/5d91cc221f10444b8d566b2ff971efc8 : SUCCESS in 10m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ce611eafd9943cd9c47acdc3e5c198b : SUCCESS in 10m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f087e904d4294fa587dcb79037dc3519 : SUCCESS in 17m 03s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a1f3bb9c12914441bdbb28787c968a5f : SUCCESS in 35m 07s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41cb3098e3234205a5a4c53f7285494c : SUCCESS in 19m 29s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/d5aed826e6e846d2b1985246fa54f2c1 : SUCCESS in 30m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/25d398b2f0464a6cb300e563b135d578 : SUCCESS in 14m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/907617b2dd514aa7ac3a1a86403f35e5 : SUCCESS in 11m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e0469c597979489f8b90be6a27f5cd75 : SUCCESS in 13m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b10e077e73c44b2d8ee4ee3c374bb6e0 : SUCCESS in 10m 49s","accounts_in_message":[],"_revision_number":3},{"id":"f20251b9fdb105fa57f02e83059c68502bb94597","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 15:38:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"03aab2108bcdb7ac2c8bf61a61a805933c7fe924","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 15:39:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12170cd115154650a8a1c1728eecc509\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2325b1f8dc3f48d485633910e14913bc : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"85ad72a561f77aea29498c26d6c917ce674aba08","revisions":{"eb72788ed00170868759e1074e78a7bc8060cb3f":{"kind":"REWORK","_number":1,"created":"2025-08-20 08:11:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/26/958026/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/958026/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/958026/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/958026/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/958026/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/958026/1"}}},"commit":{"parents":[{"commit":"7618cd375733dc5b52b1fdef85d9db8fbf4faba6","subject":"Merge \"CI: Switch from td-agent to fluent-package (LTS)\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7618cd375733dc5b52b1fdef85d9db8fbf4faba6"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-11 07:31:52.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-20 08:11:47.000000000","tz":0},"subject":"Fix known_hosts module issue on centos/rocky 10","message":"Fix known_hosts module issue on centos/rocky 10\n\nThe known_hosts module returns a failure on centos/rocky 10 when given a\ncomment line from ssh-keyscan output. Fix by excluding them with grep.\n\nChange-Id: I29e7e8a7480009fd359b8aa8b867b11900109f00\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit e30333940f6215740b22f987a50ad191e876b7d2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb72788ed00170868759e1074e78a7bc8060cb3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb72788ed00170868759e1074e78a7bc8060cb3f"}]},"branch":"refs/heads/stable/2025.1"},"f1162c8e68b9d3e5d832fb69f9f02bef80b8533e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-20 12:35:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/26/958026/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/958026/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/958026/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/958026/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/958026/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/958026/2"}}},"commit":{"parents":[{"commit":"94caafce91d4da2ee38f98739ddfc3aa129e2cbe","subject":"CI: Move failing job to experimental queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94caafce91d4da2ee38f98739ddfc3aa129e2cbe"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-11 07:31:52.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-20 12:35:00.000000000","tz":0},"subject":"Fix known_hosts module issue on centos/rocky 10","message":"Fix known_hosts module issue on centos/rocky 10\n\nThe known_hosts module returns a failure on centos/rocky 10 when given a\ncomment line from ssh-keyscan output. Fix by excluding them with grep.\n\nChange-Id: I29e7e8a7480009fd359b8aa8b867b11900109f00\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit e30333940f6215740b22f987a50ad191e876b7d2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1162c8e68b9d3e5d832fb69f9f02bef80b8533e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1162c8e68b9d3e5d832fb69f9f02bef80b8533e"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"85ad72a561f77aea29498c26d6c917ce674aba08":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-08-29 19:48:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/26/958026/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/958026/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/958026/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/958026/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/958026/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/958026/3"}}},"commit":{"parents":[{"commit":"7ff2db4867c51193989aab2d384cd7a31b157c86","subject":"Revert \"CI: Make rocky9 VM jobs non-voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ff2db4867c51193989aab2d384cd7a31b157c86"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-11 07:31:52.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-29 19:48:36.000000000","tz":0},"subject":"Fix known_hosts module issue on centos/rocky 10","message":"Fix known_hosts module issue on centos/rocky 10\n\nThe known_hosts module returns a failure on centos/rocky 10 when given a\ncomment line from ssh-keyscan output. Fix by excluding them with grep.\n\nChange-Id: I29e7e8a7480009fd359b8aa8b867b11900109f00\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit e30333940f6215740b22f987a50ad191e876b7d2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/85ad72a561f77aea29498c26d6c917ce674aba08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/85ad72a561f77aea29498c26d6c917ce674aba08"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
