)]}'
{"id":"openstack%2Fkayobe~958836","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I6abf63aec0bd11ae17c9308aec5c679b3a21d33c","project":"openstack/kayobe","branch":"stable/2025.1","hashtags":[],"change_id":"I6abf63aec0bd11ae17c9308aec5c679b3a21d33c","subject":"[DNM] Test tenks fix","status":"ABANDONED","created":"2025-08-29 07:59:37.000000000","updated":"2025-08-29 10:29:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eb7086781bd85eb71e587c18fbe135f5df3cc2b5","_number":958836,"virtual_id_number":958836,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-29 10:19:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-29 10:19:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"11b942145a699b5f0d2c160b7fe11b9f33dc8647","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-29 07:59:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db22b526eca0aa38fc5d649b16d82391e090f116","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 10:19:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d742c1d750d948fc8358b1214ede7f41\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/095bad6ab630478789d320a49f1ddb3f : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d4ae1c6319a4f31912436dbfe7cf223 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/627074508dee4b189129fb405ff51809 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7d2de16a06149379a1ed9bf981f3c06 : SUCCESS in 5m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9c2af45cfb7743c489c3b9bcf3e62b08 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0df63ec602ee4198a019263ad77e8fc5 : SUCCESS in 5m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f76fadb816294778b580a4089f853050 : SUCCESS in 8m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3e3648daf0734002bcf430a49735e2b9 : SUCCESS in 54m 09s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/031ab2d4027949fd9ada28e6483ee355 : SUCCESS in 41m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49775bc06a134a95b01560c7cf37d281 : SUCCESS in 55m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/15d11a2b22c74e5a815077d9c26df8fc : SUCCESS in 35m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8df27b97698e496ba6123193f2f0086e : SUCCESS in 1h 08m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0617367b78be4b1e8a708e2337719dc0 : SUCCESS in 9m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e84d72b5fbc45b4a2da43db48933ca5 : SUCCESS in 13m 32s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/00d2f9c36d074d349d5cdb652f3c5444 : SUCCESS in 2h 09m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9df379db150040788ba4a48452a283dd : SUCCESS in 2h 16m 05s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6af0fb8a34ef472c8b924ab1dc103e08 : SUCCESS in 1h 22m 53s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6547399a0dbb4770acedb7aa4c9b7769 : SUCCESS in 1h 08m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6dc205599a144a7b8fdf9bc31ded4015 : SUCCESS in 9m 47s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/6216bd57c8334e6ebdbaa738059a82a9 : SUCCESS in 8m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f29e04c6ac374832a5d5ec8ded6b571d : SUCCESS in 9m 15s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7d489ee627674589833d1b7fe6d1c19a : SUCCESS in 17m 55s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d2b70ca672d84ff3a854b9d51f2bbae5 : SUCCESS in 20m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85966f31e5924961aed4c76640f3c605 : SUCCESS in 31m 40s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/4a17c31cfe914783adc2766551685f55 : SUCCESS in 21m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/daebf7969e874cea990ea63008308ce0 : SUCCESS in 7m 42s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/3e325489deb2447abe3d2276e0e58241 : SUCCESS in 7m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/525158150cee4be3b66554cd20f01632 : SUCCESS in 15m 13s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/66d51a48fb014cb58686ce038791ed54 : SUCCESS in 11m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/21b77e5916f94b5a81cb9d6e44f27fe4 : SUCCESS in 30m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23aebf70f7a147be87fdd20ca2474cb1 : SUCCESS in 44m 22s","accounts_in_message":[],"_revision_number":1},{"id":"eb7086781bd85eb71e587c18fbe135f5df3cc2b5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-29 10:29:41.000000000","message":"Abandoned\n\nFix is working","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d393e85a53daca80f26e629c94e7439870c24924","revisions":{"d393e85a53daca80f26e629c94e7439870c24924":{"kind":"REWORK","_number":1,"created":"2025-08-29 07:59:37.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/36/958836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/958836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/958836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/958836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/958836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/958836/1"}}},"commit":{"parents":[{"commit":"7ff2db4867c51193989aab2d384cd7a31b157c86","subject":"Revert \"CI: Make rocky9 VM jobs non-voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ff2db4867c51193989aab2d384cd7a31b157c86"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-29 07:59:27.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-29 07:59:27.000000000","tz":120},"subject":"[DNM] Test tenks fix","message":"[DNM] Test tenks fix\n\nDepends-On: https://review.opendev.org/c/openstack/tenks/+/958827\nChange-Id: I6abf63aec0bd11ae17c9308aec5c679b3a21d33c\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d393e85a53daca80f26e629c94e7439870c24924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d393e85a53daca80f26e629c94e7439870c24924"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
