)]}'
{"id":"openstack%2Fkayobe~958965","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~Ibd2c68103e73b1ecedd1aade780f96af3d9e9ab3","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-09-01 11:48:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibd2c68103e73b1ecedd1aade780f96af3d9e9ab3","subject":"Switch to EFI-compatible CentOS Stream cloud image","status":"MERGED","created":"2025-08-30 20:27:49.000000000","updated":"2025-09-01 11:49:53.000000000","submitted":"2025-09-01 11:48:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958965","cherry_pick_of_change":958641,"cherry_pick_of_patch_set":6,"meta_rev_id":"820a4c778326d581180b558489a78568335fe224","_number":958965,"virtual_id_number":958965,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-01 11:48:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-09-01 09:17:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-08-31 08:09:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-01 08:34:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-09-01 09:17:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-30 22:55:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-31 08:09:11.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-01 08:34:02.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-09-01 09:17:24.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"d39c292bf8c6473fb4cdc133b63ce049d430b8a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-30 20:27:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"829b4cdb2f6e4d768b75e6b180f9822e3219214a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 22:55:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc50d5362cc249aa93c246f93b5d4a4f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff943fb7772446fb9ebf2156bf597028 : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45fecd4328194e48a74437f8636088ce : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16996ff30c5340f68679129010210dce : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8acb08a78363472f82c6e9e6ccb2a829 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68c6540c3fa146f885431c49da1cd716 : SUCCESS in 2m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dcbe494379cc4626871ad323ae9e121e : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fd0394e484874cc9b4fbe749f34d1082 : SUCCESS in 8m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7c2314687d08433a9498acf808a7d871 : SUCCESS in 8m 30s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/def8f5f327b14c8eaf861196f469cefd : SUCCESS in 1h 08m 01s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/48273f211c1c4c75bc7a8235b0de43d7 : SUCCESS in 56m 43s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f403e0516a6e479292f14ec1a21150d7 : SUCCESS in 28m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/95d9b18ba3254af7a867baf342856e4f : SUCCESS in 29m 20s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cfde2b781ccb42719bfb97ee51ce38b1 : SUCCESS in 32m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a8eaba770db14862874cc09c98cee63b : SUCCESS in 16m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be6e7d5c136a44828225042345f860c9 : SUCCESS in 10m 21s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f5da2acf8ce74e94a060a22373e0cf8f : SUCCESS in 2h 22m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10561df9ed8e417c8ac3ac3c672c20bd : SUCCESS in 1h 39m 21s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6fb49ea40c2445f7a9e7df44e8580558 : SUCCESS in 1h 57m 43s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b43ac401d74f4ca79d79295e8261163c : SUCCESS in 2h 18m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/21d0f8a1bc8942a3accb1aa460bba875 : SUCCESS in 8m 33s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/12b99bba078b422bac37c3f8c5c01c31 : SUCCESS in 19m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82545973dd014e539bfb8cda680b47e5 : SUCCESS in 12m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b444c8d10a614c5da03bcd20a8267a69 : SUCCESS in 8m 19s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dd412da0703a450c987769e0664a87ee : SUCCESS in 22m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c729cfe226d849f5b125fd5a73c470f7 : SUCCESS in 30m 51s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5db9098dde8141bb84afc965606a32d9 : SUCCESS in 37m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c1907e8e07d148b4a1c50fea8dba5b34 : SUCCESS in 7m 28s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/1cf06b624546481e899a869b18ec9846 : SUCCESS in 13m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fb4c89f5053454e92a231337b7f7e18 : SUCCESS in 13m 33s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/524f446de056448fbadc6ff6740f4312 : SUCCESS in 13m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c8b8909f08b148959ad3151833b01956 : SUCCESS in 14m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd366bc5eba84d77aa421612879732ca : SUCCESS in 9m 01s","accounts_in_message":[],"_revision_number":1},{"id":"40600db22c8ef24a306ba8a538ad19000c8e5529","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-31 08:09:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a565e228d5ce97c6a8387627d3cea76cee49d6c0","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-09-01 08:34:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"09c0825235c5a2eeb78e960ea774f12d3dfd0eec","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-01 09:17:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"288ff2765e1d1c08c4ad1c8649001d7e4130b13d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 09:17:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ccfa450f8864e745c8151c27f4d2e539ab66124f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 11:48:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/204f575a45d44c31a4312151b7e4c1ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d032de231be24481b225082d98f965f1 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/112573f53dc44e54a6b34e3390df52f8 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eecd27e3b22b4f1c8d6b558fce92ce9b : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ca76a2d014f45f3baf3df883abf122f : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b0b5b7c9b7374e37a3b88806005322bd : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/27f0def795ad44a08f99f89899ae114e : SUCCESS in 10m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/60d36b2742c54ebf95da8727b8243e9d : SUCCESS in 13m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a8969b72259a477c93ac17e8f8787725 : SUCCESS in 41m 41s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/1007540896774bae96ed584d8349c214 : SUCCESS in 33m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d6546dcd4e54e95b2c250d91483d47c : SUCCESS in 55m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/70ef24ca0a444a04a8d895f7e58cc3ed : SUCCESS in 29m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/64eab0d5d56d46fca6212cb032766c24 : SUCCESS in 54m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3fd56a3947d24db485b29748ebd0a0a5 : SUCCESS in 17m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca972933c25342b29e8ebfcec5a710e2 : SUCCESS in 15m 21s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d969ff56e4184e47a877a64bd67f1121 : SUCCESS in 2h 02m 19s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/809ee73076884c9fb83474f217f08651 : SUCCESS in 2h 24m 02s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e62c8747a87647298a74ac6ad9c1eef7 : SUCCESS in 1h 32m 54s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c17e7f70d6b4273ab5bf4977f17613a : SUCCESS in 2h 03m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/addfce8978c94dcc969c594d88520dc6 : SUCCESS in 16m 33s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/57d266fadcb549e08077af844dbeb94a : SUCCESS in 16m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab09b2a0b69a4b53b4dd1672a3959dc8 : SUCCESS in 21m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/64974637b3654d6381b390415380aa60 : SUCCESS in 14m 30s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4a9db48a224d40669bca780029d1bb67 : SUCCESS in 35m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef8692d9353641e6b6557a88466f2e10 : SUCCESS in 36m 30s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/48ba323c0bff47648d04c8fe77fab814 : SUCCESS in 32m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4e7c0a62c4724c8aae3c02a5d87f0a03 : SUCCESS in 16m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b4e112ef30d421596cfafac3c6256c4 : SUCCESS in 11m 44s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/00a30b617dd7478bb6ed0d211da9c1ed : SUCCESS in 15m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77d39878b9de415896d343d5dbfd64d7 : SUCCESS in 13m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0212da7d49c5eae21df2db77aaf4f8525b5ca91f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 11:48:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"820a4c778326d581180b558489a78568335fe224","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 11:49:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a0df3f946664165a19515e4d5723854\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cffba3848bf4950bf0c30ecb6198817 : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d72e9f336f034a8087564766f73b26fb : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0db0f82bf8515bee6278ea27595fb68b63fd117d","revisions":{"0db0f82bf8515bee6278ea27595fb68b63fd117d":{"kind":"REWORK","_number":1,"created":"2025-08-30 20:27:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/65/958965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/958965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/958965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/958965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/958965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/958965/1"}}},"commit":{"parents":[{"commit":"7ff2db4867c51193989aab2d384cd7a31b157c86","subject":"Revert \"CI: Make rocky9 VM jobs non-voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ff2db4867c51193989aab2d384cd7a31b157c86"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-28 21:15:32.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-30 20:27:49.000000000","tz":0},"subject":"Switch to EFI-compatible CentOS Stream cloud image","message":"Switch to EFI-compatible CentOS Stream cloud image\n\nAs explained in a recent centos-devel message [1], which is in the\ncontext of CentOS Stream 10 but also applies to 9:\n\n   Q: What is the difference between the two x86_64 qcow2 images?\n   The original (without the extra x86_64) is our original Bios only image.\n   The one with the extra x86_64 boots using EFI\n\n[1] https://lists.centos.org/hyperkitty/list/devel@lists.centos.org/thread/PQSMX2QQCOJRKMNLQSXDOFIKRGJF4QDM/\n\nCloses-Bug: #2121588\nChange-Id: Ibd2c68103e73b1ecedd1aade780f96af3d9e9ab3\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit e26b20955424619708e3d15afc39d2cdcd287739)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0db0f82bf8515bee6278ea27595fb68b63fd117d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0db0f82bf8515bee6278ea27595fb68b63fd117d"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
