)]}'
{"id":"openstack%2Fkayobe~959042","triplet_id":"openstack%2Fkayobe~master~Ie71c51ec8427e3e2acc792a8da062a69c4019a86","project":"openstack/kayobe","branch":"master","topic":"redfish_patch","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-09-04 10:14:16.000000000","reason":"Change was marked work in progress"},"38330":{"account":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"last_update":"2025-09-10 09:15:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie71c51ec8427e3e2acc792a8da062a69c4019a86","subject":"WIP: Add redfish pxe args for virtual media","status":"ABANDONED","created":"2025-09-01 12:56:08.000000000","updated":"2025-09-10 09:15:59.000000000","total_comment_count":6,"unresolved_comment_count":4,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"d56dc04d87c33d5b3731ecd326d546bb7d58ce7d","_number":959042,"virtual_id_number":959042,"owner":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-09 13:22:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}]},"reviewer_updates":[{"updated":"2025-09-01 14:40:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-02 09:01:57.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"CC"},{"updated":"2025-09-02 09:07:24.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-09-02 17:42:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-05 15:02:36.000000000","updated_by":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"reviewer":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"state":"REVIEWER"}],"messages":[{"id":"3d23b952ae0a565acf8543cad4e8bf20cba82848","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-01 12:56:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c9817671a2ca1914db99d24c9ecb162eccfab4b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-01 13:00:32.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"8a59460c0b20574d7bf3113077e1697ef8e3c8e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 14:40:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5383b2ede4a145789ede6ec293d46865\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b29e4a02d28495bb5bbf1a0d9e33708 : SUCCESS in 9m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c06e5e3f5f3c4471b73e4423672ff989 : FAILURE in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89635b9304f6482bb21165515fd8c8f1 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b453d5b4af004342b1127a80476f79ed : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ecb62eeef58c4b0dbd16d4254343c991 : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/00c3dbfb5d964ed6b75f04c32b89438f : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b4fd2465e104dd493435606c2daad14 : SUCCESS in 6m 01s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b9eb7ae7fec8451d8e0ee3c522e0669f : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d4ad1f1aa4774ff7b3c6a2ae0bc5042c : SUCCESS in 8m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c1e7e41b6c3441728a162a92a972110c : FAILURE in 8m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b9c460660c84f798e94a54fd2935669 : SUCCESS in 29m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fc727f169be14f77a54b64a437136e98 : SUCCESS in 32m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c5d71e99ac704e0b8019f62babf2978d : SUCCESS in 12m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29f82d1a201e45f1ba9d27d08f53bfc5 : SUCCESS in 9m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/559fed1ed28145f19b35a6525e5b6fcd : SUCCESS in 1h 40m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d9a9b4d82ab4b36b46c97ba69881fb2 : SUCCESS in 9m 02s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/374bbc9c94a04778a81add47284cf99d : SUCCESS in 17m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1917a0528e5d4d0dbe9d3215e5b48066 : SUCCESS in 25m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f8330e4014d14e5b98d1e771ff7da454 : SUCCESS in 14m 09s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/22f015bf30ba4edfa1632c70585c17df : SUCCESS in 13m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d7bc372d13a40b89915353621a6e75a : SUCCESS in 11m 34s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/6d51b9a238c94969a0b2d272db6201cd : SUCCESS in 11m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/04b25831bda04a0184ff8d38aad4b5a9 : SUCCESS in 13m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b03fffe33bae4fd1b8c2cb62a07177ce : SUCCESS in 9m 09s","accounts_in_message":[],"_revision_number":1},{"id":"10eaaea14c195a4e73c6ad6eed60007a1d103fa1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"real_author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2025-09-02 08:59:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_38330\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"}],"_revision_number":2},{"id":"a2a214f781d2bd5a59f94bf7595689cfb64701bd","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2025-09-02 09:01:57.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"8f747db3a2dcad0255aede0162393b4a5e62591f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-02 09:07:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9de4d35b4d4a807c0edc3fb9ae229fb11f9e2680","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-02 09:24:31.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"629657080f3b7eecc103e76f8685ddf14e33fe4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 11:06:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/52a27e65066f474298dfd1345dd0304b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05d885380df3486087bf4af5c387c22c : SUCCESS in 9m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a515f51ae2314d1ba4bb8ba4d870f18f : SUCCESS in 6m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9832619c5dd40c2b693458049e79940 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cce20c5328c44654b253bef4bdbb2081 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2088757903574027b1c81fbc0692b0a8 : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c8243b04ba2d4e848fc908e02c5f0adb : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db0eb3fc679d4ecf9cb12d4918aa1ffb : SUCCESS in 7m 41s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9bd7ff5f608a4d1f8f77214175c8d5d2 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/831b2bbad6c54dd1b822a896f4c07408 : SUCCESS in 12m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6c5879fbe1d54b3e83e458de0b0bcd4d : FAILURE in 7m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be9f7d70bfcb49d08bb5d6593b71145f : SUCCESS in 55m 26s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/72ab7b3f6a73455eabf3a826c4eefd5a : SUCCESS in 31m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/37b7aea936a143ba934f03591acf46d9 : SUCCESS in 12m 05s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45a7fc4f42fb4da0a08ffc852f0ea734 : SUCCESS in 14m 19s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f7fe23d91af492093bd64cd8b439377 : SUCCESS in 1h 37m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24cad4f3731646fdb66807cc97c1df16 : SUCCESS in 16m 00s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ccf2f874c05c4833929b8aa545fffb86 : SUCCESS in 14m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c896a2a2b3f241709239e434bf5c83a2 : SUCCESS in 28m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/546a8016455a45099bf482e7be49b71b : SUCCESS in 10m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54aa2a3e5df04728834e639fdd5cc2c9 : SUCCESS in 6m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d8a9317612674de993d4acfad9c7ba7a : SUCCESS in 11m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e08c806a215e4244be2646fc046efda9 : SUCCESS in 14m 53s","accounts_in_message":[],"_revision_number":3},{"id":"32476ad950b1e49702ac0d0bfe01c08f94750d97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-04 10:11:10.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c947b3315ac8f4175095f4fcd48444880ac2069f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-04 10:11:42.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"935ffc9c1bb524e20500272d6f17b40eda28957e","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-04 10:14:16.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":5},{"id":"990c4d0f04e102c8cc2cff0944461443c440e306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 11:46:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee7fffc88393441f8354389c116da108\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e344489141b4436db1baf425d4321f2f : SUCCESS in 8m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0e28ba5a67b94535883d5d3caa5dbdb8 : SUCCESS in 6m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aaa3114ba6e649d18a3268f63c37fa6a : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2204bfd9bfd0442abf08c3627906f7ac : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6affc0a51028446996e732fb0c04926a : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b9e4ded2625b437fa58138232403e17c : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56822a0bb7074bd5983595535db2d226 : SUCCESS in 5m 57s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/337deb894b094447a3adbfd39ce21c25 : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c77ab7cd981e4a1997337abcefc7c413 : SUCCESS in 11m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3f548fb790ba4599944c20b886372190 : FAILURE in 8m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/300d0b1e99fe46cdb546ffabae9a3358 : SUCCESS in 1h 11m 54s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6b8f0b62028640978886a17e2776a1e4 : SUCCESS in 53m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b41e57535f7c4b32ae48799bd9d4ab9e : SUCCESS in 16m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f90b158cefa4ea7949870a138069a3d : SUCCESS in 11m 42s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c56f97cb9e934a30a65e74ce7088991b : SUCCESS in 1h 06m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3abdf80988b42f1951720689f203b4b : SUCCESS in 17m 19s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4cbf0daaad344cbc8b967c54d8bd9a3f : SUCCESS in 17m 17s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16dd3e1ae9e443f4a25bd12e1d8a2fac : FAILURE in 12m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/850f30b301764314a924e80d9a3e5d6b : SUCCESS in 14m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1615c9e8127d43c38c2f286034b447f6 : SUCCESS in 10m 30s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bcc9810bb1b04dc5a6f67b57c39cba1c : SUCCESS in 10m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ff7db8256424d6f96d5f662cedbf8e7 : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":5},{"id":"a2dc9040b8e2bcee96571e7d367304331b6d7381","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-04 12:06:40.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"fc5b51bf0cf900d6cebfccaf59e830cace6a002b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 13:15:58.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c87ff8743ad482988bdc20bbae96fb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/323c76f752904b26a1dfad595093d1f6 : SUCCESS in 5m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2234857f2b9340c98eb968eb84505112 : SUCCESS in 4m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/900ba3b07c57455583f49d74237be6b2 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98896b50206c41959cd6ec38eb74c622 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bfb4017350dd4a5cb84e3e13163d22ac : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9cbf9411974f438fa4e796c630ab243b : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d98d5fdfbaf74b139c6bc089524461f3 : SUCCESS in 8m 56s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/76850c6bce3347b4b18aa55728a4436b : SUCCESS in 3m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3106b04cd95e46dab07f3fa73716ad04 : SUCCESS in 8m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1372728355fa401fbb8c8b2d9e043c4c : FAILURE in 5m 50s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/741821505b9846b98fe668328a29c512 : SUCCESS in 55m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e6da2e15b91a4b25b3f1cb07b72e4c15 : SUCCESS in 1h 07m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3f2becbe8b064bf0a0c18bef6477f143 : SUCCESS in 22m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6c6e5cb79f04ea0ab1d5ba0c803856d : SUCCESS in 12m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a8338968d3c4f9281f6a1fd6d2621ea : SUCCESS in 1h 03m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76c0847d62e54cd798e0f3ce81646ff2 : SUCCESS in 15m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b9bde907907647e8b9e878e61fbfab2a : SUCCESS in 17m 08s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d1ecc2ebdb84b2fb32187a98f8addcb : SUCCESS in 33m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bc1bbc7814b741c98581fe5ccf8f0a49 : SUCCESS in 15m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fc14937ec7a403e9628c553a1e7bd24 : SUCCESS in 7m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/61c4a81fc50649c6aab18cf13fbb8e50 : SUCCESS in 15m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/737f190adf844b48bafe8fb5f7bfff99 : SUCCESS in 9m 17s","accounts_in_message":[],"_revision_number":6},{"id":"7483086cb2974d0159e5a0ac8098a7a2dbd65d00","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-05 12:47:25.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"76d0b6f7af1e740b738c1de87126f85d5724ca25","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-05 13:37:25.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"7bdffe72806adc197752c5b6b56249dd2c3f1d43","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2025-09-05 14:53:17.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"20c8efb08d93ebfb8ea811e3c9e287cca7c7d7a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 15:01:16.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/031418cd427a43f79002ab13381140f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c09a9006772f467884b8650b0b96d9f6 : SUCCESS in 9m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1ffd444c22f14ac592dd40a7056a7c32 : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6933b01e484547fab788a73c62a555c3 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bec2db5d622648adaf748f5b6614c28e : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0056a01be92e45f2ab8bc353c6b47a92 : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/552baf7f88794b9a9f56b904aee76315 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01d05bc91988484a9ec3285db7ad620e : SUCCESS in 4m 28s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/13820657f99748c6acbfdc8605b1dd40 : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e594f66e3be8442f9cbe04bc1caf0904 : SUCCESS in 13m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/abbc6989e4664c3183dd8b764095415a : FAILURE in 5m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f7b5692c64c4bcfb40212309e0f6e33 : SUCCESS in 28m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/14fb2a0dca7b458baba6f251e9e20d72 : SUCCESS in 1h 13m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2c3d4c1f9ad344cf9f1e7d79cfb387f4 : SUCCESS in 19m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a2f91ce76ed4047a9c8d18292634f4e : SUCCESS in 8m 12s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0069a889c45474b9fcc427bc4614183 : SUCCESS in 1h 22m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/006cb89bdd804525a46b4ff6e8847adf : SUCCESS in 14m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/19b5e2581bf448f08d2a8ae2a9f9d850 : SUCCESS in 13m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/918e6aa71447437bbdba2ed2280a8c01 : SUCCESS in 30m 18s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b3304d4e04cb466cb90f42d5815cccc3 : SUCCESS in 13m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3aa8f8ca063842f592a7797f25787cf3 : SUCCESS in 14m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a4c696ad5d544d2886ca8dd49c0e9415 : SUCCESS in 15m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae6121fcfcd8496992d4ec7e03aa49d5 : SUCCESS in 9m 37s","accounts_in_message":[],"_revision_number":8},{"id":"0c10c7fc7f600a05ce6f94598a4b6ddd1b439a40","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-09-05 15:02:36.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e1a8e60db21bc056fa005d7dd2e5906eeec2f461","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2025-09-05 15:02:46.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f04013ef671a3ec037829f6320be56f5b476dd27","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2025-09-05 15:09:32.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0bae6a6a94549358ff575bc621f49017d1bc765d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-05 15:22:22.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"dcdb700eaf7c1b34aaddf18e9023f2397b8e1393","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-05 15:22:32.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"ea9c36d5b7956f683696144cd88a50af69342898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 16:32:58.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3dcb8b6aef594ef0999f885cac3fa4a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0cf2218f5d84198a8d80d64ae41508c : SUCCESS in 9m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2500973e137b431ca5d34f51c4654a90 : SUCCESS in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46d457c093974895a316a7261fcefb3a : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98acb9c3884b44a58dd42fbf19d5c17d : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5381b4ff443643be8288a028e3bba1e8 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/621ca0263c3a4c259715ea3966b1477e : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d5d425822bd451da7d9be54e53a7349 : SUCCESS in 5m 34s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/32df87876d394976bf68190a748180b6 : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2f5212e05ae24edba1ced04870b6d0f5 : SUCCESS in 6m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4267923a4c3d45f2b7b4e7effaed8c3e : FAILURE in 7m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9ea878e2192443798c4c528d5cb2843 : RETRY_LIMIT in 2m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/59ff23f1c3c24957b8ce66bb38ea6563 : RETRY_LIMIT in 6m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b9ea87da670946e898a2ef5df90f392d : RETRY_LIMIT in 7m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca7741a22a064de1b66d2c6b5c0cfef0 : RETRY_LIMIT in 2m 22s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31eca9fd43f148d088b7dc30a26c4a45 : FAILURE in 58m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54ca6a49407c427eaade2d264b50af30 : FAILURE in 6m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/201b3fe192874bd79681659be7321dfe : FAILURE in 4m 20s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e2302c081ad4ca7aeac05466e199f5e : FAILURE in 21m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e0b2b60d7ba149efad5b8e7db6f9bfaf : FAILURE in 6m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/992e33f435f346e6a7728ed35d447ea7 : POST_FAILURE in 6m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ed743ed56802415cba122c07663e7a4d : POST_FAILURE in 5m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ec58d13fb1b433ca88768e1598250c0 : FAILURE in 6m 27s","accounts_in_message":[],"_revision_number":10},{"id":"78bd54fdc31a0d59a12c0fcce0f80ddc4c82c48b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-08 09:29:59.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"5952cce28987af04bb07ac737ba4ecbd5f7548c7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-08 09:30:04.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"3d132c2ac8e841b8c68ee1858e30f40169b49b55","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-08 09:46:49.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"d4d6570989f81caa05afd68877d2d2255f572352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 10:37:40.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15bcbc45f16e4e2b81f7898e285cf15f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/635c3d72693940ddb3101c169ac70b10 : SUCCESS in 4m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cdb64446ea9c4a08a88cfa5a05666447 : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c12cf3cca15452997483ef3c0138036 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c13e88b7dd4f45f7a89826e2bd48bf07 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/472211dc8347408bb2ac59f619276a3b : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f7766b8e3b44f22a096919772b62521 : SUCCESS in 2m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa22fe97a9194ba4b5c91da3640cd5d6 : SUCCESS in 4m 29s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3da44e2c0e604ce9ab69383752d23557 : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5f3dcf15c579480d81a25956b9fa898c : SUCCESS in 11m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e06d0e85758b458895331f49c8b0e96d : FAILURE in 9m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8a04254bc5447938ba66ab8d1638325 : RETRY_LIMIT in 3m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c1c43bbfd6aa4a39b29ed226445940e0 : RETRY_LIMIT in 2m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7e2085e15bed4c99bc80d1e266a39f68 : RETRY_LIMIT in 4m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b1ac64754a044ebbf71e518c94167be : RETRY_LIMIT in 4m 09s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/625f73846ceb4a0694da7f35660b131b : FAILURE in 39m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a7307b670a14d6ca002a62d9ce2899a : FAILURE in 6m 44s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cf3e1ae3bc494f26b08ede116f175e14 : FAILURE in 6m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4327f662182049a1ba7f565ebc280d85 : FAILURE in 21m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1d7ca095647b462ca1223434c4a8f504 : FAILURE in 6m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9c29989cee548878bb8aac9a110b2bc : POST_FAILURE in 6m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4187fc8df9dd45e0af8a3d8fba856508 : POST_FAILURE in 5m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c65c583ea154824a07247a55abba063 : FAILURE in 6m 19s","accounts_in_message":[],"_revision_number":13},{"id":"dc37b82fed3d37fdf0e05661d795bba4db352d8b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-09 12:06:22.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"31c7b16f127afdb6a4bacc7a5849f746144119ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 13:22:34.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/44a4b997a2a147aab4bee8207ab6ffb0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6149862b85774294898aeb5817500686 : SUCCESS in 6m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d6864f0f7698411691b5bcdb38353485 : SUCCESS in 5m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f224113dbd149469ac0be0b92ec40f0 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eb3812f45794b5c9b18e2617d50374c : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c8f1d4de26c34cc3824c007823731dad : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a7d5317540b640ee942ce9d9c06795eb : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/945788421a92450a91fc91e339afd217 : SUCCESS in 5m 49s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aa18a1b7ce6d48b480d7e929678a0a3f : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ccda5e17270f4f07a74292684f0fd7fd : SUCCESS in 8m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d5179437cb8546e5854fb8c19b4d9327 : FAILURE in 8m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48f0b431addc41ef96d1fa15e68d7e44 : RETRY_LIMIT in 4m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/495ffcd06f844cfda586eaa3adcf9bd0 : RETRY_LIMIT in 4m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9fd64005da274df0a97c401bdbc75bd0 : RETRY_LIMIT in 4m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d28f9aaa980942318ed74937c03bce3f : RETRY_LIMIT in 4m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72c9f2e274ef4edca8ad59e4f3583d03 : FAILURE in 1h 14m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c38408f873e148b4a0770eef203ec133 : FAILURE in 7m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/888cfa156907454ea9246f68da6e239b : FAILURE in 5m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/489772503ceb428baf3de4b1c0c71cdc : FAILURE in 26m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bca4cdb9069f43548d18f9c30cd7bb23 : FAILURE in 7m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/378e04b497494ce9bbbb312fd512e171 : POST_FAILURE in 6m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bb7587a0553f452b942cca59753c9743 : POST_FAILURE in 8m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf52f516b2d74ef28e58b200342a0dda : FAILURE in 6m 32s","accounts_in_message":[],"_revision_number":14},{"id":"d56dc04d87c33d5b3731ecd326d546bb7d58ce7d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-10 09:15:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"f04bb11bc620e625dd135bca56aa62182c0b7aab","revisions":{"cc363538e820da15adff3a6178d8cb924e490406":{"kind":"REWORK","_number":1,"created":"2025-09-01 12:56:08.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/42/959042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/959042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/959042/1"}}},"commit":{"parents":[{"commit":"e26b20955424619708e3d15afc39d2cdcd287739","subject":"Switch to EFI-compatible CentOS Stream cloud image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e26b20955424619708e3d15afc39d2cdcd287739"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-01 12:41:29.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-01 12:54:43.000000000","tz":60},"subject":"WIP: Add redfish pxe args for vitrual media","message":"WIP: Add redfish pxe args for vitrual media\n\nUse ironic pxe variables for redfish to move from use of PXE boot to redfish\nDeprecated variables are:\n- kolla_ironic_pxe_append_params_default\n- kolle_ironic_pxe_append_params_extra\n- kolla_ironic_pxe_append_params\nDeprecated variables are redefined for use by redfish.\n\nChange-Id: Ie71c51ec8427e3e2acc792a8da062a69c4019a86\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc363538e820da15adff3a6178d8cb924e490406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc363538e820da15adff3a6178d8cb924e490406"}]},"branch":"refs/heads/master"},"6b07709e2c43bb7f97e979078a5826c0a7a1b0c9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-09-02 08:59:54.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"real_uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/42/959042/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/959042/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/959042/2"}}},"commit":{"parents":[{"commit":"8fb065fa129521b913c9399ed34dd99c481208e9","subject":"Merge \"CI: Rename seed-vm-efi jobs to seed-vm-q35\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8fb065fa129521b913c9399ed34dd99c481208e9"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-01 12:41:29.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-02 08:59:54.000000000","tz":0},"subject":"WIP: Add redfish pxe args for vitrual media","message":"WIP: Add redfish pxe args for vitrual media\n\nUse ironic pxe variables for redfish to move from use of PXE boot to redfish\nDeprecated variables are:\n- kolla_ironic_pxe_append_params_default\n- kolle_ironic_pxe_append_params_extra\n- kolla_ironic_pxe_append_params\nDeprecated variables are redefined for use by redfish.\n\nChange-Id: Ie71c51ec8427e3e2acc792a8da062a69c4019a86\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b07709e2c43bb7f97e979078a5826c0a7a1b0c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b07709e2c43bb7f97e979078a5826c0a7a1b0c9"}]},"branch":"refs/heads/master","description":"Rebase"},"ecc047303066745df269d41a5612e9fff54afc06":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-09-02 09:24:31.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/42/959042/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/959042/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/959042/3"}}},"commit":{"parents":[{"commit":"8fb065fa129521b913c9399ed34dd99c481208e9","subject":"Merge \"CI: Rename seed-vm-efi jobs to seed-vm-q35\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8fb065fa129521b913c9399ed34dd99c481208e9"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-01 12:41:29.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-02 09:24:31.000000000","tz":0},"subject":"WIP: Add redfish pxe args for virtual media","message":"WIP: Add redfish pxe args for virtual media\n\nUse ironic pxe variables for redfish to move from use of PXE boot to redfish\nDeprecated variables are:\n- kolla_ironic_pxe_append_params_default\n- kolle_ironic_pxe_append_params_extra\n- kolla_ironic_pxe_append_params\nDeprecated variables are redefined for use by redfish.\n\nChange-Id: Ie71c51ec8427e3e2acc792a8da062a69c4019a86\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ecc047303066745df269d41a5612e9fff54afc06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ecc047303066745df269d41a5612e9fff54afc06"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4356ee525df00d8748d4b83a3a48011635631546":{"kind":"REWORK","_number":4,"created":"2025-09-04 10:11:10.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/42/959042/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/959042/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/959042/4"}}},"commit":{"parents":[{"commit":"8fb065fa129521b913c9399ed34dd99c481208e9","subject":"Merge \"CI: Rename seed-vm-efi jobs to seed-vm-q35\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8fb065fa129521b913c9399ed34dd99c481208e9"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-01 12:41:29.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-04 10:09:31.000000000","tz":0},"subject":"WIP: Add redfish pxe args for virtual media","message":"WIP: Add redfish pxe args for virtual media\n\nUse ironic pxe variables for redfish to move from use of PXE boot to redfish\nDeprecated variables are:\n- kolla_ironic_pxe_append_params_default\n- kolle_ironic_pxe_append_params_extra\n- kolla_ironic_pxe_append_params\nDeprecated variables are redefined for use by redfish.\n\nChange-Id: Ie71c51ec8427e3e2acc792a8da062a69c4019a86\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4356ee525df00d8748d4b83a3a48011635631546"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4356ee525df00d8748d4b83a3a48011635631546"}]},"branch":"refs/heads/master"},"80744603b98a1863c951d71b85ab06160e37fbaa":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-09-04 10:11:42.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/42/959042/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/959042/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/959042/5"}}},"commit":{"parents":[{"commit":"47b9307e18cddfb3c951056352ca3bd000eeceba","subject":"Merge \"Remove rule for legacy deployment kernel update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47b9307e18cddfb3c951056352ca3bd000eeceba"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-01 12:41:29.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-04 10:11:42.000000000","tz":0},"subject":"WIP: Add redfish pxe args for virtual media","message":"WIP: Add redfish pxe args for virtual media\n\nUse ironic pxe variables for redfish to move from use of PXE boot to redfish\nDeprecated variables are:\n- kolla_ironic_pxe_append_params_default\n- kolle_ironic_pxe_append_params_extra\n- kolla_ironic_pxe_append_params\nDeprecated variables are redefined for use by redfish.\n\nChange-Id: Ie71c51ec8427e3e2acc792a8da062a69c4019a86\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80744603b98a1863c951d71b85ab06160e37fbaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80744603b98a1863c951d71b85ab06160e37fbaa"}]},"branch":"refs/heads/master","description":"Rebase"},"09a566d98610753ab475341a9cb7ca90685b692b":{"kind":"REWORK","_number":6,"created":"2025-09-04 12:06:40.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/42/959042/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/959042/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/959042/6"}}},"commit":{"parents":[{"commit":"47b9307e18cddfb3c951056352ca3bd000eeceba","subject":"Merge \"Remove rule for legacy deployment kernel update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47b9307e18cddfb3c951056352ca3bd000eeceba"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-01 12:41:29.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-04 12:06:40.000000000","tz":0},"subject":"WIP: Add redfish pxe args for virtual media","message":"WIP: Add redfish pxe args for virtual media\n\nUse ironic pxe variables for redfish to move from use of PXE boot to redfish\nDeprecated variables are:\n- kolla_ironic_pxe_append_params_default\n- kolle_ironic_pxe_append_params_extra\n- kolla_ironic_pxe_append_params\nDeprecated variables are redefined for use by redfish.\n\nChange-Id: Ie71c51ec8427e3e2acc792a8da062a69c4019a86\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09a566d98610753ab475341a9cb7ca90685b692b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09a566d98610753ab475341a9cb7ca90685b692b"}]},"branch":"refs/heads/master"},"44422c074393fc2de7f1f86c26ec2730fba8933f":{"kind":"REWORK","_number":7,"created":"2025-09-05 12:47:25.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/42/959042/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/959042/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/959042/7"}}},"commit":{"parents":[{"commit":"47b9307e18cddfb3c951056352ca3bd000eeceba","subject":"Merge \"Remove rule for legacy deployment kernel update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47b9307e18cddfb3c951056352ca3bd000eeceba"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-01 12:41:29.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-05 12:47:24.000000000","tz":60},"subject":"WIP: Add redfish pxe args for virtual media","message":"WIP: Add redfish pxe args for virtual media\n\nUse ironic pxe variables for redfish to move from use of PXE boot to redfish\nDeprecated variables are:\n- kolla_ironic_pxe_append_params_default\n- kolle_ironic_pxe_append_params_extra\n- kolla_ironic_pxe_append_params\nDeprecated variables are redefined for use by redfish.\n\nChange-Id: Ie71c51ec8427e3e2acc792a8da062a69c4019a86\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44422c074393fc2de7f1f86c26ec2730fba8933f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44422c074393fc2de7f1f86c26ec2730fba8933f"}]},"branch":"refs/heads/master"},"fa5b396c2f07a7bee5de273e2eaf6f44272ede27":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-09-05 13:37:25.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/42/959042/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/959042/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/959042/8"}}},"commit":{"parents":[{"commit":"8e977fa25acdcdd6b8ac25de505288f37c10453e","subject":"Switch infra VMs to EFI-compatible CentOS image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e977fa25acdcdd6b8ac25de505288f37c10453e"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-01 12:41:29.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-05 13:37:25.000000000","tz":0},"subject":"WIP: Add redfish pxe args for virtual media","message":"WIP: Add redfish pxe args for virtual media\n\nUse ironic pxe variables for redfish to move from use of PXE boot to redfish\nDeprecated variables are:\n- kolla_ironic_pxe_append_params_default\n- kolle_ironic_pxe_append_params_extra\n- kolla_ironic_pxe_append_params\nDeprecated variables are redefined for use by redfish.\n\nChange-Id: Ie71c51ec8427e3e2acc792a8da062a69c4019a86\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa5b396c2f07a7bee5de273e2eaf6f44272ede27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa5b396c2f07a7bee5de273e2eaf6f44272ede27"}]},"branch":"refs/heads/master","description":"Rebase"},"8ffdf8291c7be6f02580e91e0fcd0f53b92291b6":{"kind":"REWORK","_number":9,"created":"2025-09-05 15:22:22.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/42/959042/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/959042/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/959042/9"}}},"commit":{"parents":[{"commit":"8e977fa25acdcdd6b8ac25de505288f37c10453e","subject":"Switch infra VMs to EFI-compatible CentOS image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e977fa25acdcdd6b8ac25de505288f37c10453e"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-01 12:41:29.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-05 15:22:22.000000000","tz":0},"subject":"WIP: Add redfish pxe args for virtual media","message":"WIP: Add redfish pxe args for virtual media\n\nUse ironic pxe variables for redfish to move from use of PXE boot to redfish\nDeprecated variables are:\n- kolla_ironic_pxe_append_params_default\n- kolle_ironic_pxe_append_params_extra\n- kolla_ironic_pxe_append_params\nDeprecated variables are redefined for use by redfish.\n\nChange-Id: Ie71c51ec8427e3e2acc792a8da062a69c4019a86\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8ffdf8291c7be6f02580e91e0fcd0f53b92291b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8ffdf8291c7be6f02580e91e0fcd0f53b92291b6"}]},"branch":"refs/heads/master"},"9333c82a4e799600e3038870aa8c413d2f26c26b":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-09-05 15:22:32.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/42/959042/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/959042/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/959042/10"}}},"commit":{"parents":[{"commit":"43236f19c1f782e38c7a4b9ef4944b6e07128296","subject":"CI: Add jobs for infra VM testing with cloud image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43236f19c1f782e38c7a4b9ef4944b6e07128296"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-01 12:41:29.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-05 15:22:32.000000000","tz":0},"subject":"WIP: Add redfish pxe args for virtual media","message":"WIP: Add redfish pxe args for virtual media\n\nUse ironic pxe variables for redfish to move from use of PXE boot to redfish\nDeprecated variables are:\n- kolla_ironic_pxe_append_params_default\n- kolle_ironic_pxe_append_params_extra\n- kolla_ironic_pxe_append_params\nDeprecated variables are redefined for use by redfish.\n\nChange-Id: Ie71c51ec8427e3e2acc792a8da062a69c4019a86\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9333c82a4e799600e3038870aa8c413d2f26c26b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9333c82a4e799600e3038870aa8c413d2f26c26b"}]},"branch":"refs/heads/master","description":"Rebase"},"130049b3b5cc514b3b64e851193a6a32b5fc7fe3":{"kind":"REWORK","_number":11,"created":"2025-09-08 09:29:59.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/42/959042/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/959042/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/959042/11"}}},"commit":{"parents":[{"commit":"43236f19c1f782e38c7a4b9ef4944b6e07128296","subject":"CI: Add jobs for infra VM testing with cloud image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43236f19c1f782e38c7a4b9ef4944b6e07128296"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-01 12:41:29.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-08 09:29:59.000000000","tz":0},"subject":"WIP: Add redfish pxe args for virtual media","message":"WIP: Add redfish pxe args for virtual media\n\nUse ironic pxe variables for redfish to move from use of PXE boot to redfish\nDeprecated variables are:\n- kolla_ironic_pxe_append_params_default\n- kolle_ironic_pxe_append_params_extra\n- kolla_ironic_pxe_append_params\nDeprecated variables are redefined for use by redfish.\n\nChange-Id: Ie71c51ec8427e3e2acc792a8da062a69c4019a86\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/130049b3b5cc514b3b64e851193a6a32b5fc7fe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/130049b3b5cc514b3b64e851193a6a32b5fc7fe3"}]},"branch":"refs/heads/master"},"fbeb2b51ed451fcceed86f2e1334c686e3aa4259":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-09-08 09:30:04.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/42/959042/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/959042/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/959042/12"}}},"commit":{"parents":[{"commit":"c34c595dbe8c66d76f749d91319c709aee9f1548","subject":"Merge \"Fix kolla_docker_namespace comment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c34c595dbe8c66d76f749d91319c709aee9f1548"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-01 12:41:29.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-08 09:30:04.000000000","tz":0},"subject":"WIP: Add redfish pxe args for virtual media","message":"WIP: Add redfish pxe args for virtual media\n\nUse ironic pxe variables for redfish to move from use of PXE boot to redfish\nDeprecated variables are:\n- kolla_ironic_pxe_append_params_default\n- kolle_ironic_pxe_append_params_extra\n- kolla_ironic_pxe_append_params\nDeprecated variables are redefined for use by redfish.\n\nChange-Id: Ie71c51ec8427e3e2acc792a8da062a69c4019a86\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbeb2b51ed451fcceed86f2e1334c686e3aa4259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbeb2b51ed451fcceed86f2e1334c686e3aa4259"}]},"branch":"refs/heads/master","description":"Rebase"},"2897c103501cbc50490ce6d08feea6dacddd4ce7":{"kind":"REWORK","_number":13,"created":"2025-09-08 09:46:49.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/42/959042/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/959042/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/959042/13"}}},"commit":{"parents":[{"commit":"c34c595dbe8c66d76f749d91319c709aee9f1548","subject":"Merge \"Fix kolla_docker_namespace comment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c34c595dbe8c66d76f749d91319c709aee9f1548"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-01 12:41:29.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-08 09:46:49.000000000","tz":0},"subject":"WIP: Add redfish pxe args for virtual media","message":"WIP: Add redfish pxe args for virtual media\n\nUse ironic pxe variables for redfish to move from use of PXE boot to redfish\nDeprecated variables are:\n- kolla_ironic_pxe_append_params_default\n- kolle_ironic_pxe_append_params_extra\n- kolla_ironic_pxe_append_params\nDeprecated variables are redefined for use by redfish.\n\nChange-Id: Ie71c51ec8427e3e2acc792a8da062a69c4019a86\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2897c103501cbc50490ce6d08feea6dacddd4ce7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2897c103501cbc50490ce6d08feea6dacddd4ce7"}]},"branch":"refs/heads/master"},"f04bb11bc620e625dd135bca56aa62182c0b7aab":{"kind":"REWORK","_number":14,"created":"2025-09-09 12:06:22.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/42/959042/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/959042/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/959042/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/959042/14"}}},"commit":{"parents":[{"commit":"c34c595dbe8c66d76f749d91319c709aee9f1548","subject":"Merge \"Fix kolla_docker_namespace comment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c34c595dbe8c66d76f749d91319c709aee9f1548"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-01 12:41:29.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-09 12:06:21.000000000","tz":0},"subject":"WIP: Add redfish pxe args for virtual media","message":"WIP: Add redfish pxe args for virtual media\n\nUse ironic pxe variables for redfish to move from use of PXE boot to redfish\nDeprecated variables are:\n- kolla_ironic_pxe_append_params_default\n- kolle_ironic_pxe_append_params_extra\n- kolla_ironic_pxe_append_params\nDeprecated variables are redefined for use by redfish.\n\nChange-Id: Ie71c51ec8427e3e2acc792a8da062a69c4019a86\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f04bb11bc620e625dd135bca56aa62182c0b7aab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f04bb11bc620e625dd135bca56aa62182c0b7aab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
