)]}'
{"id":"openstack%2Fkayobe~959073","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~I5ab71d6c3968b852be17c02f27bbc04a400a7aaf","project":"openstack/kayobe","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-09-01 18:13:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5ab71d6c3968b852be17c02f27bbc04a400a7aaf","subject":"CI: Rename seed-vm-efi jobs to seed-vm-q35","status":"MERGED","created":"2025-09-01 15:22:12.000000000","updated":"2025-09-01 18:15:37.000000000","submitted":"2025-09-01 18:13:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959073","cherry_pick_of_change":958587,"cherry_pick_of_patch_set":2,"meta_rev_id":"5bd25b04797358f8b1a8c552ca0da8f1651031d3","_number":959073,"virtual_id_number":959073,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-01 18:13:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-09-01 18:00:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2025-09-01 16:35:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-09-01 18:00:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-01 15:22:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-01 16:35:32.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-09-01 18:00:24.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"7f0f8190d6d10571b3c544d44375171c914ab18f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-01 15:22:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* zuul.d/project.yaml","accounts_in_message":[],"_revision_number":1},{"id":"8e054f6ab36ff7146466e21cbf3ab8373d93edba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 15:22:44.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch stable/2024.2.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/kayobe/zuul.d/project.yaml@stable/2024.2\", line 63, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/kayobe/zuul.d/project.yaml@stable/2024.2\", line 64, column 1","accounts_in_message":[],"_revision_number":1},{"id":"29b6f2123db61eb00f32cda081667e1d7363a027","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-01 15:23:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6553f8f81da7705fc1af58447d8056585283b083","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-01 15:23:25.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"b55344bd688872ce15d3bd21fec0b78ceb3decb0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-01 15:24:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a11547c0be7aa24892c159ac2849c2a8e7863825","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 15:34:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f5dc4e8e3bf480dbc2ab556df34befd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1920a08108f04308a54590a2f0f06ad3 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3992d840bdc84d618c1982963ee7b692 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d6074f56ef914a6d9546c6625963f438 : SUCCESS in 5m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f61bfd2cf1ab4e41bfbee3cf137abdad : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd22e2bf63464a9fa137b44d6560a82d : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":2},{"id":"bbce3dc82f53f8916775f993083dd8e95e25ae1a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 15:52:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6dd5defaee047d4bdada62f4e06c091\n\n- kayobe-seed-vm-centos9s-q35 https://zuul.opendev.org/t/openstack/build/cff9853b6e3b4830890e8f942a51d6d1 : SUCCESS in 21m 01s (non-voting)\n- kayobe-seed-vm-rocky9-q35 https://zuul.opendev.org/t/openstack/build/908703fa556b424eb36f970ce536b731 : SUCCESS in 18m 50s\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/f2a061e1f7dd4863898605c162e80df5 : SUCCESS in 10m 41s","accounts_in_message":[],"_revision_number":2},{"id":"1d15f5f65d6f3343aa97e602c47e46df6b83f0c9","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-09-01 16:35:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f15279ceef69632c21955ec9213399a5a1c34c9d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-01 18:00:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8b79360d529069f50e1b03397bb70687c7618146","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 18:01:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"24787d5e72a56c6a1bdd07ac9e6dbfeaf9620b51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 18:13:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50b4df20445046c8a00c3534d07e0c19\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/621a18234f9a4331b6407467ef5ddd97 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ae07c3d1d704230aecda66e200fd69d : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ca7a41e8b84493b83720859a9ce98f9 : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":2},{"id":"16071ba4070fabc750b0e09e2b7b7bdc527db0b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 18:13:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5bd25b04797358f8b1a8c552ca0da8f1651031d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 18:15:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbc3536c07d84ef8a73bacfae3dc1b16\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03c63bb4cd7a42be9c5d4b5f0f3f9630 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e666c5ffac962d1681a48f602afee984cac6c264","revisions":{"eb43f5b165f04d7773a4773d1b49d016527ba2a4":{"kind":"REWORK","_number":1,"created":"2025-09-01 15:22:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/959073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/959073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/959073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/959073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/959073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/959073/1"}}},"commit":{"parents":[{"commit":"5326cf820b45b33af3e2c1904ef88aa5c27cbea8","subject":"Merge \"Switch to EFI-compatible CentOS Stream cloud image\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5326cf820b45b33af3e2c1904ef88aa5c27cbea8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-29 19:51:48.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-01 15:22:12.000000000","tz":0},"subject":"CI: Rename seed-vm-efi jobs to seed-vm-q35","message":"CI: Rename seed-vm-efi jobs to seed-vm-q35\n\nThe default seed VM boot firmware was switched to efi in 2024.2 [1], but\nCI jobs were not renamed to match.\n\nMove q35 jobs to the experimental queue, we don\u0027t need to run them on\nevery change. Also add missing kayobe-seed-vm-centos9s-q35 job.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/927015\n\nChange-Id: I5ab71d6c3968b852be17c02f27bbc04a400a7aaf\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 2228dbbcf09d52843b2178e9f3d3b00102e35ff2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb43f5b165f04d7773a4773d1b49d016527ba2a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb43f5b165f04d7773a4773d1b49d016527ba2a4"}]},"branch":"refs/heads/stable/2024.2"},"e666c5ffac962d1681a48f602afee984cac6c264":{"kind":"REWORK","_number":2,"created":"2025-09-01 15:23:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/959073/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/959073/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/959073/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/959073/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/959073/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/959073/2"}}},"commit":{"parents":[{"commit":"5326cf820b45b33af3e2c1904ef88aa5c27cbea8","subject":"Merge \"Switch to EFI-compatible CentOS Stream cloud image\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5326cf820b45b33af3e2c1904ef88aa5c27cbea8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-01 15:23:04.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-01 15:23:04.000000000","tz":120},"subject":"CI: Rename seed-vm-efi jobs to seed-vm-q35","message":"CI: Rename seed-vm-efi jobs to seed-vm-q35\n\nThe default seed VM boot firmware was switched to efi in 2024.2 [1], but\nCI jobs were not renamed to match.\n\nMove q35 jobs to the experimental queue, we don\u0027t need to run them on\nevery change. Also add missing kayobe-seed-vm-centos9s-q35 job.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/927015\n\nChange-Id: I5ab71d6c3968b852be17c02f27bbc04a400a7aaf\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 2228dbbcf09d52843b2178e9f3d3b00102e35ff2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e666c5ffac962d1681a48f602afee984cac6c264"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e666c5ffac962d1681a48f602afee984cac6c264"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
