)]}'
{"id":"openstack%2Fkayobe~959413","triplet_id":"openstack%2Fkayobe~master~I224a187372ff948634c5a039e284c4455ee24af2","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-09-05 14:21:25.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-09-05 14:35:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I224a187372ff948634c5a039e284c4455ee24af2","subject":"CI: Add jobs for infra VM testing with cloud image","status":"MERGED","created":"2025-09-04 06:49:49.000000000","updated":"2025-09-05 14:37:27.000000000","submitted":"2025-09-05 14:35:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959413","meta_rev_id":"45d0caf8177ab3c04a8201a91ae9fcf598bd4a0a","_number":959413,"virtual_id_number":959413,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-05 14:35:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-09-05 12:26:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-09-05 14:21:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-09-05 12:26:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-04 09:03:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-04 09:46:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-05 12:26:51.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"5e61b96ded9424464c7db575c2bb3b22fa99f362","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-04 06:49:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75f5598f002f39dfab337b628227124af38a9b3b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-04 07:05:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7fb84aaac2a15e243609ed30e87db5a1e2911d67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-04 07:39:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"41f89b413920c9e9b4a73b9e1d012b01969a81a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 09:46:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae79d11422fd40e2bd1662b6b70b436f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0af80f60ab3d499b825a34ad5ca772d7 : SUCCESS in 8m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/415b9e129451457698d49e619d4f1013 : SUCCESS in 6m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a5a1dbfe332450eadbe697033899c56 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ddef835334c43feb238191d63e9b751 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3314bca365744f03853dac4a72cf0225 : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17dde64592904096b0c660be7afc351f : SUCCESS in 2m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1db476075c04407d9a782dab9bdf3886 : SUCCESS in 8m 05s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d6aadb8c4a440e1920550d6b8126c2c : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fa6990f722764d5c8037084cd152ab6a : SUCCESS in 9m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/09046c617ec64ba499141e891ac805ee : SUCCESS in 10m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a298cba79b5947f686a1fdeb0d721d04 : SUCCESS in 1h 00m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/762b248afc5f4398a24ad8e5b1aac3d9 : SUCCESS in 55m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f198b26cb60e44339021f6811718926e : SUCCESS in 18m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2af30fcd138b4680860d227b6b4fdbfa : SUCCESS in 11m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e06ea5778cd426ca6a859448370cc73 : SUCCESS in 2h 04m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d67f07c014ea4ac0a17adf02252f2ea8 : SUCCESS in 15m 19s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3c7c9f8edff448fb80d62a6e1052f2a9 : SUCCESS in 15m 34s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c906c42c6fb4459ba889a7920ed126a : SUCCESS in 20m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c279f0ef22a84e17a35d1f8a77378891 : SUCCESS in 12m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da8bf169c5204e82ae3a084bd8ebb97c : SUCCESS in 12m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/75cc77d0a65741958a7dfd4a93449b86 : SUCCESS in 12m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29b6dc6f999d4b4ca848cda3458e78c0 : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":2},{"id":"118613c8c6c69d98685ec64fbfd4fae8b625fb71","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-05 12:26:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a40d82df134bde6b57c113ec3b80077538799d16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 12:29:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e3a54276317d210454580a08fbcf4c30564f3ffd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-05 14:21:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7335ca21d160c048173bde571acd12a09eb84250","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 14:35:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea8a6c86a8394a6095da731cbd0f0fc8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ea75cf9cb94386adff7857e1e33358 : SUCCESS in 9m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/685e5c5ea5be4cada80121b320dead14 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e35b2bd28ef24e5db0196cf477630a77 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75bc8c2a75d3456f8b80fcd1f4f78b5b : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e49e550878a54f4a80e42de22982e285 : SUCCESS in 5m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d67fe5960b5e490fb9ffa48655a24da2 : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/41c55abdd69a46e28d261cb2d28ddaca : SUCCESS in 6m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/149d0b2a20344069b31316ea0cd68482 : SUCCESS in 10m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10de61febcc2407ab19114dcab7705ee : SUCCESS in 30m 51s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/392d2b1cb18549c798d3cbc6577494c9 : SUCCESS in 55m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5ebeaa220dc9462bb1dd1b988fbafc0c : SUCCESS in 17m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8943e2ee81d4d4b95bcfef2fe383e92 : SUCCESS in 14m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de101b0a17ab43f8a5fad2e2a0ff0a48 : SUCCESS in 1h 53m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9329f33220d544ef9a05220dc03ca387 : SUCCESS in 8m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/11d11899017d4f6ab644c14cda29c1e1 : SUCCESS in 12m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb2364663eab4a5cbe9c5faebbb9956c : SUCCESS in 17m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f327dd0b7c33483681331f9ab3c1b4d2 : SUCCESS in 15m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ddc6c202c324f78a5afb962c92262e6 : SUCCESS in 11m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b3469c61c9644289bd3c8acaa5074a7f : SUCCESS in 13m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff1fbbd6c86b489a981aba00fcf537ef : SUCCESS in 11m 32s","accounts_in_message":[],"_revision_number":2},{"id":"52eb1daeab20fbc091acb47d7145bc7d6e5611f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 14:35:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"45d0caf8177ab3c04a8201a91ae9fcf598bd4a0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 14:37:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0649049dc2c64170884216842cadcf7f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/591c4c40c3694244b64f9ff8c49b189b : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"43236f19c1f782e38c7a4b9ef4944b6e07128296","revisions":{"eb67be6dc49f53d78c3b8df726e22ab610d20ab2":{"kind":"REWORK","_number":1,"created":"2025-09-04 06:49:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/13/959413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/13/959413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/959413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/959413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/959413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/13/959413/1"}}},"commit":{"parents":[{"commit":"8e977fa25acdcdd6b8ac25de505288f37c10453e","subject":"Switch infra VMs to EFI-compatible CentOS image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e977fa25acdcdd6b8ac25de505288f37c10453e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-04 06:49:36.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-04 06:49:36.000000000","tz":120},"subject":"CI: Add jobs for infra VM testing with cloud image","message":"CI: Add jobs for infra VM testing with cloud image\n\nChange-Id: I224a187372ff948634c5a039e284c4455ee24af2\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb67be6dc49f53d78c3b8df726e22ab610d20ab2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb67be6dc49f53d78c3b8df726e22ab610d20ab2"}]},"branch":"refs/heads/master"},"43236f19c1f782e38c7a4b9ef4944b6e07128296":{"kind":"REWORK","_number":2,"created":"2025-09-04 07:39:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/13/959413/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/13/959413/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/959413/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/959413/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/959413/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/13/959413/2"}}},"commit":{"parents":[{"commit":"8e977fa25acdcdd6b8ac25de505288f37c10453e","subject":"Switch infra VMs to EFI-compatible CentOS image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e977fa25acdcdd6b8ac25de505288f37c10453e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-04 07:39:07.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-04 07:39:07.000000000","tz":120},"subject":"CI: Add jobs for infra VM testing with cloud image","message":"CI: Add jobs for infra VM testing with cloud image\n\nChange-Id: I224a187372ff948634c5a039e284c4455ee24af2\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43236f19c1f782e38c7a4b9ef4944b6e07128296"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43236f19c1f782e38c7a4b9ef4944b6e07128296"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
