)]}'
{"id":"openstack%2Fkayobe~960153","triplet_id":"openstack%2Fkayobe~master~Ib92ae8807a5f22090d3025ff93d279004e686870","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-11-17 19:37:23.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-11-17 17:21:04.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"782":{"account":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"last_update":"2025-11-17 19:37:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib92ae8807a5f22090d3025ff93d279004e686870","subject":"Move nvidia.nvue to 1.2.9 for cumulus 5.13","status":"MERGED","created":"2025-09-08 16:39:43.000000000","updated":"2025-11-17 20:09:09.000000000","submitted":"2025-11-17 19:37:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960153","meta_rev_id":"f9a1a6e2d6b861032274624355fdc58cf4195d2d","_number":960153,"virtual_id_number":960153,"owner":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-17 19:37:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-11-17 15:57:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-11-17 17:21:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-17 17:28:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-17 17:28:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-11-17 15:57:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-11-17 15:57:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-08 16:54:03.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2025-09-08 18:24:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-08 19:12:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-17 15:57:19.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-11-17 17:28:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"72db8257048d870219ea82bac523b4395d53d823","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2025-09-08 16:39:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31a449acc590780405bcad28bf0a2c2357e93892","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-08 16:54:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ca86bafaeccce4a67a27a370df19fa3ef5073c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 18:24:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/337d19245e6d4c4cb11801c43cb97617\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40ba89bf895a4401a41bcfd2abeff08e : SUCCESS in 9m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b9217009282042de8fa8aa8cec118264 : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/802b4f1d573140eab465fff91412a880 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b58ad6faf7f64d1bad2b2d2f58bfa086 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c3c6b41aa194b93995ecfbfbf3fce36 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05baefeafa4e4e18b38b479a8d1a6608 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd27cedc478c4109b1db9829463aea45 : SUCCESS in 7m 24s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1a17bbe704bb4851b6492568784b3a2c : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/22e42fc1c450491aad5a3e497d2dde3c : SUCCESS in 8m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b57b1d581c304e0c98e00f10173cddc0 : SUCCESS in 9m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7167e1ad6224a05aa810e61e8cf8208 : SUCCESS in 56m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/17c5d3d08874400ab9e6ea7215e381ce : SUCCESS in 35m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/805009041bf34a74a948c0461876ed6c : SUCCESS in 17m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22cab155f7514e7c99b9cd28e0bc1098 : SUCCESS in 14m 36s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d568097705914e70aac4cef255a6d768 : SUCCESS in 1h 38m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d9dfba0d21f4196a3d45145972a4dd8 : SUCCESS in 13m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/36d5a07bb4df44c791ec12a916cfef26 : SUCCESS in 15m 17s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/123b0bd32a6a456fa614a489bf7876e5 : SUCCESS in 25m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8f539ba22d1849abbad5edd73c7f83c6 : SUCCESS in 7m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe8e30c005a14ece9bc4fd9af62718ee : SUCCESS in 6m 44s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/640e21dc62cb4771b0c2699081b83033 : SUCCESS in 12m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e79b08492f84b10927429c2c6849f09 : SUCCESS in 12m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a68b45bc1c6d9f6c66d369f3bdadf61d6c35e67f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-17 15:55:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ee8c58653af0aa74bec19abb56624cae9269dfac","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-17 15:57:19.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1a21659b1a6597752955b4b968f55987a2d1d877","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-17 17:21:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7ef8654428de7f204280d98196bad292f9ccd7ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 17:23:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7fd26d57ee5417c8c927a722f0176de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e48bbe8cac4845f8bb6857ca3a8fc659 : SUCCESS in 10m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/461e584d16ac40d891ac32e514442443 : SUCCESS in 7m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78911c7172ce442fb63259cdb596f3f2 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcc0f4a2bb6643b1a5756ef5c831eb82 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0da88064128944118489a3fed6f029c8 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/688cc8d4be0f49988a6425afa34c8d2b : SUCCESS in 2m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/283354aefc0343ca8eb1d06bd7975328 : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0408c1d28a0f429bbcf0f5de391cf388 : SUCCESS in 14m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9e174f60edc44cf8b4a8653607df6868 : SUCCESS in 6m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/12f58599e1ba4ba08d90ad14d6d81e18 : SUCCESS in 11m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be1b620381e34b7198e87fdae035651d : SUCCESS in 12m 34s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f5472d14ba6541b0b8dbc141dee20f87 : SUCCESS in 13m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60238dc59d2e4c259887671df82026cc : SUCCESS in 16m 47s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/57322f3e5bb9473facb7b774106a976b : SUCCESS in 33m 08s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/023d7e8b85cd46329ebfbcd24a25a3e1 : SUCCESS in 38m 23s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2ce9c2a12a394fcea5c64bda86d4676b : SUCCESS in 37m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc5a352bdcde474d986fb871f4a46acd : SUCCESS in 1h 15m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e8d5ffe879434bbb9692fe3af79c2572 : SUCCESS in 35m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/098fb08efdf044bfb65c03b78d63304a : SUCCESS in 58m 35s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9481ec3669ca481fb527c465c67e4935 : SUCCESS in 10m 20s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/24f392c6c94e4c3586502d1a254d49bd : SUCCESS in 13m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e2e82fa03e54d01a6f651ae911a3964 : SUCCESS in 8m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/89b59e7f3da94263b7d2ddd05f929745 : SUCCESS in 16m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/824166772d2d45bdaf137d48faaa78d1 : SUCCESS in 16m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c9843f540398414f800135a8149652d9 : SUCCESS in 14m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/162bced9abd049b2accf2c037d75f86d : SUCCESS in 14m 35s","accounts_in_message":[],"_revision_number":2},{"id":"523e6488549e5becd360bd3dc033d6c0e871857a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-17 17:28:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"96c6105844f33126285ed49134ae37a7601f8d73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 17:29:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"569c92dc36c2582f6a83c4fb92f8ec039854b041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 19:37:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5c150d07ff04c069377a14675e4f0d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e232aace9b4447489ff6245aaa17bf68 : SUCCESS in 9m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/635613a70dcc4142a1c7484da742d5d3 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ed1cf43837c42438425b1b5dc4e076a : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f946e02624f7421695d68304c7db9632 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2309c522186741f0b8602806196c15c6 : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/737cdb075566408e94713b035e9f6344 : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a4dfd2b76d55454cb80dc49bc6286694 : SUCCESS in 10m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f86f81a4b918400ebe321a4cfe7181c8 : SUCCESS in 12m 37s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d4e42de779764049a528ac70b80c181c : SUCCESS in 9m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e445fd1f745146198d7de297db1246e0 : SUCCESS in 12m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3ecdfeeebd624da2a10ff6cc3853d231 : SUCCESS in 18m 37s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35ab8a9cbc954fbc8b4a535a3d7c6718 : SUCCESS in 16m 10s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b3e0b02e0969408ab06e5afe3fc8ed37 : SUCCESS in 57m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/143481298e9841028e7036a074392c06 : SUCCESS in 51m 27s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7b3816f17ece4640b61cc95e03c19042 : SUCCESS in 1h 05m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3423f31846254488b77e9d33de2437a2 : SUCCESS in 1h 10m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a4470b0a6ac5467bb5c2f1c24b9f4e98 : SUCCESS in 1h 06m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7072a8708554306a6fcf1e1db6e4360 : SUCCESS in 2h 03m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3532ca6fbdd54aa4b48a9198911d2758 : SUCCESS in 15m 35s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/df91713d0c1b49338922887ff78b2475 : SUCCESS in 14m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed39fdd552284ec699eb6dcad5ff5021 : SUCCESS in 13m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/971b55dda6824da592fa76b7c5fde643 : SUCCESS in 9m 33s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b254262d2104088824772d2d4676649 : SUCCESS in 26m 05s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b2a485fca47e4e29a5ba643f68487008 : SUCCESS in 17m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4262eda9e9914c74b1fc57a040107c5c : SUCCESS in 12m 50s","accounts_in_message":[],"_revision_number":2},{"id":"b0b7f6c8016282d5ee320849a0ce736c0a40b4dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 19:37:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d5d88fe0d86cb851090dace43bebd39e093c0ec7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 19:38:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c55842a39ea43e0960207adf2bf6eae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/151fdeab6fc04aa0a039475bce6f7506 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52c41345df3e4f51a290babe65ea7245 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2},{"id":"f9a1a6e2d6b861032274624355fdc58cf4195d2d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-17 20:09:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"85a0356cbab068026ab03214a7dac7629dca294e","revisions":{"5604d5a61b98ca3081e0609447032222ec8ed142":{"kind":"REWORK","_number":1,"created":"2025-09-08 16:39:43.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/53/960153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/960153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/960153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/960153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/960153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/960153/1"}}},"commit":{"parents":[{"commit":"c34c595dbe8c66d76f749d91319c709aee9f1548","subject":"Merge \"Fix kolla_docker_namespace comment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c34c595dbe8c66d76f749d91319c709aee9f1548"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2025-09-08 16:36:54.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2025-09-08 16:39:41.000000000","tz":60},"subject":"Move nvidia.nvue to 1.2.9 for cumulus 5.13","message":"Move nvidia.nvue to 1.2.9 for cumulus 5.13\n\nChanged switch config started creating errors on cumulus 5.13\nonwards. Bumping the collection version to bring in this fix:\n\nhttps://gitlab.com/nvidia-networking/systems-engineering/nvue/-/commit/adfc6829ee0fadbcd5273a7635f26d6a79b44eab\n\nChange-Id: Ib92ae8807a5f22090d3025ff93d279004e686870\nSigned-off-by: John Garbutt \u003cjohn.garbutt@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5604d5a61b98ca3081e0609447032222ec8ed142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5604d5a61b98ca3081e0609447032222ec8ed142"}]},"branch":"refs/heads/master"},"85a0356cbab068026ab03214a7dac7629dca294e":{"kind":"REWORK","_number":2,"created":"2025-11-17 15:55:55.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/53/960153/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/960153/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/960153/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/960153/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/960153/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/960153/2"}}},"commit":{"parents":[{"commit":"c34c595dbe8c66d76f749d91319c709aee9f1548","subject":"Merge \"Fix kolla_docker_namespace comment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c34c595dbe8c66d76f749d91319c709aee9f1548"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2025-09-08 16:36:54.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-11-17 15:55:53.000000000","tz":0},"subject":"Move nvidia.nvue to 1.2.9 for cumulus 5.13","message":"Move nvidia.nvue to 1.2.9 for cumulus 5.13\n\nChanged switch config started creating errors on cumulus 5.13\nonwards. Bumping the collection version to bring in this fix:\n\nhttps://gitlab.com/nvidia-networking/systems-engineering/nvue/-/commit/adfc6829ee0fadbcd5273a7635f26d6a79b44eab\n\nChange-Id: Ib92ae8807a5f22090d3025ff93d279004e686870\nCloses-Bug: #2131677\nSigned-off-by: John Garbutt \u003cjohn.garbutt@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/85a0356cbab068026ab03214a7dac7629dca294e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/85a0356cbab068026ab03214a7dac7629dca294e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
