)]}'
{"id":"openstack%2Fkayobe~960451","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I05cf1d7fcc124651049608a588a2a6d917042bf7","project":"openstack/kayobe","branch":"stable/2025.1","hashtags":[],"change_id":"I05cf1d7fcc124651049608a588a2a6d917042bf7","subject":"[DNM] More Ironic debugging","status":"ABANDONED","created":"2025-09-11 10:31:15.000000000","updated":"2025-12-03 08:35:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"cherry_pick_of_change":960450,"cherry_pick_of_patch_set":2,"meta_rev_id":"3ffac2c89b198c1f2284705149a59d70b28a817c","_number":960451,"virtual_id_number":960451,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-09-11 12:37:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-09-11 10:31:15.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-09-11 10:31:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb0a75b65363fa5a23fbe9999bd3f3e9675da8c4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-11 10:31:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* zuul.d/project.yaml","accounts_in_message":[],"_revision_number":1},{"id":"5dd4d7fb21e33aa49776554d1e917e6e3470b086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 10:31:47.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch stable/2025.1.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/kayobe/zuul.d/project.yaml@stable/2025.1\", line 11, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/kayobe/zuul.d/project.yaml@stable/2025.1\", line 12, column 29","accounts_in_message":[],"_revision_number":1},{"id":"a0c9bd59fe00afbf7d66c48248499315b2be9b31","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-11 10:33:24.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"39dd144a0d8bde2f461088836dd40d9c829ee304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 12:37:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f9c967037c24dbc94d9874153ce60b9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59bf072e3f0a4d22b060b9303d40f1fd : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da720729b22c4fa085093c40f9985b7d : SUCCESS in 2m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1af7a77a8e04414f9da9aa415197a870 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c75e797b345a4660a12de1bbfb2de2ad : SUCCESS in 9m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7a1a054ca4b3448a8765cd1736f7bc8d : SUCCESS in 46m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45152bbbf64a4192b33de0f1784a7637 : SUCCESS in 2h 02m 23s","accounts_in_message":[],"_revision_number":2},{"id":"3ffac2c89b198c1f2284705149a59d70b28a817c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-03 08:35:27.000000000","message":"Abandoned\n\nNot needed anymore","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a6c86c0ff0105edf7ffb2a8a10f49ba8ee4d0dd1","revisions":{"76cd5cdbbf650cc2b132d57264e06ef5680339ea":{"kind":"REWORK","_number":1,"created":"2025-09-11 10:31:15.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/51/960451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/51/960451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/51/960451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/51/960451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/51/960451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/51/960451/1"}}},"commit":{"parents":[{"commit":"d20d88c5748d99ce459b4c320913d667ec127d06","subject":"Fix kolla_docker_namespace comment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d20d88c5748d99ce459b4c320913d667ec127d06"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-11 10:02:40.000000000","tz":120},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-11 10:31:15.000000000","tz":0},"subject":"[DNM] More Ironic debugging","message":"[DNM] More Ironic debugging\n\nChange-Id: I05cf1d7fcc124651049608a588a2a6d917042bf7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76cd5cdbbf650cc2b132d57264e06ef5680339ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76cd5cdbbf650cc2b132d57264e06ef5680339ea"}]},"branch":"refs/heads/stable/2025.1"},"a6c86c0ff0105edf7ffb2a8a10f49ba8ee4d0dd1":{"kind":"REWORK","_number":2,"created":"2025-09-11 10:33:24.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/51/960451/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/51/960451/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/51/960451/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/51/960451/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/51/960451/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/51/960451/2"}}},"commit":{"parents":[{"commit":"d20d88c5748d99ce459b4c320913d667ec127d06","subject":"Fix kolla_docker_namespace comment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d20d88c5748d99ce459b4c320913d667ec127d06"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-11 10:02:40.000000000","tz":120},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-11 10:33:24.000000000","tz":0},"subject":"[DNM] More Ironic debugging","message":"[DNM] More Ironic debugging\n\nChange-Id: I05cf1d7fcc124651049608a588a2a6d917042bf7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a6c86c0ff0105edf7ffb2a8a10f49ba8ee4d0dd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a6c86c0ff0105edf7ffb2a8a10f49ba8ee4d0dd1"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
