)]}'
{"id":"openstack%2Fkayobe~960884","triplet_id":"openstack%2Fkayobe~master~Ia91e5539f4338c8209fbe92739bce916c10b2b7a","project":"openstack/kayobe","branch":"master","topic":"fix-kolla-ansible-check-diff","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-09-19 20:58:36.000000000","reason":"Change was submitted"},"34212":{"account":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"last_update":"2025-09-19 20:58:36.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-09-19 20:58:36.000000000","reason":"Change was submitted"},"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2025-09-19 20:58:36.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-09-19 18:14:15.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-09-19 20:58:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia91e5539f4338c8209fbe92739bce916c10b2b7a","subject":"Explicity set --check and --diff in kolla-ansible","status":"MERGED","created":"2025-09-12 19:45:19.000000000","updated":"2025-09-19 20:59:40.000000000","submitted":"2025-09-19 20:58:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960884-fix-kolla-ansible-check-diff","meta_rev_id":"0926bc0793f5f31071e9bb9785b0fade82092e38","_number":960884,"virtual_id_number":960884,"owner":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-19 20:58:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-19 18:14:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-19 18:40:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-09-19 18:11:52.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":1,"date":"2025-09-19 18:11:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-09-19 18:11:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":2,"date":"2025-09-19 18:11:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-12 20:49:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-15 08:20:08.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-09-15 09:24:31.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-09-15 11:56:59.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"},{"updated":"2025-09-15 21:45:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-19 14:56:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"22d8989341dca957c6599da4259df20b409102a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-09-12 19:45:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77e2ff09dd63968051195fbfc14e6ad9b0ded475","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-09-12 19:52:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eafe8bcbda798198768fed892fe4d8de7316efce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 20:49:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec4197786c2147109d3bae23e993655e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8d49dd78e8e423f836b360127cd9c6b : SUCCESS in 4m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1f224b8b939e45ed80a362816788ff52 : SUCCESS in 3m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57b98d925a734e56bef107c12e73f16c : FAILURE in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dd66e11aa79459ba227fb08255f1d55 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/905d173ec31f4acfa20fb4850673d384 : FAILURE in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e7cc055e43f4aad87be75ad97caa061 : FAILURE in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ac6444829a74129a917255275e968fd : FAILURE in 8m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f11dcd58ac734f92ad18dd9620a3e227 : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1e96debb144b4486b65848df9e84b9d7 : SUCCESS in 5m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/30fbcafe2a314088953ef7b8a1e77bde : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e547b7ee5c8a4775a4d359efef4bc27e : SUCCESS in 5m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3f7ad2b772447a1969f81fb15632b1c : SUCCESS in 55m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/028cc72f5f084daea800c3da8ada69c4 : SUCCESS in 51m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/36ffcb4255314eddae39ecd2d7a20cde : SUCCESS in 19m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecdbb7ff0c8e4a35aa326b2a3911596c : SUCCESS in 7m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88a86bbe04514b259783afce0ae50d86 : SUCCESS in 48m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa733152352647908f1b9fefcd4dd1a6 : SUCCESS in 14m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/979bbc7193f5489cbab3ff7a4899c0d5 : SUCCESS in 13m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a16e117be07842a5949b0077c8aa9cbb : SUCCESS in 20m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fa183b97fdc34d539319f003e71688c6 : SUCCESS in 14m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c805926e2a645bcb373e1b01a0f3614 : SUCCESS in 12m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/02bc2790e92d4ecc87af54637f5b48b1 : SUCCESS in 11m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5439ca75bd8d4a4a8ccf24d23a4c3735 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":2},{"id":"cf8f5e79992c434634e978b8895c86f97f92dceb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-09-12 21:39:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3832dceae39295f75196fd26057023bf46c7ae39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 22:57:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a373a53e23924b52858106b322b9ca21\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8a9efa520c14894beeea65d8e559d13 : SUCCESS in 5m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4f35cf8b440445d48a4efd79f173f013 : SUCCESS in 3m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2b1454f662440e49c73ebf64196e195 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3d9999ec2a146b19a2e42cfac9e3900 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50631e4cb3d6402ebc25487781c51aee : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7ce8fd4d5ab7458aa58e49ff2b1a4d90 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/073af5b1db2344a4b323b56bc876bfd4 : SUCCESS in 6m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f7fbc7c4b704adca298bf3ec1b4e12d : SUCCESS in 2m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d871e4decedc4661900730786d317748 : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/111d997fef324b3bb944827985ea394e : SUCCESS in 14m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4a14c4fdff454327809981c490ccccc6 : SUCCESS in 6m 43s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69630fa8786a467ba615f58f491df851 : SUCCESS in 23m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5ffe6f2dd4bb40f78c7ed1d55084954b : SUCCESS in 58m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5aa235bc18b74e399a5a2e1ea3c37f86 : SUCCESS in 17m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/475e1ffd586446dfa75225d89596c42d : SUCCESS in 13m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/402b2457559f42f0a6615b53ac84c7f7 : SUCCESS in 1h 17m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/badbc163763345dd8c42228ac446d298 : SUCCESS in 14m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/00c4e8cf3187460786e7d030fcdfa344 : SUCCESS in 10m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73dacf0963f44dfba00964432b2759fd : SUCCESS in 24m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1da0c6d7374845d8a761fb7d10794d99 : SUCCESS in 14m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2d04e24051740dc8045edc09657770d : SUCCESS in 14m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ee15fe5106c64d87af13f9bdc10ece9a : SUCCESS in 13m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d79b29ef77c445589758e121894e1eb : SUCCESS in 9m 24s","accounts_in_message":[],"_revision_number":3},{"id":"83f51c7a2b5d62d6450c8618435c227d7f688441","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-15 08:20:08.000000000","message":"Patch Set 3: Code-Review+1 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6edf9d59ba905af98b8f751b4be5da103e9f903f","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-09-15 09:24:31.000000000","message":"Patch Set 3: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"87c10a6c6b6be73c742add831ba5ad0adcb7c622","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-09-15 11:53:36.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1, Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"78b3c9f33ef88c892fbcbeaa823023aa21658a62","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-09-15 11:54:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ecc022c22d52d57c49040796583815e8cc418fa0","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-15 11:56:59.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b675c5d6ccfb35819d0ea07aeb509781d4aebd32","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-09-15 11:57:17.000000000","message":"Patch Set 4: Backport-Candidate+2 Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"4930eb203e02e2671433861d9545eb5f57b14f28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 13:56:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0777eda668ca4be1824012c473169a66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9528751a8d94fe7aaaf088cbfc85fde : SUCCESS in 9m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7d36c462fa47431ba286f09ea82e60f5 : SUCCESS in 6m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a978a841e1b4458ba14e25c0db958c3 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1945e0d4e24b439f82b4a01e48079f3e : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fb4985701d14c4eb0f110eef7654509 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70d1091c2b454a79adb477c5e25d9a2d : SUCCESS in 6m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/06ad69875b474415827f98ce683f76cc : SUCCESS in 8m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42cc2e2b25af42cb9cb43a146d1120ad : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4235d1b0a9464cdfa0bb867946ee1520 : SUCCESS in 7m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7d3920061bd7443a97be6f57512f1b98 : SUCCESS in 12m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2424398cd7984e3aa06514a76c713e3b : SUCCESS in 10m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b283866c15f142e98b2a390cd279adc0 : SUCCESS in 43m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f6fa3f8ee2774e289811b08cf7fa765c : SUCCESS in 49m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/762a3188de3142c29490c265a24846ab : SUCCESS in 16m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88330bb8f28b4773b2f214a6eab6a715 : SUCCESS in 13m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f3fa2b22a0f4d73be19e154aedd9c77 : SUCCESS in 1h 38m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1843bbfe839d4083a01dfa3a96868d9b : SUCCESS in 14m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cbc35cef2ca44b12a6345bf1203de746 : SUCCESS in 15m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bcd08d79de442998ed56941ded0523a : SUCCESS in 32m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cb906f3eb80649d080b7809e36d3b4a4 : SUCCESS in 14m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1102938173d46b3ad1838bb8c8d8e3d : SUCCESS in 11m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a54aaf713ef547199b4e9f74ab7d3ebb : SUCCESS in 11m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/339653beff1141bd9184793d660a2435 : SUCCESS in 10m 54s","accounts_in_message":[],"_revision_number":4},{"id":"3d8223bfbc12ad1c17718f1ce4a74f39f557c9f1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-15 21:45:35.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6886965ec7dec8f294c41bb62ee9bc0b82b7140d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 13:53:19.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5920a7e03e38b6354f05b2ecfac02c43153d59f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 13:53:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ff888cc2ca94c9c5652ab8db0ed66e2711ba6ff0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-19 14:55:51.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Backport-Candidate+1, Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1, Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"1c0ff66ed60e29045d8c9603952d3d1f8b6bb128","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-19 14:56:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4571aaac8ae04ca73f651658dba65cb6a977acc9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-19 15:08:20.000000000","message":"Patch Set 5: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"bd6690f5ca110d9aecfa7a8d72e259ec60038779","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 17:02:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f8ad7daedcc4ca68032215845bd86fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a73e56686ec48138cfed9847f684c2f : SUCCESS in 10m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ae246342230b496bb5d12f29c0da307a : SUCCESS in 7m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aee7e2cc44fc45a4b05d8434fdf87354 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e35bd12022f449bae663e66840a2f06 : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11d35f4705ce4dc6b6aa15747ae94575 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27392cd01a7f44fb9c9720861feb16bc : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6818d3549cf4d529796c1ae9f2d1978 : SUCCESS in 6m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2ca194eaf2941219d01140ead233717 : SUCCESS in 3m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0a6caace41cc4a1b9f7ba8df65dda00a : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1311689ba7c34dd8b2a53cd2451ad383 : SUCCESS in 11m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8207b142a32345489574b8bc7b9e2404 : SUCCESS in 12m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af866fe647414f6e962f80c23f7b3d1b : SUCCESS in 58m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4b630caab6294d18b35d054ee2fd0ff9 : SUCCESS in 1h 00m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/286c13df391947f1bf507216f25f93ae : SUCCESS in 17m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46857209330943d68e2dba787bc2f095 : SUCCESS in 17m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23f754823fc648f9aafe4017367c7123 : SUCCESS in 1h 55m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c00880dae684d50894588679a2fcc10 : SUCCESS in 16m 51s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3d13a4dfe6ab4291821b534d3e85afc3 : SUCCESS in 15m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0abec75dae2413e96b01eeb9125e0e9 : SUCCESS in 16m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4196eb64d026450eb4173665ca9d5f97 : SUCCESS in 14m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eba2e7f17da2465aa73bbfdbc9e1aafa : SUCCESS in 11m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3f6939a696ce4e548f6a8d4417d28605 : SUCCESS in 12m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c94f0ab6ac74b039a2e97c79d6a0f80 : SUCCESS in 11m 23s","accounts_in_message":[],"_revision_number":5},{"id":"1ccadebb067988ddc644f0d092f7aca7a24ec04e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 17:02:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b36fd985249e8e078c11c8ff1f58f966e1508085","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-19 18:11:52.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Backport-Candidate+1, Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1, Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"ca3f13963af408dc1688dd67f48c943c6cf0f5eb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-19 18:14:15.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"c232ac4aee5e16cb5b8776b6a26d33df92208c59","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-19 18:40:22.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"93bf435ce1804490b4f65cb03a251f962da84f75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 19:49:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab0d96fd8b0c4b1b9fb0ecd05bea002e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f66ce35f6dd42ad9071a7f42807bca8 : SUCCESS in 5m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/903486baa7e149438dc52c4e6a151694 : SUCCESS in 4m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/915b9196a68f493e8eeb3dedcf2e69dc : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d1034a631e741c0b5b6491d080170b0 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/661bcb9b68604fbb9bcad0f2173b3cc8 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/782863bd70da4d538c9423f47bd3f6ea : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc23c7f3acaf45469c0aa53ed8758979 : SUCCESS in 7m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f260505196f417fbc3697ccab959609 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ea33fd0503744b05939926f6ee2ebeb9 : SUCCESS in 3m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3310b06c38634ec0894c725a14088f00 : SUCCESS in 10m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d42bd9bda3d049cba4a4de79ec0d7060 : SUCCESS in 9m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7af56ce86a8c41d1a747086a5392b37c : SUCCESS in 50m 42s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3b2402184b75474e8668ff88606169bf : SUCCESS in 42m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9286f99010ac49d49a43af743e0b8979 : SUCCESS in 21m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a80b656389744baca23e7ea5f10d2595 : SUCCESS in 13m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/084c062bcdb9473f87c7889d128add06 : SUCCESS in 1h 33m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79f82ee4e9af4d108374adeef5a3b9cd : SUCCESS in 17m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/192764559ae44178898bacb54eaabd04 : SUCCESS in 11m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3472ffecb1954522a090ff232ef6619a : SUCCESS in 19m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/db976f0072ae4ddbb702254d42e3b167 : SUCCESS in 13m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2071215d348b482594eba3e8306e562a : SUCCESS in 10m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a835679474c3441aaf2d00045b8597d3 : SUCCESS in 11m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82402dddb85b49baa77eb1e611a61604 : SUCCESS in 10m 23s","accounts_in_message":[],"_revision_number":6},{"id":"71fd13717fd004c3f3c2affdeb7b6e68623dd9d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 19:50:05.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"eb5386012bce979bbfcde9dd30094ce46fb1d8c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 20:58:36.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de4e655571fb434cbd74bf7ab47b1b37\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/befb8b60bfb542c3a9c1b53fb042b888 : SUCCESS in 8m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0efd7a52a2a6406b948274d16c11928c : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a3106500e8d4789b201f97cf4a3238b : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50784c619a0b4db2a036282731cf3a1c : SUCCESS in 5m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ebb2ee42e944995ba0bfd1ce77a693a : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4de54a8fb5f44efe81ed1803c34843a4 : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/da810b731a3b4f029522de058163fc0d : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b44c44922b5345d48c0640a73c6b7216 : SUCCESS in 12m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e5822ff1d4584a07a844cd4eefdb95be : SUCCESS in 10m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3efdf88f7cf84f85bc5d4ded0e413dd1 : SUCCESS in 48m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/98878630078d4be28bf000ae287b3fb9 : SUCCESS in 57m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f8a746790cd847ada829678db2e129c2 : SUCCESS in 17m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/078ea1d1c40d44e7859cd30bcec8114b : SUCCESS in 17m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1dbfd8383994ab988f1850e28ec8f70 : SUCCESS in 51m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf860dc3a39b4891be1329de91c0b4db : SUCCESS in 14m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e404622c619841fb8160666eb1aa778a : SUCCESS in 14m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac7475751f164d71b35a43e7c2c33a89 : SUCCESS in 26m 07s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f8cbea8984a544e89e319695250786de : SUCCESS in 14m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b59bb153b90942179d0ca181b352865c : SUCCESS in 11m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8c47710d8c9e45e69666383458861916 : SUCCESS in 12m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56fc29f937b443ffad416a8093cf09b5 : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":6},{"id":"5928f3be0205a9897881aafca8e94d2f58387591","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 20:58:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"0926bc0793f5f31071e9bb9785b0fade82092e38","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 20:59:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51f6e7a2c3fc477b9e70c7d5335e0a84\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85dd84882f254e6399528adcf4a2fb80 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7158b5d0788243eaa8da84432435de88 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f5a5ddb043248e09461996907dcc6312bd7f7dd0","revisions":{"88503bdba80f7790dedffe9e89a58913292351cc":{"kind":"REWORK","_number":1,"created":"2025-09-12 19:45:19.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/84/960884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/960884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/960884/1"}}},"commit":{"parents":[{"commit":"c136c23daee93960b6d09053b04b8e45f59d83e4","subject":"Bump MichaelRigart.interfaces to v1.15.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c136c23daee93960b6d09053b04b8e45f59d83e4"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-09-12 19:41:57.000000000","tz":60},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-09-12 19:44:39.000000000","tz":60},"subject":"Explicity set --check and --diff in kolla-ansible","message":"Explicity set --check and --diff in kolla-ansible\n\nKolla-Ansible no longer supports appending the content\nof the EXTRA_OPTS environment variable to the\nansible-playbook command. Kolla-Ansible now supports\n--check and --diff as explicit CLI arguments, so\nexplicitly pass them into kolla-ansible.\n\nChange-Id: Ia91e5539f4338c8209fbe92739bce916c10b2b7a\nSigned-off-by: Matt Anson \u003cmatta@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88503bdba80f7790dedffe9e89a58913292351cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88503bdba80f7790dedffe9e89a58913292351cc"}]},"branch":"refs/heads/master"},"f58d79aacb6d456863d0e1a1077258570dc36a6c":{"kind":"REWORK","_number":2,"created":"2025-09-12 19:52:04.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/84/960884/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/960884/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/960884/2"}}},"commit":{"parents":[{"commit":"c136c23daee93960b6d09053b04b8e45f59d83e4","subject":"Bump MichaelRigart.interfaces to v1.15.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c136c23daee93960b6d09053b04b8e45f59d83e4"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-09-12 19:41:57.000000000","tz":60},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-09-12 19:51:54.000000000","tz":60},"subject":"Explicity set --check and --diff in kolla-ansible","message":"Explicity set --check and --diff in kolla-ansible\n\nKolla-Ansible no longer supports appending the content\nof the EXTRA_OPTS environment variable to the\nansible-playbook command. Kolla-Ansible now supports\n--check and --diff as explicit CLI arguments, so\nexplicitly pass them into kolla-ansible.\n\nChange-Id: Ia91e5539f4338c8209fbe92739bce916c10b2b7a\nSigned-off-by: Matt Anson \u003cmatta@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f58d79aacb6d456863d0e1a1077258570dc36a6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f58d79aacb6d456863d0e1a1077258570dc36a6c"}]},"branch":"refs/heads/master"},"ac799cbf85a97ff2fae1c45c689af26b95500531":{"kind":"REWORK","_number":3,"created":"2025-09-12 21:39:04.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/84/960884/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/960884/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/960884/3"}}},"commit":{"parents":[{"commit":"c136c23daee93960b6d09053b04b8e45f59d83e4","subject":"Bump MichaelRigart.interfaces to v1.15.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c136c23daee93960b6d09053b04b8e45f59d83e4"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-09-12 19:41:57.000000000","tz":60},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-09-12 21:38:54.000000000","tz":60},"subject":"Explicity set --check and --diff in kolla-ansible","message":"Explicity set --check and --diff in kolla-ansible\n\nKolla-Ansible no longer supports appending the content\nof the EXTRA_OPTS environment variable to the\nansible-playbook command. Kolla-Ansible now supports\n--check and --diff as explicit CLI arguments, so\nexplicitly pass them into kolla-ansible.\n\nChange-Id: Ia91e5539f4338c8209fbe92739bce916c10b2b7a\nSigned-off-by: Matt Anson \u003cmatta@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac799cbf85a97ff2fae1c45c689af26b95500531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac799cbf85a97ff2fae1c45c689af26b95500531"}]},"branch":"refs/heads/master"},"dc5549b43376b50a306e36a9332faf8da667429a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-09-15 11:53:36.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/84/960884/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/960884/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/960884/4"}}},"commit":{"parents":[{"commit":"c136c23daee93960b6d09053b04b8e45f59d83e4","subject":"Bump MichaelRigart.interfaces to v1.15.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c136c23daee93960b6d09053b04b8e45f59d83e4"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-09-12 19:41:57.000000000","tz":60},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-09-15 11:53:03.000000000","tz":60},"subject":"Explicity set --check and --diff in kolla-ansible","message":"Explicity set --check and --diff in kolla-ansible\n\nKolla-Ansible no longer supports appending the content\nof the EXTRA_OPTS environment variable to the\nansible-playbook command. Kolla-Ansible now supports\n--check and --diff as explicit CLI arguments, so\nexplicitly pass them into kolla-ansible.\n\nCloses-Bug: #2123834\nChange-Id: Ia91e5539f4338c8209fbe92739bce916c10b2b7a\nSigned-off-by: Matt Anson \u003cmatta@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc5549b43376b50a306e36a9332faf8da667429a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc5549b43376b50a306e36a9332faf8da667429a"}]},"branch":"refs/heads/master"},"8975fb207929f30cdeb85cfd4102ab32ac0706fd":{"kind":"REWORK","_number":5,"created":"2025-09-19 14:55:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/960884/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/960884/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/960884/5"}}},"commit":{"parents":[{"commit":"44163326edf0ed33e801e9fd9644f38258225b1e","subject":"Fix disabling NetworkManager DNS management on Rocky Linux 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44163326edf0ed33e801e9fd9644f38258225b1e"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-09-12 19:41:57.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-19 14:55:31.000000000","tz":120},"subject":"Explicity set --check and --diff in kolla-ansible","message":"Explicity set --check and --diff in kolla-ansible\n\nKolla Ansible no longer supports appending the content of the EXTRA_OPTS\nenvironment variable to the ansible-playbook command. Kolla Ansible now\nsupports --check and --diff as explicit CLI arguments, so explicitly\npass them into kolla-ansible.\n\nCloses-Bug: #2123834\nChange-Id: Ia91e5539f4338c8209fbe92739bce916c10b2b7a\nSigned-off-by: Matt Anson \u003cmatta@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8975fb207929f30cdeb85cfd4102ab32ac0706fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8975fb207929f30cdeb85cfd4102ab32ac0706fd"}]},"branch":"refs/heads/master"},"f5a5ddb043248e09461996907dcc6312bd7f7dd0":{"kind":"REWORK","_number":6,"created":"2025-09-19 18:11:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/960884/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/960884/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/960884/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/960884/6"}}},"commit":{"parents":[{"commit":"d6273899b4c7c6660ef16199a1acbe2e4a85201c","subject":"Change default centos bootstrap user to cloud-user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6273899b4c7c6660ef16199a1acbe2e4a85201c"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-09-12 19:41:57.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-19 18:11:39.000000000","tz":120},"subject":"Explicity set --check and --diff in kolla-ansible","message":"Explicity set --check and --diff in kolla-ansible\n\nKolla Ansible no longer supports appending the content of the EXTRA_OPTS\nenvironment variable to the ansible-playbook command. Kolla Ansible now\nsupports --check and --diff as explicit CLI arguments, so explicitly\npass them into kolla-ansible.\n\nCloses-Bug: #2123834\nChange-Id: Ia91e5539f4338c8209fbe92739bce916c10b2b7a\nSigned-off-by: Matt Anson \u003cmatta@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5a5ddb043248e09461996907dcc6312bd7f7dd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5a5ddb043248e09461996907dcc6312bd7f7dd0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
