)]}'
{"id":"openstack%2Fkayobe~961271","triplet_id":"openstack%2Fkayobe~master~I9d9233ea576e5cf7ce4f51db12817af00400a1cb","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-11-26 20:49:07.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-11-26 14:47:13.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-11-26 20:49:07.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-11-26 20:49:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9d9233ea576e5cf7ce4f51db12817af00400a1cb","subject":"Test bare metal node inspection","status":"MERGED","created":"2025-09-15 17:34:21.000000000","updated":"2025-11-26 20:50:08.000000000","submitted":"2025-11-26 20:49:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"961271","meta_rev_id":"382038be43d840d5e22be1958644ae0eb3185e7a","_number":961271,"virtual_id_number":961271,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-26 20:49:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-11-26 14:47:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":2,"date":"2025-11-26 07:26:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-11-26 11:47:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-11-26 15:05:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-11-26 11:07:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-15 18:45:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-17 12:50:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-13 10:37:16.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-11-22 07:59:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-11-26 07:26:21.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-11-26 11:47:38.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"5c19083fb7d5353093aa86345bc8a4b33a87b02c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-15 17:34:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93ecbbf4aac833b7a9a2e4b6e046217b6f9bd1a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-15 17:48:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a9468bf07e0fbdd5978ec8368f6ce229fa485020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 18:45:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7a1306e5e7f461294e32fa114d05523\n\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a31f2f71dc84be79edce569355fe6a2 : FAILURE in 54m 20s","accounts_in_message":[],"_revision_number":2},{"id":"fe27102768fe09d3f6aba4f841c612ffb213096c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 14:27:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9a48b8420206b9b55c751d4a29aaee58ad198e70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 14:28:02.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"796332662fffd532ed0038a1e3dbdd3d495fe149","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 14:28:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f2d8d649b14fea96e2071ae7cde67d80a0005a48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 15:34:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21591c014f1148c38ac22297a193d5ee\n\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/955511b9d79149cda326653a7502944b : FAILURE in 48m 05s","accounts_in_message":[],"_revision_number":4},{"id":"156594b0332f6f8a509b8407b955b75039c2e97a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 16:05:30.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"9a6134f3c0e7c87b4bc143e5faffe4bab34015bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 16:06:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6af258705b3c579163307cd1f813ec0447a2bbfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 17:00:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/755f67ba1769498ca6485b23c4a3db3b\n\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/797bde76dcfd42df85794cc8e3cba93b : FAILURE in 48m 15s","accounts_in_message":[],"_revision_number":6},{"id":"76b046bb49fe9ded879562ec9835e3d4ae2a26b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 17:19:23.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"cc2ab8162de3ad36e3ffdb3aca039cbed96f2a5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 17:40:27.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"e39ea501a6e6b822c7a36001158dc2954477215d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 18:01:49.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"e6ef2b4ef4d4b8a1de2b4aa9ad546313efe52026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 19:03:25.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b4f0a72a6eb49d1b65cd2b1fd58b8c8\n\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4ec3be7884446e58dbc8b315d992d35 : FAILURE in 56m 08s","accounts_in_message":[],"_revision_number":9},{"id":"98c98ab23f8353c93bd746bdabd3e1c60874cb40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-10 17:07:35.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"4e8aaa19a4958a4805d18589d7c97ab6428e22f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 17:07:47.000000000","message":"Patch Set 10: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kayobe for 961271,10","accounts_in_message":[],"_revision_number":10},{"id":"47d9c4698e3562823bf442f789e732afb3836fa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-10 17:23:19.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"3e580e0239ea58f94be716d731791eb3c77401b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 18:18:15.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d39972ac3e34da1a9b316a953ea1a6d\n\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/564e7efaaa0f4e7c8478a89f6ecc3a87 : FAILURE in 50m 08s","accounts_in_message":[],"_revision_number":11},{"id":"7ac8877e3db6c001adfe965d3e625b416ca8fc2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-13 17:40:08.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"ceb58cdf4da1ef1a11e3ae2f10a85db45dc2d270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 18:27:53.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/280f4d9386e540d2b4d2c27a7781a7d8\n\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e7a002fc0e9479f9d69913733236572 : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":12},{"id":"e2f6c78de14c7707a4793b3a7ec01a1dbb0c74a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-17 08:42:19.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"fe7643f114a75e11dcb1faafe6907c8559b9b61e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-17 09:07:25.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"d97b7db4ad502a8997c1d845ec5cbc5ac590916f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-17 09:15:54.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"6281588f4f62956bd2fc4be13c3b1eb56a2f83ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 10:28:22.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89c030753d384f1ab8897efb21636bc9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84ce8b11070e4579a3581903e9188604 : SUCCESS in 5m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7335c694622549b49bfc53ef70887ad3 : SUCCESS in 6m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3203be2b46b44dabbcb186f582e20ee9 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/671f3beed1d34a4fb27261bf4eee9ad4 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33248efcbc704d0fab88d64b2dbd86d4 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/087e8dd1494c481eb4232746432fff52 : SUCCESS in 8m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b465a2aa2f1743eeb92d5667eccae737 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c6876f87653c4b8e908124df71cc19b9 : SUCCESS in 12m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ab78e6d640fb4835aff19a9beb2177ef : SUCCESS in 10m 39s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b9d83b498bd04908b7f0398ab97a27c8 : SUCCESS in 16m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a762109511024e59842754214c461073 : SUCCESS in 15m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b593f920f4ed464381fffbc28b774f09 : SUCCESS in 6m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/483eebe139ce429b86f3cf832f5e8914 : SUCCESS in 22m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c036ec06ccc2455ea12166015d13a589 : SUCCESS in 11m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62ea2dad417c424886cb598273d6ef33 : SUCCESS in 19m 44s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ad81802d0de41408cf1bcf922701d84 : SUCCESS in 32m 57s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/12023e2324634cf1a95ec52df4d51ac6 : SUCCESS in 58m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54cc58a79f7d4e3aaf0c630a5b25c662 : SUCCESS in 55m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8aaef9e43bc141b49cc5911579060eed : SUCCESS in 8m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/10e871c54c4f4382a70279c93e37223d : SUCCESS in 7m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8b896f1f12e4636a34ab610af086c65 : SUCCESS in 27m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ea1658b9a8f446ea9e041043025a8377 : SUCCESS in 15m 24s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/66ff9ce0f5a54c2b860d176eae5bf2f2 : SUCCESS in 8m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b78172784c744662ac5330ca9501bb89 : SUCCESS in 11m 23s","accounts_in_message":[],"_revision_number":15},{"id":"e91e0e15cbc5a87658b070da31d32b60eb50466b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-17 12:50:13.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"30b2dd93134f4daf134b7fb5c3bb0945c291d4e0","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-17 17:58:36.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":15},{"id":"8f7f601d09cf5784cdb07ab450cbc6f1f4a525d3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-17 17:59:33.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"555681f77b67f07410fc32e75775938dcfa4a02c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-17 18:03:54.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"43cb45f4368a2c756eed741f34c61a2d7d877d17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 20:04:45.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/150f71207cef4a0a98373ff53732f746\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/107bddade34840a1afe1e634c5329951 : SUCCESS in 9m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7962aab797de4837a00d8fbe429c33d9 : SUCCESS in 6m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b15b83fe2f446c5bd43ea17fa2ca2cb : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fe3204f651b457b9e0180ff2ba79015 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2c4dbc702504c208a37f9035cf8f530 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae20ba8f1376486a9dc17a30ea57f4a0 : SUCCESS in 6m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f73da8272569419a87cbda3dfa604e68 : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/900ca8023e124c4694f466453437ce83 : SUCCESS in 10m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/38d666f1166047149cdb4c1f89451581 : SUCCESS in 5m 46s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d7a217efb14b40c1a31b536015b45e2e : SUCCESS in 14m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e844e320612c4d2dafe690b82d7ae7d4 : SUCCESS in 13m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c514ff9d45e646a1b2620b33a45bea45 : SUCCESS in 11m 19s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/209a3840edc64ed89c4179455d3ad2e4 : SUCCESS in 9m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/159077e2910a44a1a1c17272a6b15c54 : SUCCESS in 14m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23be2ac3990944b29e4f9e109b38d2fb : SUCCESS in 11m 03s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/667c1c6337b040e69817ec5e2c441709 : SUCCESS in 49m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b90f47c18a9946a18d71d31184615691 : SUCCESS in 35m 49s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/19b489441e654f278eaf678deec56010 : SUCCESS in 38m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53fe24f5b98f485694f16e8a0b80eccc : SUCCESS in 1h 04m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/eb678c945eb44eb481e0fb12daf9e8d7 : SUCCESS in 1h 07m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83d764f546b9411e995b6954231ea769 : SUCCESS in 1h 58m 25s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b072afdef14f45d0893e5e89da92a33a : SUCCESS in 10m 03s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a5df669abeea41778f21cf7672f795a9 : SUCCESS in 13m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57035b2c535f45b08a807ab8a6103465 : SUCCESS in 16m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c36459e6036e4b96965f512910cfd1ae : SUCCESS in 17m 03s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65ef816ac8a444c8860b4409aca6df5d : SUCCESS in 29m 07s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e364d746eef8492c8b978b0d0de53dc9 : SUCCESS in 13m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/221def05434340079c9df0aae16c34a6 : SUCCESS in 14m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/226d4611641f4699b91b0d9ba3211f4b : SUCCESS in 12m 38s","accounts_in_message":[],"_revision_number":16},{"id":"8540060f4b87b00d8c0dd9bef975e116ab0fce2d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-24 09:32:03.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"0fd6e29a16d04eac7917ba8c6fdb81c1a0c18b29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 11:32:12.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d14469c593b49b28341a22780672317\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53ab5a1ae36e4c03b0b96d5f7f6a02cd : SUCCESS in 9m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1547388b354d4213bf7dee695e127d26 : SUCCESS in 3m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f9c09e78470470bbcc64213cb783ff6 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c1aa54347464230b61410f02b06e879 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7c870af0271455293e9ef5292d9299e : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60f23ffdbcf84a1f87d4bf95e646ec53 : SUCCESS in 5m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1d2c10868ef04c9ca46c71ee56c49bfd : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8370b401e00e4b609376dbf2ba643799 : SUCCESS in 10m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/067ba04b9ccc40099b368e423a7c4289 : SUCCESS in 5m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/30cebeefaf284e70931263a1bfbba5ba : SUCCESS in 16m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/aeeda97aa9af4132812f2636ac727bee : SUCCESS in 14m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/749a899e072b4868a500b781e97f4ca2 : SUCCESS in 12m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b928dc7e0fb9447d8d52d84e9bba46f9 : SUCCESS in 13m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a1dff761fabf4981a4191ad755aa92c4 : SUCCESS in 10m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77fd5bb4f8214eb5a9155ddf11b403f9 : SUCCESS in 10m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7bd512f34a264fc3829bb60faa268eba : SUCCESS in 1h 14m 06s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9c0594f5788c4ebb924eddefcf44e6bc : SUCCESS in 1h 17m 44s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1fac184acaae4bd688e979cb418bc8df : SUCCESS in 1h 08m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f71cdae4edff4eb08b2fd20d7cb5cc31 : FAILURE in 1h 38m 47s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/34de109a3673487a84e7f179b3a4e5fe : SUCCESS in 1h 06m 58s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6caa60569e4e4a089f9a1e9d7a30227b : FAILURE in 1h 14m 44s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a9267a6cd4a74a5097507e6f1b9be25a : SUCCESS in 18m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e345cd4d4a3e4ba0ad528f227fa50abe : SUCCESS in 15m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7091bcb64a064e42865b365798631384 : SUCCESS in 16m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/197d35aab38049779d5f94d92679fb5a : SUCCESS in 17m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f09b3863a3d4f1ca2c6fd6071216114 : SUCCESS in 21m 00s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ce7f65336c944a00ab1eb8a0d35b263c : SUCCESS in 7m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/73e9451b582d4f35b0cf09edf3594438 : SUCCESS in 16m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b899740077d44833a8442e962d96be50 : SUCCESS in 12m 45s","accounts_in_message":[],"_revision_number":17},{"id":"03b66000078782a5125c9c9a4e7140d0937b3fcc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-28 17:38:39.000000000","message":"Patch Set 18: Patch Set 17 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"26079f73cd24acda502a76c9afdf7a64be5288f3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-28 17:38:42.000000000","message":"Patch Set 18:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":18},{"id":"8c0791dc8d88900271d47f23e3b168324ba614c1","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-28 18:03:00.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":18},{"id":"efaf9c6d135721d807ba61c669102ef9e093faec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-28 18:13:35.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"14febf750719c5d456a53d1e88e6806e10886a30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 20:05:09.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/31fb88f1723a4e9c8c3600eb414c6716\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f2999713866493880663580edd97623 : SUCCESS in 8m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1d91cca5d2844a4fbf5dea7360b0efb6 : FAILURE in 4m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e388a17f4a4489aa8afe42de14d5105 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db4cb8834d1f40999ba5b2cbecda89b7 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62e6907a698246d0a795df6b8bb33538 : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/484470d3612643e48b5c2df50382ede7 : SUCCESS in 6m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f34dc12268b84d0d8cea3039d9368714 : SUCCESS in 3m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4aa8a185f2be4d7a8a9300e104689b45 : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e91ace74432142d1a57f2bb0cb0779dd : SUCCESS in 10m 18s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3691280ac3e04d23b8e0981a2c1ed91e : SUCCESS in 8m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d27b36fd7cee44e3890064b6993bb0f2 : SUCCESS in 12m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e221a115acda47abbde25156c5e19486 : SUCCESS in 12m 19s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/37b885cbcd1f4baf87265da5a101c604 : SUCCESS in 11m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/365ebceec5ab4e0d82f135d4b86dd194 : SUCCESS in 18m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd6395c9b8454d08843195acbaa86855 : SUCCESS in 14m 06s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a827be6247d145728bb797e908b1f1dd : SUCCESS in 34m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5c1ba2efe62f4bd3b2c21c60c055d791 : SUCCESS in 36m 05s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/68ff1b66cd1d4cfa953f606862143360 : SUCCESS in 1h 21m 50s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9b957d7e1ba45a8925b6febed5aa36f : SUCCESS in 1h 14m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/47365317dd744b42842fe43ef4d904cd : SUCCESS in 1h 04m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b423857b1a0423bb56bc12d36a6f8ee : SUCCESS in 1h 46m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5534195e09cd45979e349c44f317c0f3 : SUCCESS in 18m 57s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/821219100fe54ae28819ee956ba34838 : SUCCESS in 17m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0de97cf230f44e21864b0777cfc460d8 : SUCCESS in 17m 40s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f835a05015f54d0aa83f4bb9ca4401d6 : SUCCESS in 14m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f95c6b95e4c94907ab666efb7c5ebca3 : SUCCESS in 32m 43s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d6f40567732f4e55a93ea74d4d559680 : SUCCESS in 15m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2d13af6c136d4110b76378cf1b04b705 : SUCCESS in 16m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9122966a865a4ddb8dae165ab2f7c71a : SUCCESS in 12m 44s","accounts_in_message":[],"_revision_number":19},{"id":"9e4814f89c2dac3da678c78bca7e90d6c8576da5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-31 09:38:05.000000000","message":"Patch Set 19:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":19},{"id":"c45b800d299e97506fd8f5fb9aeec8fdfab9c4d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 11:04:37.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7dfecf1340514bed846416e6f94be49a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f3f8c6e6c824f49aea926f87eca1b1b : SUCCESS in 9m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e03d4a86f5904acbbd3c7ffc0449e3cb : SUCCESS in 5m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af9378d9df3f4830bb56277561aea4b8 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/906fe7dbea8b4f139df082b9acd2496b : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f97654383b13401e913cfe067dbe4b6a : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/00b9a1e5fc214033834b238da9a53a40 : SUCCESS in 5m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e4b76d1911e64daab0b75ca1e38709d0 : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/db2cd2b4a32940f1a3af9dafeab5a355 : SUCCESS in 11m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/82404173bbea48a6b8a50e3689128b40 : SUCCESS in 10m 25s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a7cb4e7daa264d24a50c7217c5b8e543 : SUCCESS in 15m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e5fddf36aa794b8d9698c96375e79f45 : SUCCESS in 6m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93b048a05d964bbe8b974a609fea538c : SUCCESS in 11m 30s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/06f8c815890f46aaafa1eee684119083 : SUCCESS in 9m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6f368ed3cd4749029aed00a6eda79d5d : SUCCESS in 21m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5eb5fb4d95a24161976e37135365a7a0 : SUCCESS in 7m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b3c26d2b15df4d6fb5b80470ab865826 : SUCCESS in 57m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/eae9654532a549f6b4240ee73cebbf2a : SUCCESS in 1h 04m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fd05f8ef693e41ddbe9a373c3d877551 : SUCCESS in 1h 22m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2039e75f2a6c439cb8a06bd7047103b8 : SUCCESS in 1h 10m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b43872a6aa014f03b1e37b4bf447f8d6 : SUCCESS in 1h 07m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87b66b5e9ea64bb085cad2c56766131a : FAILURE in 1h 17m 41s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a07be79a09f2467491713f20c687e049 : SUCCESS in 13m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3218af4b185c4a438ec2254305c51d61 : SUCCESS in 14m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/610db4c7525e4a4fb8caad915437d88c : SUCCESS in 18m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2f19d993867b4f32aeb69bb8fefc2510 : SUCCESS in 13m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/554443d0746449488aa9432ff109d210 : SUCCESS in 27m 52s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/71d33b3e9f0741648d52ee901c9e1261 : SUCCESS in 14m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/80e16a1d60f74e919d423a05f80edc39 : SUCCESS in 15m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63e7c35b2ab34911babfd21b35b9ffd1 : SUCCESS in 12m 27s","accounts_in_message":[],"_revision_number":19},{"id":"75bd567897afc3be5a89fd0f7405788da395be3c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-03 09:50:49.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"b548c9b087f18dafa8fdce780cd8c6ed81754bd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 11:43:10.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e35b05b93e874fdf930fd3dd88fb1b5c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f203f9b20512456c8dc499d4664a57ed : SUCCESS in 8m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a652c0a922df4e2db0c230509c702c20 : SUCCESS in 7m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b652ed102bbe4642a50d7f675c9f4fd2 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4dfc8fd31d14e05938843993e2574b9 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd7ec4055c97457199605bbea8011e59 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a284d795a73453eb2810710ea08feb9 : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2db5825eeef34406a1d1bf85ba4873cb : SUCCESS in 5m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ced3dca011164bafb8dc6d9d22f09b0e : SUCCESS in 12m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7c06a391ee9347009f622354072a0f94 : SUCCESS in 9m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8c5ca6ac8df2477ea33661fa59489bb5 : SUCCESS in 11m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3ac6b04da0ee4c1a931fe3e4768726a7 : SUCCESS in 6m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be31e6ff451144559a41a18e4708a130 : SUCCESS in 12m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d460bb2555634c609b1a708ccde89954 : SUCCESS in 9m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/96f47684fa104340980bd13b3585fddd : SUCCESS in 18m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb1576e88e89427e90b8e3707c059c89 : SUCCESS in 7m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/23797e1f3f6d488a90331f265d3490c2 : SUCCESS in 1h 13m 14s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b58a936c28914afe8257974b2080d31f : SUCCESS in 37m 31s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0e183662715e4a04a8803c1759555329 : SUCCESS in 55m 53s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50c6bedf202b4b38a3a6a1ae0da77c35 : SUCCESS in 30m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e20b690da1274263a51d4f3ec1611ef3 : SUCCESS in 1h 16m 09s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bce85682bfe042efbe614846b074d7f4 : SUCCESS in 1h 46m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a04703e7e84d4c9e84ec553c8c875fec : SUCCESS in 18m 59s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/7854cb2d4dab4b99bbe4d0e0cdf484b4 : SUCCESS in 16m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3d47c780ca248e0ad95d0631ec39a30 : SUCCESS in 8m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f04e4a74721f421fafdf29550995ab4b : SUCCESS in 14m 33s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7367296ec59f401399c5c3a97eede70f : SUCCESS in 29m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7c34fef46a8349d484cb84265c7d790d : SUCCESS in 8m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8810bcf62ef9456aac53ce94ff025a97 : SUCCESS in 11m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a9741a7e81d4d549baa6970acc97feb : SUCCESS in 9m 27s","accounts_in_message":[],"_revision_number":19},{"id":"cf280073b499ba9651ef4ebac3f952e9f4d3429e","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-13 10:37:16.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"55347a502f53d0e52466ba6e4e086f2c45d69a49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-14 08:07:38.000000000","message":"Patch Set 20: Patch Set 19 was rebased on behalf of \u003cGERRIT_ACCOUNT_28048\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"_revision_number":20},{"id":"3e47d73de432635e7af93a86abd55d29675c4810","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-14 08:25:33.000000000","message":"Patch Set 21: Patch Set 20 was rebased on behalf of \u003cGERRIT_ACCOUNT_28048\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"_revision_number":21},{"id":"0c5c35387af1cc18b0ed7c3804e6773aad52832e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 10:15:28.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e79225d61274b0d9d3575db07a8bb20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68aaf934150a4bafadc16ae3804697ed : SUCCESS in 9m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4a9c843c7607419f993ca168d4fce003 : FAILURE in 4m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6673351eb4554003ac97f77dd2eebe73 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39e53c62d75f4633a38a7992d6eb336f : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c2f5b5ff84b4be3b2d52b5a9c8033f8 : SUCCESS in 7m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f123356c195048ceb04b21e079f7bec0 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1f5275102dc94210a9922998e592eb25 : SUCCESS in 12m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cd2f051ff3024aaf95cd124136e9bd0c : FAILURE in 8m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fc9e09df0deb4f4b90068b62ac351071 : SUCCESS in 16m 18s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b04d0c9925f44cd893a223b323a8697 : SUCCESS in 7m 20s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7e7ebff2df944a7ba581af7f1b920441 : SUCCESS in 15m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2606fdc7900c49b19a97431159d20d5e : SUCCESS in 14m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1b767e9276764339ad4a8f9b7dacc4ea : FAILURE in 10m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6830ce2603f04886aba8f1493cbbdc09 : FAILURE in 6m 46s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/73a6aa5d1df948218987c43cd3ef040a : FAILURE in 4m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60e2e6f3e7f44c89a3dfdbfb5c8731ae : FAILURE in 12m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c9d880fbaf8b47feb8b38c59c282a927 : FAILURE in 10m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10d65d7b6e1b490db6cfb000232c8b92 : SUCCESS in 1h 43m 47s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e25e3654b1db44ab8f029ffdf65c43b8 : FAILURE in 10m 43s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4d1c7c29501d4ce881e46a7b3f253428 : FAILURE in 14m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a89a5ead7654bcf8e011e2fe985fa75 : FAILURE in 17m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/53f832185f24438898361e1abb1127ad : FAILURE in 14m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24e37f4da85e49c293600ae2d223b82c : SUCCESS in 26m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d2cdce0a7b3a4ebaafd93acdf0f39e9e : SUCCESS in 17m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bdef9090ae54e66932dc9f8789df907 : SUCCESS in 12m 43s","accounts_in_message":[],"_revision_number":21},{"id":"e8dfd1da89486330633988f879a209ef85056247","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-14 11:54:43.000000000","message":"Patch Set 22: Patch Set 21 was rebased on behalf of \u003cGERRIT_ACCOUNT_28048\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"_revision_number":22},{"id":"c64152cb7e791a56407aace8b0f5c3e7cd0351b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 13:17:24.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28a4b924d7d44dfcb029e7c838a0390c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28569e914772484b92041eaec798e3ae : SUCCESS in 4m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b9904be81b924575b3d34450fca6ae4b : SUCCESS in 8m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5dc3ea8ac194ab7927b2df4d6f2edfb : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6123a5401f2450db240324a95b82a10 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b16e58e0e7e4ee2b0f4be026686cc6e : SUCCESS in 3m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/988142a65adb420fa316dbb3596f4dbc : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cc1535eb169141ef84ce037233276423 : SUCCESS in 10m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3f16f3f2a5a546abb0acd58bf85b40ea : SUCCESS in 11m 26s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/db3ae4f546cd4ba58fdc9530f519842c : SUCCESS in 15m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10c432076bc24fc590d91af02fe2b3aa : SUCCESS in 16m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5f6be2cbdbae43eaac2e62bfa3e202f7 : SUCCESS in 10m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/627d7417ad114387a482948a8f2aa1d3 : SUCCESS in 9m 25s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c3795a47459c458996d5e475d2690ba2 : SUCCESS in 43m 46s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9fa96c2f44db4f5ab75a3c0e3eecb397 : SUCCESS in 1h 20m 25s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ef19417054bc4395a91a7a7dbc5a4144 : SUCCESS in 38m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d736657754e429782adee67fd6aece4 : SUCCESS in 32m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/21d19888186d434dbda2b6cfa42f1b9f : SUCCESS in 40m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f13d88a0b6548b0a2299727505f52ec : SUCCESS in 55m 41s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9ad0133fbfb64a33aedd85a0be9f6458 : SUCCESS in 9m 18s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a17cee0f8afd436fa1bdc366368ed935 : SUCCESS in 11m 34s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95016d430c304498969e81381e790a2f : SUCCESS in 7m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/460076f2611744aaaafd7fc3bed52368 : SUCCESS in 9m 18s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/220a91dfc681484e939996c0ac7e37cc : FAILURE in 14m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bbca724f0bbc49858f122b6036821fc8 : SUCCESS in 16m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb97ded451874aadb3dfe4841a205edf : SUCCESS in 12m 53s","accounts_in_message":[],"_revision_number":22},{"id":"a780ba2dd727736db06383ec45a1c61ad9412bb0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-17 11:21:15.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"0a0f84f8dc88598aa6b10235d86e6319973666a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 12:15:24.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73fc6771cd944ea681ed4a8775ca89cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f36b993066654f23bbe9e445db61deb6 : SUCCESS in 6m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bf2efc1e84f646459bff8d77ec104d9b : SUCCESS in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7bf241433d94d86841357a6aca8a56e : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20d9d6c6a64647a5ac0b7d5221e99469 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a94470bc24eb4ad1b6ae8806e5a30885 : SUCCESS in 3m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a38238d00d8c4c7fb95f11985b17c667 : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2d41d9ef7b4245db83404b71a735d61a : SUCCESS in 12m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a9f3cec98a1d4f82a4e46d6c1d0b9b4b : SUCCESS in 9m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5823777e2dde4fa08156b31366b107c1 : SUCCESS in 9m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02bfd085f0db44c58ff014922dc9ebd1 : SUCCESS in 12m 29s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fec1036b1dc3440b8fa5ddcff92ed0a5 : SUCCESS in 12m 54s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62f3c97f05fb456eb33f4db39b917109 : SUCCESS in 17m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/22c71f6f586c4dc1abf023645418a00e : SUCCESS in 37m 01s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/22ff1fc0315241d7b501cf22c3e9f491 : SUCCESS in 41m 06s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e95816838f9845dc8d146aa8f05f3064 : SUCCESS in 43m 22s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2164ba8af62e4fc8a200fad1a4ea6ce8 : SUCCESS in 32m 24s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d60ab59a9dbc4e9aa500eb18b78df240 : SUCCESS in 39m 14s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a392efa1e53c4b8bbc5f29c887e8812d : SUCCESS in 52m 49s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cfef0f0cf41a40958a0e2e4f937e1dff : SUCCESS in 11m 08s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4b5bb6ff269a47469b6193e54d92b992 : SUCCESS in 10m 52s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d2a822e6c8047ea9c979024ce250546 : SUCCESS in 19m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e88d71487df44ae0ab8a4715ff8d2b3d : SUCCESS in 15m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2822b26361314fa79492d362d886277f : SUCCESS in 29m 10s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/452f05a231404ad8a98f0025bd2743df : SUCCESS in 8m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee59604a74f943aaad4ee4afcdfab662 : SUCCESS in 15m 34s","accounts_in_message":[],"_revision_number":22},{"id":"11286e9ba0966fd6f1f164e412c27b965ca2b938","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-17 12:22:13.000000000","message":"Patch Set 22: Review-Priority+1","accounts_in_message":[],"_revision_number":22},{"id":"b200923e3625f49005bdaa39836633a1c2584efc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-17 12:37:21.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"24238dc2d93d6fd5a907e226cc811ae4c40265a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-21 09:54:44.000000000","message":"Patch Set 23: Patch Set 22 was rebased on behalf of \u003cGERRIT_ACCOUNT_28048\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"_revision_number":23},{"id":"d81afa7b93ab10947a9747c10513112f87da5ba3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-21 10:44:30.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":24},{"id":"39508ab8a7b08302982a1517ecd50fbadd6a732e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-21 12:28:05.000000000","message":"Uploaded patch set 25: New patch set was added with same tree, parent tree, and commit message as Patch Set 24.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":25},{"id":"54908b2dd5a9c9edf3a734db4d240d7c052de5ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 13:53:38.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e218ac4d65af4492a1dcf2e13a704bc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc91b826f5df4be6a8426ca6214cc0e4 : SUCCESS in 5m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/33c6c218c116425b80b0d3160ce07c4a : SUCCESS in 4m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c27cf1f89144711acd7e18d800fcc75 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e25d56eca9945e281869007ece1d629 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c632d4a09f5c4a3e8e940d925864ab78 : SUCCESS in 7m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1970f0325be34dcb9b7059bc7bfd6d4f : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/db4f218e01ef442982dc50c5b7d0df8e : SUCCESS in 7m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e14ad83fa3a846dab4d0c2571fa3f155 : SUCCESS in 9m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9504f33641bb42928dccae80af722804 : SUCCESS in 15m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5856c1ddcc8a4155bf53c7cd530bde84 : SUCCESS in 11m 34s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/dce28649199f4c1ca6973a68bfc85e7f : SUCCESS in 23m 29s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f3fbf454a93432380112de168be98f3 : SUCCESS in 15m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/bbb7cbecb0164481b4f7aefceeb62d26 : SUCCESS in 1h 18m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0cbf1b853b6848d6b3fca10fd7463a2d : SUCCESS in 38m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c9824cf4792a43969be17c9bf49fc6a8 : SUCCESS in 42m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8683f396c23649a4838a3cb1da5b90fa : SUCCESS in 1h 06m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a6ad7c4703b14a45b911f8068cc416e0 : SUCCESS in 39m 14s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a446853b3374a73aa541e0ccaeef83a : SUCCESS in 1h 18m 32s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a63fe060e8144d82a5cdfe5f3bab3884 : SUCCESS in 13m 40s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/919c88fab98f44e98efe6b06e4fb3515 : SUCCESS in 10m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/626016bf8bab4260ba6e3bd2a92620c0 : SUCCESS in 16m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/188d6ee4856f4d1eb83ff86d94dfe2fb : SUCCESS in 15m 03s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b1a2117f12f4028be975609eca8e7ec : SUCCESS in 27m 52s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/278b53105b2643c582532ff510b3fd44 : SUCCESS in 10m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd3bcd9c33614f909bae3d780ff0b6c7 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":25},{"id":"6936330c0da0f6291f3880bda9626d83628dd12c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-21 20:19:35.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"167bea424634fb7135a48db7dfc969ef8d20116b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 22:02:59.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/776dc21f6b7049c3bc9c1631e0b94290\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd2b518fd3cb41ea87c0f2d434c27f7f : SUCCESS in 9m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/44b44e3e179948ca8bb438805e5ded4a : SUCCESS in 8m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36ac6e5427e74666a43f96135e34bf27 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15c8a66e25454930afd0b97f5bfcd208 : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab0872d88bd34ff78859d042afc86fac : SUCCESS in 7m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/38ec84270eac418d89ca91632b32a770 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4174a1c88c374800855619fadca2186d : SUCCESS in 14m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/21c5c1586dcc42f5aca23f93a5f6e7c3 : SUCCESS in 10m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/759b0359b6574cf58b359c6194027ada : SUCCESS in 14m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/856499ed1d7f4277b43595f6d6d8c7ce : SUCCESS in 12m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c435e572e5974fe59150c406fe760cfe : SUCCESS in 10m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b54324a9d70d45c387745edd938fc6f3 : SUCCESS in 14m 37s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6ebe913c472c447a9697d0c799e08ba3 : POST_FAILURE in 1h 07m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/24c58dc4dc3b474d9ae1fef257fd1c79 : SUCCESS in 1h 23m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d7273add424f4dbcb20bffdcb0356638 : FAILURE in 1h 03m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af0c36bf93cc40099ca03ad6a4e437f7 : SUCCESS in 1h 05m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/17c154c146be4bf9b539a6d5e67f041f : SUCCESS in 34m 58s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57f7dd7c3d2d4fa988b3950455abfed7 : SUCCESS in 1h 38m 42s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/19f6d6e083dc470193fe8bf089ac34ef : SUCCESS in 9m 23s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/54a335c0c7904a95bee0502cee588143 : SUCCESS in 8m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16c64ce146ce49debf6cc4b8f4c26336 : SUCCESS in 17m 15s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5bd0cd5e506d40b892788521a6e63421 : SUCCESS in 17m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/addfdf6013404615bd2c161a6093010d : SUCCESS in 32m 29s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ca0f92393dbc41938e7b6facf8e2d8ce : SUCCESS in 18m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddf535ad2977496ba8c2bf279947dde0 : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":26},{"id":"83793e4c5c4aa055de5e794f8140df333be5440f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-22 07:59:40.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"c536b8227bb6b6d2a2ec2b4fbdc9d950dc6a4426","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 09:54:51.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf7fc5a9bf4348d0b653baf4940b6266\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c14ba9fa415c42079e58202f0e2b25fc : SUCCESS in 4m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c1b3d735274042e7a1ce825711ed5d36 : SUCCESS in 7m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/309cd2144e16482c99f9d8f24f06bfac : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05f5b742c5a5433a92aec88675f63b59 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4484f130975d4152aa2b45e3e49748b7 : SUCCESS in 7m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c59f6c2bc28f4a47b0b4ba39e772ab53 : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cea82514707c4e29a43aa56f662e47eb : SUCCESS in 11m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/54d97d5da260457fad628d5305df064b : SUCCESS in 6m 51s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/235e47e7050445bf92fc22b614c93d0e : SUCCESS in 13m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b740631044d24e038bb5f62a804b33f6 : SUCCESS in 12m 57s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/19daec7bff3f4e9e8c6d94df35111c0a : SUCCESS in 20m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e54578dae5584699ac321cfe86be3db2 : SUCCESS in 10m 43s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/11af959f4d364443905a3b21ac19b7d8 : SUCCESS in 41m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8d540b96c4c543239a1e0d989c450cc6 : SUCCESS in 1h 06m 30s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2a18844632474e919810c764f371b904 : SUCCESS in 1h 13m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c5316b5480a405b8f76e5c4fd8f59ef : SUCCESS in 1h 10m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4e72473b25e84adfa8cfe48331633b6d : SUCCESS in 1h 10m 46s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9b7521931de4728a129c005b41c1bab : SUCCESS in 1h 50m 09s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7767fad3e57e4b42b057a0bb672bcaa7 : SUCCESS in 17m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d7d9c9c9cd7d4ca9a3421c790cf0b7d4 : SUCCESS in 13m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6697f05e43f40809b3ba5fc5264d60b : SUCCESS in 18m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/118fd472724f4958a0e8fa3bc030e61d : SUCCESS in 15m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f1cb6f00eff4a5191ff31e93324af6b : SUCCESS in 29m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1f04a050f53b4f06a4f1dcec90f4c33a : SUCCESS in 12m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a28c32eb88604f3ab0452796df100449 : SUCCESS in 13m 27s","accounts_in_message":[],"_revision_number":26},{"id":"ecc0347695a6b0e87a795a3d83c0f09899117bef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-22 14:43:41.000000000","message":"Uploaded patch set 27.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"0d5336697e1a8cf3f53b0ec635af1454318f77c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 16:38:11.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78a3b655a18a474da7629358ba413ae0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd8b8dcec1e74f12a15ec950c9aaa928 : SUCCESS in 4m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5efdf56e62dd4ece9bdb5b0a568615aa : SUCCESS in 6m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6c815a7ef674f139e2038aa9ced9be0 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/953c39c7085e400880f10150f156663c : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5905d4a1c2b48f6b41a8114f9898914 : SUCCESS in 4m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3ecb41b2da774297b233be0c1541e57f : SUCCESS in 4m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a92054ebe2904cd89a25ec0d18160579 : SUCCESS in 12m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ad4884e8a64546f98ca651734ddd4ab2 : SUCCESS in 11m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/aec83cef428b4a319a2aa6ea297da728 : SUCCESS in 17m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca5ee328acc644068998c52067e24c8e : SUCCESS in 11m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0103832196a04c87aca130ba3b6429c0 : SUCCESS in 15m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b21fcc654c2d44ef9e13be013606d770 : SUCCESS in 15m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3cd60c53f3a044268cf2837e5550f0f6 : SUCCESS in 50m 56s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c092c0c19058424c8b78f8f91d2a8f38 : SUCCESS in 54m 28s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/df200904876940839dcb435350fd8e7b : FAILURE in 1h 42m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9a5b5c6763c4cc385806a9e7a64e84a : SUCCESS in 1h 07m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/234ebc1e5d5d426f963221625f7ec32f : SUCCESS in 1h 13m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be8a1e91009c4f6faf45a27d1c5838ae : SUCCESS in 1h 49m 46s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d760be32aa044e03ab475382b6cbb52f : SUCCESS in 21m 07s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2c9e60f4812f477ea8e008cee5703de7 : SUCCESS in 8m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/875945d402f544c59e2ac63376d39e27 : SUCCESS in 7m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3ab759d9a3f54060baab0e8ccd7257f3 : SUCCESS in 12m 03s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33c49abed5994b068e74bbf1e0e758e1 : SUCCESS in 31m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dc4dd3392bd04fd48e89e9afb4c86c1c : SUCCESS in 7m 58s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54fd5630b5274912b62267382df90c6c : SUCCESS in 14m 22s","accounts_in_message":[],"_revision_number":27},{"id":"a147e02feaa136eae8e58d3bfac1125acf9ced1c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-23 07:43:18.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"0b61a45a79411c07ad2768b63580886bd4b8c292","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 09:39:52.000000000","message":"Patch Set 27:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f5b0dbf2b1c41059f04b7ca9750f595\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/144e6426942d440c911b2e13954aaa43 : SUCCESS in 5m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dbfbbaa701364f76be93587daba4bd8b : SUCCESS in 7m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c941b7a0d2c54563bf27dccbab009367 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed03f4970a2b4ea988cdaabd845456b6 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46c29108354645e7b9a585368006596b : SUCCESS in 9m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/939e6903aed74dd39ad3b8629fa7f709 : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/03814247dcb341f4ad372841809ad35f : SUCCESS in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/978562865cf14673af66f3cbc6644039 : SUCCESS in 9m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d810ff7102c7455db0c20223bc5584b3 : SUCCESS in 16m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01548bf86f2f41c7b248074cbfe449b1 : SUCCESS in 10m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c7abab18599841babc18f3ea8e733bac : SUCCESS in 18m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bca39076e9446858ab13bdc53bf156e : SUCCESS in 16m 21s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/63a577357d9945dd857db8e399432352 : SUCCESS in 1h 18m 43s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3f067f4b19f946da859f0f28a823286a : SUCCESS in 38m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/78410c2256524561a5714f68345f6afd : FAILURE in 1h 02m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18b3508174bf4960b6894e550b2c2cd3 : SUCCESS in 58m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1ab8f9c0f4d14ce690440f526a1aaecf : SUCCESS in 1h 07m 35s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e078d8564fb468a9b295101e96d0080 : SUCCESS in 1h 51m 34s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ab97a3fe052a494898c4a3834e117d9b : SUCCESS in 17m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ea28685761c04610ad2cf3dce73743b5 : SUCCESS in 22m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1dd08f9f37af407191ee144ee712f546 : SUCCESS in 16m 58s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cfd66e5b83394a34a2fa61c47fd03039 : SUCCESS in 16m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb33addf54424fe4bf7fc5f47083108f : SUCCESS in 27m 32s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3493fd44c35b4c89ae2487b7ddfee3b4 : SUCCESS in 14m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62add048fc294ac7a5e5500cbfaa2f6e : SUCCESS in 13m 08s","accounts_in_message":[],"_revision_number":27},{"id":"69e7022376ac2d5c4f9bda2a7c55da22eb861c16","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-23 20:50:53.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"86a8d011cbe3ff7c8a520c4a685a3f10d8e33b91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 22:22:08.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dfab4b13ca44910bcca8796fceafe77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec94b3daa864db183b431e2cc5b5efd : SUCCESS in 7m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e6e72155fcf440c08d1acc0370705756 : SUCCESS in 5m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cbb2b192423425d9955827480a5b01d : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24726065427f499ba519b7621b3d794b : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c066f850474049d5a9562db6ea1efd52 : SUCCESS in 6m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/249faad465d54d69bf252716d74da013 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b7b385520c5e4df0a23b45ac0636986a : SUCCESS in 12m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/97e98303a803412bbd441fc7052d458f : SUCCESS in 6m 40s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/812c88248ec647a18ee225db212d7509 : SUCCESS in 7m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9673e1c442a84171935bc4dd299bd9e8 : SUCCESS in 16m 23s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/21df2e12e61a488a950d229150ae340c : SUCCESS in 15m 54s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2ff05ab9b3e4b8baa99152ceb59429a : SUCCESS in 18m 17s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d743d5a8f67544da938c6ee48e642e73 : SUCCESS in 36m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4421b4d08d0e4b079bf4b850018d45c9 : SUCCESS in 55m 13s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6cd7197115e44540b82f748037eeea4f : SUCCESS in 1h 26m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddf131c43722472ea097f769e06ebd4a : SUCCESS in 57m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dff2ed4bfbf64ab394964c119f4d9f4b : SUCCESS in 1h 16m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c736ba66e3a149908226774425d59833 : SUCCESS in 51m 47s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a9a2891ae0f440bb8a3a483d9b44c691 : SUCCESS in 9m 13s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9d7305ccc2044eb9a897bf2c17ea81e0 : SUCCESS in 15m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5082c5a5cf8f4a8ea9d268bff0f16172 : SUCCESS in 12m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8629e42ecf1941f69228d69753353371 : SUCCESS in 15m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7af685204a564deb8e5a78060f5da0dc : SUCCESS in 31m 22s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/414ad3c51df44c009ac3152ef787749a : SUCCESS in 11m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f719e2ecd5346d5881a44e89d8292ce : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":27},{"id":"e4716910e6032faf759dc11ee1b93aee371704d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-25 21:08:19.000000000","message":"Patch Set 28: Patch Set 27 was rebased\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"30c948f2832128e80bfda1df9a8be64164261fab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 22:51:49.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4b1b984cf50479eb2940062ba35d53a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab01792dcfa24ab4a238e6f066bc4b06 : SUCCESS in 5m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/392f6218ff554f5c872bfde6a71fdf11 : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ad311a0e4054bc08bf4af7461429ada : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0de873d733b24ffb9d87e364dc03ca8e : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da589f1bcfb34d7898c61f94a44833eb : SUCCESS in 6m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/759edafc00994f9e86b00e1a2b733362 : SUCCESS in 3m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f5640fb826524450a2bc28d4db9f5dab : SUCCESS in 12m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fe0d1ccf61304c24b03696a509ed2114 : SUCCESS in 9m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9c54301cef944efb83c50ed7601eaf90 : SUCCESS in 8m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab8998852e5542f4af4d54ab573a2b0b : SUCCESS in 14m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/08b5610248bd47528f61600a04d468d7 : SUCCESS in 17m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdba1829c4b9438bb28c46c002d37da4 : SUCCESS in 14m 32s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0d7955f8e68d4e2a887bb2a9a1fe1226 : SUCCESS in 1h 15m 40s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/09af43684e984e458f27a20e13d95e0d : SUCCESS in 1h 20m 23s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b335398dc7a34026a4ce67b6235f56cf : SUCCESS in 1h 06m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/219845b5c670451d834d1c2c82aca2a1 : SUCCESS in 1h 14m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6d046e40626f43b6ab9c0b2e7df9314a : SUCCESS in 59m 07s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd369b2f5c484661ade88318f148bbc1 : SUCCESS in 1h 29m 08s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/318f9639150a4c02acb59bccb5a63f2a : SUCCESS in 22m 28s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3e58965c66a24420ad733efe86244e75 : SUCCESS in 15m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b44e2296a405490bbe854d905335a676 : SUCCESS in 16m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/33308c083b88494ab5aa194668f0a7b7 : SUCCESS in 14m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc0b69ed67484d06a2c0e914895f1bae : SUCCESS in 27m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4d3cbfbad81b44deb3992f6a7477a70d : SUCCESS in 17m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85a752be92674577bd8c0b17e011a2e0 : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":28},{"id":"fc4d528d247edd03bb4f00ffcdcd2efb71a4d846","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-26 07:26:21.000000000","message":"Patch Set 28: Code-Review+2","accounts_in_message":[],"_revision_number":28},{"id":"71c2041064bcbaa3d0c5c414876aa2d6584d66e9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-26 11:07:49.000000000","message":"Patch Set 28: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"e65789775241e46ccea37ec9969e58a3f3c31981","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-11-26 11:47:38.000000000","message":"Patch Set 28: Code-Review+2","accounts_in_message":[],"_revision_number":28},{"id":"111a9792e7f781bca3216789fb59028b850bc36a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-26 14:47:13.000000000","message":"Patch Set 28: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"48dd2c6ece8c878478f33b907b6ec77b32ab6894","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-11-26 15:05:15.000000000","message":"Patch Set 28: Workflow+1","accounts_in_message":[],"_revision_number":28},{"id":"d062b2bb7b5a8b8787eaba794d1d5057f30fa0d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 15:05:34.000000000","message":"Patch Set 28: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":28},{"id":"905156a43e53023007948bfd7c463760ece44254","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 17:07:23.000000000","message":"Patch Set 28: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9dcd31c330947148dfbdfe8a9333f82\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43eb10d495aa40eca02dd734bbf00c7b : SUCCESS in 6m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3bfaccb6f2a74d6aa08392d8e53e2292 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a344dfb774ff472f92347b2bd07a2b39 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8f9f151ee7a942099418e9ee3935fb42 : SUCCESS in 3m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/770ea198d1eb49988a77b545631c3f2c : SUCCESS in 7m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/77e4c36172f644fe814c7316f1c40554 : SUCCESS in 12m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e577f1f9101540de933cc82103e45f7d : SUCCESS in 11m 42s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/00b6a5bc07854e13b40b31b5bf420db1 : SUCCESS in 7m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa0fd02375644cf5a09732838929caff : SUCCESS in 12m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6e7f258af60c431ca1013a19648c6095 : SUCCESS in 20m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be03ca8f266846f5bccf15d0c4c21bc9 : SUCCESS in 19m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ba86157291dc44ad821387954093bc64 : SUCCESS in 37m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ce1bce3afecb4e35bfa6b254a26eef5c : SUCCESS in 1h 02m 25s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4bc288ee7faa461486057ae0de6ba079 : SUCCESS in 41m 08s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d690044b3424ba8b8c8c31f4dd67b46 : SUCCESS in 33m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a5025e40f12e4223afc3d96575514a77 : SUCCESS in 1h 09m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3103f9af110c417fa74ec6a2b7de93db : SUCCESS in 1h 57m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/da726e9f915b419aa81afdedeb084604 : FAILURE in 6m 32s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9f8654a401c7427e8d9ebf4c69ebc6b4 : SUCCESS in 10m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6426c40546514a10aa8c8d4c556b2ef8 : SUCCESS in 18m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e7e3b756313c4ab996ed562db791c449 : SUCCESS in 14m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f683aca223b48f6a3eb26eb9563bc3d : SUCCESS in 14m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d0afde1843b64ba19bf906ec4ecc28f0 : SUCCESS in 16m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cd93343c09d465d945cdbe9fc5d1d53 : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":28},{"id":"bb8fb4c66fd72cb77b3ef5c9b78a7517d2442e24","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-11-26 17:23:15.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"e87a515f66fbc6ea93c2d240098a62728bb8ebe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 19:13:15.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae79d18ae31c4f71b58dedb78c7c58f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ce91a2c9bdf4b9ea7272706ac77666f : SUCCESS in 9m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fde1fd5921e14ac2988c208130c26e81 : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81379f8265514630950acebb4a1eb5b0 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a84027f94bf4ce9a1aba612c10208fa : SUCCESS in 5m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c5de85e3bcf14995a48590928fc94acd : SUCCESS in 7m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8c6778a2f37b46cab2d3e77f570ff204 : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/91da743b5a7c4d08a0f09d47a665162a : SUCCESS in 11m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3974fe1ebd3647fb8234f9f19c6d9be2 : SUCCESS in 7m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d54f628d9ff84bc8a086ec5067413485 : SUCCESS in 8m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6480ceeca9614294b370a05d638c5b5a : SUCCESS in 12m 12s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3b299abf2ed549a1a3f7325154c0a87d : SUCCESS in 24m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30f0171cf14f4075a0322aa57d07923f : SUCCESS in 7m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/33f68791b5584bc5928c4e14e6cad773 : SUCCESS in 39m 08s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3e4566b01a0541da8bd1c8ed427e6bff : SUCCESS in 1h 12m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d00552c83f3142f89db6e6ba10140f79 : SUCCESS in 40m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59275b58553c41078a7c2403911a9a26 : SUCCESS in 37m 28s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c47d18e6087a486c8aea7b0e443420b4 : SUCCESS in 1h 13m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/186c455b853d45fe85f06bc46598cee5 : SUCCESS in 1h 45m 13s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/22ce060c09cc4d4ea7d19085021692fb : SUCCESS in 22m 30s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/abd27335fa384d6ba3225fe815965d23 : SUCCESS in 9m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3668a5d60b204af69fb852b9beaa62bc : SUCCESS in 9m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9995c80c1a3149bfbdf58447bda46dad : SUCCESS in 15m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13d6620a831b4d158ae6f434b75ac816 : SUCCESS in 28m 09s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/69ec3a560cc94b6b902537347de0daa2 : SUCCESS in 8m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c358d687118a49a1954206a7cce88bd2 : SUCCESS in 13m 21s","accounts_in_message":[],"_revision_number":28},{"id":"dbb8a7456eae220785c08119142ad6208c40a226","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 19:13:29.000000000","message":"Patch Set 28: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":28},{"id":"e42e7a678e59b8abd1f0335bbae6fab802ae199f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 20:49:06.000000000","message":"Patch Set 28: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fc61eec11e843b3a55a2c13f11a346c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0970493ed9b145119a409cee35137089 : SUCCESS in 4m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b2f2f533a6b54aecbd7fa3722011bc7e : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f310af706924b4fbfbcf9be9ea0d7fc : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c2e541d83f7a464cb8f026a5d001c8d5 : SUCCESS in 7m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6813a30103e4441dbfe8cfd484203351 : SUCCESS in 2m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1321c88d5cd74f2ba33abf8a5f93a8a3 : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/76af83d54fc148afbe885cba68f4165a : SUCCESS in 11m 05s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/54bc34b46a7f4de6b677f38ccf605504 : SUCCESS in 11m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fa10554e39a42d990d7ab1d61dce343 : SUCCESS in 15m 09s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5c0898693269464d90a826ada4991d6f : SUCCESS in 11m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/281c22e5f2064869acaa55579f1e2a74 : SUCCESS in 13m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b1f0f8d6aefa4f4292c84767553c046d : SUCCESS in 38m 25s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/595162ea763747c9acf349f440b66557 : SUCCESS in 1h 04m 32s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/38fe15a3232642e7acf4a8a8ce523edd : SUCCESS in 58m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6bd031204c547b9bfd2a75b3ea1214f : SUCCESS in 1h 19m 32s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9cf8e6fd918544c5857dc02aed17342c : SUCCESS in 35m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1db7f987f95d404bb6d8e1fecb129e08 : SUCCESS in 1h 34m 04s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/246faf89e2d24cbebe90a0ef6fbaddfa : SUCCESS in 14m 34s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cd75b9a63d1a466ba9e502ac3dc4dba8 : SUCCESS in 10m 34s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bcc3ff7715545ef8b7cabff18f068ca : SUCCESS in 19m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6826e418f0af40d8beb0bfcd14f512a5 : SUCCESS in 7m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a5422249990489ab819e010eef81b64 : SUCCESS in 25m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/26bd7a182237421c9134366d1b9e434b : SUCCESS in 8m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70967c3fac8f46d2aef8aa291c804d3f : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":28},{"id":"d56e2eec8d8182bfda22a3b9398a2116a3047e3a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 20:49:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":28},{"id":"382038be43d840d5e22be1958644ae0eb3185e7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 20:50:08.000000000","message":"Patch Set 28:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66640eafb58e4f0fa7a95d70a6aab00d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/574ca4f4348342d68022767c46321de3 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":28}],"current_revision_number":28,"current_revision":"03c849c6cab0045a78257eb37db95c75cc1c42c2","revisions":{"ab90792d3121b69aad87c411b9e3b402c89af780":{"kind":"REWORK","_number":1,"created":"2025-09-15 17:34:21.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/1"}}},"commit":{"parents":[{"commit":"9fc779384c1c4dedec8545f2332c2d9adba18c5e","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fc779384c1c4dedec8545f2332c2d9adba18c5e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:34:12.000000000","tz":0},"subject":"DNM: Test inspection","message":"DNM: Test inspection\n\nTODO ironic.conf\n\n[DEFAULT]\nenabled_inspect_interfaces \u003d redfish,no-inspect,agent\n\n[inspector]\nrequire_managed_boot \u003d False\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab90792d3121b69aad87c411b9e3b402c89af780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab90792d3121b69aad87c411b9e3b402c89af780"}]},"branch":"refs/heads/master"},"2eb791c704df316abc861c54152f09f552393182":{"kind":"REWORK","_number":2,"created":"2025-09-15 17:48:36.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/2"}}},"commit":{"parents":[{"commit":"9fc779384c1c4dedec8545f2332c2d9adba18c5e","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fc779384c1c4dedec8545f2332c2d9adba18c5e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:48:26.000000000","tz":0},"subject":"DNM: Test inspection","message":"DNM: Test inspection\n\nTODO ironic.conf\n\n[DEFAULT]\nenabled_inspect_interfaces \u003d redfish,no-inspect,agent\n\n[inspector]\nrequire_managed_boot \u003d False\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2eb791c704df316abc861c54152f09f552393182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2eb791c704df316abc861c54152f09f552393182"}]},"branch":"refs/heads/master"},"e1ba2d2284ea5278017bba7714327e7bbc83a12d":{"kind":"REWORK","_number":3,"created":"2025-09-19 14:27:38.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/3"}}},"commit":{"parents":[{"commit":"9fc779384c1c4dedec8545f2332c2d9adba18c5e","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fc779384c1c4dedec8545f2332c2d9adba18c5e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-19 14:24:15.000000000","tz":0},"subject":"DNM: Test inspection","message":"DNM: Test inspection\n\nTODO ironic.conf\n\n[DEFAULT]\nenabled_inspect_interfaces \u003d redfish,no-inspect,agent\n\n[inspector]\nrequire_managed_boot \u003d False\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1ba2d2284ea5278017bba7714327e7bbc83a12d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1ba2d2284ea5278017bba7714327e7bbc83a12d"}]},"branch":"refs/heads/master"},"770a4e29a5a105c1ce904877c4133747d4d7d087":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-09-19 14:28:02.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/4"}}},"commit":{"parents":[{"commit":"7e08d4fec30c316cbc94723fdaac1a4a20b990da","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e08d4fec30c316cbc94723fdaac1a4a20b990da"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-19 14:28:02.000000000","tz":0},"subject":"DNM: Test inspection","message":"DNM: Test inspection\n\nTODO ironic.conf\n\n[DEFAULT]\nenabled_inspect_interfaces \u003d redfish,no-inspect,agent\n\n[inspector]\nrequire_managed_boot \u003d False\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/770a4e29a5a105c1ce904877c4133747d4d7d087"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/770a4e29a5a105c1ce904877c4133747d4d7d087"}]},"branch":"refs/heads/master","description":"Rebase"},"c4fd5ec9044e6d1953424f890c4a71910c477b91":{"kind":"REWORK","_number":5,"created":"2025-09-19 16:05:30.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/5"}}},"commit":{"parents":[{"commit":"e94d50850fa06a059c8aa4e05045c84f610aec7d","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e94d50850fa06a059c8aa4e05045c84f610aec7d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-19 15:58:55.000000000","tz":0},"subject":"DNM: Test inspection","message":"DNM: Test inspection\n\nTODO ironic.conf\n\n[DEFAULT]\nenabled_inspect_interfaces \u003d redfish,no-inspect,agent\n\n[inspector]\nrequire_managed_boot \u003d False\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4fd5ec9044e6d1953424f890c4a71910c477b91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4fd5ec9044e6d1953424f890c4a71910c477b91"}]},"branch":"refs/heads/master"},"39c70f8f9a64465b38c9316f0b3614944c8fbcc3":{"kind":"REWORK","_number":6,"created":"2025-09-19 16:06:33.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/6"}}},"commit":{"parents":[{"commit":"a724be65242c7e339c77b38d0ba1b5f9566c34ac","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a724be65242c7e339c77b38d0ba1b5f9566c34ac"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-19 16:04:59.000000000","tz":0},"subject":"DNM: Test inspection","message":"DNM: Test inspection\n\nTODO ironic.conf\n\n[DEFAULT]\nenabled_inspect_interfaces \u003d redfish,no-inspect,agent\n\n[inspector]\nrequire_managed_boot \u003d False\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/39c70f8f9a64465b38c9316f0b3614944c8fbcc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/39c70f8f9a64465b38c9316f0b3614944c8fbcc3"}]},"branch":"refs/heads/master"},"2b5c7c06bf11aae6d94bfc5f176aeee1af6efe9e":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-09-19 17:19:23.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/7"}}},"commit":{"parents":[{"commit":"fb8ce9bc8a9d4d3ba8c5651075d228f670038f15","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb8ce9bc8a9d4d3ba8c5651075d228f670038f15"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-19 17:18:59.000000000","tz":0},"subject":"DNM: Test inspection","message":"DNM: Test inspection\n\nTODO ironic.conf\n\n[DEFAULT]\nenabled_inspect_interfaces \u003d redfish,no-inspect,agent\n\n[inspector]\nrequire_managed_boot \u003d False\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2b5c7c06bf11aae6d94bfc5f176aeee1af6efe9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2b5c7c06bf11aae6d94bfc5f176aeee1af6efe9e"}]},"branch":"refs/heads/master"},"38c442245c2814d0fad33ff1935d25ec8c512cc2":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-09-19 17:40:27.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/8"}}},"commit":{"parents":[{"commit":"6338b74834d13abd2fa5593a5a3a4b039806153d","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6338b74834d13abd2fa5593a5a3a4b039806153d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-19 17:40:27.000000000","tz":0},"subject":"DNM: Test inspection","message":"DNM: Test inspection\n\nTODO ironic.conf\n\n[DEFAULT]\nenabled_inspect_interfaces \u003d redfish,no-inspect,agent\n\n[inspector]\nrequire_managed_boot \u003d False\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38c442245c2814d0fad33ff1935d25ec8c512cc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38c442245c2814d0fad33ff1935d25ec8c512cc2"}]},"branch":"refs/heads/master","description":"Rebase"},"36cd3f04e18d02ae61ed7d89f75154cce9da75b9":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-09-19 18:01:49.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/9"}}},"commit":{"parents":[{"commit":"b40d835eb6de5dda292d994386a271c8b892ac05","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b40d835eb6de5dda292d994386a271c8b892ac05"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-19 18:01:49.000000000","tz":0},"subject":"DNM: Test inspection","message":"DNM: Test inspection\n\nTODO ironic.conf\n\n[DEFAULT]\nenabled_inspect_interfaces \u003d redfish,no-inspect,agent\n\n[inspector]\nrequire_managed_boot \u003d False\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36cd3f04e18d02ae61ed7d89f75154cce9da75b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36cd3f04e18d02ae61ed7d89f75154cce9da75b9"}]},"branch":"refs/heads/master","description":"Rebase"},"108a1ae0be4ccde1f65b66225e2baa34c08033da":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-10-10 17:07:35.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/10"}}},"commit":{"parents":[{"commit":"a5830e8948a75790c5b42c4ea7bd1caf2191b1f1","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5830e8948a75790c5b42c4ea7bd1caf2191b1f1"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-10 17:07:35.000000000","tz":0},"subject":"DNM: Test inspection","message":"DNM: Test inspection\n\nTODO ironic.conf\n\n[DEFAULT]\nenabled_inspect_interfaces \u003d redfish,no-inspect,agent\n\n[inspector]\nrequire_managed_boot \u003d False\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/108a1ae0be4ccde1f65b66225e2baa34c08033da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/108a1ae0be4ccde1f65b66225e2baa34c08033da"}]},"branch":"refs/heads/master","description":"Rebase"},"8dfd2cfb9bd3d3e340eeaa267f5a0bca3673772d":{"kind":"REWORK","_number":11,"created":"2025-10-10 17:23:19.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/11"}}},"commit":{"parents":[{"commit":"b65c11a3e0b5c47239316ff308c897ee07109ff1","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b65c11a3e0b5c47239316ff308c897ee07109ff1"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-10 17:23:10.000000000","tz":60},"subject":"DNM: Test inspection","message":"DNM: Test inspection\n\nTODO ironic.conf\n\n[DEFAULT]\nenabled_inspect_interfaces \u003d redfish,no-inspect,agent\n\n[inspector]\nrequire_managed_boot \u003d False\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8dfd2cfb9bd3d3e340eeaa267f5a0bca3673772d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8dfd2cfb9bd3d3e340eeaa267f5a0bca3673772d"}]},"branch":"refs/heads/master"},"204a3948d352e8f77a2c4ae258ab90082cc22da3":{"kind":"REWORK","_number":12,"created":"2025-10-13 17:40:08.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/12"}}},"commit":{"parents":[{"commit":"b65c11a3e0b5c47239316ff308c897ee07109ff1","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b65c11a3e0b5c47239316ff308c897ee07109ff1"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-13 17:40:08.000000000","tz":0},"subject":"DNM: Test inspection","message":"DNM: Test inspection\n\nTODO ironic.conf\n\n[DEFAULT]\nenabled_inspect_interfaces \u003d redfish,no-inspect,agent\n\n[inspector]\nrequire_managed_boot \u003d False\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/204a3948d352e8f77a2c4ae258ab90082cc22da3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/204a3948d352e8f77a2c4ae258ab90082cc22da3"}]},"branch":"refs/heads/master"},"5e4f696e73bd91aa380200e60022ee87e0e020a5":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-10-17 08:42:19.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/13"}}},"commit":{"parents":[{"commit":"f6434fac5cbcbb89f30b2befff52ec123ec948d2","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6434fac5cbcbb89f30b2befff52ec123ec948d2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-17 08:42:19.000000000","tz":0},"subject":"DNM: Test inspection","message":"DNM: Test inspection\n\nTODO ironic.conf\n\n[DEFAULT]\nenabled_inspect_interfaces \u003d redfish,no-inspect,agent\n\n[inspector]\nrequire_managed_boot \u003d False\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e4f696e73bd91aa380200e60022ee87e0e020a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e4f696e73bd91aa380200e60022ee87e0e020a5"}]},"branch":"refs/heads/master","description":"Rebase"},"5bd8be46cefcfb306018e74e68af6b8743ef2932":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-10-17 09:07:25.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/14"}}},"commit":{"parents":[{"commit":"0664805d461e73d7f44caaf6f301a2dc9c59e9c7","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0664805d461e73d7f44caaf6f301a2dc9c59e9c7"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-17 09:07:25.000000000","tz":0},"subject":"DNM: Test inspection","message":"DNM: Test inspection\n\nTODO ironic.conf\n\n[DEFAULT]\nenabled_inspect_interfaces \u003d redfish,no-inspect,agent\n\n[inspector]\nrequire_managed_boot \u003d False\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5bd8be46cefcfb306018e74e68af6b8743ef2932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5bd8be46cefcfb306018e74e68af6b8743ef2932"}]},"branch":"refs/heads/master","description":"Rebase"},"b2d33ba95f44b48a560720dbb20a90dfeae3e258":{"kind":"REWORK","_number":15,"created":"2025-10-17 09:15:54.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/15"}}},"commit":{"parents":[{"commit":"0664805d461e73d7f44caaf6f301a2dc9c59e9c7","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0664805d461e73d7f44caaf6f301a2dc9c59e9c7"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-17 09:14:20.000000000","tz":60},"subject":"Test overcloud inspection","message":"Test overcloud inspection\n\nAdds a test to inspect a node in the baremetal-compute inventory group.\nWe also print all of the inspection rules and collect the inspection\ndata for debugging purposes.\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2d33ba95f44b48a560720dbb20a90dfeae3e258"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2d33ba95f44b48a560720dbb20a90dfeae3e258"}]},"branch":"refs/heads/master"},"769e4a8ee212c909c8a537fe539e9cb710dd4ee1":{"kind":"REWORK","_number":16,"created":"2025-10-17 18:03:54.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/16"}}},"commit":{"parents":[{"commit":"0664805d461e73d7f44caaf6f301a2dc9c59e9c7","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0664805d461e73d7f44caaf6f301a2dc9c59e9c7"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-17 18:03:54.000000000","tz":0},"subject":"Test overcloud inspection","message":"Test overcloud inspection\n\nAdds a test to inspect a node in the baremetal-compute inventory group.\nWe also print all of the inspection rules and collect the inspection\ndata for debugging purposes.\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/769e4a8ee212c909c8a537fe539e9cb710dd4ee1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/769e4a8ee212c909c8a537fe539e9cb710dd4ee1"}]},"branch":"refs/heads/master"},"0274dc2e61f2460794a77548f6875af09407a4c0":{"kind":"REWORK","_number":17,"created":"2025-10-24 09:32:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/17"}}},"commit":{"parents":[{"commit":"0664805d461e73d7f44caaf6f301a2dc9c59e9c7","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0664805d461e73d7f44caaf6f301a2dc9c59e9c7"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-24 09:32:03.000000000","tz":0},"subject":"Test overcloud inspection","message":"Test overcloud inspection\n\nAdds a test to inspect a node in the baremetal-compute inventory group.\nWe also print all of the inspection rules and collect the inspection\ndata for debugging purposes.\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0274dc2e61f2460794a77548f6875af09407a4c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0274dc2e61f2460794a77548f6875af09407a4c0"}]},"branch":"refs/heads/master"},"4f7437ca96c9d5f60eebf4da94ae130579c7a8fe":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2025-10-28 17:38:39.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/18"}}},"commit":{"parents":[{"commit":"7d60c1ac801d9d297a71d2f84ccc8a5b695330ad","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d60c1ac801d9d297a71d2f84ccc8a5b695330ad"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-28 17:38:39.000000000","tz":0},"subject":"Test overcloud inspection","message":"Test overcloud inspection\n\nAdds a test to inspect a node in the baremetal-compute inventory group.\nWe also print all of the inspection rules and collect the inspection\ndata for debugging purposes.\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f7437ca96c9d5f60eebf4da94ae130579c7a8fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f7437ca96c9d5f60eebf4da94ae130579c7a8fe"}]},"branch":"refs/heads/master","description":"Rebase"},"2ccfbec17c4c75e687d8e10e65f09e9ebbc73c4a":{"kind":"REWORK","_number":19,"created":"2025-10-28 18:13:35.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/71/961271/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/19"}}},"commit":{"parents":[{"commit":"7d60c1ac801d9d297a71d2f84ccc8a5b695330ad","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d60c1ac801d9d297a71d2f84ccc8a5b695330ad"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-28 18:13:21.000000000","tz":0},"subject":"Test overcloud inspection","message":"Test overcloud inspection\n\nAdds a test to inspect a node in the baremetal-compute inventory group.\nWe also print all of the inspection rules and collect the inspection\ndata for debugging purposes.\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ccfbec17c4c75e687d8e10e65f09e9ebbc73c4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ccfbec17c4c75e687d8e10e65f09e9ebbc73c4a"}]},"branch":"refs/heads/master"},"46e8b6cbd68a4e33c73a4de935cf043aeed6409d":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2025-11-14 08:07:38.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/71/961271/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/20"}}},"commit":{"parents":[{"commit":"d672aa1d7bb56d9ba3e64f682bbcd1798c7c8ff1","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d672aa1d7bb56d9ba3e64f682bbcd1798c7c8ff1"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-11-14 08:07:38.000000000","tz":0},"subject":"Test overcloud inspection","message":"Test overcloud inspection\n\nAdds a test to inspect a node in the baremetal-compute inventory group.\nWe also print all of the inspection rules and collect the inspection\ndata for debugging purposes.\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46e8b6cbd68a4e33c73a4de935cf043aeed6409d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46e8b6cbd68a4e33c73a4de935cf043aeed6409d"}]},"branch":"refs/heads/master","description":"Rebase"},"4c12671fb553acfbf9186744cc6b39827a0d3a3f":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2025-11-14 08:25:33.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/71/961271/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/21"}}},"commit":{"parents":[{"commit":"2888b585c0e161a336efb70b0e3f0698d554d720","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2888b585c0e161a336efb70b0e3f0698d554d720"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-11-14 08:25:33.000000000","tz":0},"subject":"Test overcloud inspection","message":"Test overcloud inspection\n\nAdds a test to inspect a node in the baremetal-compute inventory group.\nWe also print all of the inspection rules and collect the inspection\ndata for debugging purposes.\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c12671fb553acfbf9186744cc6b39827a0d3a3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c12671fb553acfbf9186744cc6b39827a0d3a3f"}]},"branch":"refs/heads/master","description":"Rebase"},"044bbd9d0c9b91bfeeeb65563516d6086200d59a":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2025-11-14 11:54:43.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/71/961271/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/22"}}},"commit":{"parents":[{"commit":"75d93c8da05484e23ffa83ba26fc1b8c10be5ec8","subject":"Support built-in inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/75d93c8da05484e23ffa83ba26fc1b8c10be5ec8"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-11-14 11:54:43.000000000","tz":0},"subject":"Test overcloud inspection","message":"Test overcloud inspection\n\nAdds a test to inspect a node in the baremetal-compute inventory group.\nWe also print all of the inspection rules and collect the inspection\ndata for debugging purposes.\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/044bbd9d0c9b91bfeeeb65563516d6086200d59a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/044bbd9d0c9b91bfeeeb65563516d6086200d59a"}]},"branch":"refs/heads/master","description":"Rebase"},"75f9f9c8faa2828b969b817dadea938eeb9951b3":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2025-11-21 09:54:44.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/961271/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/23"}}},"commit":{"parents":[{"commit":"d39bfabe0a6506be375601e2f658a3a8c897704b","subject":"Fix issues with inspection rules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d39bfabe0a6506be375601e2f658a3a8c897704b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-11-21 09:54:44.000000000","tz":0},"subject":"Test overcloud inspection","message":"Test overcloud inspection\n\nAdds a test to inspect a node in the baremetal-compute inventory group.\nWe also print all of the inspection rules and collect the inspection\ndata for debugging purposes.\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/75f9f9c8faa2828b969b817dadea938eeb9951b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/75f9f9c8faa2828b969b817dadea938eeb9951b3"}]},"branch":"refs/heads/master","description":"Rebase"},"86b08a6174c52397c0ff568725f520ad2cbcec9a":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2025-11-21 10:44:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/961271/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/24"}}},"commit":{"parents":[{"commit":"946fa55a10736fe9b5a97ddb1e26d75eb3d9eceb","subject":"Fix issues with inspection rules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/946fa55a10736fe9b5a97ddb1e26d75eb3d9eceb"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-21 10:44:22.000000000","tz":60},"subject":"Test overcloud inspection","message":"Test overcloud inspection\n\nAdds a test to inspect a node in the baremetal-compute inventory group.\nWe also print all of the inspection rules and collect the inspection\ndata for debugging purposes.\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86b08a6174c52397c0ff568725f520ad2cbcec9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86b08a6174c52397c0ff568725f520ad2cbcec9a"}]},"branch":"refs/heads/master"},"440d3fbab62ee6a80125213a47ca353d3623bc05":{"kind":"NO_CHANGE","_number":25,"created":"2025-11-21 12:28:05.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/961271/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/25"}}},"commit":{"parents":[{"commit":"3f1b6ff0b5099a340199c11aed384bcaac49a480","subject":"Fix issues with inspection rules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f1b6ff0b5099a340199c11aed384bcaac49a480"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-21 12:26:54.000000000","tz":60},"subject":"Test overcloud inspection","message":"Test overcloud inspection\n\nAdds a test to inspect a node in the baremetal-compute inventory group.\nWe also print all of the inspection rules and collect the inspection\ndata for debugging purposes.\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/440d3fbab62ee6a80125213a47ca353d3623bc05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/440d3fbab62ee6a80125213a47ca353d3623bc05"}]},"branch":"refs/heads/master"},"8a6a7a1b21dc9ea276c37fa3b57920335c88d6a0":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2025-11-21 20:19:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/961271/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/26"}}},"commit":{"parents":[{"commit":"4275c90eb511930dd9b327fcd3970824b70399b4","subject":"Fix issues with inspection rules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4275c90eb511930dd9b327fcd3970824b70399b4"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-21 20:18:50.000000000","tz":60},"subject":"Test overcloud inspection","message":"Test overcloud inspection\n\nAdds a test to inspect a node in the baremetal-compute inventory group.\nWe also print all of the inspection rules and collect the inspection\ndata for debugging purposes.\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a6a7a1b21dc9ea276c37fa3b57920335c88d6a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a6a7a1b21dc9ea276c37fa3b57920335c88d6a0"}]},"branch":"refs/heads/master"},"3cc5c7e0db47be55847e456fac457dbbf1e45024":{"kind":"REWORK","_number":27,"created":"2025-11-22 14:43:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/961271/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/27"}}},"commit":{"parents":[{"commit":"4275c90eb511930dd9b327fcd3970824b70399b4","subject":"Fix issues with inspection rules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4275c90eb511930dd9b327fcd3970824b70399b4"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-22 14:40:34.000000000","tz":60},"subject":"Test bare metal node inspection","message":"Test bare metal node inspection\n\nAdds a test to inspect a node in the baremetal-compute inventory group.\nWe also print all of the inspection rules and collect the inspection\ndata for debugging purposes.\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3cc5c7e0db47be55847e456fac457dbbf1e45024"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3cc5c7e0db47be55847e456fac457dbbf1e45024"}]},"branch":"refs/heads/master"},"03c849c6cab0045a78257eb37db95c75cc1c42c2":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2025-11-25 21:08:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/961271/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/961271/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/961271/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/961271/28"}}},"commit":{"parents":[{"commit":"2858b253b147c6198b112c3437ac23bb364d6192","subject":"Document moving releases to unmaintained and EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2858b253b147c6198b112c3437ac23bb364d6192"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-15 17:27:46.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-25 21:08:19.000000000","tz":0},"subject":"Test bare metal node inspection","message":"Test bare metal node inspection\n\nAdds a test to inspect a node in the baremetal-compute inventory group.\nWe also print all of the inspection rules and collect the inspection\ndata for debugging purposes.\n\nChange-Id: I9d9233ea576e5cf7ce4f51db12817af00400a1cb\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03c849c6cab0045a78257eb37db95c75cc1c42c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03c849c6cab0045a78257eb37db95c75cc1c42c2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
