)]}'
{"id":"openstack%2Fkayobe~961552","triplet_id":"openstack%2Fkayobe~master~I9e968eb26747d690f1dd0cf138497311e9360cff","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-09-18 09:16:46.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-09-18 11:13:42.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-09-18 08:42:45.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I9e968eb26747d690f1dd0cf138497311e9360cff","subject":"Fix disabling NetworkManager DNS management on Rocky Linux 10","status":"MERGED","created":"2025-09-17 21:06:21.000000000","updated":"2025-09-18 11:14:34.000000000","submitted":"2025-09-18 11:13:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961552","meta_rev_id":"28cb9b39eb0edba220aabf833d2130954c53e170","_number":961552,"virtual_id_number":961552,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-18 11:13:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-09-18 09:16:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-09-18 09:27:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-18 08:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-09-18 09:16:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-17 22:46:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-18 08:16:53.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-09-18 09:08:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-09-18 09:27:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"e989175a85c6fa9403ead1cd23ed22e0471494f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-17 21:06:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"feb657eec75380e2804e7c92e03a240848b7d9fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 22:46:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d28a2a49ff214c0eb8f0f0c91c86e7b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea713558cba643b593ec9dfc906e8b0e : SUCCESS in 9m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/92d4376ed3cd4be392f042a553e2678d : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7dfc9b9fd9742cf9e4bbb935d95e752 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38ad8eea24b6460a859a6684365a39c0 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e55b281b9a34d40ac11a813df81b3d4 : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/465ae3d3504043d4bd58476f6b6ef7b3 : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cdf1e06b384c46389840b185a5e27c67 : SUCCESS in 4m 31s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/301fe4bfd8f94b8a90632ca4cf170dce : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f7e1613760c744d7a2794eb7ad051147 : SUCCESS in 12m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c791e6bca2d3476db9baed4381576a6d : SUCCESS in 11m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a1c6682c3274576be7f2813b22bd474 : SUCCESS in 26m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5accf5995eb04f9189a3cb7bc203670a : SUCCESS in 48m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/64eadb751e244bf2bfe9b202897e80c4 : SUCCESS in 17m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f8b8c5150644e08b43286a013ec5d87 : SUCCESS in 13m 03s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5350b99a42a24e90962a3eb357800c91 : SUCCESS in 1h 35m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b0d253c26b045bd958f50c5fede2bc7 : SUCCESS in 8m 24s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3ad2a53c85d2429495d8c365d961250e : SUCCESS in 6m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a111fcb6a014c11a6bfb43188a1ed83 : SUCCESS in 26m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1808defaa3714e8da1948a1a14cba0be : SUCCESS in 8m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fa95d5960284433a9e2d636e4b69fd5 : SUCCESS in 11m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2031fd83e6914689b6d18f68328f1261 : SUCCESS in 12m 18s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a950883339214d0ab1f9c774dfcd21fe : SUCCESS in 13m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a6f9ed4f0f2de6f9ec66e1dac8b1858158136290","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-09-18 08:16:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b00deb8c8df03dc63e2cf913e61dac5c8c1e2f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-18 08:33:50.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a24d6c62b7c268058aa5f91e1830c9d85f8b3a1d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-18 08:34:01.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"81131837a8694d80d9bdb52cf734076a73d91289","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-09-18 08:42:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e67c40fa86944d3b1406dcb8463a02fb20c204d8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-18 09:16:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"39e6b92a0ac84a1c7bcc92461402d2e7eef11c24","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-18 09:27:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ee2c8a6a5c1a0c5a7114c074b44d40fd47bc116b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 09:29:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1453bc5e40d34d7e983e667881cd641e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d19adef499c42258b1d87b2a2cccc7f : SUCCESS in 9m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2552ed7d6ab04de2b4cf7411472a72b7 : SUCCESS in 7m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/670bd15fcd1e4746ad321069fbe20bc8 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb06bd8c3764bcd9e4bf5ad4ed87b76 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5dc320039c8e4fa08b4629e10b9335ca : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26419e67dcb747b28335aff875f5b23e : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a4a0ccd55a0c4b01a64540675a3c523b : SUCCESS in 7m 19s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/43a45808d3644e739c4de0d6f5202963 : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/75ea3d92ac9448f0b3aff8562965480b : SUCCESS in 9m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dd48ee5d77274697b91a9ad60a9434e0 : SUCCESS in 12m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d5a551e79774367b8b17ff687a2942f : SUCCESS in 44m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8516ea65928045899384feec4d94285d : SUCCESS in 49m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/98e72d3a9aca4573918b62a19a86399a : SUCCESS in 16m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c315342919d415496e8e2f1d50952b4 : SUCCESS in 6m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d27c0730e86746caa1bef9e790c12fa6 : SUCCESS in 48m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/142ccdf450a949c2a7094fadc6b679af : SUCCESS in 14m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fd1e3d7dd8794dc38419b63f74ec7326 : SUCCESS in 14m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b72c997e9824b4cadefc7477e30accb : SUCCESS in 17m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a1fdea3e260b45c5a7f5998838f6dc11 : SUCCESS in 12m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be790e8438bc49dca4b42367fdcf8b29 : SUCCESS in 6m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b2a30a17eb0b453ab944fe96c1a1f2f9 : SUCCESS in 14m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0c5251f5ca545dda85830a363a90919 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":2},{"id":"c959522120e505e2e01aa6c25d63a039e1573b50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 09:29:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"82eb655d191b8ec47041f5bfad6d6533febc35d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 11:13:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7efcbe9ec236493aaf4c6d32aeca5913\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a32217518f84c1791087815c4db5605 : SUCCESS in 5m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/06aed4d1acdf4798839d068a5017b53a : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c84aff47084f48ddabd82dbc0487bd41 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15cc79c879104efba05384fbe5262e2c : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/31b41b01d9e946d887a793f6269dc148 : SUCCESS in 4m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/22ca360e08474bd194832c41c1c7f0c9 : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/04879fd22a314a62ae1b7fd08dedee61 : SUCCESS in 9m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aeada59dde2746c2914fbbc17fda7579 : SUCCESS in 12m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d422a7d5ef44171a27e691883f582fe : SUCCESS in 59m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8c36a94127b94dd6b93b266b6b86f848 : SUCCESS in 47m 49s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fdeb486db4f74137b920ea4c42885d33 : SUCCESS in 17m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c4f505f0d5d4d54a959c4a660a23e19 : SUCCESS in 13m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffe256c611644cadb42fdcd4bc9924cf : SUCCESS in 1h 29m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f1adb247be3430f9852282918e79275 : SUCCESS in 10m 22s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/501bb4bdf11341599eca7f6d23ddac4f : SUCCESS in 12m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b97c15665eb748fa9bc40b40cf88adfd : SUCCESS in 26m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/51e10baaa36446e5ab8cf7aca3953414 : SUCCESS in 14m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/747b831644b4433784819ce0571ce7b1 : SUCCESS in 10m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/adb7cf3ca7264e8eb0c4aa62826342e3 : SUCCESS in 8m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2bfb3650d5f45c8822c917520ab31b2 : SUCCESS in 9m 16s","accounts_in_message":[],"_revision_number":2},{"id":"5136ac5e9348ee3d5f85d9b765eae77bc76be033","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 11:13:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"28cb9b39eb0edba220aabf833d2130954c53e170","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 11:14:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28bca14da5a9496a9267db9d13026209\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4621761cc36647bc84fb943504b8313d : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"44163326edf0ed33e801e9fd9644f38258225b1e","revisions":{"5f1719520071fd9a7bc1ad5b52e94288478fa98e":{"kind":"REWORK","_number":1,"created":"2025-09-17 21:06:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/961552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/961552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/961552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/961552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/961552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/961552/1"}}},"commit":{"parents":[{"commit":"c136c23daee93960b6d09053b04b8e45f59d83e4","subject":"Bump MichaelRigart.interfaces to v1.15.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c136c23daee93960b6d09053b04b8e45f59d83e4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-17 21:05:48.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-17 21:05:48.000000000","tz":120},"subject":"Fix disabling NetworkManager DNS management on Rocky Linux 10","message":"Fix disabling NetworkManager DNS management on Rocky Linux 10\n\nChange-Id: I9e968eb26747d690f1dd0cf138497311e9360cff\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f1719520071fd9a7bc1ad5b52e94288478fa98e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f1719520071fd9a7bc1ad5b52e94288478fa98e"}]},"branch":"refs/heads/master"},"44163326edf0ed33e801e9fd9644f38258225b1e":{"kind":"REWORK","_number":2,"created":"2025-09-18 08:33:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/961552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/961552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/961552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/961552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/961552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/961552/2"}}},"commit":{"parents":[{"commit":"c136c23daee93960b6d09053b04b8e45f59d83e4","subject":"Bump MichaelRigart.interfaces to v1.15.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c136c23daee93960b6d09053b04b8e45f59d83e4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-17 21:05:48.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-18 08:33:44.000000000","tz":0},"subject":"Fix disabling NetworkManager DNS management on Rocky Linux 10","message":"Fix disabling NetworkManager DNS management on Rocky Linux 10\n\nChange-Id: I9e968eb26747d690f1dd0cf138497311e9360cff\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44163326edf0ed33e801e9fd9644f38258225b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44163326edf0ed33e801e9fd9644f38258225b1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
