)]}'
{"id":"openstack%2Fkayobe~961608","triplet_id":"openstack%2Fkayobe~master~I3a4279178a39a4ad9f834832f3616acaff092212","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-09-19 13:10:27.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-09-19 12:43:31.000000000","reason":"\u003cGERRIT_ACCOUNT_17669\u003e replied on the change","reason_account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-09-19 12:53:36.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-09-19 14:21:58.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-09-19 17:55:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a4279178a39a4ad9f834832f3616acaff092212","subject":"Change default centos bootstrap user to cloud-user","status":"MERGED","created":"2025-09-18 08:21:24.000000000","updated":"2025-09-19 17:56:45.000000000","submitted":"2025-09-19 17:55:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961608","meta_rev_id":"7d4cb91ad050993aa8fcaddce2cd0ef04c1215f6","_number":961608,"virtual_id_number":961608,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-19 17:55:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-09-19 13:10:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2025-09-19 12:43:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-09-19 12:53:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-09-19 13:10:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-09-19 13:10:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-18 09:08:57.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-09-18 09:08:57.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-09-18 09:25:58.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2025-09-18 10:14:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-18 13:10:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-19 11:54:42.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"15e9289cbc2af2abab3161723f3e0509444e4366","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-18 08:21:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dfa5f89c772f8ae7c038dda60f7cc7257d93097","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-18 09:25:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a2c080c4235af7ba98abcdf1d3c32ee64b7991b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 10:14:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5eb0edd9fb9d406b9cdf689aa434a70b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12e8c4d1e7de4b30a31bdf29056dcab3 : SUCCESS in 6m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2dbfca9dd9e7407187657ab6df90dde8 : SUCCESS in 3m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d27d9450a9df4142b9fa69481092d954 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82ebfa1d91fb42708e92c51115468c54 : FAILURE in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cac3d7b3c3874cb192fd24e4c2b88d93 : SUCCESS in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2b8a4ed0973498e90af6bd7499a6abb : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/323b7cdba4da4b28bd1220fd9a3a9b4a : SUCCESS in 8m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e83e9007e46341328fde31ff15f53538 : SUCCESS in 2m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/084ab8d14d594a3ea8223a85743f7830 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d41cc8e8418d4708bbd5b501fb2a34a5 : SUCCESS in 10m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c92ad269bbc4447baed249a5d04dfb6d : SUCCESS in 10m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7a458012e584e4b9710f0a716eae83e : SUCCESS in 36m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0f71324426024ac1b09afe29b88fbdcd : SUCCESS in 24m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/26e4676a853240dfa8b503567f6ebae6 : SUCCESS in 24m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d98320b1e5843fd9fa6673b17c30072 : SUCCESS in 7m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e491e13922ab4d4281228ba7b8125493 : SUCCESS in 1h 50m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6f2cf471ce34e689d48383c7123e5e2 : SUCCESS in 8m 21s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3c1b555c6f764e9ba070473a8fd77cbf : SUCCESS in 16m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ccd22704d4b14726803c69aafe807823 : SUCCESS in 31m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1aa6a504a02a4ae99c16619912714631 : SUCCESS in 12m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96f41c086c5c42a6baf99b8c302e7c6a : SUCCESS in 13m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/46a7f58d5e3a4a369dc0d706110b36a9 : SUCCESS in 12m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6d37a53ad314ef28dade04d9a7b1c70 : SUCCESS in 13m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d076532d8e3dbe14585b7fb395d4aa2b8193ed35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-18 11:56:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5dd4de75dbb055b4243d06dbfed41b07690db87c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-18 11:56:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c7614df2e77ff8678efd3d3fbb82115b930c0571","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-18 13:10:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"db20c854acd356a1ef1c83f87960d9f631e99ac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 13:55:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1e9885053d0429888a020a15def494d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2898d53b7f7456f8b2503e0c4552a9b : SUCCESS in 9m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ec978ddcc45044c09520004e708edc39 : SUCCESS in 6m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce710caffe1049f79a85f59194794f26 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e127c796ad4b42e8aa31eb0ae0ac4e8f : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b73f7444e3414276a5034e913f46afc5 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/00c774b6f9e84971ab9054f3a8fe07a7 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/404c0c94aa2342feb27900de74b9d6ef : SUCCESS in 7m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84d8e07afa4f4ee5a0c8419bb42d1a23 : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f6d9080982684f5e81682e7d7b6254b1 : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5de5f608c4ba40f9a51947fe29a915e6 : SUCCESS in 10m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d90d35bc2c4548bf9d62eb7087c8179b : SUCCESS in 7m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b0ff65e02324932bf0adeb029f68997 : SUCCESS in 57m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4aac8f9231b64d3ba4ffcbef1a61fa20 : SUCCESS in 49m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3e2cc5ae90ad4b0688445e9696e3762d : SUCCESS in 12m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0001224136843b9881b67c97ccf6d7f : SUCCESS in 14m 28s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab796522ada94a7591189c86d66feaba : SUCCESS in 1h 40m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d075ed97515641939119080446eaf290 : SUCCESS in 14m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c30d9acd46e04293a97d97dbd939d9fb : SUCCESS in 12m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c31d72215371403b823a94703d133dfd : SUCCESS in 23m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f0fb0291191b4aaa95d605a9ca3a3151 : SUCCESS in 15m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a78e4c1272d74e85804395b09a89b98e : SUCCESS in 12m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1cd9e711bfc54b52a419bd3657a9f0d6 : SUCCESS in 11m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf54bddfae77433f9dffce11bcf477c3 : SUCCESS in 10m 49s","accounts_in_message":[],"_revision_number":2},{"id":"f2cdf98a8ecc5115b49a96ba6071a29be3fafed6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-18 14:02:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"64185bc61c7eda3658d762ebded50d35fff8304c","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 15:06:06.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3374f316e5144bc9b297d356a3040979\n\n- kayobe-infra-vm-centos9s-cloud-image https://zuul.opendev.org/t/openstack/build/301b803330e84d9c95ec4273deb543a8 : SUCCESS in 24m 41s (non-voting)\n- kayobe-infra-vm-rocky9-cloud-image https://zuul.opendev.org/t/openstack/build/c5720dec70de481293650b4e58143dcf : FAILURE in 20m 01s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/fedaefa385354c509de572ee0ae627c7 : FAILURE in 14m 46s (non-voting)\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e5ea8bb0191b411190540434aae59e12 : FAILURE in 20m 12s (non-voting)\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2b48fd31dd2d4f34939458f8ca0e191f : SUCCESS in 17m 54s (non-voting)\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/14e874c1eb0746ef848a65f73fc4f0db : FAILURE in 26m 58s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/59e4c32b4fd2486d9a621a25254db210 : FAILURE in 27m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/efa987d7cb23409e9a756415cb875725 : FAILURE in 32m 35s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/552ac620424b42f59a2e7e4378e63bff : FAILURE in 55m 07s\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/f0773ba1493c4becaae784dbb3dd69b7 : FAILURE in 18m 08s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/abd456f353074472bfd10b2f3b9cfe0b : FAILURE in 21m 04s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/ebaec95e76bc4925bbe883f2654f9f8b : FAILURE in 10m 51s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/943f62077e274c069017b0310176ef86 : SUCCESS in 38m 20s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/113dae852d4b444993edc0ecf502878b : SUCCESS in 48m 29s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ce908f24b25643afa9ad70f6daf2eeaa : SUCCESS in 17m 03s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/8c22058afb874202bfa7aee5a6ae3a41 : SUCCESS in 16m 43s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/164813e0fe3e4facb207de189cef254f : SUCCESS in 35m 49s\n- kayobe-seed-vm-centos9s-cloud-image https://zuul.opendev.org/t/openstack/build/06120f32e9ef4e71b9fa8592f22dc2a0 : FAILURE in 20m 55s (non-voting)\n- kayobe-seed-vm-centos9s-q35 https://zuul.opendev.org/t/openstack/build/13f6635cde154a2590402cfd28e32360 : SUCCESS in 12m 02s (non-voting)\n- kayobe-seed-vm-rocky9-cloud-image https://zuul.opendev.org/t/openstack/build/4f47963a88d44ef2aa18c0cbb47cc6b9 : SUCCESS in 19m 43s\n- kayobe-seed-vm-rocky9-q35 https://zuul.opendev.org/t/openstack/build/8c03dbee68f74f51ad50e5ba07b0feb2 : SUCCESS in 13m 38s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/f069924a9e9f4be9a83fcfffd61e4483 : FAILURE in 16m 47s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/8c807c5e0f8148c9a012a6215a8bfa1b : SUCCESS in 12m 02s","accounts_in_message":[],"_revision_number":2},{"id":"9b1555f877cedac23fa1892e73155313713b35ad","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 10:03:15.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a7de37c8fcf586e9810769da38d4087e4e0f6b67","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-09-19 11:54:42.000000000","message":"Patch Set 2: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5b088c924c0829a6f4397d1292c835fa5254b7f9","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-09-19 12:01:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f2a7287d22f71145856d795aa80f0f2e68c0db31","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-19 12:29:52.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"c917aa863f44bb37204907b355c32076959f2ab2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-19 12:38:10.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d32498ab8a792f25b7bbb7813c95e1e5133f034b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-19 12:38:32.000000000","message":"Patch Set 2: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"92d797d9e560e1472a92f4e6b3ce47b811e6a5cb","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-09-19 12:43:31.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9a72b94ca0dedc2f0c6cbd6ef549dd1392ba387c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-19 12:53:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"cbe2efdeec385c8ce0259e3798859154e9849b84","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 13:10:27.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a8b2ff377572fd42b6f232de96dfdba94f503a9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 14:14:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4dd5f18b6fe1481cbc0127deebd7db87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66196ebe3dd746c992d27f42db80b4f8 : SUCCESS in 9m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8179ccc6ca064c6b9932b9a75b50e285 : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e2ae64cce774991976d8c91e5b716cb : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a4cbdb3f515412a8fa2aa50fe7401fc : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9cc2653713744d1be5d9094159c3ddc : SUCCESS in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9b0b60ecb474dfba52792aed781bf92 : SUCCESS in 5m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f17f4dc641eb486c957236a3cce8ac7a : SUCCESS in 7m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/041a3d72edf1499787b4c11d26e3a545 : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/40ca8b210c76424194cb40f9b70a5bdc : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a5572e5fa44f4a75bcdac437089bf9a5 : SUCCESS in 13m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/84543392e0304135839f3d5813423809 : SUCCESS in 10m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b2cba4dea8b479385acdbfba00de6fc : SUCCESS in 48m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7d9ce7afe9eb470e8f86a318b345c1ac : FAILURE in 14m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6a815b0a30104d00b19d1bc08394ef73 : SUCCESS in 21m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b89ce6c07cd463bae138c9393cc57bc : SUCCESS in 12m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a97b3a190ea458895dad5431a860fcb : SUCCESS in 1h 30m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f186bf112d3402eb7f23693f13f90dc : SUCCESS in 7m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fe389a02f1ab42268e0dcd3b6473b487 : SUCCESS in 8m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48d34ced830347e4a43bf2b7dfa8f1a6 : SUCCESS in 22m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/deece30b78984a78b98c72a2ad69af18 : SUCCESS in 11m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75d8c0de7b894ef9aefaf8c907d9d354 : SUCCESS in 12m 38s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9bef96fbdbe94b24b99ebd9bf1776183 : SUCCESS in 12m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10734f131f1b454fb501d82f97d14a1e : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":3},{"id":"8b2d54f03aa375a293e9b9bad96392ab392c6fe6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-19 14:21:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4d973f10cf2973f08486112526024ffce75f474a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 16:18:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b043280b9d74040abe83042cd0c9621\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fae88af9c61d472fa486e109a986b1a8 : SUCCESS in 9m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d7c71e66690f4bc893ba5d6a65831c20 : SUCCESS in 4m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/532e0becae9046538d0582b08b280f8d : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d6ea9dd2a3d4e12a52e6264bb7e8fe6 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/840ebb8179304715aace450762686ef2 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f28f71e36164d3c9b9952340987667f : SUCCESS in 6m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/69879a45557f43d3b517ccb22b43ae41 : SUCCESS in 8m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b89a8c70d569429a82e9e5e541c81066 : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/68eb4d9201924404aa0a13682e4b3a95 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d4bb865a7d8c48f98d1f01f884748fe9 : SUCCESS in 6m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e963d72726834b068b1e753f4c0784de : SUCCESS in 11m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/612e3d96f7ea49458f1937f0f2635b64 : SUCCESS in 52m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a74832a0dec340a7be1a19e1c27d9694 : SUCCESS in 46m 59s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9ffdd0db80304909964b899b0b8ddb10 : SUCCESS in 15m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68edae318b504a748a6f1df7e3439e61 : SUCCESS in 12m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed3eb24a56cf484db18fd34febeea0da : SUCCESS in 1h 32m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce369d4959db415ea8ecf7d38c3942f9 : SUCCESS in 13m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/96e55f9aec654c9d92a58638f938f595 : SUCCESS in 14m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c324e282c8e41a9a3753fa2afcd45a9 : SUCCESS in 26m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/069a7a37b1ee422d984485b470a41667 : SUCCESS in 13m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1158df20ccb545bd9132211d12444635 : SUCCESS in 15m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fa859f33f7914b7c90fcdb506b74c002 : SUCCESS in 11m 22s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9b0a03226cc4f7ca28d897fff3038df : SUCCESS in 11m 15s","accounts_in_message":[],"_revision_number":3},{"id":"9759053d8126fff5d9a59acd3a1ffab069fda07b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 16:18:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d21d2e6b299b3055814ff06dcbf7d409c3c87109","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 17:55:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a93178f3632347f6be43105f50877346\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f3e90e730234d51808bf82d438b4ecc : SUCCESS in 9m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f91f8a1ba162486da57ed326dc766707 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f8248dab0954df68cc27d61ed6d387b : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d70dbe439a846d386055a3434123851 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/04aaadaa49534e6ab270be18b1030089 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed76763a7b0d47c7bcd63d454ff24582 : SUCCESS in 4m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/339436f75c8d480492151e56638a39cf : SUCCESS in 6m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b9dd27ee71f84a9988b041e100b42c4e : SUCCESS in 10m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/774103db0f314c6c814d401ab9b396c2 : SUCCESS in 9m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1df2585254c54efe9a1d3df907d1ef49 : SUCCESS in 46m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/808c70a9cded42348685eb6962091bd8 : SUCCESS in 44m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a6be513edaf44e2ab3a570891bbad373 : SUCCESS in 22m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfac6a8da8ee478bb300c7601857830d : SUCCESS in 7m 08s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d18ce95a3594f90bf56366b241ab426 : SUCCESS in 1h 31m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50f08358a68746a08db989c5c1407c89 : SUCCESS in 15m 21s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3707b81c47bf43d58f1a3d6b06ec48dd : SUCCESS in 12m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ec6aad46dc642d19141e6346f7a43b8 : SUCCESS in 19m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b8f4028d511d4e0590eefbfafdf6a4cb : SUCCESS in 13m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe1b674dedd747f4a416168ed8c56848 : SUCCESS in 11m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1acc9edc5ae64d75bc7cd3e9142190c4 : SUCCESS in 12m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7a3a79078c4465b8459f57bb3293a0d : SUCCESS in 9m 18s","accounts_in_message":[],"_revision_number":3},{"id":"d7fb44dc08f1f9c57e9b86c7bbb51f69b646ee91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 17:55:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7d4cb91ad050993aa8fcaddce2cd0ef04c1215f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 17:56:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3aee275fc28d4e04b9bba78f39024b74\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f81b59895f2431487f2ababb3dcdb16 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b92f5cb481314799b59074bd493eaed0 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d6273899b4c7c6660ef16199a1acbe2e4a85201c","revisions":{"c0330b45ec91ba751152733bba19ec887a6c8bbe":{"kind":"REWORK","_number":1,"created":"2025-09-18 08:21:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/08/961608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/08/961608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/961608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/961608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/961608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/08/961608/1"}}},"commit":{"parents":[{"commit":"c136c23daee93960b6d09053b04b8e45f59d83e4","subject":"Bump MichaelRigart.interfaces to v1.15.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c136c23daee93960b6d09053b04b8e45f59d83e4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-18 08:21:15.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-18 08:21:15.000000000","tz":120},"subject":"Change default centos bootstrap user to cloud-user","message":"Change default centos bootstrap user to cloud-user\n\nThis matches the default user in CentOS GenericCloud images.\n\nChange-Id: I3a4279178a39a4ad9f834832f3616acaff092212\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0330b45ec91ba751152733bba19ec887a6c8bbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0330b45ec91ba751152733bba19ec887a6c8bbe"}]},"branch":"refs/heads/master"},"c0899f211cf3b1e834b342eafecdd19b1a9370dc":{"kind":"REWORK","_number":2,"created":"2025-09-18 11:56:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/08/961608/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/08/961608/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/961608/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/961608/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/961608/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/08/961608/2"}}},"commit":{"parents":[{"commit":"44163326edf0ed33e801e9fd9644f38258225b1e","subject":"Fix disabling NetworkManager DNS management on Rocky Linux 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44163326edf0ed33e801e9fd9644f38258225b1e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-18 11:55:00.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-18 11:55:00.000000000","tz":120},"subject":"Change default centos bootstrap user to cloud-user","message":"Change default centos bootstrap user to cloud-user\n\nThis matches the default user in CentOS GenericCloud images.\n\nChange-Id: I3a4279178a39a4ad9f834832f3616acaff092212\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0899f211cf3b1e834b342eafecdd19b1a9370dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0899f211cf3b1e834b342eafecdd19b1a9370dc"}]},"branch":"refs/heads/master"},"d6273899b4c7c6660ef16199a1acbe2e4a85201c":{"kind":"REWORK","_number":3,"created":"2025-09-19 12:38:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/08/961608/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/08/961608/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/961608/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/961608/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/961608/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/08/961608/3"}}},"commit":{"parents":[{"commit":"44163326edf0ed33e801e9fd9644f38258225b1e","subject":"Fix disabling NetworkManager DNS management on Rocky Linux 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44163326edf0ed33e801e9fd9644f38258225b1e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-19 12:34:02.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-19 12:34:02.000000000","tz":120},"subject":"Change default centos bootstrap user to cloud-user","message":"Change default centos bootstrap user to cloud-user\n\nThis matches the default user in CentOS GenericCloud images.\n\nChange-Id: I3a4279178a39a4ad9f834832f3616acaff092212\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6273899b4c7c6660ef16199a1acbe2e4a85201c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6273899b4c7c6660ef16199a1acbe2e4a85201c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
