)]}'
{"id":"openstack%2Fkayobe~961832","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I3a4279178a39a4ad9f834832f3616acaff092212","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-09-25 14:15:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a4279178a39a4ad9f834832f3616acaff092212","subject":"Change default centos bootstrap user to cloud-user","status":"MERGED","created":"2025-09-19 19:27:49.000000000","updated":"2025-09-25 14:16:34.000000000","submitted":"2025-09-25 14:15:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961832","cherry_pick_of_change":961608,"cherry_pick_of_patch_set":3,"meta_rev_id":"80a69229a5b8b903ca78d2e2717ab380ff627d34","_number":961832,"virtual_id_number":961832,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-25 14:15:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2025-09-24 10:35:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-09-19 21:07:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-23 13:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2025-09-24 10:35:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-19 21:07:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-19 21:49:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-23 13:02:43.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-09-24 10:35:00.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-09-25 09:35:50.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"}],"messages":[{"id":"75e4da61e6704f7cee0338f099dd1c262e15ff8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-19 19:27:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"07d0e134bbc262119a91cb55909f5ba600f0535a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-19 21:07:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"66d7de9105dc7ff946512172d81dea9d0f763dcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 21:49:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/131ce197a8c84866accd59f1a19e2f67\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dca4dea025c4aeaafae58a0655dd2f1 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb05a77c5ed845939aba21f34ba1f9e1 : SUCCESS in 5m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88070a7f69174bada226179f3dc3080f : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d41893a7e0649d6bfa7fa7ac82bae42 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01ec76821d0f430eab3f8ffcc2ce3b39 : SUCCESS in 3m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fc5ff8dc077e46c3818738dac464d663 : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/57b1d40065604479899c05717e4e511b : SUCCESS in 10m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/36ed644e6ff34cce9bfbe868fc77b6b7 : SUCCESS in 9m 02s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6abdb54e1a0a4575b458050284c8866b : SUCCESS in 1h 02m 10s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/76e9dd4080b94ebda46652f886ce7e53 : SUCCESS in 48m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1e037ba94e247eca78901b227b87b26 : SUCCESS in 35m 26s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9192c23e3c324fdaa9e4b41ea86469e7 : SUCCESS in 1h 08m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f35891e3a9f3442a9a6ab88bafd30009 : SUCCESS in 1h 04m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8aaf7bdfa8b549479576023a10c79b8b : SUCCESS in 22m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a8a6be129ec4c21a043c7fccbd10234 : SUCCESS in 14m 10s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0f1a569dc1664d438cedfc58f9f1a9f7 : SUCCESS in 2h 12m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24e29f8cbc7b42658612bdd64f6b7c9b : SUCCESS in 2h 01m 00s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f1cf070c03624747a50d8b0c6072906c : SUCCESS in 2h 17m 05s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82998147102540c28aea8078ab985e72 : SUCCESS in 2h 17m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8c00f2ef0a1c4f92928551c489cd921e : SUCCESS in 19m 29s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/73a17068d92b4b1588bf99441e7c4b07 : SUCCESS in 15m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c8e88726f444744a49255744eacd10c : SUCCESS in 15m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6e5cd1d2a12c4dd591244e43f4682357 : SUCCESS in 17m 53s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/aa4732e4ca104157a23fe051a690d648 : SUCCESS in 32m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e2157a52fca4796a9e3c5c21b81b932 : SUCCESS in 33m 59s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/edb02277eb9740cb9fba6428c61ea8e1 : SUCCESS in 34m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1bc921af120c4f32b08670e39094ca23 : SUCCESS in 13m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3478876cbf749839cab757b65b2fc3c : SUCCESS in 11m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ed893019b6794130b1b47bf97addee39 : SUCCESS in 12m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/120de1b8a283480ab513e3fb76efde9b : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5f731e177fa2409cbbfacb73950d2d19a77c7f67","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-09-23 13:02:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2631771683c9d29adc8b47e5723c30a5390e9587","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-09-24 10:35:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e3ad0674cba0a425c045e12422b7baa2c54fb1ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 10:35:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39c6f08676412570776dfbae4c42cc5fcc18b8d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 13:08:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/85c0e1b2873f486b9cbff203e8085ca8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdc9c950a07a4f4fa9f6dbb3f5460fff : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52c32a2291f84942ac5c5160aa882e84 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25402ba8877b4e8fa9ccb9263a42e18c : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d116ac5be82248758e4bd37145b5c4c3 : SUCCESS in 3m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7c66d8ff7e3e4e1c8c9924b7b3d30e7c : SUCCESS in 3m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/16dbfedbf4b141e18bd24421a4e3fc7f : SUCCESS in 7m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5e0ee4f9579a4178ae29272348061e84 : SUCCESS in 11m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/33c872f8e2394ccb98d706d420774bab : SUCCESS in 1h 02m 58s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/e6c4b61478954991840fd6b3fc7417d6 : SUCCESS in 59m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cc6102044034ce0882fd22604382293 : SUCCESS in 50m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3822188fdae647b89456a969d7a54922 : SUCCESS in 57m 52s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/905b01dd5f7949f8b6c772975f23de1e : SUCCESS in 1h 05m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c3676f7ed7bd484080699238225c154a : SUCCESS in 16m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bdc2a51befc4df2a6a7dd8d1768eebc : SUCCESS in 16m 31s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/23ed2c758802408c933ad6d701f80293 : SUCCESS in 2h 27m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3a3e5669d5b4c18ad36191a52501dc0 : SUCCESS in 1h 56m 54s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1f3dc2f454294c81bc0e096c8953eefa : SUCCESS in 2h 05m 58s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6034bd5b59d04f879be72aa8869c605b : SUCCESS in 2h 04m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/16741f8e798f4757a9c9de2beae920cd : FAILURE in 51m 06s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/089efa5a92c44113aae86c0620f6bb68 : FAILURE in 50m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67498954f9244777a9fa5b31399440c0 : FAILURE in 46m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1d98ec747f334ee1b8822d65525b3e4f : SUCCESS in 13m 32s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4917040c889843508b37aca054d6d189 : SUCCESS in 27m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d37eccbb8cfb467daf5a45cf8fa45926 : FAILURE in 33m 04s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/06325e3bceb64c27b3dae1b3c7ea83a2 : TIMED_OUT in 1h 32m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/59617e6476a74ec3838a60e2d9518f2e : SUCCESS in 12m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d04e656f6ca6403d98d57d20d23e1583 : SUCCESS in 10m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/90d7f59fb8fa4f2781b2107431ea40b6 : SUCCESS in 11m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b0bb9d85d3e48ecb6e4d8e69303cec9 : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9d01e6d2592e42816cc6c8dc9f2a51e6e6829896","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-25 09:35:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2713a3a8e6b403ed97eefa9d356d78c70f305f78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 11:38:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef72aa16385d4f13bd60195ceb8981d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b65a97de4eba45d78cadfa0d9e741384 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aeb09e6e0374aadacba3e18f20538f1 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c0060528e54d4b1a9931381d50f23b28 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a1fcefb9d79400daca74bfd62c0c176 : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8205f2dd78b14843ac48e85ae9151e21 : SUCCESS in 4m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bf2e10e7eb5d4ac787837a641d5107d5 : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/35f817e70c41451592c8bdb41df48019 : SUCCESS in 11m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7b246bad5dc54ed38dee24685daa9222 : SUCCESS in 10m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c04877d4682542499266892a6f4b3a2e : SUCCESS in 1h 10m 02s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/931374cbf6a54fd9ae785d7cdef7a4a7 : SUCCESS in 1h 03m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2f3c7a1fc7a457eaa4d1c6e9e2d873f : SUCCESS in 59m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6ce9eb7799c04bb1b03fc7cc02162a10 : SUCCESS in 53m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/717cb6dd48d14fbeb340e33233e44bcd : SUCCESS in 1h 07m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4636d0c790874ac0b744ea11acde4eaa : SUCCESS in 18m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee1da41690534273b1b51ded7f011927 : SUCCESS in 12m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/325660edcd764f87820d64b558cdccc7 : SUCCESS in 1h 14m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9473053ec5824911964495e3a8f229bd : SUCCESS in 2h 00m 32s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/81c23c5247324c1488b42665f32573c8 : SUCCESS in 1h 15m 31s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c450812bb0a4577af2c841ac7ffe90a : SUCCESS in 1h 40m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c13b3b7e78824189a072332225176aba : SUCCESS in 19m 46s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/8f402d6e846f47aab4969ff19643b639 : SUCCESS in 16m 01s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02aae09351ae4db88ac061d84391c2f8 : SUCCESS in 18m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e020d658dda94761b836049f627d0f91 : SUCCESS in 17m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fc0f8406acb4453db21e84eae5d5b563 : SUCCESS in 38m 18s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4d810ae8fdb44eabe68e7882f25d5c0 : SUCCESS in 38m 41s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0ff7a8bc95f94102919e007c0a9ec634 : SUCCESS in 34m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ffd5daad4f644d198f426a685c22509b : SUCCESS in 12m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f8a4bf9f3ff4cb1afae76630bce68fe : SUCCESS in 11m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/791580f4f9744c979077e56b8ed7c161 : SUCCESS in 9m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b4c641322054ec7884504841a6bc006 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9e375f14552e947385810f96d98cedc20672f74b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 11:38:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"690daae1dc9d1f4ff6c6994ca7a6d40699e3d556","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 14:15:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/466597355c144c5a988ee861b5aaeb42\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9a998816580431f97ee81033a228ba4 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7aef2ac778024df691ef9c6a36a45e5e : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ff8c9d7d7ef46268a4ffb9bc6589767 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c747d15a948d4e4ea47142f7e01550f1 : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca52dc94a504460c83434285b17646aa : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e5c923ee84ed4975bb497a4223d52214 : SUCCESS in 5m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/36293320de8c487aa8a28ce4c415003f : SUCCESS in 9m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/50c2f43a760b407a99024e340db6b195 : SUCCESS in 39m 31s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/201a19426ee34268b184a7f2f57f910d : SUCCESS in 1h 01m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4e5c8379c2a459a9789d33d3f79505f : SUCCESS in 57m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c780a0747d4a4b1a998c90ec6cd111f4 : SUCCESS in 55m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/27dea6e8cea0412393879f717989940a : SUCCESS in 1h 00m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/17c4e69bcd8f41aca13416cd885d1dc9 : SUCCESS in 19m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc7f502eea8343c38b01adebe2fb69d4 : SUCCESS in 6m 39s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b0687a2e027f4131875f349244668983 : SUCCESS in 2h 32m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/956d3e5f09e34f29959e804ea9ce09d1 : SUCCESS in 1h 11m 09s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/49b1a852176b494cafc0cd96b112271f : SUCCESS in 1h 40m 07s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/555806d684ec41dbbd86d531bab800db : SUCCESS in 1h 28m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/254250f765594ea8a2ce1cdd65766e82 : SUCCESS in 10m 48s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/3c5ac2ebcc83484d9105d843e47fad7e : SUCCESS in 16m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/130740d3996a4defb16e6318718543c1 : SUCCESS in 14m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ab5d67a657c54b068e870052acbb8491 : SUCCESS in 15m 05s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/605e4d2074274327b1edfb2b7f5691e2 : SUCCESS in 38m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04819dbb3b50441b8b658431945e2b50 : SUCCESS in 31m 40s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ff367328a9cf4a3dbce9cc89c88bf2f8 : SUCCESS in 31m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3eb724dc3818451aae26f07aa042ef81 : SUCCESS in 13m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b349ac214174fb1a03e79ca252a3f2d : SUCCESS in 13m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/948dda2f2fac4ce58d64e6886b9789ab : SUCCESS in 12m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0f76a302f644b8182da50f1769457c9 : SUCCESS in 11m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fdbecae222d261baff59548b572ab6f5841dd8a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 14:15:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"80a69229a5b8b903ca78d2e2717ab380ff627d34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 14:16:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c624c868a8d4c9c9b677b11c3a642b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cccb65bdd08240bb83255b6a42c9aa78 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fb4de1fcf274f3c932f35afadeb4849 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"edc2ed899ee43a7febd3246f65e89ca088613e99","revisions":{"edc2ed899ee43a7febd3246f65e89ca088613e99":{"kind":"REWORK","_number":1,"created":"2025-09-19 19:27:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/32/961832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/961832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/961832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/961832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/961832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/961832/1"}}},"commit":{"parents":[{"commit":"94302ba8f73d7a1637ac4c066c07285a163d1aed","subject":"Fix disabling NetworkManager DNS management on Rocky Linux 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94302ba8f73d7a1637ac4c066c07285a163d1aed"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-19 12:34:02.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-19 19:27:49.000000000","tz":0},"subject":"Change default centos bootstrap user to cloud-user","message":"Change default centos bootstrap user to cloud-user\n\nThis matches the default user in CentOS GenericCloud images.\n\nChange-Id: I3a4279178a39a4ad9f834832f3616acaff092212\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit d6273899b4c7c6660ef16199a1acbe2e4a85201c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/edc2ed899ee43a7febd3246f65e89ca088613e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/edc2ed899ee43a7febd3246f65e89ca088613e99"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
