)]}'
{"id":"openstack%2Fkayobe~961835","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~I3a4279178a39a4ad9f834832f3616acaff092212","project":"openstack/kayobe","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-09-24 13:12:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a4279178a39a4ad9f834832f3616acaff092212","subject":"Change default centos bootstrap user to cloud-user","status":"MERGED","created":"2025-09-19 19:47:14.000000000","updated":"2025-09-24 13:13:37.000000000","submitted":"2025-09-24 13:12:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961835","cherry_pick_of_change":961608,"cherry_pick_of_patch_set":3,"meta_rev_id":"f309decd9bdc0ba696b8c6f906a9fc66d14aef4f","_number":961835,"virtual_id_number":961835,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-24 13:12:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2025-09-24 10:35:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-09-19 21:07:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-23 13:03:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2025-09-24 10:35:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-19 21:07:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-19 21:45:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-23 13:03:21.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-09-24 10:35:20.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"dd554b3174729989c10fcdf26d5c9ab911288746","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-19 19:47:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a7516293bae4d0a80d3bf5efacc6f92af3e21439","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-19 21:07:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e8021ffb6f966f3ef80d5093cd94da27fa716650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 21:45:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/068197041d9d4800a8563e86770d543c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f0cc4ebc1d642dbb490b12e1a17a95f : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a92f625a10874342a4fd24f001a362a7 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bbdae907841e46ef886e7af2a434ec7d : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/150ec601394c4cfbba87c39ff0c6456e : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a252493d1014c979402ad887c1cdacc : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a31a4866cfa4bd7a1a502482267be2b : SUCCESS in 4m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ed2bb88d1a5045f4ba15b41aa35f98ea : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ed54c7d1915249eeb95bfc9eef8aae95 : SUCCESS in 10m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/94ab813724ee4c848ae8e81c0969b654 : SUCCESS in 9m 41s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/54eb0ea8334545d4af395164f7765697 : SUCCESS in 1h 03m 31s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9f34c817d9544826ac96fae8c53fce60 : SUCCESS in 50m 54s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/497cd96de3954c8da95346b09ab155a9 : SUCCESS in 57m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bd9e8ae8a11048f9944aadfe0707e4a9 : SUCCESS in 1h 02m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/072cf27ecd8349fb9c36ed2a29433d88 : SUCCESS in 17m 33s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/74da6a62e6464116b012f744196fc721 : SUCCESS in 19m 17s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/694a7d2ce5514f518884aa267325f7a8 : SUCCESS in 16m 20s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7a4b129921f04fd38ced75d0f65a9d36 : SUCCESS in 1h 53m 05s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c76330b2bcb40e88d09370473581671 : SUCCESS in 1h 35m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/19ce161643ec4a0395e6a687427f54bd : SUCCESS in 18m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0efe5b6847fb4fa5931585c666ebc5bb : SUCCESS in 19m 57s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d5d241ceeec0421bbc13d06607f81f87 : SUCCESS in 34m 07s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4299d90ba95f4d1c82129e6d6cb78350 : SUCCESS in 23m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4cf936593cdd4317ab91c998fa24fa3f : SUCCESS in 20m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48072487bb60483fab844e17180a8edf : SUCCESS in 14m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f6f8dff8eef14b53b56b36d2bfa19d85 : SUCCESS in 15m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bdd6e7576e74927b1a3581dbb74e4c6 : SUCCESS in 10m 35s","accounts_in_message":[],"_revision_number":1},{"id":"20db0c036ae668614e4f3eef5ff7bf36e167047e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-09-23 13:03:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"29311b322749d166279e0e0948418a5009bf0090","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-09-24 10:35:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dc59199f6a55ad6b4eae5e4d5dee3eafccc73c07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 10:36:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b60e0d7cc9b697221fe647c7d4a142b9d481af0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 13:12:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3930311f2fb2474bba04c3ee1e2a3099\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4264b0abf6ec46e18d288283620625cc : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb6527927fed4f8ba428fbe27657120f : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e01e36da5303451e816118243f70478a : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2007644b28e41bc832c5fc437288b8b : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/26832f1b6649447fb7cfe5e44cc551c1 : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b60c9335f1e94afda70c12f54caabe81 : SUCCESS in 10m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c379b2f93efc49059b59eff7203965d3 : SUCCESS in 12m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c4fa6e9493f6499286c07b9c6b3a407a : SUCCESS in 54m 50s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d26ae1f83644337a794c71eb16b371d : SUCCESS in 57m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b30e0714c97840d99f75bcaafac42a50 : SUCCESS in 1h 10m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ea00f311564c4dc7ad734ab331ddfe9f : SUCCESS in 18m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db8fdc6b8ccc43d498ce89f908c6518a : SUCCESS in 7m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4c2073f53df4440cb68c04de347b1dd0 : SUCCESS in 1h 55m 46s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/242009d500e14c599db93c04ebfba802 : SUCCESS in 1h 57m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/05626f42c9144e4589fcbbebf0c8fc0d : SUCCESS in 22m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f76c825cf8ec4dc2ba7ca5b3d536d689 : SUCCESS in 12m 06s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/26de5396ae7344ef88de7feea2b48e04 : SUCCESS in 38m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b42d0c5ab6a46198e6e996521552c81 : SUCCESS in 17m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5ea37539cdbe42799a1e714c532738c1 : SUCCESS in 15m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/041c6d021fa54a39b0517d2e00148c6c : SUCCESS in 9m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b0856e8fb7484d0b9b75720d32adf78c : SUCCESS in 14m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7748888ac6104fceb23be7e488cb34d4 : SUCCESS in 13m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a1107c132aa230a31d02ee0419c915b400944840","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 13:12:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f309decd9bdc0ba696b8c6f906a9fc66d14aef4f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 13:13:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a123acb729f14f4fa056c009343f6276\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8921c3d01364ff9966f287e4a987917 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33882328f28b47aeae4b3f701bba4765 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed977f8400e5b1a398dc4c713e4a3c806fb47715","revisions":{"ed977f8400e5b1a398dc4c713e4a3c806fb47715":{"kind":"REWORK","_number":1,"created":"2025-09-19 19:47:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/35/961835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/35/961835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/961835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/961835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/961835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/35/961835/1"}}},"commit":{"parents":[{"commit":"602a47a7fff32236ced0d655d6653454222e8e15","subject":"Merge \"CI: Add jobs for infra VM testing with cloud image\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/602a47a7fff32236ced0d655d6653454222e8e15"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-19 12:34:02.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-19 19:47:14.000000000","tz":0},"subject":"Change default centos bootstrap user to cloud-user","message":"Change default centos bootstrap user to cloud-user\n\nThis matches the default user in CentOS GenericCloud images.\n\nChange-Id: I3a4279178a39a4ad9f834832f3616acaff092212\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit d6273899b4c7c6660ef16199a1acbe2e4a85201c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed977f8400e5b1a398dc4c713e4a3c806fb47715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed977f8400e5b1a398dc4c713e4a3c806fb47715"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
