)]}'
{"id":"openstack%2Fkayobe~961836","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~I3a4279178a39a4ad9f834832f3616acaff092212","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-09-25 14:15:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a4279178a39a4ad9f834832f3616acaff092212","subject":"Change default centos bootstrap user to cloud-user","status":"MERGED","created":"2025-09-19 19:55:06.000000000","updated":"2025-09-25 14:16:32.000000000","submitted":"2025-09-25 14:15:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961836","cherry_pick_of_change":961608,"cherry_pick_of_patch_set":3,"meta_rev_id":"e2567f041ffef5cde4362f8089d05da53bae8c82","_number":961836,"virtual_id_number":961836,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-25 14:15:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2025-09-24 10:35:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-09-19 21:07:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-23 13:03:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2025-09-24 10:35:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-19 21:07:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-19 22:23:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-23 13:03:48.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-09-24 10:35:53.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-09-25 09:35:02.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"}],"messages":[{"id":"b7f1785f4bc48362f907438192dd7cbc1d941f3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-19 19:55:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"948c1aaa3f1d6d1c7bc421137bae794616f0b519","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-19 21:07:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fc1dc3907700e6c2249af54d7073e01e017ad4bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 22:23:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd7ef094dee743ce888ed7d87e041c37\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5232d597a32a479287b99cc6cbd8a867 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a97ea088f7a4ed38434c651ff649db6 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56448f0b40d84dec9dc771df5afc9601 : SUCCESS in 4m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dbe762feeb7f4be885b26ea5f326a9d7 : SUCCESS in 3m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/af13a15fb4a74f7faa561d0acbcf2088 : SUCCESS in 11m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/60e1cd174dd048db9ac8c14400f87efe : SUCCESS in 9m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/89f80e442e8e4f4a89340148e555f133 : SUCCESS in 53m 15s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3c7e4de0bc3d41cebbc28a34a499c135 : SUCCESS in 47m 30s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a753da0a4b1e454b9b2fa2eead0212ea : SUCCESS in 59m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed8bdc5037034e6bb8a594fa101d0daf : SUCCESS in 35m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8bedce5e586345a6b67f00831c88f8d0 : SUCCESS in 1h 04m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/64aaab4f3ab74d778845883abc474519 : SUCCESS in 13m 59s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0d1a3d9e6d5640e3b2358babee9cba49 : SUCCESS in 21m 37s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5bf74ca3e9624ab7ae7704155e41a554 : SUCCESS in 11m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65c9dfaf0161454d96b4f7dbe6a546d7 : SUCCESS in 11m 45s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a878d3af50ef4bebbc12acd5d0fc8609 : SUCCESS in 2h 19m 52s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0dbf11c090d24247ab36a29adb370d83 : SUCCESS in 2h 11m 29s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1562e36fdc3e405184f83d9157b7307b : SUCCESS in 19m 16s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5d03229d6bc4d879f7cee631448c8ba : SUCCESS in 15m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/308c5e6b6bc54a0dad1dc16be774e337 : SUCCESS in 19m 59s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8c761715b10d4d4dbb74fe1d2ea8ca12 : SUCCESS in 28m 14s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa4c7290d664419781dc5ef83ec34f1c : SUCCESS in 25m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/23faa1bf89b642db8e39eac4df6589a5 : SUCCESS in 17m 00s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/8deaca06467a425b8e023ebda986c485 : SUCCESS in 16m 23s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5f549463e164f8899d2816535ca93e8 : SUCCESS in 14m 01s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/e8b302de93f6483797ac510a1ceaabc0 : SUCCESS in 14m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4354e7b970a340bb886a6fab4caadce1 : SUCCESS in 19m 35s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/a8ba6a8c75d44a9498a4d9e2037a889d : SUCCESS in 14m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6fd401db3d1c42a6b0bb0614a9757cb7 : SUCCESS in 17m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0c6e2ffaab24c6984c7ebf537a99565 : SUCCESS in 13m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c24b1266d2b9406587d4169a5e2feba4 : SUCCESS in 14m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ffeada1016b4e9599c4c8a0c72a5128 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b703570c63149bfb920646bd907b9ba : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4002c389c6de487dbaaf36198859fd2b : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/022a4531a7b24119a14a4941ee428662 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c407b2882529b1d75a3e632b8ca881f8b3171910","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-09-23 13:03:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9e1bfb7b1c4ba8bca1ab6a42287cbb9dff5823e3","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-09-24 10:35:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2b304ba58bad823341ece74a049965d861957fe4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 10:36:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ff265c3b492440d276461854de9bce131d7b64f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 13:12:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e917fb0221b4b3b9913aab94c763532\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c75b3fdbaae43fe9d3b80f8036a7fbe : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f8cb31a8e0142a2b85d349a39fa16b3 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e1cbc9a6d95f49fc81856215ce4eaa47 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6af0a2ebe5e84d01aa1d7f36e26247ac : SUCCESS in 8m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4ffea815d9f0402ba08c4aea2273d774 : SUCCESS in 8m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a50fb20004964149a02330117e033c94 : SUCCESS in 1h 01m 53s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9875a4cf605b4987bd5e47d4e051fe27 : SUCCESS in 56m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e606cca1282d4f4bbdbdf42ab35d156d : FAILURE in 1h 02m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/360c10bebeda4ed5a3bc258528283017 : SUCCESS in 1h 04m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9054739082f44faabc47deaf738b5330 : SUCCESS in 16m 47s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52ae35bbbb50499894c9062d316b6a3c : SUCCESS in 14m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43a0b01db4c445e789a0cb5236b97785 : SUCCESS in 15m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/485d256bf04048af8ce8faf7e9622033 : SUCCESS in 16m 36s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/caddc000885c44c2aa881778f97bae03 : SUCCESS in 12m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fb5116813474c92b3c2af41afa2cca5 : SUCCESS in 12m 30s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/adf953581d344967bc03d368c85f8535 : SUCCESS in 21m 26s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b2c824ebb154b478e1cd954f2d0d939 : SUCCESS in 19m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e3bc7071dc3f454c914901e42ff68738 : SUCCESS in 18m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/213748351d694ad59b4afe5bded59de5 : SUCCESS in 14m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d941e3d8bb74c1d8f6fd1ff65cda6fd : SUCCESS in 10m 38s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b8ecb20a09124de9a04f914ad71a59b2 : SUCCESS in 16m 58s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f52c0d1a0ace42f7811e1d0791955296 : SUCCESS in 13m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58b2d9eb74814fb1b57e44e194bf7672 : SUCCESS in 13m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2dd582821ea4a1deab66b0a75c7722431aa30537","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-25 09:35:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f44ff03059a0d179654258bb2083b3e035a640fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 11:38:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aa5230774d04e9eba7ee34465d2f1da\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f76ef6e70d194722a6285276264339d2 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/144de266a89049afb43ebd330bc8626c : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25cfaccafe004bce8c56d03f7fe886ab : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/af978589af22451eb56d5dcc01c32648 : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4e17c28a0c444afb895b0c5dff3a3386 : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/04fae3ef083c44e588601fe6ee607034 : SUCCESS in 9m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0a9b01dedf814ce1ae31fb6e8ff12b20 : SUCCESS in 38m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/88e3c50c70eb41dba7c69404820fef80 : FAILURE in 8m 44s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/636f2dccffe0407fb6b908617df3a573 : SUCCESS in 34m 22s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1dc21aa83e8f41a79d7c22f4a5436620 : SUCCESS in 1h 08m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/73b80b8afc204364bbd8a63f8a4a1e64 : SUCCESS in 41m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/323361a8fb21465e9e2f164805d5b8d8 : SUCCESS in 16m 31s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8a8ae36c508242a1aec63fba2558f4ba : FAILURE in 9m 41s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7efa94393f5a4dccb57fa6730a884179 : SUCCESS in 15m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c97d3cc6945c4f13a7b792f0e544a206 : SUCCESS in 16m 48s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ad60a65215604e0a8499e24fe290c564 : SUCCESS in 1h 58m 07s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af128502c2774ce6aa865119c2695b69 : SUCCESS in 1h 54m 47s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1f8a920ba12b4024913471616d676665 : SUCCESS in 19m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de28b19090174d8fbce90dbbb98bdd15 : SUCCESS in 16m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d73c933eb9034e9c8422b1544449fa17 : SUCCESS in 18m 15s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8a78a0e75a0a419d8e926257b995b24f : SUCCESS in 15m 06s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1aa6a363217f4b50b5707e7525582818 : SUCCESS in 12m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c74b0d253b864ff18d811c975f9ae83d : SUCCESS in 16m 37s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/1818aa9b567f4a26b3292edba825a0d5 : SUCCESS in 15m 58s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1a4165cadf141d9bae6546814a64fc5 : SUCCESS in 14m 30s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/4aabc9e970b147f997ada8df0999ba2b : SUCCESS in 14m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/672a8310287240c39f65a8ef637f8c36 : SUCCESS in 12m 32s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/0546c938fb034f45a096e94cdf9196d0 : SUCCESS in 9m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1e2b5d48e39943099778e87407f2dd94 : SUCCESS in 16m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/589342598fed43efb318b774739394d6 : SUCCESS in 14m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b165de9cc1041568059a6d2f83c7b25 : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12b8367d0f754e6fbff9fcefbb05248d : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a191cbafa1754349b0162ec7201ef54b : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27c9a6db9ffb42eebbc1f4d50b47817d : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23fe817053494c7dacfc658b1af12f0e : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"dcc3342d55b16b4316b8902369013cad02cb84cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 11:39:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d2cf0c3d84ca3a590653d059209c9cbba137795","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 14:15:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6221b41703ba4ec2b8b0d7117867688c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f1e2932ad784c418e8570415934ea01 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72cf08257e4148978a92ca2d4494da37 : SUCCESS in 3m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fa710da5bec54f6183d0617e28ded2e7 : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/578419ea0d924d28b9e735c4d8f0d553 : SUCCESS in 10m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9100ee0de93a4b4097fbad3d325120fc : SUCCESS in 9m 53s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/89ed789acec34b8eb27d42313612ecba : SUCCESS in 1h 06m 02s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/891ff581a87244c7ad90037962b70b42 : SUCCESS in 1h 03m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9ac4cad4b474587a8bae443e4a3e1c2 : SUCCESS in 54m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/47bbb6d2630d4ab99d056e5beec33a6b : SUCCESS in 1h 09m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4187745a94f5428ba80341d536600f7b : SUCCESS in 17m 11s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d291f8d43424672ab6286dce317d9fd : SUCCESS in 9m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba5e7be1bb9040fc81a79cc3197f136b : SUCCESS in 13m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/54e2585d5b5a4830ac9eed4cf0ca822c : SUCCESS in 19m 38s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b81c5b956da4d1b822eeb63a13baac2 : SUCCESS in 17m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09d6088958ef4a48b997b16470b44713 : SUCCESS in 11m 03s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/97b3819dd793410581cf4ec615785b6c : SUCCESS in 27m 37s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f1ae335e2a1401b8251cfa4aa59bea9 : SUCCESS in 24m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/891e4699014344ae9311bae8a8a6fd8f : SUCCESS in 16m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a89f2fd525ef43acbeb9bd4bf2afccb7 : SUCCESS in 14m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcb215ba9832406ba4d3a2451950c129 : SUCCESS in 18m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7db250944e5e493cbea207595f23ae6e : SUCCESS in 9m 59s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c06f0fe868e42f2b9c39094c7ca25a1 : SUCCESS in 13m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba12562723c64f3996b0618cb710415f : SUCCESS in 15m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c03a22b5249e559d40c5f9ccd6699d834a54843b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 14:15:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e2567f041ffef5cde4362f8089d05da53bae8c82","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 14:16:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ccde35f97614853bb2044cd322fb962\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ca63f65f22b4dd6b988e91ef3bcc749 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/164fcaaaa2d440be89292ba55b535838 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e1fbb42088296ddb4146866cb43da91a37389020","revisions":{"e1fbb42088296ddb4146866cb43da91a37389020":{"kind":"REWORK","_number":1,"created":"2025-09-19 19:55:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/36/961836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/961836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/961836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/961836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/961836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/961836/1"}}},"commit":{"parents":[{"commit":"b9d04d587d67359aad9e89b15a98fc114821a8a6","subject":"Fix kolla_docker_namespace comment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9d04d587d67359aad9e89b15a98fc114821a8a6"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-19 12:34:02.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-19 19:55:06.000000000","tz":0},"subject":"Change default centos bootstrap user to cloud-user","message":"Change default centos bootstrap user to cloud-user\n\nThis matches the default user in CentOS GenericCloud images.\n\nChange-Id: I3a4279178a39a4ad9f834832f3616acaff092212\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit d6273899b4c7c6660ef16199a1acbe2e4a85201c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1fbb42088296ddb4146866cb43da91a37389020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1fbb42088296ddb4146866cb43da91a37389020"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
