)]}'
{"id":"openstack%2Fkayobe~962267","triplet_id":"openstack%2Fkayobe~master~I3ab49d28804733733474565d33cc7b564b203ba6","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-11-13 14:53:20.000000000","reason":"Change was submitted"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-11-13 14:53:20.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-11-13 13:07:15.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-11-13 14:53:20.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-11-12 08:46:13.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I3ab49d28804733733474565d33cc7b564b203ba6","subject":"Bump up Ansible supported versions to 11.x/12.x","status":"MERGED","created":"2025-09-25 10:43:34.000000000","updated":"2025-11-13 14:56:39.000000000","submitted":"2025-11-13 14:53:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"962267","meta_rev_id":"1c5d862508cc113939e573b2ca022b6b66f93c50","_number":962267,"virtual_id_number":962267,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-13 14:53:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-11-13 13:07:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2025-11-13 13:09:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-13 14:11:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-11-13 13:14:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-13 14:11:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-11-13 13:07:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-11-13 13:07:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-13 14:11:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-25 10:55:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-09-25 11:22:19.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-25 11:44:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-20 12:06:59.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"CC"},{"updated":"2025-11-12 14:20:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-11-12 15:59:32.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"3a88c554da8718709e2271a8e69d78ae95a5269f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-25 10:43:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89e67f2c6b25730135f271d5e3bb2d39e88447eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 11:44:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c6636eec745494db5ec41c644215c1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/745955e30c834ebab2c37e49b899bbf9 : SUCCESS in 9m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/173d92a2641d4632ae3e7f28e6123e68 : FAILURE in 6m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9fca8829c0b4734b5f3d7160720336c : FAILURE in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8aec3476da347cbb7213e317ec029e2 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/484611737ca5442a862ac772457476a6 : FAILURE in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee9640e1a1144b7e8ca475690b169dae : FAILURE in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/40173fdb5c59489ab3796c5c990951de : FAILURE in 5m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3915459fc6c248caa5e29f6a7f41ac1c : SUCCESS in 2m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a3c13dcbf7574e619dcf438a7767f4e2 : FAILURE in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/587f2ce9910a435ab1d80f87955957d5 : FAILURE in 5m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8e4b0ad7705a4e758e91ab2f1b4a77f4 : FAILURE in 6m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7df66b4d10a04ed9999ca2375deb00ea : FAILURE in 6m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/72ffe7f890ad4bf8afc40549662d3538 : POST_FAILURE in 8m 22s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5afd547228b8401891955b397402aa44 : FAILURE in 6m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d45280bf74f4f5bb78a2c1a74f850b8 : FAILURE in 7m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4deb9a4263bf433d8418f23b9698d3a0 : FAILURE in 59m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6404a627ffb4024a55603309d345207 : FAILURE in 7m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e9853839297349b68b77825f07a9a449 : FAILURE in 7m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3c9b263ae4044a1a0a04c8d502520ef : FAILURE in 24m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/89fb81256bdf48e5a1b20c9a1b3d6f07 : FAILURE in 10m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45319e4b057341488695317f6173be23 : POST_FAILURE in 11m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0a4eb626ffc44ca29e760dc17c9e7c01 : POST_FAILURE in 13m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a4b01dc3a4c439fa105d1b71f5364d0 : FAILURE in 7m 53s\n\nWarning:\n  Comments left for invalid file ansible/kolla-ansible.yml\n  Comments left for invalid file ansible/roles/kolla-ansible/tasks/install.yml\n  Comments left for invalid file ansible/roles/nclu-switch/tasks/main.yml\n  Comments left for invalid file ansible/seed-manage-containers.yml","accounts_in_message":[],"_revision_number":1},{"id":"ebacb758c5d40e6a5ad665765b76811525519c92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-29 15:20:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"90067aab80271a1b86b0ba90680389608589231a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 16:28:02.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72378b04c2a145f3958ac7ee33cf6332\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7b74ba62a254a5799d9e058e228863f : SUCCESS in 10m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8675492b5f534a5490d98fb0af5870a4 : FAILURE in 6m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/567599c148974b45a7275d2d8f67c940 : FAILURE in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b429e1239c9480c8523f06e77df33c2 : FAILURE in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25803a33c6d340128780825b8b8c889b : FAILURE in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98302bf6954d4cd8bd457cbb7e450680 : FAILURE in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2c783eac86345b1bbe711484b7369bb : FAILURE in 7m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f767124c4441427e9883384b1aa2950b : SUCCESS in 6m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3bdf3224d96347bf8ef36380a161fb8c : FAILURE in 8m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5122a689d4ba4487bfb557db368269ed : FAILURE in 10m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c37119f5bd044cc1bb875f57c0cbb62c : FAILURE in 6m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd0a530d7cff42368eac8ee855f06639 : RETRY_LIMIT in 3m 45s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b8c40db917544701bba3a4c3fdd38d19 : RETRY_LIMIT in 5m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3cb54754853747068c2d23de4f6c787e : RETRY_LIMIT in 5m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02207ec2fa0b49a1a3c9ba2d52e4473b : RETRY_LIMIT in 3m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f601f4698404484e8533079c4a55d5de : FAILURE in 1h 01m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3804259374fb4461879e16002e266763 : FAILURE in 6m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a5ab3bd3419d4467bbe662b420cfe0e9 : FAILURE in 8m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78369645863e47128ea6a6346f26c524 : FAILURE in 17m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c3de4ede7ebd4daaabbed701ef035576 : FAILURE in 6m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e92d9273c9334d1bbb1e2e3bbd4a585c : POST_FAILURE in 7m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d27e4a41540f466ab612f255c547ea45 : POST_FAILURE in 7m 27s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d57fc65c35804af694778c5ac7c9694b : FAILURE in 5m 52s\n\nWarning:\n  Comments left for invalid file ansible/kolla-ansible.yml\n  Comments left for invalid file ansible/roles/kolla-ansible/tasks/install.yml\n  Comments left for invalid file ansible/roles/nclu-switch/tasks/main.yml\n  Comments left for invalid file ansible/seed-manage-containers.yml","accounts_in_message":[],"_revision_number":2},{"id":"75fce76b4c6c3f0aecfd1f842f9929f620169f8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-29 16:48:53.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7edf43e133f5d53eb78d8fd66fc46b412c9c4a8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-29 17:16:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3416438886c353b713bbfabb5e455dd902ecc8d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 17:16:48.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][0][\u0027openstack-tox-py310\u0027]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      queue: kolla\n      templates:\n        - ansible-role-jobs\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n  ...\n\n  in \"openstack/kayobe/zuul.d/project.yaml@master\", line 2","accounts_in_message":[],"_revision_number":4},{"id":"0c70aebad9e246f5d477c6202deb92dcee6570b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-10 16:56:57.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"297202aeb9e64086d1eff433f28bb3e7536756fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 16:57:15.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][0][\u0027openstack-tox-py310\u0027]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      queue: kolla\n      templates:\n        - ansible-role-jobs\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n  ...\n\n  in \"openstack/kayobe/zuul.d/project.yaml@master\", line 2","accounts_in_message":[],"_revision_number":5},{"id":"a97c17b83f5d86e61887f5e3dc4fb08802860625","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-10 16:59:42.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"58c91d45ce3df5978399185c39768fa05a16f097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 17:35:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0da62d13a89e4a3d870ab98f23047a48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6466522526f94f119f8957fa56487b25 : SUCCESS in 9m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ec3249bdf53047ba8d3d6b8c357742ef : FAILURE in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4036c17cf47d4863a7e835355e2a3f77 : FAILURE in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5a9b78b38794241a65422929eb19477 : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed1cbec9eba74cd78a45bf28f7db25d1 : FAILURE in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7add66dca33347959a57b09af81ad7ec : FAILURE in 5m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfd33722a0044e6ab9d2f9fe064e3a30 : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/797a0656cdc24181bf14e51422f82383 : FAILURE in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5177ce0af98443f29ff9ddf395dd362d : FAILURE in 9m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/12c1cd313bcc45468e758447d6f652d4 : FAILURE in 6m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/981c4585a702414392b70ffcda504a6c : POST_FAILURE in 9m 44s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e0b24e13570a49dab10493dc39646f4c : POST_FAILURE in 5m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d52ed0c440ae46a287061d809f8ff132 : FAILURE in 9m 29s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/815911122adf432b92f9633b75ea603a : FAILURE in 13m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/50eb98fdb36f4769b5153e3c19c99b11 : FAILURE in 10m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15a5fc0d5c264f45afa580fdc207bf63 : FAILURE in 5m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f5bc573ed0d4f819f64ca391eeacb7d : FAILURE in 5m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/955000a90245437a9d764803ea073411 : POST_FAILURE in 11m 25s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb8ff708be064bd194efd94834319b8b : FAILURE in 34m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/724b991ed80f46098c9c447a3a7f7a43 : FAILURE in 7m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d2d335b00ac04c04b028b75fb0aedcdc : FAILURE in 8m 07s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/990da928a5f6478ebfa47ec15a1bfa25 : FAILURE in 12m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dbb7faf03ba14fbd8bf373292b1f3b30 : FAILURE in 6m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/214d35bbe0994df8907c88f602488bd7 : FAILURE in 10m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f37be59bef854b39adf5e6e063c1fd36 : POST_FAILURE in 9m 49s\n\nWarning:\n  Comments left for invalid file ansible/kolla-ansible.yml\n  Comments left for invalid file ansible/roles/kolla-ansible/tasks/install.yml\n  Comments left for invalid file ansible/roles/nclu-switch/tasks/main.yml\n  Comments left for invalid file ansible/seed-manage-containers.yml","accounts_in_message":[],"_revision_number":6},{"id":"1a4d3ba09a9262f9a6dada9d618ab7ef6ae52916","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-13 18:18:47.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"62be9b113c15b21c47446a519426e454cd86a92c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 18:50:26.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef200120937945f591070faea32a835b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceccaf1a5745459fb39a77b0162a951d : SUCCESS in 5m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4b2d4a2fb5f64157962a161316f71713 : FAILURE in 3m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0aedffe5625f4cadbc79fb013b0a7ba0 : FAILURE in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07e34bf39a234190908e7ded9d4368fb : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e240fcc624248a280dd0e275f92dc58 : FAILURE in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe1f62969368404a9efe84dc677a8705 : FAILURE in 6m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb80367c390f4d1faee056cd6304da86 : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3b3a30b675ee4a97838f77891b842314 : FAILURE in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3c14f3d0ecc74c038164774073193e30 : FAILURE in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c2ff1fa284fe45b7bb6e37e7417aa558 : FAILURE in 6m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8c136b53ccc143e5aaf5637cc8a9e828 : POST_FAILURE in 9m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e98fe449b7884ba3bdb1e6691129a643 : POST_FAILURE in 10m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cd30c18c2a04e278265f5e430a34fe0 : FAILURE in 9m 20s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e495942a15654407a98721ea50366df6 : SUCCESS in 21m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/56bd0c72e66442c286b484be078aa0c1 : SUCCESS in 10m 09s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de98b7f366c5434db3feec54eb8e9987 : SUCCESS in 18m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cf9b76ebc4c422699bdc850e1223aae : FAILURE in 8m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/52c6d8dc1c0e4042a7a99e6cb41dd9ce : POST_FAILURE in 6m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3ac852e69e24ef5bf11882091ccd5f8 : FAILURE in 30m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bc43576f70e412e9933ba7c78adfaeb : FAILURE in 10m 00s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5604853fdf804508abe0c9d8d0a03a3b : FAILURE in 10m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a9b9aa07a244f1e9cc221d313293e94 : FAILURE in 19m 27s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bc7597b769664c5d8f302fcde2b95ddf : FAILURE in 13m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e84e59958bf74183b348729ac2e0ada6 : FAILURE in 11m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bfd177f0eb345f3b72ead8a5961607c : POST_FAILURE in 9m 18s\n\nWarning:\n  Comments left for invalid file ansible/kolla-ansible.yml\n  Comments left for invalid file ansible/roles/kolla-ansible/tasks/install.yml\n  Comments left for invalid file ansible/roles/nclu-switch/tasks/main.yml\n  Comments left for invalid file ansible/seed-manage-containers.yml","accounts_in_message":[],"_revision_number":7},{"id":"29c40ca783dfc7045131e0051404fa410fcf7412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-17 17:51:24.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"426d9c49e9fafe7f44cbc2d7b177a9b0be80aa27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 19:27:40.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/888b2adc69744210917e31086069fa4d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5645fdd482e947568cb425ecf256f57d : SUCCESS in 9m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aea205539cda41859d100e0e32759a7a : FAILURE in 6m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b9901cbaf194ae6a65348e1c6c0f79d : FAILURE in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c45f95e226846a6b82c623329741151 : FAILURE in 5m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7860b36e84484679b47e80837115fe3c : FAILURE in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01352aba4a3948d29b6cd0860f44f851 : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8ad13728ca7746dca21de2e7824c4cac : FAILURE in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7948bbc73e3c448193431a196ee69f3b : SUCCESS in 11m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7815740d2e234903b1e46a401872ce7f : FAILURE in 7m 03s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/52334c631e60429d8c5a4e650db0210e : POST_FAILURE in 7m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fcb4e8d7433440d49fd571a00195ad00 : POST_FAILURE in 6m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c5055c7071a4a5dad539a479aaede1d : FAILURE in 7m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9d6c8cfde6524309b87faed4a7efc957 : FAILURE in 10m 24s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5560ee30f25647638f40572bbc347fd7 : FAILURE in 8m 12s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6cff0f256a7e485db63e7f9b80594640 : FAILURE in 7m 39s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c3f1976c2c0f4f86997c1886fa4f745b : POST_FAILURE in 9m 45s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2626554f91ed4e008ec88fddf7455b41 : FAILURE in 9m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0a7b680a438f4291947671543a53c7e4 : FAILURE in 7m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e0c7c66d07a4ee8834b8b0fff63b259 : FAILURE in 8m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aa3ecd0806f24f98b85ffec64cb19c65 : POST_FAILURE in 7m 36s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cd956b5425b438fbc710d8b403a49be : SUCCESS in 1h 31m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5132916797734207a25d020df8f2ad61 : FAILURE in 10m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0bb2400bc00e448ba312b4efcede7d6e : FAILURE in 7m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eed489f6e4184fd494ddb25a0221204c : FAILURE in 7m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/65e255da33604f609940b8dbe49cd4f3 : FAILURE in 7m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2184bb2226c245078e37cc1382c7be59 : SUCCESS in 32m 56s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a66d66e2129e48419dfc9e72a293cd65 : FAILURE in 3m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/397b0eb996e54cb793004132064ceeff : FAILURE in 7m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b64f1322d7f4d84be70630aee381073 : POST_FAILURE in 8m 43s\n\nWarning:\n  Comments left for invalid file ansible/kolla-ansible.yml\n  Comments left for invalid file ansible/roles/kolla-ansible/tasks/install.yml\n  Comments left for invalid file ansible/roles/ssh-known-host/tasks/main.yml\n  Comments left for invalid file ansible/seed-manage-containers.yml\n  Comments left for invalid file kayobe/plugins/action/kolla_ansible_host_vars.py","accounts_in_message":[],"_revision_number":8},{"id":"cf0263f9b6c95269f5dfd0742255ce374df7ae41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-20 08:50:55.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"d9c8568ccab01d279ef1498fbc40aac3fd862d71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 09:08:58.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f48a2894d11e4948b62dba397c2888ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10eca6a067924d748ae11369a8f0a729 : SUCCESS in 8m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/383dc8eaf5994098b7f45e5b20442669 : FAILURE in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5182cf98bfa149c2ac5920290a17c56c : FAILURE in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9643f3832b324482a28a61b3bfd8c9d2 : FAILURE in 2m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97fbacbe46b34aeaab79c62bbe1d24a9 : FAILURE in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75aa30fd7cf346e0ac4dd48d99a77736 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/85cf882eb74c4db29265ab844e39e3b7 : FAILURE in 2m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1cb0646e56614cf99128bf74ce422cf4 : SUCCESS in 12m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5956bbe270264cac875160564d9cdfb8 : FAILURE in 7m 08s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0027ea135a3d4422ba85d0a5262f2fa5 : POST_FAILURE in 5m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/aec5380538a247b09cd21664c86eef34 : POST_FAILURE in 8m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/313da568737b415c8e874971e96f105d : FAILURE in 7m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5cbe725d83e04ef8ba8702c44bf27bce : FAILURE in 4m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c4c30d55c6974c62a0591ea7e789d873 : FAILURE in 7m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cb7774bc233452387b5dbfbad18d879 : FAILURE in 10m 28s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/012aeef7b28845fc98c76fef31c91913 : POST_FAILURE in 9m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c4e0ea4eae9c46e0adef34001226d12f : FAILURE in 6m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e77bad9033bb4a01bfd015734076f8bd : FAILURE in 5m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f268a54e47084a57923dfaf85da47a1c : FAILURE in 8m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d58f3ab2cce24068bb625f353f746e53 : POST_FAILURE in 7m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/891978bdfc6f42e08a1f47c9be4b8cc6 : FAILURE in 12m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/adfa4f53c1ad4645a9807175181af2a4 : FAILURE in 4m 39s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5ef0101277574c33b573a18913ec1852 : FAILURE in 4m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85df4bc61db44c2586be2476fef821ff : FAILURE in 7m 41s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ea9569df7677484593a1618b7c3b57c2 : FAILURE in 8m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2be16bef6ea94cc89891925662afce67 : FAILURE in 12m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c2929852f09747cc97a1a158b8ff254e : FAILURE in 6m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ebb0eb05b23741ff8a147670dac6005f : FAILURE in 8m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d661b1034a524eb6a10b6d9066140b3d : POST_FAILURE in 5m 59s\n\nWarning:\n  Comments left for invalid file ansible/kolla-ansible.yml\n  Comments left for invalid file ansible/roles/kolla-ansible/tasks/install.yml\n  Comments left for invalid file ansible/roles/ssh-known-host/tasks/main.yml\n  Comments left for invalid file ansible/seed-manage-containers.yml\n  Comments left for invalid file kayobe/plugins/action/kolla_ansible_host_vars.py","accounts_in_message":[],"_revision_number":9},{"id":"cd24f55622323ca26a566781f133989d7b132b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-20 09:29:49.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"3a90b7ac0c509f68610a197f62cdf8d013ab3aa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-20 10:59:24.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"a48fee2f55abb4f116a642d848dc5f85c443f28c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-20 11:29:42.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"2c0dc8c829e959aef15e02e61fc4d8cac7c8571e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-20 12:06:59.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"f6738db69f9087f9199506dcfb31d398405f99c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 13:43:48.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/285cb00406cb47e99b69a29510005948\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9637c03374f041839bc1ca2a342d8db4 : SUCCESS in 9m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fa879c77b460469c8b2ee308e6ebfbc8 : FAILURE in 4m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0818219b51b746b487b4af36b2bde89d : FAILURE in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daa85b51f3eb43ca8974da9a3e4aca61 : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a4d2cc1c73854fbb9279a5e3c086b8f7 : FAILURE in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab7dc7ed94e6407592827238fc200cab : SUCCESS in 2m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/deb92e0bc4b44443af54501a2d41e10c : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/30d0489dfa7244aa858544a6de4ba2e1 : SUCCESS in 6m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e17a049358f14f78a015e6c7bffdbf30 : FAILURE in 6m 53s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/46b3e2740e9341a694798b7545b87da2 : POST_FAILURE in 9m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/25c7d0fe872e45be978b4ccf2e480daa : POST_FAILURE in 11m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f060a697e3c64d28ad3bf1bc1a2e3fd9 : FAILURE in 10m 00s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/47416297c69b4c1e8ee87ff573760c10 : SUCCESS in 10m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/796a06586534400b9f9d2b9a369e97a3 : SUCCESS in 19m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00f697b40a44454a8e3380497ad42f8a : SUCCESS in 18m 32s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3ae1198badee4b92adfd67912648950a : POST_FAILURE in 39m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ce85904f4d744a9890e5d79153a19381 : TIMED_OUT in 2h 01m 50s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e2a39d5c544949dd9bcf625139ec3b9e : FAILURE in 12m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82f85ba0344b45ff974c3acfbe7db9c7 : FAILURE in 15m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e2e5e00720db4b09842292c33bbe8392 : POST_FAILURE in 28m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a7fa4ebb4454a3da86977cb61dcbc15 : FAILURE in 31m 32s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/55a9d37f44c64226ad0a6a8367d9a07b : SUCCESS in 8m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/79b32e53e6bc4ec09fe978c7098c7de9 : SUCCESS in 8m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18b1e00ccb35473dada7d22001df410a : FAILURE in 11m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/627c45d7e6cb4f708f0e7fd4c44eac93 : SUCCESS in 9m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b09148955d0148b3adfb5c4fab71e0c4 : SUCCESS in 15m 28s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1e3d3b0b4b6844b99bad81180adf03da : FAILURE in 13m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1422e160debd462da6e90484e40ccd14 : FAILURE in 10m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48e5971b4f1748f6baa0671d166d0bc0 : POST_FAILURE in 9m 02s\n\nWarning:\n  Comments left for invalid file ansible/kolla-ansible.yml\n  Comments left for invalid file ansible/roles/kolla-ansible/tasks/install.yml\n  Comments left for invalid file ansible/seed-manage-containers.yml","accounts_in_message":[],"_revision_number":12},{"id":"72ef853a104bdcbe027c325ce86f7d76a2c61f8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-20 16:14:32.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"995a2798421e7462fbaa67635b61cd56506dfbbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-20 16:23:03.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"4fecec088bacc4bdd1697e94419f725b3bbd293d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-20 16:27:34.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"9c3868a332bf8909e452f13de18a5d059537e416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 18:34:02.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ea774e497c74eb38f6fa81c79861d1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54de9e53db5d4d2b840ab292cafcb7c8 : SUCCESS in 8m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f9ff2cece5934666a6a7bdddbc5499b6 : FAILURE in 6m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5d252c2649a4b68b518e7ace3a477cd : FAILURE in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dffb79a591d4d98bd2a837f85a1eec0 : FAILURE in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4bb610ec04c846cea9b714b6cf85c8ec : FAILURE in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bac8870aaaf74b209df905742c373dcb : SUCCESS in 2m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0abcf6d828434513800a555b42de35c3 : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e74f7c729afe47a3baaaa3885da38c6b : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/55b7d600f9ee428385c018cbd8c7b162 : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/aa92237873a74997a2164879ea38e9dd : POST_FAILURE in 10m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/37a1f87debe74d31bc2359a2469372dc : POST_FAILURE in 8m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec9dbcd5c15f4fb2a0a8cd9cd6c6d719 : FAILURE in 9m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a18a08411ae24011a33da0920fb3e718 : SUCCESS in 17m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b7baa1240ab4427f91e51f285d828071 : SUCCESS in 10m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/304eb10b3a314d4b812fdee206238b24 : SUCCESS in 8m 34s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3b80b5d0d23d40a7879374d6c7f85d34 : TIMED_OUT in 2h 01m 53s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9bd4cce1aaff41918aa4f068f12cf594 : FAILURE in 30m 25s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/cf0a3589363d48b990ee43ecbdbefde3 : FAILURE in 8m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54971a5c7bf840c3b3a2df44e87a76de : FAILURE in 12m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ead1e6f32b7e4c54927fa634077b0a6a : TIMED_OUT in 2h 01m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c08ef3193f7447c7bd239d9d984cdfde : FAILURE in 14m 46s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/37828e1493f7479b8e97b1e0da2d7d69 : SUCCESS in 9m 54s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/67bc1f93ee26455f82941d461ae30cb4 : SUCCESS in 9m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7179f8475da243f9b15db9641039812b : SUCCESS in 42m 06s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f619c95612654ef6bfb686d4922263ff : SUCCESS in 7m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cac56e49a0c2409b9e42c1fac0385945 : SUCCESS in 14m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2ba0fece8e8745a29f40aebfc21e7d49 : FAILURE in 13m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ad6229098c67489db309ebd90d3f2910 : FAILURE in 11m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21641a6cadeb4ba39d2beaa5bfe5ae0f : POST_FAILURE in 12m 01s\n\nWarning:\n  Comments left for invalid file ansible/kolla-ansible.yml\n  Comments left for invalid file ansible/roles/kolla-ansible/tasks/install.yml\n  Comments left for invalid file ansible/seed-manage-containers.yml\n  Comments left for invalid file kayobe/plugins/action/template_content.py","accounts_in_message":[],"_revision_number":14},{"id":"be9337cefb615c1b390900839fa1fe1ee6daa0df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-20 18:35:07.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"9ead9992525ff0f0a458dcf5f6b9e982edea404c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 19:35:49.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97ca74f032164b12a98a00dcfac4e0ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2412333ed8f74cdf8652d02f66d28626 : SUCCESS in 9m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0b41bea76d37406cba442ce5989fdc59 : FAILURE in 5m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f0843c6ad80492c96df6889cb086609 : FAILURE in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69e2dc0978034904be84967ad8fa7f4c : FAILURE in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/365f0e33e961465eb043201843823250 : FAILURE in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7d0255c193a4e47b09b3146957137e4 : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3cae8e0d69484350806f4bef1b258b37 : SUCCESS in 3m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/02f09efd5d8c471783e0942e57481309 : SUCCESS in 7m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b3a5d9e478b94ca4b7a06ddae179671f : SUCCESS in 13m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a6ac469f15394f9da0289424a38a7d63 : POST_FAILURE in 10m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a68f7a211b464dd898ff600800f64db5 : POST_FAILURE in 12m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb114eaf36614d8f9a4b271b1748143b : FAILURE in 6m 09s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f9d62b4d652045d5b4e2a48afa362bac : SUCCESS in 22m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f0260ab764d347cfb33cb9f22f6efecf : SUCCESS in 20m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52209dd0cea143a19a9931c58b121f9e : SUCCESS in 15m 21s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e157a76c1b9a47629ee52da098b8bb14 : POST_FAILURE in 42m 54s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/beaec04a192c45d497ea8c47ea478a9c : SUCCESS in 43m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7254365567f041fca685c3a30c521d93 : FAILURE in 12m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/673920c4cb974b879c834ac9f88aab6e : FAILURE in 11m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3cba8578255747a49408f894d5fb9d7d : POST_FAILURE in 40m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/472c9c784445470d8b31ab4310800fc3 : FAILURE in 15m 32s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3ee2e26f017646e59bb31f6d9462b19f : SUCCESS in 10m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6ed02d2b61624e488471717649a5e1d6 : SUCCESS in 10m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d0eec38133349109ee7fecede48c96f : FAILURE in 12m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f6114a1842a14ee585d1938fb6b3556c : SUCCESS in 55m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/077f067d374c47c2a92f80b40f8d306b : SUCCESS in 15m 22s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/75bf5bbfb5a746a4a46de6e152970254 : FAILURE in 11m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b61c014e6dff4525a846067a9d304c83 : FAILURE in 11m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24124247a82c46958770ef42b209246a : POST_FAILURE in 7m 30s\n\nWarning:\n  Comments left for invalid file ansible/kolla-ansible.yml\n  Comments left for invalid file ansible/roles/kolla-ansible/tasks/install.yml\n  Comments left for invalid file ansible/seed-manage-containers.yml\n  Comments left for invalid file kayobe/plugins/action/template_content.py","accounts_in_message":[],"_revision_number":15},{"id":"6dcc4ffbad2e62eb8d028d6c96e91537e66aa236","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-21 16:57:38.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"afe4149a13b96977581983f66b88b4ec8ed8cc1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 18:49:27.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9eb4962a38a4d79a5a8104ab5cd9497\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e124f6c5c9044ef28a61eb88bd6f90ac : SUCCESS in 9m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/65e904c9c490429a8b38883ce6c051bc : FAILURE in 4m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3da3e9416534edaac8d00748c9d2ad6 : FAILURE in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4f3be7bc7fa42beb19e43f2844c6350 : FAILURE in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6bdd50113d2d4b219f610fa99d40a3a8 : FAILURE in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfc8056c1cc244b4bfcd35ea090260e1 : SUCCESS in 2m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/62091e04df434602830c662cdef7c1a2 : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a33fe7380ad942278fb19acf1db10089 : SUCCESS in 11m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4777d1a633364ffda9df01e83c575097 : SUCCESS in 12m 25s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bff1848233a149b09da98149a41efa1c : POST_FAILURE in 6m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f56048ecf051485697b0e1752034fe17 : POST_FAILURE in 11m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d12e3cf571814e7787314360e63b8ad7 : FAILURE in 9m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8a6c53918e6d40398b858fb09602b8ec : SUCCESS in 16m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c9015b529d524f058873d6041ebc7f1d : SUCCESS in 15m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b3228b3016d4c99b896cf661fed90cb : SUCCESS in 10m 09s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/97ec250f2e2a4dc88d18639bf2e531ec : POST_FAILURE in 40m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/03da13ce770e42a09a8e129708125693 : SUCCESS in 28m 35s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/186673f9bdd04373b1c96474254066ac : FAILURE in 18m 18s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3dc2bc9c09a84df590631de335afa480 : SUCCESS in 52m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a2d5db254ad6462f8f671e567ffb14b3 : POST_FAILURE in 50m 55s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cd4f69d97474b5f8cc76fb53abe971a : SUCCESS in 1h 46m 09s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ef76455e11e44452ad033fa0c0b9effe : SUCCESS in 12m 18s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cdc0be4c66474dceb8b67d35432c2ed6 : SUCCESS in 9m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70f4b7cfa09e4a00a0f46695843848d0 : SUCCESS in 13m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8bd02c172a894b8d8d95297ec1f9f20e : SUCCESS in 9m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/694d63ccd4b64f5ab3a7bf535cb70323 : SUCCESS in 21m 14s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/704a45e6b45c47709eb5d2318e7aa22f : FAILURE in 9m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/70e9f55b9b1d4704b1572280b4d8c70a : FAILURE in 8m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49924f1b1b4e4f8eadeb820d6d958106 : POST_FAILURE in 9m 25s\n\nWarning:\n  Comments left for invalid file ansible/kolla-ansible.yml\n  Comments left for invalid file ansible/roles/kolla-ansible/tasks/install.yml\n  Comments left for invalid file ansible/seed-manage-containers.yml\n  Comments left for invalid file kayobe/plugins/action/template_content.py","accounts_in_message":[],"_revision_number":15},{"id":"f56a91f675169dc3672e2f55c2fe5a3dba8fad44","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-24 09:35:41.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"35509cd62d8aa575872f4dd93c6d9ef8ec7c80ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 11:47:10.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75168f8e11984d5d8b6dfbb08c296234\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ae49bcfa0f44cfca6ec65e62e218b23 : SUCCESS in 7m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d460ca91c5984984b9e7e5849be16e3e : FAILURE in 7m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7adca8b1ba3a406496f07b19c1f373d0 : FAILURE in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdf63728ffa3464098ba753149a120da : FAILURE in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/923b10d70bfc41339731c75410137f35 : FAILURE in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/104fe2681e1548fe8d8b0cc187a91ce5 : SUCCESS in 4m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dbdf005f25d540298e6322d4230f99eb : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bce396ab48a44630856f7e158b2d677f : SUCCESS in 11m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c8848eb8d73474192cd0ab5ac3b2503 : SUCCESS in 6m 42s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f4934cb25156444192974c4771055601 : POST_FAILURE in 14m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c3dccb655f304ba9b3ca77ac832860c8 : POST_FAILURE in 11m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44c014b45cb548ca9acffe03ad45ba3b : FAILURE in 9m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/763324d7307949699c293a7db9add48c : SUCCESS in 17m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/af8cc1a0b0a549d09046c38eb3b440a5 : SUCCESS in 9m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d0df382f6864dbba14670214b1f8444 : SUCCESS in 11m 52s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/db86cae08fe643cdb23b4ebbd4969e45 : POST_FAILURE in 26m 01s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/03ec090fdfcc481e8b45e4115effc2d4 : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1e6f6607c67048bf8d3deb2c048773af : FAILURE in 18m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9989470b50654e6aa433df8035ae00c6 : SUCCESS in 50m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3d8b22ba315f4035a99bf4c1748b97a9 : POST_FAILURE in 24m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a99f60f2c234513ab5f20a4cd46a1e2 : FAILURE in 1h 52m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7f76b33da5c34d1cab8d358690675654 : SUCCESS in 7m 54s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9f6eefa428144adcbfde4d52abc2d621 : SUCCESS in 8m 08s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5856040124484e50ad68d22450841810 : SUCCESS in 15m 36s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bf974eb0818e4d96bfe49446253458c3 : SUCCESS in 13m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b435ce15219145848a9007a160d1c51a : SUCCESS in 17m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3cfd6f51fe2746ca8e8313d087a11774 : FAILURE in 15m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/92758f75aeac479fb3bda264d829a3df : FAILURE in 7m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2944dba36b2439c8909fbded7c64514 : POST_FAILURE in 7m 23s\n\nWarning:\n  Comments left for invalid file ansible/kolla-ansible.yml\n  Comments left for invalid file ansible/roles/kolla-ansible/tasks/install.yml\n  Comments left for invalid file ansible/seed-manage-containers.yml\n  Comments left for invalid file kayobe/plugins/action/template_content.py","accounts_in_message":[],"_revision_number":15},{"id":"3e64dbded0674130c0a5a8dc922877529ecc9f12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-24 15:51:01.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"194abc684d0140e9092454d23baadd07ce8595a7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-24 16:30:19.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"a96693d650293482f53eb216de92cb054c7e898d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 17:29:20.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a866a54454940e7836de11329f5b543\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a1131815e57458e9eb387c888f7e02f : SUCCESS in 10m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/df8665802a734cecbf85b1c15327c379 : FAILURE in 8m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/213a1f9167054cca97701497a07c41d7 : FAILURE in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bdd9b9e7e7149ddae5d891782056f14 : FAILURE in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc6d2c063677469c8fff9a3c92d2a895 : FAILURE in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11f803cc920741ca83f3604762c3eb63 : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2521596509ea45ddabbce23a1e25fef5 : SUCCESS in 3m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f9daf9d46a11481ba609c82f86d93ed5 : SUCCESS in 7m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/15cf43e54db54678bd41a80717fc7649 : SUCCESS in 6m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/83b62ed9f8f041939cd0743ab90bef6c : FAILURE in 6m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a6391501ece74c39ba525016d5a1b43e : FAILURE in 13m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ab5f3363b564d2a8a8a245810bb503c : FAILURE in 8m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d677853b3e2845029841eab834f9a626 : SUCCESS in 18m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b8000e76482842af9cbc919b78af338b : SUCCESS in 10m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ab07d1459ad48f6b01bf06fb7dc4b0f : SUCCESS in 9m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/950bd0bd0da04d9abf860ba294d5bc71 : SUCCESS in 27m 14s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/24e7d13c264043bea7da5de7d3c19933 : SUCCESS in 46m 42s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/40040a3f96b044cdac6a3f4a630e6054 : FAILURE in 8m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc51ecf448954c2fafb6ef524fa57616 : SUCCESS in 50m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3847917595c44113854b141114fb9068 : SUCCESS in 25m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/946857cac99a42a8b284f04ea461484c : SUCCESS in 1h 32m 28s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/db3e7e3f0b3a4f3da38146b5595b40ab : SUCCESS in 8m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8c9647d0b2d1454c9d3b68310fe02dfa : SUCCESS in 18m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19cac086b7294344937d6d2c5e2dccd3 : SUCCESS in 16m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f0eb42f60bbf4c9c8228a5a096d991a1 : SUCCESS in 17m 17s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0283c8b79b144c4987f4d41d1bd22905 : SUCCESS in 30m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9df691887e1d472892a1599d24a0c5d3 : FAILURE in 16m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ce1442c3ea044343a710640718bac465 : FAILURE in 6m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f846a84c4eba46b59bb17cddf0195fe1 : FAILURE in 10m 38s\n\nWarning:\n  Comments left for invalid file ansible/kolla-ansible.yml\n  Comments left for invalid file ansible/roles/kolla-ansible/tasks/install.yml\n  Comments left for invalid file ansible/seed-manage-containers.yml\n  Comments left for invalid file kayobe/plugins/action/template_content.py","accounts_in_message":[],"_revision_number":16},{"id":"f5b2a047779a198046cc71a1776acf27ccb749e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-24 19:28:58.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"66cf5bbd1a7eaccc09405ee307434f6be294df30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 21:14:23.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5f2dec722bb4663b2a73626468f87e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad21fde7f5884936834911e085fc2ea2 : SUCCESS in 9m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/71602533451846619b0bf36309123a3b : FAILURE in 7m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42cf610286444578aca3f25e083b0df6 : FAILURE in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1420363ff3b94ca18729da99c88ffa7c : FAILURE in 5m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b31584e0ed942d4ad6ceba8b321807c : FAILURE in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a376733d9fea4fc88f96c466c1398f19 : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/24ad6d2fc4284d56bf4d30192ccd644e : SUCCESS in 7m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f7ac5cc06d1a4cb69a0eb4f4e4755dcd : SUCCESS in 11m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f0ee9ca451de46c1ad1568b2b91536c6 : SUCCESS in 11m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bbc21de13daa4d2ea68065312abc6c99 : FAILURE in 15m 44s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c6ddf165796542d7881abdc0d73ea070 : FAILURE in 14m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59409a024c564d43aa82580015201842 : FAILURE in 10m 23s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/def40636dc074c86913a5c4d56bf91b0 : SUCCESS in 23m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1f1a495426bd4851a1b84a17f8d8e971 : SUCCESS in 18m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/997c47b2927849b4b81c277796cfbf58 : SUCCESS in 16m 11s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/861136204d8c41b790d09d9669e7fdc7 : SUCCESS in 26m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/260bab2dbfc5404b88f5f769d5756e10 : SUCCESS in 28m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c617e72d7d4445f6a7643a5e1865e043 : SUCCESS in 1h 10m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9811b200980047bd942f3b09807ee6be : SUCCESS in 53m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8975d6aab4e041ee894b9eb9b4a601c6 : SUCCESS in 49m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f49e7d3ceb34de49fe037a4ff4881ba : SUCCESS in 1h 38m 59s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a85f928c0cc846d5aeb654a1aa0b3c6a : SUCCESS in 8m 48s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/038a267efdc349348762214d75f7beaa : SUCCESS in 13m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e768598c87684455acc705bba06ca561 : SUCCESS in 14m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ceb01fb1791c458285b3bf79660722c0 : SUCCESS in 14m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2282f0a3fd0d4738b7fd718b611ecf56 : FAILURE in 27m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7db3368889f3447eb1718cc3f347eb44 : FAILURE in 15m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/726f3b42cdf3450cbee9ff3272ba6a4e : FAILURE in 12m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e92ee72604904b4cbd6de48786300d7e : FAILURE in 12m 41s\n\nWarning:\n  Comments left for invalid file ansible/kolla-ansible.yml\n  Comments left for invalid file ansible/roles/kolla-ansible/tasks/install.yml\n  Comments left for invalid file kayobe/plugins/action/template_content.py","accounts_in_message":[],"_revision_number":17},{"id":"4d099b33ae53a9f6de8a0fa0a68c34b48ca92ca7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-07 10:59:59.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"af13817bcb7827ca049294828aa361b9d56023e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-07 18:52:47.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"993af1cb7b8698ca51702dbbdb3d63863ae53a91","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-07 18:54:46.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"6c626627a287da4a395b8001b95fea6909da87bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 20:49:18.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fa51c92c5f14bf4929d28043e958c71\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8c644659453418eb8e706623fbfe363 : SUCCESS in 9m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3ed00a92717d4508a64a718765fa8e51 : SUCCESS in 7m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0781e8a8f541421db8dffa7502f0e5ee : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4149e612a18b46e686be163e84a879dd : SUCCESS in 5m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f51f0277fba426d833ad04211b4d4e9 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30764deecbd140a190c6669b83ce023c : SUCCESS in 2m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/05bce6eb636444dc9a351bd5618a0cab : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/732f291a31fb4c488a7be037b29282db : SUCCESS in 8m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/733551e5a59c4b10b5342bb7fa578b23 : SUCCESS in 8m 54s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a27f47f0ae324cecac5b3a226b18f33d : FAILURE in 7m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c20183973fa9442a8d1aaffc3388ab91 : FAILURE in 13m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c39e789e052d4d859d6d5b6b2efaee3d : FAILURE in 9m 56s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7e1f2272348249138ccda315f613faa3 : SUCCESS in 20m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1c670048b8dd47c888936d0051b4c641 : SUCCESS in 24m 37s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63f5a63bb58c4f89a553408c92d70592 : SUCCESS in 16m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/eca942d83cf74621981e4dc3ccfb0622 : SUCCESS in 1h 01m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f8f63cce56c345f68ef980347239bef5 : SUCCESS in 1h 03m 28s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/bbffdf5dac4b4c6bb5e0e47723488898 : SUCCESS in 57m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e80c1b0a77f4252aa1125d3329c1c51 : SUCCESS in 27m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7ffc7616f2b5455e9db5fc88f8de2e4d : SUCCESS in 48m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d20f2d568077464c92a1ae7942629bf1 : SUCCESS in 1h 53m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/023fd19a119f49028897ba68d08d5d61 : SUCCESS in 11m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0c832d4222044c859374062a642a3f07 : SUCCESS in 19m 08s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1448c90f699d43b28860bd29a707e72c : SUCCESS in 9m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b01bfbee774b4cef8d24e4cdc2fc662c : SUCCESS in 9m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b6a59e3ffa44b308cc8b82f1b1fcdb7 : SUCCESS in 29m 23s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/708f6b2048bb471a87bb97a436774a92 : FAILURE in 16m 03s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/163a412d0b83414c9499f974f2110b46 : FAILURE in 15m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12956b3deadd46b2be7da1bd8e3e7b01 : FAILURE in 11m 48s","accounts_in_message":[],"_revision_number":18},{"id":"5b1c6e91605799f89b372ddf37833d9eae1b0889","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-07 21:48:54.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"7e6244b3daa05f49a768a7d35a0729cd64fa1337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 23:46:17.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9eff15606534a30bc8cf812b2fe8bae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb06cea0aea14c6f85684ce1e5ae2a66 : SUCCESS in 6m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5eeef1433e1346ccad413ce243d18279 : SUCCESS in 7m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3817bdd2f2e5456da951ced5cb42099e : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f457398b5b2d40a285243c71264760d8 : SUCCESS in 8m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dc41cfb65c434d64a27843c9c4fe64e2 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1284350e58645e9bfd46ad732a7ed78 : SUCCESS in 3m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6781c17a84db47908dc32ef77d63eb16 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/48b6c9f5ec06438ba4510061859fac6e : SUCCESS in 12m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/096e480f2f874d9ab7040b8565fd5620 : SUCCESS in 11m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5cb9136b77374ef29d2e279488c1b22b : FAILURE in 15m 30s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b0f31212192e4f3db31593c8b4148026 : FAILURE in 15m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed9d688b57a047dab46754b7b9b641cd : FAILURE in 11m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/56b106182f56486ab10f9c8f326e7c47 : SUCCESS in 23m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/afad782f75e74da19cf6766a9e597825 : SUCCESS in 20m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b17a71ec459d46b7ade0806a417a1336 : SUCCESS in 15m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b0b37014c3dd435fabb41dd106b9aa1b : SUCCESS in 30m 17s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b7c70b0d33e04a71a811f2424d8a927e : SUCCESS in 32m 55s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1a8d521937f34953b862880042d14217 : SUCCESS in 34m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/168bbf7328ca40d29b931b8cbea796f4 : SUCCESS in 50m 26s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2edeabe725f7421c89cce2234446e210 : SUCCESS in 45m 46s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae21290cb1984795a0eec9a632a0cd89 : SUCCESS in 1h 51m 30s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/16f28f81b104481c8825f2a796ef54f8 : SUCCESS in 12m 48s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9a52d71d781b455495c2ed50d449dd5b : SUCCESS in 12m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc33b8edee9049c9bbce67f23c3bc9d2 : SUCCESS in 18m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/21eb842a7f8649628cbdba368dfa3dc6 : SUCCESS in 14m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9759951566c4d4dbfb83574df58ce31 : SUCCESS in 28m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/471e01e12b9c4f3097341b0f9134fa1c : FAILURE in 10m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/16fb87c2453042ada7c8f9e1234018ca : FAILURE in 10m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c66b52b94bde45e4953628edc7804f2e : FAILURE in 11m 13s","accounts_in_message":[],"_revision_number":18},{"id":"ee09235614e9edb18d22f3771d7426de4034de32","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-08 13:10:24.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"26143008f0c680cdc1a92f17302814a6095fd19d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-08 14:50:16.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1c92d2d4a254e0794f020da90bd4c4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f0dec75d14847c19884094d83ef9af7 : SUCCESS in 9m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/39720acb3abd45d796a3b105153a49da : SUCCESS in 6m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4faecc4395ea48ffab89cddda49eca15 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/569b7dc93a6e4e7d83976c4a1e05c68c : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4103afb9619842aca15ebad3dc247810 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ed8134b8ae44677b78cc6291620eecb : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d65a858834084651be113c35d0f1c988 : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a1bc1d09dc7943b5bbc4dedd6320e3d8 : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bf30a34f2aee4801bb2f0673ea264294 : SUCCESS in 11m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4e2a1fd8666c467e88dfe59824555fd2 : SUCCESS in 16m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ba7fa1171f2a48478202fcd01b198872 : SUCCESS in 13m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02ee8c2fac6444889bdb45e31bf8b67d : SUCCESS in 12m 26s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2af57b401b4c4e9a9134b0616a92aa7e : SUCCESS in 11m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/33af2b1b344045da9e2c218fe9820b17 : SUCCESS in 21m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb243d5678384b8da7095489c2a7cdb2 : SUCCESS in 17m 03s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4d9d83dcdf1b45ed86dee3c48069f1c8 : SUCCESS in 30m 59s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/dcc16af5441f470b83e19235356fd187 : SUCCESS in 47m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/150f487ecdb345338dc416899db22627 : SUCCESS in 47m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcd56ec7754b466c8b4d22a846642c04 : SUCCESS in 49m 28s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dc6a0dfd6e874a5f90304ec4ab04a455 : SUCCESS in 49m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a7914b473024151aca00dc5adca3053 : SUCCESS in 1h 34m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a022d8bfbf414672a438eb8881e95bbb : SUCCESS in 13m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a637717b6d0a4e8686be24671d38ceeb : SUCCESS in 9m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/297819f365a64a0188dbf66ea1460dbb : SUCCESS in 17m 19s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ed26aed69a51407e850d956a97c7ad51 : SUCCESS in 14m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e60e4ac400754a8fac3de4e91a71f6d7 : SUCCESS in 30m 14s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/eccc7c1eff38473ebf331341a5b831dc : SUCCESS in 12m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a9dcbb7021c747f28bf3398d717068c1 : SUCCESS in 14m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e66a9c3215e947bba771396ae09d84eb : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":18},{"id":"07be0fd7bdda1e5c62dac3d1c98ecd591ae5fdfb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-12 08:46:13.000000000","message":"Patch Set 18: Review-Priority+1","accounts_in_message":[],"_revision_number":18},{"id":"97e7753868379146c1295c2a921b1600f316b370","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-12 08:46:26.000000000","message":"Patch Set 18: Review-Priority+2","accounts_in_message":[],"_revision_number":18},{"id":"817d913333a6cc01cc93ac89e933fd268ebccd2c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-12 14:20:47.000000000","message":"Patch Set 18: Backport-Candidate-1 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"3c411ee29f1acc6e6a25aaf33e21d74684b4e1df","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-11-12 15:59:32.000000000","message":"Patch Set 18: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":18},{"id":"8dd77bdeb959190783625b7d89716c806fc1768e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-11-13 10:00:35.000000000","message":"Uploaded patch set 19.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"c69beef64f8cf298878f93d8ead24930c55c1e3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 11:47:53.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14a3d2c0bf654c45b68fabaad5d04fae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5440a550553496aa20f2dfe7f1e685e : SUCCESS in 8m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b959a99ab91f4825814b47bc8d17be9a : SUCCESS in 6m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee7e4af1bd77472c915b77cbab604cd1 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c54046fa61ec4ebfba71abe514ca5d66 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ac02fd818bb4d748d59138224870fdc : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3220e42e1ed845a98d07e195abd5dfa1 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f87784cc4524420789aa33539170fa2e : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b846d13ac8e49d6963fb8d242a3adc7 : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d648229a4657411281c2d2c60092a283 : SUCCESS in 9m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c47b52ca3bc944df9da70257f4ae7341 : SUCCESS in 11m 40s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/708f8e76a4f142a5b600335fcffb2203 : SUCCESS in 9m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9543ec9de337471597bc04b603faa26f : SUCCESS in 12m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d187b59bf576445189b66d93c1b657d3 : SUCCESS in 16m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8813d8a6b48542cc952997673fe9ccf4 : SUCCESS in 1h 02m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1643b53298db477695b5759543c73327 : SUCCESS in 28m 54s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7df918fb5a0546dca9616d29b71c456f : SUCCESS in 1h 01m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38d15b96207b4975b71748f04ab7e029 : SUCCESS in 40m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/83a8a7c6cd9045be88c287f4af08ea99 : SUCCESS in 31m 35s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75c312bcc00843d78c5964f03cfec8a9 : SUCCESS in 1h 42m 31s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5fd319d7a76e4e83af078992cebe64ef : SUCCESS in 10m 31s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d8b5ec59bea74ca293094eaf818abc39 : SUCCESS in 9m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e524effada23457db4da3c72fcd5f2c8 : SUCCESS in 8m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/59c3f8020bb84fa0a315ce5e9a1af847 : SUCCESS in 14m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/053f6a3aced5454f8d3b01c50394e2d3 : SUCCESS in 31m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/45d82ab6047c4829a585b64929a7a374 : SUCCESS in 8m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68dcfcdf2d404e20a318cb4a23df8e5c : SUCCESS in 10m 13s","accounts_in_message":[],"_revision_number":19},{"id":"44d94a12f84859d0d48dbca061bd0914de122776","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-11-13 13:06:45.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"09ab01e22e2f15a37ebff7712bd0eaf87759bc00","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-13 13:07:15.000000000","message":"Patch Set 19: Backport-Candidate-1 Code-Review+2 Review-Priority+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"09eee055ed1057a829355cbd1bce48e45183d38c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-11-13 13:09:04.000000000","message":"Patch Set 19: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"b3c7f7ff031579c2fdb87038cdc6ef79aac7e447","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-13 13:14:20.000000000","message":"Patch Set 19: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"e7cbccf1e8e6e32584641858ab31bb4741dc30a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 13:15:14.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"15994fad4f6442c706d0c04c39a7c77e8f8323b8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-13 14:11:26.000000000","message":"Patch Set 19: Code-Review+2 Review-Priority+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"f7c162eb8b14c66c5d3a39cb7266c3001acf2652","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 14:53:20.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8aa95816d2d42c285a30c42c1a6c4ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd073cdf94cb40ba87eca43e4b73d036 : SUCCESS in 6m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1592aa18a190417abd5f317dd2a1877d : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8046fa07f6104fb5b4c76515b39549fe : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c3a6e4787a294d83abb07d037a79aa94 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/349fcb3abe6b43cc9aa67010c3f20e15 : SUCCESS in 2m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3abc12c3bf96404ca408c32ff504e272 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/916b74bd6c174e468e7097b74bc30057 : SUCCESS in 11m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5f4f4c3b57a241e59481ef0ca0c29cf4 : SUCCESS in 6m 20s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6cbb1c153f2c4d4ba30b4891097ccb99 : SUCCESS in 15m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be3dd5b3ebfd49eb803ff9d3e207efee : SUCCESS in 11m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ac8abf0e6c2b4f7a836b2eb18adae6d7 : SUCCESS in 25m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f499c0c7a0e4361871f319e19169869 : SUCCESS in 21m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8652e4d9a0f9433ea1ce3db3a5b8926c : SUCCESS in 57m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3c68586c8ea74f85bfdea28d4256d90b : SUCCESS in 35m 02s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/eafbf3c480cf4267a8ae77f01a374a6b : SUCCESS in 32m 08s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2746598052c34bdeadf3494c58c4f55b : SUCCESS in 53m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/db86d034c11a4ce0b5908710a92472a3 : SUCCESS in 58m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/137153b2aaf44eaeac034bc7d75e2d3d : SUCCESS in 1h 33m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5fe5e667bfba440c9726659e1525343c : SUCCESS in 15m 47s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d51192f7c45c4d97a49cef6562c37247 : SUCCESS in 11m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcfc2d8f785440a0a6d5f305e0ae5e89 : SUCCESS in 16m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dd9f71ebf48641d4a8c9433358ec99a6 : SUCCESS in 12m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e18ca38e8574c57873c79df1346e96b : SUCCESS in 28m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4be270f78103463fb1a6524ab9949f9c : SUCCESS in 9m 58s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/006621d1d3ec4ebb90b1dfb7a3c36879 : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":19},{"id":"b6c678a068a93c0b5b175ab90115378f420e062e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 14:53:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":19},{"id":"1c5d862508cc113939e573b2ca022b6b66f93c50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 14:56:39.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bfac277b66144ef9e76404727c7c0b1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3212948e644a4f99af867bfd38a21fa6 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/950bc9dea956411d88fd2bed62b8077b : SUCCESS in 40s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"2f845652dd6d70c67f9454c2b09f983076bbca92","revisions":{"0799d39e671d2910e3c4b5d1d1c49d6dd638ded6":{"kind":"REWORK","_number":1,"created":"2025-09-25 10:43:34.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/1"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:43:27.000000000","tz":60},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0799d39e671d2910e3c4b5d1d1c49d6dd638ded6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0799d39e671d2910e3c4b5d1d1c49d6dd638ded6"}]},"branch":"refs/heads/master"},"33cc8b0d927a1dd30ee2a12bac8b18dc96d65ee3":{"kind":"REWORK","_number":2,"created":"2025-09-29 15:20:35.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/2"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-29 15:20:16.000000000","tz":60},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33cc8b0d927a1dd30ee2a12bac8b18dc96d65ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33cc8b0d927a1dd30ee2a12bac8b18dc96d65ee3"}]},"branch":"refs/heads/master"},"6565000e5199b82d02d1e184ea495a8c731eadc8":{"kind":"REWORK","_number":3,"created":"2025-09-29 16:48:53.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/3"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-29 16:48:53.000000000","tz":0},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6565000e5199b82d02d1e184ea495a8c731eadc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6565000e5199b82d02d1e184ea495a8c731eadc8"}]},"branch":"refs/heads/master"},"b37a28de0b23cde8b577ac5fc9461dcdc75c3e6e":{"kind":"REWORK","_number":4,"created":"2025-09-29 17:16:15.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/4"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-29 17:16:10.000000000","tz":60},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b37a28de0b23cde8b577ac5fc9461dcdc75c3e6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b37a28de0b23cde8b577ac5fc9461dcdc75c3e6e"}]},"branch":"refs/heads/master"},"cd4bab5e2efbf564d56443533aa5805ad70a1834":{"kind":"REWORK","_number":5,"created":"2025-10-10 16:56:57.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/5"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-10 16:56:57.000000000","tz":0},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd4bab5e2efbf564d56443533aa5805ad70a1834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd4bab5e2efbf564d56443533aa5805ad70a1834"}]},"branch":"refs/heads/master"},"c4de7962cbd28aac07c56041d5a88aa094e3bd13":{"kind":"REWORK","_number":6,"created":"2025-10-10 16:59:42.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/6"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-10 16:59:42.000000000","tz":0},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4de7962cbd28aac07c56041d5a88aa094e3bd13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4de7962cbd28aac07c56041d5a88aa094e3bd13"}]},"branch":"refs/heads/master"},"e2b244e36766e6e8e6610813df0b4279462826e8":{"kind":"REWORK","_number":7,"created":"2025-10-13 18:18:47.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/7"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-13 18:18:39.000000000","tz":60},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2b244e36766e6e8e6610813df0b4279462826e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2b244e36766e6e8e6610813df0b4279462826e8"}]},"branch":"refs/heads/master"},"093eb6f76b72bd5c93591927a4d5dd5789c7a664":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-10-17 17:51:24.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/8"}}},"commit":{"parents":[{"commit":"4d45119bfd6e5eccc2e1eee6615f9c848c19c5e3","subject":"WIP: Resolve trust issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d45119bfd6e5eccc2e1eee6615f9c848c19c5e3"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-17 17:51:14.000000000","tz":0},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/093eb6f76b72bd5c93591927a4d5dd5789c7a664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/093eb6f76b72bd5c93591927a4d5dd5789c7a664"}]},"branch":"refs/heads/master"},"48e32063f8fce56a7b0c61c769a8528975e61e3f":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-10-20 08:50:55.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/9"}}},"commit":{"parents":[{"commit":"38f5d328c4b9bf8c3c587e89f5e2f1550c234264","subject":"WIP: Resolve trust issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38f5d328c4b9bf8c3c587e89f5e2f1550c234264"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-20 08:50:55.000000000","tz":0},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48e32063f8fce56a7b0c61c769a8528975e61e3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48e32063f8fce56a7b0c61c769a8528975e61e3f"}]},"branch":"refs/heads/master","description":"Rebase"},"02ea250cd81adb3c33f2a85893442c7e0b180b18":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-10-20 09:29:49.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/10"}}},"commit":{"parents":[{"commit":"e3aba1e9a5f01ef3c46e9388d888169087bfb175","subject":"WIP: Resolve trust issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3aba1e9a5f01ef3c46e9388d888169087bfb175"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-20 09:29:49.000000000","tz":0},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02ea250cd81adb3c33f2a85893442c7e0b180b18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02ea250cd81adb3c33f2a85893442c7e0b180b18"}]},"branch":"refs/heads/master","description":"Rebase"},"262330b51ad1846127a70fdddf8587269c53680d":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-10-20 10:59:24.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/11"}}},"commit":{"parents":[{"commit":"14bb5893fa058326cb57350bbfcccce49d487667","subject":"WIP: Resolve trust issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/14bb5893fa058326cb57350bbfcccce49d487667"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-20 10:57:42.000000000","tz":0},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/262330b51ad1846127a70fdddf8587269c53680d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/262330b51ad1846127a70fdddf8587269c53680d"}]},"branch":"refs/heads/master"},"5166307949be6bf14e7b3e2d0f7173b3c05fb609":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-10-20 11:29:42.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/12"}}},"commit":{"parents":[{"commit":"d91037b101b7c36934203e70e36015aa98c67ec6","subject":"WIP: Resolve trust issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d91037b101b7c36934203e70e36015aa98c67ec6"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-20 11:29:35.000000000","tz":0},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5166307949be6bf14e7b3e2d0f7173b3c05fb609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5166307949be6bf14e7b3e2d0f7173b3c05fb609"}]},"branch":"refs/heads/master"},"f17662440443167ac252c376ad34b61828138bae":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-10-20 16:14:32.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/13"}}},"commit":{"parents":[{"commit":"ebc3bcc8704fb5a71f5081738424ef583c91b428","subject":"WIP: Resolve trust issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebc3bcc8704fb5a71f5081738424ef583c91b428"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-20 16:14:06.000000000","tz":0},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f17662440443167ac252c376ad34b61828138bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f17662440443167ac252c376ad34b61828138bae"}]},"branch":"refs/heads/master"},"cd69d7fe421a81b556d655f3e3e32c284da12969":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-10-20 16:23:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/14"}}},"commit":{"parents":[{"commit":"d44a4081a69975b67ebbeaaabc9452e6bd9efbd2","subject":"WIP: Resolve trust issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d44a4081a69975b67ebbeaaabc9452e6bd9efbd2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-20 16:22:49.000000000","tz":0},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd69d7fe421a81b556d655f3e3e32c284da12969"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd69d7fe421a81b556d655f3e3e32c284da12969"}]},"branch":"refs/heads/master"},"3fbaa5454609fe561136fc3adc830c0be0ad67cc":{"kind":"REWORK","_number":15,"created":"2025-10-20 18:35:07.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/15"}}},"commit":{"parents":[{"commit":"d44a4081a69975b67ebbeaaabc9452e6bd9efbd2","subject":"WIP: Resolve trust issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d44a4081a69975b67ebbeaaabc9452e6bd9efbd2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-20 18:34:52.000000000","tz":0},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3fbaa5454609fe561136fc3adc830c0be0ad67cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3fbaa5454609fe561136fc3adc830c0be0ad67cc"}]},"branch":"refs/heads/master"},"b5bfb8d98659671d93fc0c37a140616d078be407":{"kind":"REWORK","_number":16,"created":"2025-10-24 15:51:01.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/16"}}},"commit":{"parents":[{"commit":"d44a4081a69975b67ebbeaaabc9452e6bd9efbd2","subject":"WIP: Resolve trust issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d44a4081a69975b67ebbeaaabc9452e6bd9efbd2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-24 15:50:14.000000000","tz":0},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5bfb8d98659671d93fc0c37a140616d078be407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5bfb8d98659671d93fc0c37a140616d078be407"}]},"branch":"refs/heads/master"},"f51e4f7f288b6bf06ebda090337e38afb5f03518":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2025-10-24 19:28:58.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/17"}}},"commit":{"parents":[{"commit":"6f8fd00c384078aa6cbd5f9cb0467563bc55f844","subject":"WIP: Resolve trust issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f8fd00c384078aa6cbd5f9cb0467563bc55f844"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-10-24 19:28:48.000000000","tz":0},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f51e4f7f288b6bf06ebda090337e38afb5f03518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f51e4f7f288b6bf06ebda090337e38afb5f03518"}]},"branch":"refs/heads/master"},"8601c0104c07c7e331e8474216729a95e76a6bc2":{"kind":"REWORK","_number":18,"created":"2025-11-07 18:52:47.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/962267/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/18"}}},"commit":{"parents":[{"commit":"2f8cc7bec13180e247a4491eb5ae69621a50bf4e","subject":"Replace local_action with delegate to localhost","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f8cc7bec13180e247a4491eb5ae69621a50bf4e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-11-07 18:52:37.000000000","tz":0},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8601c0104c07c7e331e8474216729a95e76a6bc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8601c0104c07c7e331e8474216729a95e76a6bc2"}]},"branch":"refs/heads/master"},"2f845652dd6d70c67f9454c2b09f983076bbca92":{"kind":"REWORK","_number":19,"created":"2025-11-13 10:00:35.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/67/962267/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/962267/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/962267/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/962267/19"}}},"commit":{"parents":[{"commit":"2f8cc7bec13180e247a4491eb5ae69621a50bf4e","subject":"Replace local_action with delegate to localhost","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f8cc7bec13180e247a4491eb5ae69621a50bf4e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 10:18:28.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2025-11-13 09:54:09.000000000","tz":0},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.19.x) and minimum to 11.x (ansible-core 2.18.x).\n\nChange-Id: I3ab49d28804733733474565d33cc7b564b203ba6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f845652dd6d70c67f9454c2b09f983076bbca92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f845652dd6d70c67f9454c2b09f983076bbca92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
