)]}'
{"id":"openstack%2Fkayobe~962285","triplet_id":"openstack%2Fkayobe~master~Id0938c3fef6ccb9ceb556ce68a442393eabf2624","project":"openstack/kayobe","branch":"master","topic":"ipa_patch","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-10-13 17:25:03.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-10-13 18:53:32.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-10-13 18:53:32.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-10-13 12:24:57.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"38330":{"account":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"last_update":"2025-10-13 18:53:32.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-10-13 09:17:47.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-10-13 18:53:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id0938c3fef6ccb9ceb556ce68a442393eabf2624","subject":"Change IPA compression algorithm from default gzip to zstd -19","status":"MERGED","created":"2025-09-25 14:25:01.000000000","updated":"2025-10-13 18:54:29.000000000","submitted":"2025-10-13 18:53:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":42,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962285-ipa_patch","meta_rev_id":"84c534cbbb9f182b2c93464b8a9f50ea42883373","_number":962285,"virtual_id_number":962285,"owner":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-13 18:53:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-10-13 17:25:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-10-13 12:24:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-10-13 09:17:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-10-13 17:25:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-25 14:25:47.000000000","updated_by":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"CC"},{"updated":"2025-09-25 15:25:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-25 15:39:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-09-25 17:58:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-10-02 13:54:12.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-10-02 13:57:36.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-10-02 14:02:38.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"CC"},{"updated":"2025-10-02 15:26:06.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-10-10 13:30:03.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2025-10-13 07:43:02.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-10-13 17:25:03.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"c5817d5eac28ba2e7e36a51fdd44be97d47d4d75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-25 14:25:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0610c33f9bf355ee98ad2e0b2a673d7f3ec075c5","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-25 14:25:54.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"15b90cfdac4c8d69632aac270d8fda75ab36c114","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 15:25:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eef78cfd582d4a4f894b452386d05c25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc8c0a45f1d7496ba4f5b5960018d3c7 : SUCCESS in 8m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9ff869958f64411da0ff9db8269fa78e : SUCCESS in 6m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d938d2ff92c04a6191fa86882b429d67 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f52f81ef65d42bdb5a1f13335568fda : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de308c1d4aa2483a980b045373306c80 : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3342d1c0122549c7b025223f2e34f3be : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e908a8a27ac4bd6846d26b90d3694ab : SUCCESS in 5m 36s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/042f9ec7bae74956acca712ff6d2cc80 : SUCCESS in 3m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/28c0c01689f04605a6faa84371cb6e44 : SUCCESS in 6m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f18afe5c33334b0aa6e2b138659afa10 : SUCCESS in 6m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29c1603c26554c4397b961b121a81db6 : SUCCESS in 28m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b79754cd4ba04e65ae820512425f7560 : SUCCESS in 30m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c03f2cd984b944e397ef55e4c33d3916 : SUCCESS in 13m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c94415525cf44d1a41cb2c4ead56b46 : SUCCESS in 10m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/787bfc02397349cf80081fd71b5979b9 : SUCCESS in 58m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/648316e6415243fe8c2d91eddd098233 : SUCCESS in 8m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5ee407b978d94b8388d42c4ccd6964d1 : SUCCESS in 7m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/378318c751994afb9b68e4d3bd57470f : SUCCESS in 25m 03s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3a54a10049e74a6c83576058ef675654 : SUCCESS in 7m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd39faffe6da442ea1e3cae03091aa3c : SUCCESS in 11m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ff3759004bbf4891b41beba60cea76bb : SUCCESS in 17m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/817cd8252a534576bceda320c0c23028 : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d6e7b28201f019638fce41887159af6fc353a003","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-25 15:39:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67650ef5cdf449c6d9e18e447e5b521d3e67f31f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-30 10:39:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0cb7a66178fed17835e4c4b3e77c23107819db1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 12:04:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b34f9c6b04c6433caad02657da243ce3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd655a7e07374cb3aa885f660cd14f81 : SUCCESS in 9m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/817fd8183d25400783a0ff61d1a4ac8c : SUCCESS in 6m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f87390802d047d38970e3fc9d065e2b : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06009e88dea244fb8a0f57532c3292a1 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/318fdca958024d7c90fd6c7d6fda4a0c : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28a6511f74df4a50b4ca56eeb9188862 : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7b2bb81e66c4f458e3c92a1d8708711 : SUCCESS in 5m 05s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dc25b0745a46454ebe2bda1371e9a57c : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc92455690f246d59e6b29760f3450cc : SUCCESS in 7m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6b7e8c7fa36e43dc89ca1c979a100ee2 : SUCCESS in 7m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5c8cb3edd434f439c0f2567258955ef : SUCCESS in 36m 20s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f33887c702a64243a70701517d93f1b1 : SUCCESS in 47m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e6fa8a07e95044048b7a0f1dda9b047a : SUCCESS in 17m 33s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/64bdec69afef446680e47d6dd99a9b96 : SUCCESS in 17m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f5f010120ef4a948fca121fb2972d88 : SUCCESS in 13m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a10645ae019749f583799a87a968cdcc : SUCCESS in 1h 23m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18ff5d9deeec44e9a19b1d969dd5cf1f : SUCCESS in 17m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/11785953e17b49fd99ed90ea706539cb : SUCCESS in 9m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a21083d5f30e4244b058b01eb0b1fdd0 : SUCCESS in 23m 24s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c769ec9e4b82446299236a51dafe53a3 : SUCCESS in 10m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a6db0c2eb524a16b96b366126e35389 : SUCCESS in 13m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f2a96e10a7be47c088c472adeade5776 : SUCCESS in 8m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ddb5719c0a846da83301280be5c3f07 : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e9c3c0b34edbe1498b744294a0f5d4afc04059a9","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-02 08:44:32.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"69fab3fde582c70397febc8931c766b19ab9591a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-02 13:55:58.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"98747f6b489367965ee1cc489ebb295b17107572","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 13:57:36.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6d2d7b2c3e3e5f8d25e8ea12456d8a88292168c0","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-02 14:02:38.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"69efda1661bb3aa4631ccaef1131c37556718f2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-02 15:01:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"c1d5f0aea2732774bf9d387928a8dbf2a4b76a5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-02 15:01:53.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"6eafc2543eec40f551b43a6dcb39ca7d434e6a54","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-02 15:02:18.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"55b56e2646c895439312dfcdfaa771e53903fb76","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-02 15:26:06.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9dc200948af8d71d10f2124084440d3261a67396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 16:36:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99a1ac000c8947a3819906db0590d925\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3a86ad252f34920b8e4705e1ab41e1b : SUCCESS in 9m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a5c7eb56c17c45fbbdc69f4cb7974845 : SUCCESS in 6m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/958ca573e8dd41f38ac3cbbbcecdea04 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a644455fd964fd9897787e1a2b8063c : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b8b78c883824ca49d911aadd7a42980 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e588d9ba49c9477d934590fd92bfa00f : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d6abfe61d4784d2fb602248eb7baf1a2 : SUCCESS in 8m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9d28e1d62d84385937f5c561f137777 : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/891e60807c3a42218327c0d5b588f2b1 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6de6955f17f74adbb67c1f5a21bc96e3 : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c0d59a3e663c4b859457e74bc5cb336f : SUCCESS in 5m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/61f74b2f181d45b998192fe5953c3912 : SUCCESS in 15m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b1248d43f1f442e391666c541540f859 : SUCCESS in 10m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c6457fdc36947a089387ea73bd887a5 : SUCCESS in 13m 43s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/dc97d606952f4a41b1dfad21af30afcc : SUCCESS in 9m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f59f65bd53344d209bbef84558361c56 : SUCCESS in 18m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db8a47fe17a84f7fa78f1b18480e5fdd : SUCCESS in 13m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3735a43bb3634c2f8adbd1dac36567f0 : SUCCESS in 49m 51s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/36a4be33f0054ca1be7a7eb309054cf3 : SUCCESS in 46m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84c11bee2ebb42919021629da951f5e7 : SUCCESS in 1h 19m 52s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8d6380c12a940afab38a353300fddc6 : SUCCESS in 15m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bf491a614a85477eb4ed62b412bfcc6c : SUCCESS in 15m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0f6d28caafb4d458bddfd27d8f066e7 : SUCCESS in 28m 24s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1f8910799dbd4aeb88c77986466b7641 : SUCCESS in 12m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9e4199f7f50b4309a6da39fea99a5070 : SUCCESS in 8m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/837990d12b1e455cb2f1de84a39ca0c2 : SUCCESS in 11m 41s","accounts_in_message":[],"_revision_number":5},{"id":"0ec416ad0e1a10dc291b4d8b04477417e436873d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-03 09:05:26.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"5b0103c2a5587c3d26eb8eca91f5f66fbeb234bf","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-03 09:46:58.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"4b499a53cda14cbee9052534457e3ee58eb75464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 10:36:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d5bbb1c7b48444e8ce486420a435b3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5909c95dfe8e4560ba033238b9e5e064 : SUCCESS in 8m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c29f8d2e3f0c4aafa0ee8f043f0341a8 : SUCCESS in 6m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57f1e61f8e104474b9edf2e8ff5ba6e1 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7be7e2bdf8594d7e939ca2e025ae89d4 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0b3643389804ee8952214ec7be95049 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d2406db865104aa082c80489dbd39a8d : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56d8d84d76904fe6be55fa3c6e9bbeac : SUCCESS in 5m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fecd01a7a97e4151a413291878556651 : SUCCESS in 3m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/58b46fc9eaee4b60ba3dc1c2437c74e6 : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/17ccb4c3693e4c9099d5974b82e38c12 : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3b8ce58d034e45f9a2e2a8c5aa8c1657 : SUCCESS in 11m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7c66984c6c6a4d0588cf5a83b7a4a231 : SUCCESS in 13m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cd0c595b883b41ab96a5391ebf1a2237 : SUCCESS in 14m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e688908111247fd8f15c3ab4b99bd46 : SUCCESS in 10m 46s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9bd5eeafd4f04132a711547c05136715 : SUCCESS in 14m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9f1facc0d3be4ff9bb454cfd8e85549f : SUCCESS in 22m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70bbcb30eba640089155124aabe41946 : SUCCESS in 12m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a76df2e223a4ec684180d47306cc72b : SUCCESS in 49m 44s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e08c3b939de84d26aed2723c3f3b21d8 : SUCCESS in 46m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23ab764c1b954f4eb72d2881c7b8a086 : SUCCESS in 1h 29m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3f51829a1234faea07828610087a105 : SUCCESS in 15m 00s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7181852da12745ddbe72bdf7a0c68feb : SUCCESS in 10m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f35e01390bd34c35a1f32812e4112e4b : SUCCESS in 13m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1c4fb10418034d1eb685621bd851b433 : SUCCESS in 10m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b60d4128e4904446964d56f8eae01b1d : SUCCESS in 12m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2da328eddfde44999abacdbce23b1d5b : SUCCESS in 11m 27s","accounts_in_message":[],"_revision_number":6},{"id":"2217baa47adf8c262bdb7dce443f64228c33c2f0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-03 10:42:10.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"facd2445dd1abfaebb3d5a65d07c11657aaa6e9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 10:42:24.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"320ff86eded9fe9c3b4b8cfe18f69112085210bd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-03 10:54:32.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"bad496751089936c143317485656ec6f775c3f33","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-03 10:55:22.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"0f4bc891c5b04d074635a3315013daa41642399d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-03 10:55:26.000000000","message":"Patch Set 6: -Workflow","accounts_in_message":[],"_revision_number":6},{"id":"27a4ca3bb3f8603a67bb4d0e91be46cb9bde9552","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 10:55:26.000000000","message":"Patch Set 6: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":6},{"id":"05dbbb035e2b2530098e7b3d231b8e290b946e78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-03 12:26:38.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n","accounts_in_message":[],"_revision_number":7},{"id":"bdfb7c2bc52fc6c80bc7a86f68908d7c2e4365cf","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-03 12:31:33.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"34761593c619964157fa1ef06b04a08595b502fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 14:14:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14bb0abe295c4231b0a8a03f2c82c65a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f0a7eb12f4493291c29e1f230027f3 : SUCCESS in 9m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7c39146dee7149e69446768f2c14c872 : SUCCESS in 3m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88df2e5856524bf9b125e0453a35f5a1 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fc84da8491247c0b27becb33bb7b10c : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08c04cf6c103421884d5a59cce2525c8 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ba9454ed0a3406c90212fc15c738638 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd6e5cf42aea4a99a6be80b0da7c253d : SUCCESS in 3m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f102fb9233cd4f67811ca9936ad2bd0c : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/81fc0c23674c4a038544bd48c24ea511 : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fddd931b00e048daa67df86f889790dc : SUCCESS in 11m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/77cfdf34ebd343f180dfaf09e61527a4 : SUCCESS in 11m 05s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c93c4fcafe5444c9ad0a91d09ff553a3 : SUCCESS in 12m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5966a8b2df614cdb9195e5eb7a25ca03 : SUCCESS in 13m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3033f5d1f1b421d827fe06fa3f3f156 : SUCCESS in 11m 02s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5eea52ad9ff84302bde16cbdcd187d73 : SUCCESS in 10m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d16f33872d3047bc88591b3595664f43 : SUCCESS in 17m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19eb64367fbf40c083222e7d014c9e12 : SUCCESS in 14m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8da6426ea2e44479a2341fbc708b4d67 : SUCCESS in 45m 18s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7e6fc2cfd7f4428e95bfa4ac56e908d0 : SUCCESS in 51m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09477b2b124844a5b114273279482243 : SUCCESS in 1h 42m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b22aca7583c473388ac080eb8f461fa : SUCCESS in 15m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9d1af0b4777f4327804a2e1484a737b5 : SUCCESS in 12m 00s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/653776b6f01e4af69344adc26d27ab57 : SUCCESS in 16m 07s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/69b3a71870db4b998f105688e7b4fdb2 : SUCCESS in 13m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/588769fe79c0463785998f0d9c8aaf02 : SUCCESS in 13m 40s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ee1e963158c4abd8b4b6f7806068eaf : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":7},{"id":"f99fb7b3ef97a3612da06e9264b10b539a736219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-03 14:16:44.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"e2405683bd2d96e3882fa2602fdbc0f7209d44b1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-03 14:16:48.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"a01cd65b1ee1cc4f18989fe9b8b01eae75e72719","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 14:42:48.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/53f08e3b66de44c2bf62a6f4557a61a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58373e1d92194c348d02ac97ee0830d5 : SUCCESS in 8m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/70319bc2e2d74979b8960c89efe95b3a : SUCCESS in 7m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01398bae26cb4be396731cb5dbfbb758 : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/887c8a0186f640b0a81124d2c3ec1397 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf8e942bdc3e4e1bafe1479abed82c5e : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff68837fc6424405aa2bb931b2e982a7 : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04a31b2f719e4c678b0d96eae5eb9bae : SUCCESS in 6m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b720908d39f4b638fa6c43c684fc4cc : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fa9954430cbf4cfc819897d0c00a68c1 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/171496dd673a41b9910aeebeb3fb1ace : SUCCESS in 7m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/69f6147da4904e1089358022350c6d5b : SUCCESS in 10m 40s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d297bd27f4cf4fc6ac2002a19de67c27 : SUCCESS in 7m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/90adbf12faf94e598e043588d30f4e45 : SUCCESS in 13m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/970305f691ae4e488901e1bdde71c4e1 : SUCCESS in 11m 38s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5f020da89c0745b4a0e9bc5d823a00eb : SUCCESS in 15m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/cb89e04f33354675bd4cf35818d52937 : SUCCESS in 19m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/812c6525458a4eeb9a6fcbae47f91559 : SUCCESS in 13m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ba01e4d6d04419c8e65826a048e1b89 : FAILURE in 18m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/59a8e5d846fb41fc96b4451a3a28408c : FAILURE in 16m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0dcb3729cb51445490d5cab422bf5f5b : FAILURE in 19m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6428b372587a44398cd443801376745d : SUCCESS in 13m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b6b3086ba19149b292b386e57bc0367d : SUCCESS in 14m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2b27bd8b0b8461bb9e24f64a0212a26 : SUCCESS in 15m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/33472aed943c4979992652a86af1000f : SUCCESS in 14m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/870653c0b6e4442c96df2df9a4dbb78b : SUCCESS in 7m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bee4cac18c34be5a553c5862a7520f9 : SUCCESS in 14m 04s","accounts_in_message":[],"_revision_number":8},{"id":"167b65a5232bb730227b9751377299abe544b324","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-04 06:11:26.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5caf33e563c1cfe9b8e181f34858c33d35885c24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-04 06:29:34.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb2437210c2e4b0fbb13ec364ccfa6bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a917529cd13439d88d1a7dad389dc9e : NODE_FAILURE in 0s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fcdb93b3a0164778832f024d9568a218 : NODE_FAILURE in 0s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/262e27ae9d2b4693a2e2b4a05928bbb5 : NODE_FAILURE in 0s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d9841552fc14bddb24bb72ebd4d0e9b : NODE_FAILURE in 0s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29b5b5d3faa648b2bfbcf69f0f1f87b3 : NODE_FAILURE in 0s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de161b76ad0e45b68510d4b0fc84ddda : NODE_FAILURE in 0s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0d2da25449b4b44a9c650b7b587d3ff : NODE_FAILURE in 0s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ab31676a8754adc937231843135f41c : NODE_FAILURE in 0s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/81a85f53943549b087f01c55d62e67d7 : NODE_FAILURE in 0s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/687665f8f9e04f089352575a59c263d0 : SUCCESS in 10m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e88507a58ddd41f69e9718608c8c9f63 : NODE_FAILURE in 0s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/84aac295eb0e468796853acaf7e3a043 : NODE_FAILURE in 0s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0f91343bb7fc4604836dce46f3b1f036 : NODE_FAILURE in 0s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/717962249bb34c0e9854f956bce083a6 : NODE_FAILURE in 0s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3adf8a2b1ea24dc4a30837b0fc7310a8 : NODE_FAILURE in 0s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7e0fdf3c4df6416dac2f58d6594e1895 : SUCCESS in 17m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/573a7f04445a4286833c1279223fa477 : NODE_FAILURE in 0s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13e911059f1c4da3a1d4094a0e77258c : NODE_FAILURE in 0s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/835eaf75cf214268956e5553182aef89 : NODE_FAILURE in 0s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4259c71648ce45e5a01a71da11adbf05 : NODE_FAILURE in 0s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa1d4dee0cbe41159921dd37339b511e : NODE_FAILURE in 0s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ef6e2a9361df4cfd86ede96df7eb222e : NODE_FAILURE in 0s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca12ee59bcae4b1d8ac8a519a20bd353 : NODE_FAILURE in 0s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c88b4f13d1f444219893f35cd8ca252d : NODE_FAILURE in 0s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/85b12efb74b5411a8d2ce2fc0ce0d7b0 : NODE_FAILURE in 0s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/796a4bab89444367b9f9d3489a5a75cb : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":8},{"id":"e874bd071308011375b022f1e6d684b22fb88b99","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-04 09:14:58.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"fe53c0ac5d7077ab66d1e41deeb85748f5b69b13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-04 10:58:30.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8867e6b2457e471da9cc91861af27ebd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26d1bd0274e2407b95e8a9ef45907738 : SUCCESS in 4m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/918ce0633eba42a5ba137d44dea57743 : SUCCESS in 6m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eafb501de122401e90a99614afd62349 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e506136f89143d6a7c305a3100041be : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3dfa68d76d724808a8912b7c0487d7f3 : SUCCESS in 2m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f6404cffe96496197ab4d91d4fabb23 : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da797e4dd20343c7b816781350d13141 : SUCCESS in 5m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad18eb7f283942ff9738516e1a67811f : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3bfdaf9ded96423d9dd96c4e5c4d14a2 : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d8f8284ab5ec4ab781154710e8b35fae : SUCCESS in 8m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ed8796c324cf411a9cdcef0a65e40ec6 : SUCCESS in 8m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/078d33cf6fc643f7ac43c4bba18dad2c : SUCCESS in 10m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7d5e804a47144271ad6a2669d00eef8b : SUCCESS in 13m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5627a52cf49840cfbccb92b5487f34b7 : SUCCESS in 5m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3422ca173f34467083ec0f00261842b5 : SUCCESS in 14m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9a4ea53609fa469589588101a5d7d7a6 : SUCCESS in 22m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00c2e9f029284e98a876a4244fb121f1 : SUCCESS in 12m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea54a993fc164332910b42e79ced9bee : SUCCESS in 40m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8f086bca61f84ae682515cccfe501161 : SUCCESS in 59m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/999fd756cc7e40b4aedb886ecd8ce90d : SUCCESS in 1h 38m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e956850ca02740aabc28f5662a7a04d7 : SUCCESS in 14m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9c97ce19292441eaa458f79f30ae63cf : SUCCESS in 12m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4af1c889ca744c8bac70b1b6c4e62dad : SUCCESS in 25m 56s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5b752c763d534f75bd95d09655352733 : SUCCESS in 11m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2da6a10327844a789f24f8df9797269c : SUCCESS in 13m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/939fd67b172042d7bd5a9d0ca55fdf09 : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":8},{"id":"ca05cdea7729c5b6f56fe4c6602db82be9fffa99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-08 08:40:06.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"7c21560db4eadf4f459235b50ca41d8e0787cb02","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-08 08:40:20.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ece7795845ceb3217b5818d3410a882fc692168a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-08 09:52:04.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"1eb33fac1b68e5741de8e8dcccd22099cfd42225","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-08 09:58:32.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"61e8490bef7b025bf792751506552b4ab7ca0ce4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 11:21:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3986828eb20649979182dc9fe6c0e060\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23c01a725aeb479a81ed823ccd695009 : SUCCESS in 4m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fa07dd6b8d0c4e11b1008122d7556ace : SUCCESS in 3m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7814d16458904de6bc17c9f79544dfc0 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f073f86a29b7468d83d048c1d03b5f31 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7b9704d3ee443a58e34bfac2f374522 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3462229a4b1d4ee395fcffdd481918db : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ae6280901514c899505508d2d76c8b5 : SUCCESS in 8m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bdf4809b075435b9c95844375509a14 : SUCCESS in 3m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bad1a21ad0c642139d9cd85cd6a07d08 : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f5dcca9d0e5847d99a3c79e3a91c9353 : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5973c7bff9f14338aeb1d40c0b0e2397 : SUCCESS in 5m 46s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/442db26426a244bc982e03ae25801713 : SUCCESS in 15m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/48ec87fe3e8f4a73aaddd883988c9691 : SUCCESS in 9m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b2362ce15084db2bec28aaceca5de41 : SUCCESS in 13m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/054f4790da1f4ac78b73aaa289b2b3a4 : SUCCESS in 14m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a693f8089eb44305a28b7aee4eb677c1 : SUCCESS in 22m 46s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/894a704e1b3c4f8dbe4d11f878f613d0 : SUCCESS in 8m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38e8f483d4464eda8fc6f15b4907f8d0 : SUCCESS in 54m 32s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1c234fd512414025b593fbb2d640c628 : SUCCESS in 54m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae8edfe20f5f47a9849c8bdce72b3b16 : SUCCESS in 1h 27m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b2df183ef924def8ba2ec7e6de646ce : SUCCESS in 16m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d426004b9baf4dcdbec5b1bc7024ca82 : SUCCESS in 14m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e140c9fe2dd4ddd94d0d6e354747cac : SUCCESS in 31m 21s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dd748b21993d432eaa7a2380b2cbb098 : SUCCESS in 7m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/50951968fad2458bbb2f5aa7e0eb3bb6 : SUCCESS in 13m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6af23b47dcf9483199cde762db22d418 : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":10},{"id":"9f6c6d4b501616ee5cbae2c9ba682adeeeea2d13","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-08 14:32:28.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0eed2d248924e45044ba640f94d34af76ddbd8bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-08 14:33:53.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"d7bae81939777f66ed5c45a291a6a35025c0ab81","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-08 14:36:52.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"e47c171da57270439084bf72b37fb45a6dd8a32e","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-08 14:39:01.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a7fec987244c2a4a95615267bf4800a493cfc0d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 15:52:24.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e67cabfc36434624a1ea6c755c4b217a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0742605b3cb49c0ae99f80b77808e1a : SUCCESS in 6m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a95d60c1de6c47139ff2525b6bb225ee : SUCCESS in 7m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1e770d01ee9470b8066f4a2a710060f : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e24ecf299d34389974d162d031f7f34 : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8e9787f43464699aecf0a2ce376122e : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65bd8a0e2bd947e2a5e2f91918041c81 : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07193b39f47c4c4e84d9497b8f5a7ce9 : SUCCESS in 5m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4af99201f4004ac5955d30e91be9f7c2 : SUCCESS in 4m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8662458f811b46a1b201af613220c093 : SUCCESS in 3m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8cd285baa9c1409cae3e9147dc2bf6dc : SUCCESS in 11m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d5b8486860584744968fd541dca2101a : SUCCESS in 5m 48s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f35e7be6cadf41e8a2efdb6976b2d014 : SUCCESS in 16m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e2eb6515a7a2412dab6676ab61dfe383 : SUCCESS in 15m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ccaf31c4dd8466985c9dfadf952f88f : SUCCESS in 6m 09s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f5e8772de7d044aba0a7cbf4c64721b3 : SUCCESS in 9m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/446588439a6840d98d7becc01ee0e063 : SUCCESS in 16m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b4aed4013bf4290bc24966dbb9a28d8 : SUCCESS in 11m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a78740ac3d34db78c403977cbb7793d : SUCCESS in 47m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/61968ea35954459b9e066b70147d4c60 : SUCCESS in 42m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b91b02de600a48da854b8d13e3aa0817 : SUCCESS in 1h 17m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79da4f9ac7734a14af26dc029cf6d930 : SUCCESS in 14m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/32be4ee1ebc54c46b233c1f2421cd26a : SUCCESS in 12m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d854c4a849d1457b89a4064d4e87c376 : SUCCESS in 26m 00s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c7f3440bb7de4989911878b4e52e1142 : SUCCESS in 16m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5aadcfb3e1e64c6da1c169ab8a55b463 : FAILURE in 9m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d43f56682a7a49d8b4423a5f290b7dc1 : SUCCESS in 11m 25s","accounts_in_message":[],"_revision_number":11},{"id":"3b7343ec01826e82af9e372782f406f54734ee2e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-08 19:01:01.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"43fffb5e0025afa34e741bb5a04b2eddaa6a80c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-09 10:21:35.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"8d2f9c60292a712948bef79c76ae68480d354858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 12:06:38.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62af3863f2d141889c797f01e60336d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec712dbef88745acb1bd799d52c1d661 : SUCCESS in 10m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b1c31308268e4739bbd1b56eaee218ca : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce8512a4e5fb4471a69428db929ddfe9 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b754b97207044e3943a7892cdbce899 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2465f8792c8493dba545e2bdbac2278 : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b0b26c593b1401dae3b0db692480bc7 : SUCCESS in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/362a46d5ea464a44a326624f2b20d0bd : SUCCESS in 3m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa7f2bff7182418aa2f8cca703c8e55a : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fce49161f25d4ef2966fa94149cdeb6b : SUCCESS in 3m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/930c11738c33471bbe0c97b4c1ae480e : SUCCESS in 5m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/12b7e4e1da9a40f882b8fb90b6f2a5ea : SUCCESS in 10m 09s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3d945ec1704f43d2b95e28b71c58d0d0 : SUCCESS in 12m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/927033424a6b49528ef1563d2838a77b : SUCCESS in 12m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bbe96c862ed400f8bb8fc870afc3ec3 : SUCCESS in 14m 21s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5cd6e9e143b744878241d8efa9645503 : SUCCESS in 9m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a73d1a2b7c4547b39c1db88b48806e3e : SUCCESS in 19m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b942572dcb1643679a62aab54ebd54f2 : SUCCESS in 6m 44s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7eb22febb2e64981a441e57ecaa91e22 : SUCCESS in 23m 34s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d899d8bb901943139f36f01dd7500d38 : SUCCESS in 30m 09s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4536a7ee7268459dae4af587c191ef87 : SUCCESS in 1h 40m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab87138fc0b341569cfdc0b541fcafbd : SUCCESS in 14m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1a07406af9b040568724b789789a4cc0 : SUCCESS in 13m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c5a02fc9dbd4dbcb4f45436a835d0c3 : SUCCESS in 18m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d07326255fb04bd7bacf8dcf0d7fc11d : SUCCESS in 7m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1623b8df39ad434998af800f4bde7de7 : SUCCESS in 14m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0edd2699eeda45c48cd8082d1eca2768 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":12},{"id":"dc039b5f76a750297453e8bb819368464569fe46","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-10 13:30:03.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8ab10e0edaf1609026e50128603cac9eadc45a66","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-10 13:32:42.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"af983e67237e9db49ce321eedff5459f62a9d11d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-10 13:34:39.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e9e5cc49838a3a1972d12a7246cb736796753433","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-10 13:52:28.000000000","message":"Patch Set 12:\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"0b0be9d046a55239c1368aa3fe1ebbdbd2e4a907","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-10 14:55:09.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"ccea8d50554a566e128c8027c981ab0d1b70950f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-10 14:56:45.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"1de10ace27ec841996fe3629e3b3ccf360c56774","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-10 14:57:18.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"1b19a009c08075f1fa07e7c7824985205ebdd5d7","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-10 14:58:13.000000000","message":"Patch Set 15:\n\n(5 comments)","accounts_in_message":[],"_revision_number":15},{"id":"a9520f7ddd80a3fa07866f4cb08b99fefc183173","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-10 15:00:28.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"7394205f00105ae26c2e59dc072ecfa6aad54896","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-10 16:15:31.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"5a4f0ed9749501b99282b029f451851d047f5e63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 16:37:38.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/923a939c8829400cb2585cad7bfcb5ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b386308cef844eda066b1e50a297e2b : SUCCESS in 4m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1803e0152acb449daede54c130a34de2 : SUCCESS in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1352cc64139f4a14aa41aef2d43d63d8 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41c02329ae9b4f42bf94718719440f2f : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f77df9833c2249f1b4c1b2b4fac0389e : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff194da85a3a4859bc9f7e7f6507887b : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c413a6e21c56474197a29b92b2e98e33 : SUCCESS in 6m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46c18afddaab490ca3b9c57e3d3a26ff : SUCCESS in 2m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8e884535448c440daa05e5f13cafbc14 : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aa9ebe3c4bcd4da2b64f4c2cc1521770 : SUCCESS in 10m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a05bb238ec3847409f406ee222021931 : SUCCESS in 6m 04s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d3920c179da042e680b7590c541f7e1d : SUCCESS in 17m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7a7373ffdfc343318143d83b7b6c7167 : SUCCESS in 13m 47s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fccff4b37a474f8a88f8d04b2e2e1aeb : SUCCESS in 12m 38s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b3a6cf7e380045769b431986f761962b : SUCCESS in 23m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fd5e306decae48e091be48f8eded6421 : SUCCESS in 15m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fbeda085c114a6092f40b06681a2395 : SUCCESS in 16m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b4bf26bc04b471ba7bfa89c4e0a8201 : SUCCESS in 50m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/faa6b0bb632641fa8de66caed365d2b2 : SUCCESS in 50m 28s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e73cc9e4c724bbb8d37d8b3d2976f94 : SUCCESS in 1h 37m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b63d13768a8d4b0988c4dfb3c0737f5d : SUCCESS in 16m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/26d66e39ab0d4fb59469d163f29e5dd4 : SUCCESS in 13m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61245f1e541f45bb99014a90deea07b0 : SUCCESS in 27m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/faadcd12b6b647958d3ae9e2659fa3de : SUCCESS in 16m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7915390d06b748cf9068b5651e618233 : SUCCESS in 11m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91b0492ec0e34383bd43ef070f73c636 : SUCCESS in 12m 39s","accounts_in_message":[],"_revision_number":15},{"id":"c5088451471b7b446275566e54c8969e738a169e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-10 16:51:21.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"d5ca7e367d0459c5181bb04a1e58fc4579c29b19","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-13 07:43:02.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"c9ed94b44e5312868477124b0664c6eba15b10b6","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-13 08:25:06.000000000","message":"Patch Set 15: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"3cc46334e8870199258b3fde981fe3eb5a2b42a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-13 08:36:57.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"d0cf10b0ab0d4b31b120871b9744e49c8e3bda81","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-13 08:37:52.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"0040b350faa86f930c0d09d0f6b11bdc7f440462","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-13 08:41:53.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"e13da6f2fd623bb27dd2bbfbf307e9fcbbc49bbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-13 08:43:51.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":17},{"id":"3e796858dc28253ed6ee9137a515eb49e31c3cff","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-13 08:44:44.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"5fe087eb557631c595cabd5a9ab8e2865a29eda8","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-13 08:48:17.000000000","message":"Patch Set 17: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"0cfac14df15bfdb2aac932f5c29e4509807e831d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-13 09:10:05.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":18},{"id":"2e1d0d200d9307d3cce61800621ca28002acfaf6","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-13 09:10:24.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"ea002ab8b8a5eb9bcaaf5b30fe54f934909e401b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-13 09:17:47.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"063b30a28b8871f32edd168ccf2d139ba4b517a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 10:45:25.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f0790d537c34c77b422032af1d55f5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/644888a608ef49ffa35f285cf86b0e02 : SUCCESS in 7m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/27487ff0d9ca4220b81c816ae7d8b71c : SUCCESS in 6m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/143e2819c02642f199a69c02051f30eb : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03996109b0a9455695b65e40ad305560 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1abe9f9ce0a4d168626c94fd60aaff7 : SUCCESS in 2m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b11083d82a24edeb445934543ad79a2 : SUCCESS in 5m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1003d4749ead4363b9309b17daee5549 : SUCCESS in 4m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1df1f9f7dbc4e26b201b781f5f7a9d7 : SUCCESS in 4m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ec7b53bdd4574422a680412b3d069ddd : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ea1857f2ac9042fd974f3db02d08c319 : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2cc6fe56712a401b9e69f9157a87412e : SUCCESS in 9m 25s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d610ea60e1bf42dc9d80c9fcfcf81731 : SUCCESS in 16m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3ced39ee56494cb4b4678faa71ee165c : SUCCESS in 13m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dab37560be2c4833aba95adea106cabd : SUCCESS in 10m 26s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d37a2f97e656479f8856142020497a42 : SUCCESS in 16m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b6d0336d600b4d399e040a3dd4a4b0d5 : SUCCESS in 18m 19s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f92269a98e44989b24564f3d951e3d7 : SUCCESS in 13m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8d112b23ebc4132bfb9848a509184f7 : SUCCESS in 31m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cf91596640574afd96a93c5f5d15c603 : SUCCESS in 57m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/776634996f35454ea2bfb78b5098fd67 : SUCCESS in 1h 33m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5934253f34c48f3aa17d6106f64b46e : SUCCESS in 12m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f9a7e250659448d68b1052d1a83a1472 : SUCCESS in 12m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcf642fd493c4869a4a0db89045cf455 : SUCCESS in 22m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/77f02bc0c715416cb31f5c8d5d7a3815 : SUCCESS in 13m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6d1fd98d4e17438a963401976ae04c9c : SUCCESS in 13m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4db5d32cb9246b2be4e0c0a32f6465f : SUCCESS in 12m 31s","accounts_in_message":[],"_revision_number":18},{"id":"bbb3c0906c31069bd1f6e2422158374184a316d9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-13 12:24:57.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"d572300c22be90aa5857e2fc94801b37341bfd4a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-13 17:25:03.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"4cec8aebadd8cc6a031274f1d14f7e0dd250706d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 17:25:18.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"e459bb1a3b2a6879e800d58a8b72d5cb1147df66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 18:53:32.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9660abf98f54650a02d1a9ce6d81cae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85872d8ecc5c478497c6bd7025bb3f7f : SUCCESS in 9m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7678743316124b278738e4c0f1ed3b30 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eca3708e251845a29f5b5d82caa615ad : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04dc88ea48244711bfa27b1697cb02d4 : SUCCESS in 6m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c605e10cfb634261b245253cfeb81ce3 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e50060a7a1c4221aab3aee2d4ededb9 : SUCCESS in 4m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/591aacd1dfa64941b65613d0b7b3ca60 : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/33263b323f9b4b66a49c44712806452e : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8e63e554807f40679b52e9b024ef90e2 : SUCCESS in 7m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3c6f4194a9ef4ebf9c1c16fe0e7aa0c7 : SUCCESS in 16m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3c81a7aeae9448d2af51d82259d4a712 : SUCCESS in 7m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ceb49dc17de340e4b66379666fe5b1e6 : SUCCESS in 13m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/55e0eb8da2a44bb8852d750c45549fc4 : SUCCESS in 9m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/11c6fde87450435b9b13db865188dc87 : SUCCESS in 14m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67ed5b04ab8f4db3af0fe6395c698087 : SUCCESS in 7m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b19e10d370147c69a7c37924b294e95 : SUCCESS in 27m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/189b1d1a30de4b4bb5f8e05237b44d0b : SUCCESS in 51m 55s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e946158e1ee64ad7bee738829066546a : SUCCESS in 1h 27m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/099fba84552f41ccaae6e7186671689f : SUCCESS in 9m 06s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fd3f53d106c44621904639b91b792a6c : SUCCESS in 16m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a78129a06764e0fb223c01fdef24f26 : SUCCESS in 26m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4ed875b9d5d04960be0ad48381148f42 : SUCCESS in 13m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a89301f2da9c44ef8fadd0bce5d0fb01 : SUCCESS in 13m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f14ae1f0c7a243e99d9de094cd902e1d : SUCCESS in 11m 48s","accounts_in_message":[],"_revision_number":18},{"id":"50585b983608148cd1ae3f4eba97cbb0757b6a8c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 18:53:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":18},{"id":"84c534cbbb9f182b2c93464b8a9f50ea42883373","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 18:54:29.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9dea77d3e6e4197bb8d12c36ace40c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97429db1f6ca48b6b180426fc8625466 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edf0f1b2c983494d87ef447f0a6f4ea3 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"6986fea5983268448a529d03b7babae8976cb898","revisions":{"e1e9345c9b6a5ac88249a4ee4cc9c6ad5fbfe05a":{"kind":"REWORK","_number":1,"created":"2025-09-25 14:25:01.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/85/962285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/1"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"subject":"WIP: Change IPA compression algorithm from default gzip to zstd -19","message":"WIP: Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This change will reduce\nthe transfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1e9345c9b6a5ac88249a4ee4cc9c6ad5fbfe05a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1e9345c9b6a5ac88249a4ee4cc9c6ad5fbfe05a"}]},"branch":"refs/heads/master"},"bf1591c7e957fe91b55788000fb6d7266e090917":{"kind":"REWORK","_number":2,"created":"2025-09-30 10:39:25.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/85/962285/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/2"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-30 10:39:16.000000000","tz":60},"subject":"WIP: Change IPA compression algorithm from default gzip to zstd -19","message":"WIP: Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This change will reduce\nthe transfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf1591c7e957fe91b55788000fb6d7266e090917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf1591c7e957fe91b55788000fb6d7266e090917"}]},"branch":"refs/heads/master"},"0a8bd6c1f50f81cbd734b9d11f7ffec881e9268a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-10-02 13:55:58.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/85/962285/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/3"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-10-02 13:55:58.000000000","tz":0},"subject":"Change IPA compression algorithm from default gzip to zstd -19","message":"Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This change will reduce\nthe transfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a8bd6c1f50f81cbd734b9d11f7ffec881e9268a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a8bd6c1f50f81cbd734b9d11f7ffec881e9268a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c0f1e030e002c3215b3da1b043bb62d39b8e4453":{"kind":"REWORK","_number":4,"created":"2025-10-02 15:01:27.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/85/962285/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/4"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-02 15:00:25.000000000","tz":60},"subject":"WIP: Change IPA compression algorithm from default gzip to zstd -19","message":"WIP: Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This reduces the\ntransfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0f1e030e002c3215b3da1b043bb62d39b8e4453"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0f1e030e002c3215b3da1b043bb62d39b8e4453"}]},"branch":"refs/heads/master"},"832675b434ab010c72ffa48b4724016e9857ba6b":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-10-02 15:01:53.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/85/962285/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/5"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-02 15:01:43.000000000","tz":60},"subject":"Change IPA compression algorithm from default gzip to zstd -19","message":"Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This reduces the\ntransfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/832675b434ab010c72ffa48b4724016e9857ba6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/832675b434ab010c72ffa48b4724016e9857ba6b"}]},"branch":"refs/heads/master"},"801361910e2f60df4d18bedf3bb3b00e3910cb51":{"kind":"REWORK","_number":6,"created":"2025-10-03 09:05:26.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/85/962285/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/6"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-03 09:05:26.000000000","tz":0},"subject":"Change IPA compression algorithm from default gzip to zstd -19","message":"Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This reduces the\ntransfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/801361910e2f60df4d18bedf3bb3b00e3910cb51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/801361910e2f60df4d18bedf3bb3b00e3910cb51"}]},"branch":"refs/heads/master"},"e8ba765eee165144ba3a1236230ea7adf82a04a2":{"kind":"REWORK","_number":7,"created":"2025-10-03 12:26:38.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/85/962285/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/7"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-03 12:26:21.000000000","tz":60},"subject":"Change IPA compression algorithm from default gzip to zstd -19","message":"Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This reduces the\ntransfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8ba765eee165144ba3a1236230ea7adf82a04a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8ba765eee165144ba3a1236230ea7adf82a04a2"}]},"branch":"refs/heads/master"},"07985fa114b6b568da1e0f7c7904b9840910940c":{"kind":"REWORK","_number":8,"created":"2025-10-03 14:16:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/962285/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/8"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-03 14:16:44.000000000","tz":0},"subject":"Change IPA compression algorithm from default gzip to zstd -19","message":"Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This reduces the\ntransfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07985fa114b6b568da1e0f7c7904b9840910940c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07985fa114b6b568da1e0f7c7904b9840910940c"}]},"branch":"refs/heads/master"},"abcd0d70bcf73b25932f6f992733125d30c00ee2":{"kind":"REWORK","_number":9,"created":"2025-10-08 08:40:06.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/85/962285/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/9"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-08 08:39:54.000000000","tz":60},"subject":"Change IPA compression algorithm from default gzip to zstd -19","message":"Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This reduces the\ntransfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abcd0d70bcf73b25932f6f992733125d30c00ee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abcd0d70bcf73b25932f6f992733125d30c00ee2"}]},"branch":"refs/heads/master"},"4a266451e930ab3eaa71a960a99dd04fd79b580a":{"kind":"REWORK","_number":10,"created":"2025-10-08 09:52:04.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/85/962285/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/10"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-08 09:52:04.000000000","tz":0},"subject":"Change IPA compression algorithm from default gzip to zstd -19","message":"Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This reduces the\ntransfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a266451e930ab3eaa71a960a99dd04fd79b580a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a266451e930ab3eaa71a960a99dd04fd79b580a"}]},"branch":"refs/heads/master"},"5c47c03cdd5ace37c1d21cd24fe72b2b8b890446":{"kind":"REWORK","_number":11,"created":"2025-10-08 14:33:53.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/85/962285/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/11"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-08 14:33:53.000000000","tz":0},"subject":"Change IPA compression algorithm from default gzip to zstd -19","message":"Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This reduces the\ntransfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c47c03cdd5ace37c1d21cd24fe72b2b8b890446"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c47c03cdd5ace37c1d21cd24fe72b2b8b890446"}]},"branch":"refs/heads/master"},"806215909eda533409b0ab339b2a862480354704":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-10-09 10:21:35.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/85/962285/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/12"}}},"commit":{"parents":[{"commit":"5bffc12c085aa6f0468fc3a61e414af44a60df73","subject":"Bump Ceph package repository to Squid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5bffc12c085aa6f0468fc3a61e414af44a60df73"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-09 10:21:35.000000000","tz":0},"subject":"Change IPA compression algorithm from default gzip to zstd -19","message":"Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This reduces the\ntransfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/806215909eda533409b0ab339b2a862480354704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/806215909eda533409b0ab339b2a862480354704"}]},"branch":"refs/heads/master","description":"Rebase"},"7d2ec66b90e3222ff8bd5d3219343c0fb92e32d2":{"kind":"REWORK","_number":13,"created":"2025-10-10 14:55:09.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/85/962285/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/13"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-10 14:55:01.000000000","tz":60},"subject":"Change IPA compression algorithm from default gzip to zstd -19","message":"Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This reduces the\ntransfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d2ec66b90e3222ff8bd5d3219343c0fb92e32d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d2ec66b90e3222ff8bd5d3219343c0fb92e32d2"}]},"branch":"refs/heads/master"},"3b5ca87bd08c2d6101e23ed47c8197eeff01055b":{"kind":"REWORK","_number":14,"created":"2025-10-10 14:56:45.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/85/962285/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/14"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-10 14:56:45.000000000","tz":0},"subject":"Change IPA compression algorithm from default gzip to zstd -19","message":"Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This reduces the\ntransfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b5ca87bd08c2d6101e23ed47c8197eeff01055b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b5ca87bd08c2d6101e23ed47c8197eeff01055b"}]},"branch":"refs/heads/master"},"67801527722bfe7f7efe16f1584e55df8c029f0a":{"kind":"REWORK","_number":15,"created":"2025-10-10 14:57:18.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/85/962285/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/15"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-10 14:57:18.000000000","tz":0},"subject":"Change IPA compression algorithm from default gzip to zstd -19","message":"Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This reduces the\ntransfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67801527722bfe7f7efe16f1584e55df8c029f0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67801527722bfe7f7efe16f1584e55df8c029f0a"}]},"branch":"refs/heads/master"},"a2a317880ecd46538ab811d76baa17900002721e":{"kind":"REWORK","_number":16,"created":"2025-10-13 08:36:57.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/85/962285/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/16"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-13 08:36:45.000000000","tz":60},"subject":"Change IPA compression algorithm from default gzip to zstd -19","message":"Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This reduces the\ntransfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2a317880ecd46538ab811d76baa17900002721e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2a317880ecd46538ab811d76baa17900002721e"}]},"branch":"refs/heads/master"},"9b9a25a51a2ac7991a60e686abb386e19e6d16f9":{"kind":"REWORK","_number":17,"created":"2025-10-13 08:43:51.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/85/962285/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/17"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-13 08:43:44.000000000","tz":60},"subject":"Change IPA compression algorithm from default gzip to zstd -19","message":"Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This reduces the\ntransfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b9a25a51a2ac7991a60e686abb386e19e6d16f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b9a25a51a2ac7991a60e686abb386e19e6d16f9"}]},"branch":"refs/heads/master"},"6986fea5983268448a529d03b7babae8976cb898":{"kind":"REWORK","_number":18,"created":"2025-10-13 09:10:05.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/85/962285/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/962285/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/962285/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/962285/18"}}},"commit":{"parents":[{"commit":"751d52ca443caaee592c83082b27c3fbc802b1c8","subject":"Merge \"ironic: Add kolla_ironic_inspection_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/751d52ca443caaee592c83082b27c3fbc802b1c8"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-25 14:21:54.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-13 09:09:52.000000000","tz":60},"subject":"Change IPA compression algorithm from default gzip to zstd -19","message":"Change IPA compression algorithm from default gzip to zstd -19\n\nUpdates the IPA (Ironic Python Agent) compression algorithm from\nthe default gzip to use zstd -19 instead. This reduces the\ntransfer size of the IPA boot ISO from the Ironic conductor\nto the baremetal nodes during provisioning.\n\nChange-Id: Id0938c3fef6ccb9ceb556ce68a442393eabf2624\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6986fea5983268448a529d03b7babae8976cb898"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6986fea5983268448a529d03b7babae8976cb898"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
