)]}'
{"id":"openstack%2Fkayobe~962380","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I1b71990af38a8fd6c6258a49d2b94e16496fb274","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-09-26 21:49:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1b71990af38a8fd6c6258a49d2b94e16496fb274","subject":"Network connectivity: Fix regression with noip","status":"MERGED","created":"2025-09-26 16:40:52.000000000","updated":"2025-09-26 21:49:53.000000000","submitted":"2025-09-26 21:49:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962380","cherry_pick_of_change":962251,"cherry_pick_of_patch_set":3,"meta_rev_id":"c00b04a95b10cbaca0b3f70f397034c06af0d6a5","_number":962380,"virtual_id_number":962380,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-26 21:49:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-09-26 17:06:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-09-26 17:18:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-09-26 17:06:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-26 16:40:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2025-09-26 17:06:31.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-09-26 17:18:22.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-26 19:17:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c26446fd5ee9e4846aac5f09a695412f30abcdb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-26 16:40:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"60d0336c37ba1d0caccdd84a8030af98a255d0c2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-26 17:06:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6f860c7267e4c4ba047e9034befb9b433b348ff2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-26 17:18:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f190414fd224e4f10c6b64d0e1a2a0d522735b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 19:17:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/898e811237754c1da032b808ab68f1a1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69abc2494e514888a5149dd0a68af602 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d236574bbc1433398191c56c9ea1bbd : SUCCESS in 5m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/37e1a1ba1c2241b282f02fbc48186e2e : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe38351ef1764946bf2d2311986493ba : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5b283b72fe74774821bc7f9f35894cf : SUCCESS in 3m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3cba83176f204c94bf4808ee33920c6b : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0c9960f4bc474b378d53bee86c212699 : SUCCESS in 13m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c98683f75b764964be60e1c6af0727ed : SUCCESS in 7m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8057ceaf1f5a4ca3a63b237bfdafbabe : SUCCESS in 59m 21s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/087b67430e2641e3a9985d9ba6cd4932 : SUCCESS in 1h 09m 44s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af01d39df24f47178f0a3fede91963e3 : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/396f84cb7c5444edb79575ec1e4245d7 : SUCCESS in 55m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a534ddea076e4efaa10a70829d5ce81c : SUCCESS in 1h 06m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/88a8a9b1800a4623900b37f499fb8f6e : SUCCESS in 21m 19s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb56af0bf2144ade8b34106280d5d1af : SUCCESS in 8m 41s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ae15caf0d93f4e00921470386aac1b36 : SUCCESS in 2h 26m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c53601c5504d4f40a8077247a12621e8 : SUCCESS in 2h 16m 16s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0267ca3749464902876e9033fc9dd0c4 : SUCCESS in 1h 51m 52s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91349b21143d412bbc3a1d61cd84c90a : SUCCESS in 2h 04m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/17c63dbc392d48508ada56822f23977b : SUCCESS in 8m 30s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/5a1f196059ac4303a4b106f1e7b35b7e : SUCCESS in 15m 01s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61eae83150ca41939573925411e3906a : SUCCESS in 18m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/312478e5df9947c8b03dc5dab1e434c1 : SUCCESS in 14m 28s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c5d4dcf8cc3e4235890aad517e5685f4 : SUCCESS in 23m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aee717e6fa0f492eb6ca4a9d0d168546 : SUCCESS in 22m 58s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/dcee3ad08bec464b8fbf324f69e63954 : SUCCESS in 36m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c45ba5a7e1604a3f92b5b783a18c8705 : SUCCESS in 11m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa2c57dafb794706bc92dec395730253 : SUCCESS in 11m 24s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/967d4eed14294d14871bc848ac1283b4 : SUCCESS in 13m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80a577a8c0f444acbbe202d8eccfd4b1 : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b8dae76bceabd3ecedc434d2d0f6ad20cc9d7e46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 19:18:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"739b62ef1ac01a9f8d0af5f1c0d6f4d2f7b44a20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 21:49:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ec5adfb5670418396a1279cff1bb6ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/121e19bc83e6456b9de295d21f1a0d5f : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/072b969151644230a934f06470abb159 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/810ce62b035349a79d0de812ee1792c4 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9836e7898a3549ab8b1ae92715490c5e : SUCCESS in 2m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/68265fe0517240f9bbb0551a26d91cf4 : SUCCESS in 3m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1bfb1d9c0bbe4ae09af7060d69f06566 : SUCCESS in 5m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/95f2326b40394104b0ff361ebc5a20db : SUCCESS in 12m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d29e60b07c294ad288550992e4505d87 : SUCCESS in 43m 50s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/27dc85b489d84b49b69b98a51da2572b : SUCCESS in 46m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a682d177cf94fbd895f24e23907b6bd : SUCCESS in 32m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/09c29815102041b5aa15fe2ad2c809f7 : SUCCESS in 29m 05s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/918016bf0c354b2b9c7a51da53ce5dc8 : SUCCESS in 47m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e0f7ac5e8543409f95d9643163baeff9 : SUCCESS in 17m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf134ced5ea347d297d00fad9a9cbe67 : SUCCESS in 7m 07s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/eecafc04ce894f8a8a3648db2fd49c14 : SUCCESS in 2h 28m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7e051645b2044338ed4cda18f10a0a3 : SUCCESS in 1h 40m 46s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/d1e7afc2aa594ba88dbd9b093d924397 : SUCCESS in 2h 19m 48s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9da992bc92e49e7a14425b45438b099 : SUCCESS in 1h 11m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/631e10e875fb43b8a5f2d247f3d7ba8e : SUCCESS in 9m 55s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/88209aab310e4140bf3e24c8533d0e31 : SUCCESS in 19m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/989559f32cfa4228b5ab14affc7e3733 : SUCCESS in 14m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/65b1af66056e447b8f6d80dc814d0a35 : SUCCESS in 16m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cc13bb2bab834f448f0fac76fb2d11e1 : SUCCESS in 31m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34328b0b042f4783bdfcf97a9a5325cf : SUCCESS in 31m 32s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a2439cc990144ea3868fd9fba049cb41 : SUCCESS in 27m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/61478da1ff214158b42b83a217be9742 : SUCCESS in 12m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c19bd1f7d05e41cf8623ee699571d4d0 : SUCCESS in 9m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7720a6801a1849abad7ebe4fce317071 : SUCCESS in 13m 22s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3924b74d0d8d497e882aad1d0bb3afe1 : SUCCESS in 11m 43s","accounts_in_message":[],"_revision_number":1},{"id":"ab711ebe05e3478b21275fddd25cf845de47e5eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 21:49:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c00b04a95b10cbaca0b3f70f397034c06af0d6a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 21:49:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/807de32a6f34437993e913e65feeb04f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6a1ba912acd435384026f2daf7f49d4 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47e85910079044fa930a0b6de3b7e003 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8b39a2f42433e3acea3835fbd67d71054d77724","revisions":{"b8b39a2f42433e3acea3835fbd67d71054d77724":{"kind":"REWORK","_number":1,"created":"2025-09-26 16:40:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/80/962380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/962380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/962380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/962380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/962380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/962380/1"}}},"commit":{"parents":[{"commit":"24890d844cd170396c56ea49fbb1a52d69ae903b","subject":"Merge \"Change default centos bootstrap user to cloud-user\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24890d844cd170396c56ea49fbb1a52d69ae903b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 08:34:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-26 16:40:52.000000000","tz":0},"subject":"Network connectivity: Fix regression with noip","message":"Network connectivity: Fix regression with noip\n\nWhen using no_ip on all hosts on a given network, the \u003cnetwork\u003e_ips key\nwill not exist in network-allocations.yml at all, so we need to tolerate\nthis key not existing.\n\nCloses-Bug: #2125560\nChange-Id: I1b71990af38a8fd6c6258a49d2b94e16496fb274\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit f376ae079519c0425d76e53022801754560ac7e7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8b39a2f42433e3acea3835fbd67d71054d77724"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8b39a2f42433e3acea3835fbd67d71054d77724"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
