)]}'
{"id":"openstack%2Fkayobe~962382","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~I1b71990af38a8fd6c6258a49d2b94e16496fb274","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-09-26 20:18:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1b71990af38a8fd6c6258a49d2b94e16496fb274","subject":"Network connectivity: Fix regression with noip","status":"MERGED","created":"2025-09-26 16:41:41.000000000","updated":"2025-09-26 20:19:23.000000000","submitted":"2025-09-26 20:18:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962382","cherry_pick_of_change":962251,"cherry_pick_of_patch_set":3,"meta_rev_id":"2a68647c69c0f035aa6f30a7e0e7998528b041f6","_number":962382,"virtual_id_number":962382,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-26 20:18:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-09-26 17:06:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-09-26 17:18:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-09-26 17:06:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-26 16:41:41.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2025-09-26 17:06:37.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-09-26 17:18:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-26 19:04:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c5b0cd996561c6d40862b8b4501e77d7f446b5c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-26 16:41:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a171633623d92f551d78e9d93fb066307bc93516","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-26 17:06:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7272cd7fee0d4511bf31366873d3f9fbc2d4f4c3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-26 17:18:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"eb36748bc0c1a899d9dd027a1f7ebdada71e2283","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 19:04:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ef1f5e8171a4dac8ba6a98c9205e8c4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/561ef8387e6a4ba5867fea1ea382dfb1 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec8580eb20a6495baf8585b00e939da7 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c8f9e375c674507b644807d1e3d1cb4 : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cc7c2ee274e246bb8380ef07a8f8c704 : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0f175f8894ab466cbffe2093cf6ab6ed : SUCCESS in 12m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1ff02e2eb25a493c85602e0ee19b8c13 : SUCCESS in 9m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cb9540716db346e38029a2a22b633fa3 : SUCCESS in 1h 07m 39s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ed0ca9f40c73462686683aff7d9ee1de : FAILURE in 8m 31s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ca8e9951d3140b286fe9dc48c6f8455 : SUCCESS in 45m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5c3540eddf14bdc963cb62de8fcdc5b : SUCCESS in 59m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/18df2938af2945889f974cecc5a4797e : SUCCESS in 1h 11m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a2265c638d0a4e43972e26c94200eaff : SUCCESS in 18m 39s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/744c5c73236c4b1f93f458a90e3c5139 : FAILURE in 7m 10s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00a13c60343f4fc69578d3a76b84ea47 : SUCCESS in 14m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbaff575bacb44a1a73631153bd32fa0 : SUCCESS in 11m 10s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/813d59dcb0c9446c95588fb29a8aec7d : SUCCESS in 2h 17m 47s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bbde791a7a549638ed47e2f3b1bc563 : SUCCESS in 57m 30s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/491ae95d19bc4b51969e10f7dcfbdedb : SUCCESS in 22m 53s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e8a3767974d458384422eda975036f5 : SUCCESS in 17m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cebc3051068242ada42add948a2d3847 : SUCCESS in 17m 29s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/730d9044ae404a818c56999488c799a6 : SUCCESS in 31m 16s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d4b2f832d8148e7bca88029f123fb34 : SUCCESS in 23m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b68cb31e536a455fb8ee9af70572a609 : SUCCESS in 16m 28s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/9433df91a8904c8a841d0ca533994e42 : SUCCESS in 16m 28s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eba15fb1140349c89cb0f1bd70c085d4 : SUCCESS in 12m 35s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/3960ca9ba69a4dae9c8c18c6c6e231dd : SUCCESS in 13m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7ecf77659194a9cbc6937e6dc3b5dd8 : SUCCESS in 16m 42s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/47697f50088c4be8b8dae2ac6f3bf7f4 : SUCCESS in 14m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dd1a21c723344b028a4731381f222709 : SUCCESS in 16m 35s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a021aa086ef4c128ab979893d1f2f06 : SUCCESS in 9m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8d6631dc190427a89f1f35637a8cc10 : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/648f7e987f844b06922ecb9b509fdc6c : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31560df7edb34e3aa38f16e3a6acd384 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc9b1a0a2f444d0b80bc5ade1c4d603c : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52ae1d7924114545ab5dcdf23638840f : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"475e8071412555f71c326ba9729c984d1cb35e68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 19:05:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3071f184584aad254546edfe30d657a3a44dbfad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 20:18:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e460ce5c05b4b5d94bdad8871993287\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f113ac2a94a42b89b0a7daa1c1fff0a : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a410bccdd74046c28f83871f3017a018 : SUCCESS in 3m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6905c54d181b410dadb840f682b90f40 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c22b8f39790c444d94e9170702d7b47d : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7e76deeee40547a1bca6e45abcce5d85 : SUCCESS in 10m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7b6a7fab9d884b65815bb48d8c411794 : SUCCESS in 1h 00m 55s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62db414a74d641adb474c591be68dee2 : SUCCESS in 56m 58s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5eba1b95759d4faea4ced443c9a6790f : SUCCESS in 55m 28s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6e100d3cc8734a1c9342cb5912513376 : SUCCESS in 1h 08m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/914e969423284e13834a56fe063b1fbb : SUCCESS in 10m 33s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e037806f24d64fe5a19d5ddec41a32bd : SUCCESS in 8m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34adf07e153f405084734c57c621aa03 : SUCCESS in 14m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6cb4e531f88b4ef48f17bccc24002913 : SUCCESS in 11m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b994df052fe7421a833fb2b4e7655d2d : SUCCESS in 18m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a19fb38749d4323bb1cffce8c3f2489 : SUCCESS in 17m 32s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/cdec4ab118824bfa8e0183b987084706 : SUCCESS in 29m 32s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98d3f7447e6e4732b94769abe01c9a92 : SUCCESS in 25m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6fc359ce3ad8489f8a80ee294e5eed93 : SUCCESS in 16m 31s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb3c2fe07f304f3c9fed7071b0d19455 : SUCCESS in 9m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dea2bfe554a6466c97b59a3bf59b2f7a : SUCCESS in 15m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/917e7729d91e4e96a7e15b3de85b5cfb : SUCCESS in 10m 16s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/677c6d096e7647a6816cb7cfb65b3351 : SUCCESS in 8m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08d146e7e3c04c4e83d45d057c9a21b6 : SUCCESS in 14m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0e7807d3bbecfa2d275871926ddb0a124ca010dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 20:18:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2a68647c69c0f035aa6f30a7e0e7998528b041f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 20:19:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bdf1cf2e17043578e12eda194fbb8d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b167a4896db46d0b3f678bb52fd4963 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/692f2690ab3b451987ff7f494e020fce : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b862814b86b7dcdcd8c33070657c826ae0102eca","revisions":{"b862814b86b7dcdcd8c33070657c826ae0102eca":{"kind":"REWORK","_number":1,"created":"2025-09-26 16:41:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/962382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/962382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/962382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/962382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/962382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/962382/1"}}},"commit":{"parents":[{"commit":"e1fbb42088296ddb4146866cb43da91a37389020","subject":"Change default centos bootstrap user to cloud-user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1fbb42088296ddb4146866cb43da91a37389020"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-25 08:34:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-26 16:41:41.000000000","tz":0},"subject":"Network connectivity: Fix regression with noip","message":"Network connectivity: Fix regression with noip\n\nWhen using no_ip on all hosts on a given network, the \u003cnetwork\u003e_ips key\nwill not exist in network-allocations.yml at all, so we need to tolerate\nthis key not existing.\n\nCloses-Bug: #2125560\nChange-Id: I1b71990af38a8fd6c6258a49d2b94e16496fb274\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit f376ae079519c0425d76e53022801754560ac7e7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b862814b86b7dcdcd8c33070657c826ae0102eca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b862814b86b7dcdcd8c33070657c826ae0102eca"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
