)]}'
{"id":"openstack%2Fkayobe~962473","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I002cadf4c14325bbb10bbfd2efeeb87084bd4314","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-09-29 12:57:29.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-09-29 14:33:08.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-09-29 15:26:59.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-09-29 12:02:41.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I002cadf4c14325bbb10bbfd2efeeb87084bd4314","subject":"CI: Fix Ubuntu cloud image tests","status":"MERGED","created":"2025-09-29 09:15:53.000000000","updated":"2025-09-29 15:28:00.000000000","submitted":"2025-09-29 15:26:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962473","cherry_pick_of_change":961783,"cherry_pick_of_patch_set":5,"meta_rev_id":"ff9cea6a7d84e5922aaf0d7d04c10369d7581a04","_number":962473,"virtual_id_number":962473,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-29 15:26:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-09-29 12:57:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-09-29 14:33:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-29 12:02:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-09-29 12:57:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-29 10:55:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-29 11:06:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-29 11:32:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-29 11:39:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-09-29 11:39:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"21d305dda47c64a787e9a9745dd4986e2af8bdbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-29 09:15:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"177c48ba5750711fa0c413636de4386a69628efd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-29 10:05:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1539d96ebb3df88ef305d6523e8bad27bb147c0a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 11:06:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d52624907f04c019b219f949385c32a\n\n- kayobe-infra-vm-centos9s-cloud-image https://zuul.opendev.org/t/openstack/build/11cbfb7cee8b4fa8910e54c217c3add8 : SUCCESS in 20m 51s (non-voting)\n- kayobe-infra-vm-rocky9-cloud-image https://zuul.opendev.org/t/openstack/build/70e08fa585a6499da128668d6900ab27 : SUCCESS in 16m 09s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/d87c3b09d8bc443eb8be85661bbcaf31 : SUCCESS in 7m 17s (non-voting)\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8e337c16c25a406bb56dd32b20e2e30c : SUCCESS in 57m 49s (non-voting)\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d7eb11630595479387fee2bf9c16b4b0 : SUCCESS in 9m 32s (non-voting)\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/9555f8deec344cbbb67f73cc950ecfdb : SUCCESS in 54m 44s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1e3f63aa4a4f418db29d4aecb4b5db9d : SUCCESS in 35m 30s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/ff05c33f8f0f4241baa0a515d73dbae4 : SUCCESS in 33m 22s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a24b352f6d4c420e8b0cc69347295511 : SUCCESS in 46m 31s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7d85267a0fbd44b4add45c34d8c50e38 : SUCCESS in 31m 34s\n- kayobe-seed-vm-centos9s-cloud-image https://zuul.opendev.org/t/openstack/build/389adf89ace441b9a4581a0319ce3bf9 : SUCCESS in 9m 33s (non-voting)\n- kayobe-seed-vm-centos9s-q35 https://zuul.opendev.org/t/openstack/build/c074b225a9cf45128ad1c1138b6f21c9 : SUCCESS in 7m 27s (non-voting)\n- kayobe-seed-vm-rocky9-cloud-image https://zuul.opendev.org/t/openstack/build/72442e753256413b9d6557f122ed2992 : SUCCESS in 10m 17s\n- kayobe-seed-vm-rocky9-q35 https://zuul.opendev.org/t/openstack/build/2c4f942ed1bf42b5881fa08676b5d75b : SUCCESS in 13m 24s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/8f79a33666424ad588bc040185a9ce54 : SUCCESS in 15m 43s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/de7ca430ce694eeda9ca1284b52fe9ab : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e8537de4defb0cee01e5f4e2ad088e005deff27c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 11:32:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ba90153723c431ab94030aaa456de7d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba7b096c6eae4c7a88de437a46b570b4 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7add2be377224373b85244d47f0d3537 : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ceca898c5d947778e62d1aff063ca2d : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20babb7592ec4dc186d068b294efb7de : SUCCESS in 10m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a93769d7a76b4df9a53684993cc3619b : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/51355f941d364b258c05617f187c5673 : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a0e30482a6db4273925e878c4e5dcc49 : SUCCESS in 10m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f2a87ff332b5497ea47382f0ccb9b569 : SUCCESS in 1h 02m 00s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/b4f38e63385445db8f7274fbb8e73f1f : SUCCESS in 1h 06m 50s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b50a10f95cf94547b718fa513f8971b0 : SUCCESS in 48m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fcb6c464b93d471b982ff6395113c053 : SUCCESS in 52m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7f133a07a8c644298e0d575227dc89ad : SUCCESS in 1h 16m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a5a94061adbb41b9b2d4a4fbdf52c62e : SUCCESS in 16m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46d06cb1766b42feb99334a13b2a2343 : SUCCESS in 12m 55s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f9c47ca90f3641c59012b867b1ae5152 : SUCCESS in 1h 58m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c3eec39efe24063880510e211e7e393 : SUCCESS in 1h 15m 14s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e5fccaa69aae4afaa7ce7762b5eb77a6 : SUCCESS in 1h 15m 49s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9cefcbfc2c84b98b78ed86cc6ca80b1 : SUCCESS in 2h 11m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fd94754d319349e4af827132eb6687ed : SUCCESS in 17m 27s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/44513195dac348fcac71d48f071bec6e : SUCCESS in 16m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fd4368f497b4d459b1764146fd0619c : SUCCESS in 14m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/94c42183d37f48e8970268d527c21406 : SUCCESS in 14m 35s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a6b391f847d644e788f1fbb3aae9aad7 : SUCCESS in 28m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c906e39ea6a64358923da165a4423bdb : SUCCESS in 31m 15s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/bed8f8cdabbd42e29cf1b1a1b7d6e88f : SUCCESS in 33m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3f4b4be14721445e9bd137c0ccd6a7c9 : SUCCESS in 7m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83181100666f440ab0ae725a2cea0263 : SUCCESS in 12m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fa8d4770280240c2be77a52b04fd9e4e : SUCCESS in 10m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/848646b3b6064ac3902480f562abd350 : SUCCESS in 11m 03s","accounts_in_message":[],"_revision_number":1},{"id":"52736041c08f1d975cdd54cb1fbfce6ebd12dc44","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-09-29 12:02:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"77ededcc500b98b28165308515b5591aae67658a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-29 12:57:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5e50dad6ba8a8cb63f64026ccd23029252f4b875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 12:57:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f74655db660573d0ce0a5cbb09067d53522755d9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-29 14:33:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5b1cfde2d145480722ac5293cb51257d0fc007b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:26:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c70d666323a94a6585e0ca6ccfdd2a2f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a987e0688d754656ad7276a543b52e46 : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/297e7040fb074adb90c53970e8c23546 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baa4260b6d1043d788429b2553aa16af : SUCCESS in 6m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a0dfbd125e584b868dead735a3bb1b2e : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8fb7b7b05d954d8fa2293e5944ca71c5 : SUCCESS in 11m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e486da63394b487ea75de6653b9b0967 : SUCCESS in 7m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/82b52b9b26544a1eb8e44894e3b93f9e : SUCCESS in 1h 11m 43s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/140c9d25ca5d4b42a438b61dd125b47c : SUCCESS in 34m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a915cb19164640b991f43a7cf1e725ef : SUCCESS in 53m 49s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/807912bcad1c4cfaa9ebc03ab52babc8 : SUCCESS in 31m 20s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b1be8f138c9f4f6c8678c01773325b12 : SUCCESS in 35m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8f0fc01eb75f40bc97be8beb9e5dbc84 : SUCCESS in 17m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5731b807f215426daee8c8347df30675 : SUCCESS in 8m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7c9bd628853842a0a15c18ce271133cc : SUCCESS in 2h 24m 09s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/309a3132bfe04d7c999b7893cbdca785 : SUCCESS in 1h 11m 37s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/53f8d1531020432885b0bb49f69c2c54 : SUCCESS in 1h 16m 09s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16dac4dbb25d464194a31817a4e037fe : SUCCESS in 2h 00m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/19903df325f0474da8ad06c67cc21f19 : SUCCESS in 17m 08s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/7377262faa004707bcfb336a3068cfe6 : SUCCESS in 17m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62fba0faa75e46878c645e8b4ee8c451 : SUCCESS in 15m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/33df7e7800524a8f90d9a2fcb244f96b : SUCCESS in 15m 00s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3f2f0b44a36f424daaaf75b18820be2d : SUCCESS in 25m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e50fd384086473fbd7574d81e084566 : SUCCESS in 28m 46s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/d920a40e5ee4476bad7ac1518e8fbc86 : SUCCESS in 34m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/076586f6ac4548febd6f9c6a237fe005 : SUCCESS in 15m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6577c414746844efa145a0ddc5a9f29a : SUCCESS in 8m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d72a8b0d93234be3a0eb82a53e60aa01 : SUCCESS in 8m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70361a97486045ae82179b99602b01ea : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5dd37d14ee2327b243789336f1505d37b9fa9fd4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:26:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ff9cea6a7d84e5922aaf0d7d04c10369d7581a04","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:28:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ac90df198bd47018b308ad2617d4d6c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3461a1ab562141678d2cf002dc886145 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d045b16bf9b3196f8565f4c44005c7ea0c5de640","revisions":{"d045b16bf9b3196f8565f4c44005c7ea0c5de640":{"kind":"REWORK","_number":1,"created":"2025-09-29 09:15:53.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/962473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/962473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/962473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/962473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/962473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/962473/1"}}},"commit":{"parents":[{"commit":"61501f72c9eb9bb210b91ef82a9ff28aaaba0091","subject":"Fix markup syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/61501f72c9eb9bb210b91ef82a9ff28aaaba0091"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-19 13:59:23.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-29 09:15:53.000000000","tz":0},"subject":"CI: Fix Ubuntu cloud image tests","message":"CI: Fix Ubuntu cloud image tests\n\nUbuntu VMs use the ens2 interface instead of eth0.\n\nWe also increase the VM provisioning timeout, since 360 seconds is not\nalways enough to reach the point where OpenSSH is ready.\n\nChange-Id: I002cadf4c14325bbb10bbfd2efeeb87084bd4314\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit b491c276e789a1c1a0dd262059f3ef8bc2ba538e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d045b16bf9b3196f8565f4c44005c7ea0c5de640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d045b16bf9b3196f8565f4c44005c7ea0c5de640"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
