)]}'
{"id":"openstack%2Fkayobe~962496","triplet_id":"openstack%2Fkayobe~master~I4da3046a521cf5e5b93526bc767a5eca63b6f2ed","project":"openstack/kayobe","branch":"master","topic":"ironic_NTP","attention_set":{"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-11-17 16:33:21.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-10-20 11:39:53.000000000","reason":"\u003cGERRIT_ACCOUNT_17669\u003e replied on the change","reason_account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},"38330":{"account":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"last_update":"2025-10-20 08:54:19.000000000","reason":"A robot voted negatively on a label"},"36238":{"account":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"last_update":"2025-10-13 15:25:43.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e mentioned you in a comment","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-10-15 12:46:50.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-11-17 15:58:05.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}},"hashtags":[],"change_id":"I4da3046a521cf5e5b93526bc767a5eca63b6f2ed","subject":"Add default config for the ipa-ntp-server","status":"NEW","created":"2025-09-29 14:21:55.000000000","updated":"2025-11-17 16:33:21.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":17,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"ec7fe46e9a5304c8b80752ce5aab699c5dcdb015","_number":962496,"virtual_id_number":962496,"owner":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-10-20 08:54:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"disliked":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":2,"date":"2025-11-17 15:12:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":-1,"date":"2025-10-20 08:26:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2025-11-17 16:33:21.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-29 16:03:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-03 08:30:31.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-10-03 21:23:35.000000000","updated_by":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"reviewer":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"state":"CC"},{"updated":"2025-10-04 09:38:11.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-10-10 12:49:35.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"CC"},{"updated":"2025-10-15 12:13:17.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"}],"messages":[{"id":"1f5ea1fff5f121dcd37b080538fbcc40f830042c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-09-29 14:21:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"773ef7ffd4ea334a7f8c9255885468bb58086e37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 16:03:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82ad801920614531a4539cc64edf4066\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b7a3e8b3d5941c59badd128c6c65a00 : SUCCESS in 9m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fa17bdd56c56420aa6aedf0c97a1c2a6 : SUCCESS in 3m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb95b6b5a4a44997bd2774a98c948bb0 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7b84cf08d4841f2aed8c7a0753bc3b1 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d795b55792744c9b8f8d39abbfa42576 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4db8157b13d4c509b26275dbc36561c : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b6d48c76bf446c4aefc4053b211b77f : SUCCESS in 8m 56s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7e16b01e8adf4710ae06d35ee611ea4b : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1fbd9530222342d69a59bd4ea7d50fee : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/08c527afcb2a4e2fa9fe8eafef477f2b : SUCCESS in 10m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f860367087448f093e90d80c21c48ca : SUCCESS in 55m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bcd9594ea0754f028fcae4bf0d2964bb : SUCCESS in 50m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b364a34ec2a048d39841684bac68d4f1 : SUCCESS in 20m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45f1398b91fc4f1888766d0abf610680 : SUCCESS in 13m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5ee51920d514d62b4ec6fad594575d2 : SUCCESS in 1h 34m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59b5d239b39344518cf85db5c7e9c14c : SUCCESS in 14m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/721a2ba3f3fb47879ae7368d18194a9b : SUCCESS in 16m 17s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddf715e5ead24f959d83e961a3db5066 : SUCCESS in 27m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/15cd98c7096b46aea3d3058523151c18 : SUCCESS in 14m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa2e9ee9dee540cebac86f27da55b4a5 : SUCCESS in 13m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/aa21dd5e538d485288d9963823f8df5c : SUCCESS in 13m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bbe9ae1aa5d4a3b9873a4f56202f7e6 : SUCCESS in 9m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d71231f9ae6a06a485e360caff78d46e6fbd9785","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"real_author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2025-10-02 09:51:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_38330\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"}],"_revision_number":2},{"id":"b64557d00053cd9f662be8fca31206f1cd236324","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2025-10-02 09:52:18.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"601921620dabad9d97dc7d1fbe467a3405b10db8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 11:50:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a324168c0d2b429d8ef5fd4ce9932095\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3e128f3d77d46f6b224c174976c61b1 : SUCCESS in 9m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d53fe6cafc1749b985662fad756cd1fd : SUCCESS in 4m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b77e7a3870a4e82ada9f2ced0aa2299 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74956f38198b4a66be1231c070166285 : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f54d001fa3344aa4bab8e7b4b5db7bb7 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0dea5542ce264f7bbc10b0f993fc2ff2 : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7dab8505e414a329ed7a8a53e2fbc2c : SUCCESS in 3m 52s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c52992b839ac4c60b715afcf41882606 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8fed64880dcc48e39582ac27c67ea368 : SUCCESS in 12m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f336642e3cbf4c6ab271fe51136e4fbb : SUCCESS in 9m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a3a6a61ea160442482a262d18310505d : SUCCESS in 16m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/20c0c79930cb4617867e4083038a06e3 : SUCCESS in 6m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cefc842fcd5444b58e6d9e7e2722ed73 : SUCCESS in 5m 42s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a04e78983dbe4dd092789363dc4fd8e6 : SUCCESS in 12m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c7989a0d79fb48ee8e40937531f8f01d : SUCCESS in 17m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc3507e02f1e432dad574a0137a3f8c9 : SUCCESS in 16m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a2b7af30d58448ba40bef2b86f77998 : SUCCESS in 54m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d606618d87ea4544b9fd76cc3700d891 : SUCCESS in 31m 55s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cdb602e0277430398b079e0a52cdf42 : SUCCESS in 1h 54m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/154402c68a0d43dd9a3702daebe38ce8 : SUCCESS in 13m 00s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5819c59156384fbcb98107ee4793cd0f : SUCCESS in 11m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/463f106537df4c62a899a94352f993da : SUCCESS in 22m 28s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ca4585fd89cc41038c9274c4b9daa780 : SUCCESS in 17m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1c76189bee4348ac9ba507fa5e7e423c : SUCCESS in 11m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4cf32dfe6a943249382bedbd90d7e93 : SUCCESS in 14m 26s","accounts_in_message":[],"_revision_number":3},{"id":"5c932c86bde676b0a9a0e2fcd4647c9b7177ee45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-02 12:56:34.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f552dbdd3f0e6d251b9edfffc86764479c2472b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 14:41:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1b692f454ec47adb6c0e3ee9193854d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0b372f00eeb4cf289f269ebd5e32088 : SUCCESS in 10m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b06b4332653b44309db3f76982f0382b : SUCCESS in 7m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a36ed1e779df48fdbb732c32d4584094 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/567203a0fd7e4d6180ba249136049a42 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ffc71bc6b79431885e5521b74558965 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ecc9370bfa14803a3917d347896841b : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38a8b890909a47e89f4e2305882b62c4 : SUCCESS in 7m 07s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bea197bd0b5f4b8f9d7b45bf21634af7 : SUCCESS in 5m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f98f3a2cf1d8401080e7da92fa7eec56 : SUCCESS in 12m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5c7619edaf734571807e7fa75cf15cd3 : SUCCESS in 5m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0dec8adca3844bb19ea5da80c28c3bbe : SUCCESS in 15m 52s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/490bd1a04bfa48b2862e98dddb76ef95 : SUCCESS in 15m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e86322a872449de8fd81937544c6a5b : SUCCESS in 12m 09s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b1558f24b24d456dad69f54b29aa1867 : SUCCESS in 21m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6c555528ff5f42d48dcc73bd4538dacb : SUCCESS in 21m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59170d6dc54f4cddbb81e0b13d36629f : SUCCESS in 14m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75aefb4f8bdd40748bbcc6c30d6d03b6 : SUCCESS in 56m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f7ea95b39da14b1881685c2779e7f694 : SUCCESS in 50m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d509dab287554d918141b8c85320599d : SUCCESS in 1h 39m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cfb5a130ba347a48c32ac749e33d473 : SUCCESS in 13m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/64f16a1ef26b4292a780e835183d40f4 : SUCCESS in 12m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2143649dab77417aaf486321f1db1cc0 : SUCCESS in 31m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ecb99617681848d691204651007f9c5a : SUCCESS in 14m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ecfabcf993bd411285dbda6ce28203db : SUCCESS in 14m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/745e126cf0474eb3ba9aca7b911a1354 : SUCCESS in 12m 08s","accounts_in_message":[],"_revision_number":4},{"id":"cab7271076d4a2d5c90af99659e2e56c9dff5bf2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-03 08:30:31.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"861272f0aae9fc637e69c5addea4ce8a391ea2ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-03 09:51:48.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"ececa9ef525ce36427201db6862c7524afabab62","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-03 09:52:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ee9d2f212654d465a63c5c044b5cd41bda654809","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 11:46:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/348e316baf644253bf42f47cf06a9b0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d502a482af24683abbf39090a6bfa79 : SUCCESS in 5m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0c9f77697d29437b99ec200bbb1411c6 : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ddf4bb6af6e4ce69fef455afd2283ab : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f58ca2aaa524f23a1809c9338e7cbfd : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/662d0e7fe44642d68bec1282c43483b3 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4bad63dc7cc84fb3bc5bbebc9a094d7c : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/810fdba7cb15496f9fd36ef0c64953f2 : SUCCESS in 7m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f65f5105f934c0fa410eaa1b00b25c2 : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0dd00b5e975d464d9bd5487a871b17a7 : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b62199bef9a44513b7e905d45ef73204 : SUCCESS in 5m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7d1e77430f2642dbad27de794ef5e321 : SUCCESS in 9m 26s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5e20588620364d289c5d3740513be3b5 : SUCCESS in 15m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7de89eba90e5489aaca4baccd5ac9a8f : SUCCESS in 8m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b68efbe130d49f8842b1f016f7ac0d4 : SUCCESS in 12m 50s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1520a07d44334c42a18c7a5f25913d38 : SUCCESS in 17m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3b368c1fdd414f3c9a097f44562bf77c : SUCCESS in 11m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08c473410b9e49eab82b45cc0d3e3a60 : SUCCESS in 14m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b09575287ff44c99cb35fb099fc10c4 : SUCCESS in 48m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8325d5595ef34c83a990aac5e21f2e99 : SUCCESS in 42m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f2d275a3e0740d68e263399eb863277 : SUCCESS in 1h 52m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1d947123aa54ed3a533ed3b8e5e2180 : SUCCESS in 10m 44s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/592e00eb4b17434eac8bb59466474592 : SUCCESS in 6m 25s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/312f9b4b9a424744b235da81e2e6f7d0 : SUCCESS in 14m 45s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b8998ef337354e81b34299a7b128021a : SUCCESS in 8m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/58dd67982df842e28193ac508366f044 : SUCCESS in 16m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5b2e04d29f24a4b8cf1005b243d93b8 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":5},{"id":"9478ffc27f47e73534ef40338fb72e9b128b155c","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-10-03 21:23:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"95044eabb62adc16aa63626d983123896230f04f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-08 08:18:36.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"36b1bf8846f613337f60db5af2915420cdc7d59a","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-08 08:18:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5b361f3557ab722d646dce25662bf4aaffcb8bc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 09:58:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c011d24a5d4b4d3382d3dbc74d5b66e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72f2589f290b4c4992212ac17f4736ac : SUCCESS in 10m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/82e5a8359ccb4506bb0df54c4173b8b3 : SUCCESS in 6m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbb26fa15d5941d5873b4c5f76ed8831 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b80563b167a416d9504a033a2d7f7b5 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31bb212e60114bf0ae9f63d24a6a051f : SUCCESS in 5m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/159a509e607e4aa9b8ea3200bc443461 : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe8eff0196e949138d8a1d20b5a466a5 : SUCCESS in 5m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2abf9df548d240f7be54f9d9cb4b49f7 : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4631689cfad94b2d973e06caaf498f36 : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/47377f9577284f479a9dcecf74000c90 : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2fd9dbcdfa9f499ea02b30183f06d559 : SUCCESS in 12m 20s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e5080a50e1024094a99d6abd3ede6562 : SUCCESS in 15m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eed09ac114e54e0d94328aeeea0309fe : SUCCESS in 10m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dad6dc51abbb426abdd50acac1372bca : SUCCESS in 9m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4c9f7af2d2b2492388013588806abc74 : SUCCESS in 14m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e3341818adc14cef8e134d05cc0c4818 : SUCCESS in 19m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e70eaeb68e9c4d26b9711d2c51fdb40d : SUCCESS in 17m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97d207ae608f41beadb9bb6709ce14dd : SUCCESS in 44m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a5bef507e7e4459eafce09269d84c738 : SUCCESS in 31m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/343cae9d26524c84be41d1d2d0db3ff5 : SUCCESS in 1h 34m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/147fd47367394b5d97eccda1f064be6a : SUCCESS in 9m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7271846177f34ac9bbd2482ad5746450 : SUCCESS in 13m 13s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/084e09463f7c484995b2aae8bfc18c41 : SUCCESS in 31m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/47cadf1a8a984b6d981e014fb75854cc : SUCCESS in 11m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/341c66d3135e4daaa82c0a5dc80f184a : SUCCESS in 7m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/798f263ed51a4bcd9ea3284249433d16 : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":6},{"id":"58c70d2fbf1cfda064bb41a3e5ab7453e66bf111","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-10 09:21:03.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"1a24cb6c2dd235459a8d753d4b50eb13433dadd0","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-10 12:49:35.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1fcafeabe0494b14b6e7b3687efb9851aa2f1c5d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-10 13:15:27.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"46c3104cc39cdd05c5b17b80487e330fd15ae23b","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-13 09:18:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"008725b154e9d31051d94233bf3471d256e699b0","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-10-13 13:02:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"463e68651e0c533956997242f5e100605da2e7e2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-13 15:25:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9191661ba5af0eb098d170c684ae911a8dfb99ed","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-14 10:22:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f7eb573cfdafb717ef924b9f5e477e2c3489bfac","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-14 10:23:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"031464324dd6b61379d790a4165246ab5da17992","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-10-15 12:13:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d5ff0ffc5cade7d2713409e5e269ca53ff960ff9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-15 12:46:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0797fd953294d843883f63cd3164dc78b7238d30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2025-10-20 08:26:09.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"822015c5cf503cb7ce14bd13a130981df8e8b433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 08:54:19.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0b9c970a323465f815fde6ffd71e86b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e42bea69f91a4f84a8ec768f7df9c755 : SUCCESS in 9m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/318cb0caecf44cc899c0fb56bf784ab8 : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6099356588234995bebfb737d19b88ba : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cd37242921e48daa693f9abe2da2e0d : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33a953c23eff4c43940b521bccfd92b3 : SUCCESS in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/416e89fbfff346118686c3c0aaed6685 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d63c6730887a42d88cc182edd1a7d26c : SUCCESS in 3m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/90ad35bfd9f34890a23a53fbe677e8dc : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7d835851b4b04faeb751bd442d9b5c80 : SUCCESS in 11m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1a3f2787ad5c475a880d6e934781c1bd : FAILURE in 4m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/faf979cf0bac4126af2f9d4ce8f870a9 : SUCCESS in 16m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/170c926602a54e30833cba5f9dc68711 : SUCCESS in 13m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3105051dcd034f9fb71c101711448f55 : SUCCESS in 13m 28s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a70219ca0197440694238e67443fa390 : FAILURE in 8m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/36d22934ea57433d9e8da72e0f0d6d13 : FAILURE in 16m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b893b6211a145d78f5d97d0be87ecf1 : FAILURE in 13m 26s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d80fba41fd46405d9f2fea80231f4a1a : FAILURE in 11m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/889b8c2f75d841db8f13e290aa333aab : FAILURE in 12m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/330c44be55024211b488644b75f73547 : FAILURE in 26m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/585d95764dfa405787850fa8b40be655 : FAILURE in 10m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/992e9c8415b94f2eb009ac5162f9d46a : FAILURE in 20m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fec71607430149caa639aabaefb2fc37 : FAILURE in 14m 03s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/400bd2691783457c8ce1416a4292997a : FAILURE in 12m 33s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0fa5590e3e9040c98e1e355ecca903e0 : FAILURE in 7m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/337c8b186fdf49118c5b53ca84a8c3bb : FAILURE in 8m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b9e339af0b884c5b9f6a3867c9a8c1b0 : FAILURE in 5m 08s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24eb543a31014f85880b6c18f1038b87 : FAILURE in 10m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3d9c4ddae8fb405bb0a4cba56a95afeb : SUCCESS in 12m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/64b6db5d28584b999a9edce3bbee71e5 : SUCCESS in 13m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e094df8833814565bc81b8ec7a3e514d : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":7},{"id":"484571e81332d20fe08f6fb14eec0657fc49b01b","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-20 11:39:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2cd9cde4c277e648110b0837ce40d163e0e6d6a4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-17 15:12:20.000000000","message":"Patch Set 7: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0510c7730bf801d8949efdc46ae95b5ad9545a33","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-11-17 15:45:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"aaf3580c1688553bc3222f80bc85e6b127e00918","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-17 15:58:05.000000000","message":"Patch Set 7: -Review-Priority","accounts_in_message":[],"_revision_number":7},{"id":"ec7fe46e9a5304c8b80752ce5aab699c5dcdb015","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-17 16:33:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d762e71ffe96f5df2b33773ae4ae86340d166054","revisions":{"9ece1be97be2dd4f39f1d2f540175cee413078e6":{"kind":"REWORK","_number":1,"created":"2025-09-29 14:21:55.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/96/962496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/962496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/962496/1"}}},"commit":{"parents":[{"commit":"8c2382006a590f41a4ade20a49c364274819ee50","subject":"Merge \"Configure tuned on Ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c2382006a590f41a4ade20a49c364274819ee50"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-29 14:19:20.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-29 14:19:20.000000000","tz":60},"subject":"WIP: Add default config for the ipa-ntp-server","message":"WIP: Add default config for the ipa-ntp-server\n\nAdding default configuration to Kolla Ansible for the\nipa-ntp-server kernal command line argument used by\nIPA during porvising and cleaning.\n\nChange-Id: I4da3046a521cf5e5b93526bc767a5eca63b6f2ed\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ece1be97be2dd4f39f1d2f540175cee413078e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ece1be97be2dd4f39f1d2f540175cee413078e6"}]},"branch":"refs/heads/master"},"a9757232cc8519f46a5babbfa83cc2b7667a984c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-02 09:51:55.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"real_uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/96/962496/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/962496/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/962496/2"}}},"commit":{"parents":[{"commit":"12197a2cc9babaa86b03fb0412667b416e61480a","subject":"Add redfish pxe args for virtual media","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12197a2cc9babaa86b03fb0412667b416e61480a"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-29 14:19:20.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-02 09:51:55.000000000","tz":0},"subject":"WIP: Add default config for the ipa-ntp-server","message":"WIP: Add default config for the ipa-ntp-server\n\nAdding default configuration to Kolla Ansible for the\nipa-ntp-server kernal command line argument used by\nIPA during porvising and cleaning.\n\nChange-Id: I4da3046a521cf5e5b93526bc767a5eca63b6f2ed\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9757232cc8519f46a5babbfa83cc2b7667a984c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9757232cc8519f46a5babbfa83cc2b7667a984c"}]},"branch":"refs/heads/master","description":"Rebase"},"42ca3439ae501add9939e61765dc637ead8ecf27":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-10-02 09:52:18.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/96/962496/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/962496/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/962496/3"}}},"commit":{"parents":[{"commit":"12197a2cc9babaa86b03fb0412667b416e61480a","subject":"Add redfish pxe args for virtual media","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12197a2cc9babaa86b03fb0412667b416e61480a"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-29 14:19:20.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2025-10-02 09:52:18.000000000","tz":0},"subject":"Add default config for the ipa-ntp-server","message":"Add default config for the ipa-ntp-server\n\nAdding default configuration to Kolla Ansible for the\nipa-ntp-server kernal command line argument used by\nIPA during porvising and cleaning.\n\nChange-Id: I4da3046a521cf5e5b93526bc767a5eca63b6f2ed\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42ca3439ae501add9939e61765dc637ead8ecf27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42ca3439ae501add9939e61765dc637ead8ecf27"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b7cc8c5fe9378e8afea2adb34f2f56d64b522ad0":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2025-10-02 12:56:34.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/96/962496/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/962496/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/962496/4"}}},"commit":{"parents":[{"commit":"8c2382006a590f41a4ade20a49c364274819ee50","subject":"Merge \"Configure tuned on Ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c2382006a590f41a4ade20a49c364274819ee50"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-29 14:19:20.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-02 10:36:45.000000000","tz":60},"subject":"Add default config for the ipa-ntp-server","message":"Add default config for the ipa-ntp-server\n\nAdding default configuration to Kolla Ansible for the\nipa-ntp-server kernal command line argument used by\nIPA during porvising and cleaning. This ensures\ncorrect system time during certificate creation.\nPreventing issues where Ironic fails to communicate\nwith the Ironic Python Agent due to invalid\ncertificate timestamps\n\nChange-Id: I4da3046a521cf5e5b93526bc767a5eca63b6f2ed\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7cc8c5fe9378e8afea2adb34f2f56d64b522ad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7cc8c5fe9378e8afea2adb34f2f56d64b522ad0"}]},"branch":"refs/heads/master"},"e3d745f2801f60b6c262fb864f6f067748949953":{"kind":"REWORK","_number":5,"created":"2025-10-03 09:51:48.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/96/962496/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/962496/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/962496/5"}}},"commit":{"parents":[{"commit":"8c2382006a590f41a4ade20a49c364274819ee50","subject":"Merge \"Configure tuned on Ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c2382006a590f41a4ade20a49c364274819ee50"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-29 14:19:20.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-03 09:51:37.000000000","tz":60},"subject":"Add default config for the ipa-ntp-server","message":"Add default config for the ipa-ntp-server\n\nAdding default configuration to Kolla Ansible for the\nipa-ntp-server kernal command line argument used by\nIPA during porvising and cleaning. This ensures\ncorrect system time during certificate creation.\nPreventing issues where Ironic fails to communicate\nwith the Ironic Python Agent due to invalid\ncertificate timestamps\n\nChange-Id: I4da3046a521cf5e5b93526bc767a5eca63b6f2ed\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3d745f2801f60b6c262fb864f6f067748949953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3d745f2801f60b6c262fb864f6f067748949953"}]},"branch":"refs/heads/master"},"f0ebbd5afe1b97200d04056946e5cab341deb567":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-10-08 08:18:36.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/96/962496/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/962496/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/962496/6"}}},"commit":{"parents":[{"commit":"26b33e1076232c5474a6848aafce94733be60865","subject":"Add redfish pxe args for virtual media","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/26b33e1076232c5474a6848aafce94733be60865"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-29 14:19:20.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-08 08:18:36.000000000","tz":0},"subject":"Add default config for the ipa-ntp-server","message":"Add default config for the ipa-ntp-server\n\nAdding default configuration to Kolla Ansible for the\nipa-ntp-server kernal command line argument used by\nIPA during porvising and cleaning. This ensures\ncorrect system time during certificate creation.\nPreventing issues where Ironic fails to communicate\nwith the Ironic Python Agent due to invalid\ncertificate timestamps\n\nChange-Id: I4da3046a521cf5e5b93526bc767a5eca63b6f2ed\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f0ebbd5afe1b97200d04056946e5cab341deb567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f0ebbd5afe1b97200d04056946e5cab341deb567"}]},"branch":"refs/heads/master","description":"Rebase"},"d762e71ffe96f5df2b33773ae4ae86340d166054":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-10-20 08:26:09.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/96/962496/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/962496/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/962496/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/962496/7"}}},"commit":{"parents":[{"commit":"33f39fa6cace1f7b408163f7d5faa901a6491be4","subject":"Merge \"Remove default(omit) from Bifrost inventory template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33f39fa6cace1f7b408163f7d5faa901a6491be4"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-09-29 14:19:20.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2025-10-20 08:26:09.000000000","tz":0},"subject":"Add default config for the ipa-ntp-server","message":"Add default config for the ipa-ntp-server\n\nAdding default configuration to Kolla Ansible for the\nipa-ntp-server kernal command line argument used by\nIPA during porvising and cleaning. This ensures\ncorrect system time during certificate creation.\nPreventing issues where Ironic fails to communicate\nwith the Ironic Python Agent due to invalid\ncertificate timestamps\n\nChange-Id: I4da3046a521cf5e5b93526bc767a5eca63b6f2ed\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d762e71ffe96f5df2b33773ae4ae86340d166054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d762e71ffe96f5df2b33773ae4ae86340d166054"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
