)]}'
{"id":"openstack%2Fkayobe~962523","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~I002cadf4c14325bbb10bbfd2efeeb87084bd4314","project":"openstack/kayobe","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-10-10 09:05:45.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-09-29 17:58:16.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-10-10 15:36:03.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I002cadf4c14325bbb10bbfd2efeeb87084bd4314","subject":"CI: Fix Ubuntu cloud image tests","status":"MERGED","created":"2025-09-29 15:46:15.000000000","updated":"2025-10-10 19:34:39.000000000","submitted":"2025-10-10 19:33:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962523","meta_rev_id":"bad978e23508a28dcd4567588b7a6a5be712353e","_number":962523,"virtual_id_number":962523,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-10 19:33:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-10-10 09:05:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-09-29 17:58:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-30 13:57:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-10-10 09:05:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-29 17:01:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-29 17:45:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-30 13:57:30.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-10-07 11:37:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"106cab75555be09db4b9c0adf1cbcea3d065a71e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-29 15:46:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40467ba6405e772345b7e5049853c525e671ef7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 17:45:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09255cf6e8a64fca8d3694c49a7442ec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8e821a7821d4eff9e3a6eb35f22a95c : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31a5338be1a440f8a240aeb220406d9c : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb009a41014742a8b85070e5bdb79da2 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ef9dc226ec7d4d498f079f64c912ff52 : SUCCESS in 2m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b11b01a781de4afeae01c32fed45ada5 : SUCCESS in 7m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f04f3dfc957446678f8d954d20f45928 : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/818433bcf3c84604ab633e90de4a07a6 : SUCCESS in 10m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc930c8648ab45dea815e92caf18f3f4 : SUCCESS in 9m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d51414c2a53a44b191b4dbea4d236e0a : SUCCESS in 1h 00m 44s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b7ca65cedced4027999114a085778043 : FAILURE in 8m 54s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f094bcb79c2442948673603792ef3bcf : SUCCESS in 48m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d750559346ed4fdc8146cb6416725693 : SUCCESS in 1h 07m 32s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/64bb637c983f4d42bf5639b3f3210760 : SUCCESS in 18m 21s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/90f9a566e8fc48a290ce37ff5fd62d32 : FAILURE in 7m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54376c287e2042578c600362fa9b3436 : SUCCESS in 13m 35s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0c292731dd9c47d6b0e5ae8883571354 : SUCCESS in 1h 01m 03s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b017bcf090f4bfba5a6a7264384587d : SUCCESS in 1h 38m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a64d5887df8d44339305f58dad6e5cd4 : SUCCESS in 21m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/993c7ddf4b514ec380ee381cbb1ff4c3 : SUCCESS in 19m 00s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e01693fd67c7434b8ac61c0a9b1329ed : SUCCESS in 34m 41s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47a388de517a44f58aa28ef07f6e17b7 : SUCCESS in 29m 03s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a2ee12a53225452583a85d5d21f488ca : SUCCESS in 17m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bc4765534e449939764741336d3a088 : SUCCESS in 9m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/60df1a3df1bc4ad4b08acb097ee399cf : SUCCESS in 17m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3147f5f083854ac8bf2ac8458500b091 : SUCCESS in 14m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a0077b2b8d320ae2f1557fa1c40b5ea95282ff82","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-29 17:58:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2e80855f2cc38149a7d1261592b93d8eab582185","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-30 11:09:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d10ec6f4fbdaa673fc6de739546731af908db270","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 12:02:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c2140d0ad6a4906bd0ec11cf50a375e\n\n- kayobe-infra-vm-centos9s-cloud-image https://zuul.opendev.org/t/openstack/build/7a0fbb581f5f43fca3da7476b06dd32c : FAILURE in 8m 29s (non-voting)\n- kayobe-infra-vm-rocky9-cloud-image https://zuul.opendev.org/t/openstack/build/233a6c3e51d74036a091d0d809e899ab : SUCCESS in 22m 49s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/175ce11c0a9f41bda9f5c7c212c0e3a6 : SUCCESS in 11m 41s (non-voting)\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/5b975006235d40d5b0646d8c61803725 : FAILURE in 5m 00s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f1ee1c5bd3a64c6eaccc46ea6ba7f70b : SUCCESS in 44m 34s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5e6a7e5715d4be795c4e22854732cef : SUCCESS in 47m 44s\n- kayobe-seed-vm-centos9s-cloud-image https://zuul.opendev.org/t/openstack/build/da3ba55e9da54790bf82a2684676a2ee : FAILURE in 8m 22s (non-voting)\n- kayobe-seed-vm-centos9s-q35 https://zuul.opendev.org/t/openstack/build/06d3528071d741c5911c462e5e6bd65e : FAILURE in 9m 02s (non-voting)\n- kayobe-seed-vm-rocky9-cloud-image https://zuul.opendev.org/t/openstack/build/8165511f14bd4e97b916e39b8c98a0eb : SUCCESS in 24m 38s\n- kayobe-seed-vm-rocky9-q35 https://zuul.opendev.org/t/openstack/build/75adf24886d742169438bcf7a0a0af7b : SUCCESS in 16m 59s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/b17ae8fc519b47a3a844e29c0ef823e8 : SUCCESS in 11m 48s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/e59b188343944cf1a5afcbbcd69d4cf1 : SUCCESS in 9m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ebdd46ac8e15ebbd5ec0d1f15d66456fd161b4ec","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-09-30 13:57:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9b5847aba8e85331fbbbb9566a7e8e508def0beb","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-10 09:05:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12d58656105533ee26df004dbbf12e907359814d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 09:06:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6b45a5fc52680e69ed4316fdc54c4b35a5e28f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 11:21:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f8815d784764bbf957b481bf9e2ec8e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/766a5dd2455a41a0b1af502a85656112 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61a42fd7688d413980970ae5cd0f56f7 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d028196d4a148cc8c4320beb84ed813 : SUCCESS in 7m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d1a516e293ab4fe3a45dc65d830034f2 : SUCCESS in 7m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/284dc7da0b884faa81488241a71e2935 : SUCCESS in 7m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/68e349f4a16340bf9e6f3ad2f8cb29ec : SUCCESS in 13m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e391caf6a7eb43139d729dba73f7f1a0 : SUCCESS in 34m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb2f6a9b1252450f9f46351d7bfab3e0 : SUCCESS in 1h 01m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/755a8993d1cb4db1a0e58792dda45aa7 : SUCCESS in 1h 22m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/58afeccb225b4f7482da8d8b13039482 : SUCCESS in 19m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f09eb96afee04b76bc92f97fd0166cfb : SUCCESS in 16m 45s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1489a924a304468eaed77b435279810c : SUCCESS in 1h 38m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81ab337814d147e69870c917b2f57fc5 : SUCCESS in 1h 35m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a8c3166ac97942bdbfcf8f37eef57831 : SUCCESS in 20m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4462f7db23df4ce7adf117e2565025f2 : SUCCESS in 16m 47s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e99ba8fce2334c7e8fc5595ae7498882 : FAILURE in 19m 18s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93d69d8bab664ceb9f8adf8d4c073313 : SUCCESS in 23m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c39528d423214faeb48c419cbacfac41 : SUCCESS in 11m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c00821595f52475481c9e36d33a15acb : SUCCESS in 16m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6a23437c57eb428297f7128a9a0b4a55 : SUCCESS in 19m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f203903bd9641238b3047a907574eca : SUCCESS in 18m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2e188e47f79e70dd5fc0612cdaf8e6a097793842","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-10 11:29:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"406017e8cf7dfeb2d7030fb89be33b6f76ad568d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 13:13:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e76b0198c55c471eace27265d22ca4e0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fce72d47a54a41a39a94de4b43189da4 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40063da996344b3fad9a36b6acfaac8f : SUCCESS in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89f2f1f56f87421b85d1c52c554c3da9 : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d3cf45579139478e8239c41f27d71e5d : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41c6a69a9fc14f309ec7945d6e5de9da : SUCCESS in 10m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/db2ec1582bb34fe28e664ae88309e889 : SUCCESS in 6m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc8d25a200ed4ad9b61622df58db8918 : SUCCESS in 9m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/25b6ba9981604ca29265915a93c0cd49 : SUCCESS in 9m 10s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7d0dbafe3df14e44989f964fdbf88fd3 : SUCCESS in 1h 00m 46s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/fe855c463b9d44daa403775c505a813c : FAILURE in 9m 05s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/118bf4d865d941baadd379c15e4f6e6d : SUCCESS in 1h 01m 20s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ff865363472744238935e6e3e68200b3 : SUCCESS in 1h 10m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7b947fe42d2c4b1ca2953c026f8e6d2a : SUCCESS in 10m 54s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ca06b4896b7f41d28600839b340ed1cf : FAILURE in 9m 25s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fbf5f4cccc34c0cb327c576cea44052 : SUCCESS in 13m 46s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2e816c112aff4280bebd506e6fce8e3f : SUCCESS in 1h 02m 06s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3984086a1064bab8c5828f908f55801 : SUCCESS in 1h 38m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/58fcc5cf4a2e44a38e988cfa2201c260 : SUCCESS in 20m 01s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/141d8640a4ed420e8daa81c87e11faba : SUCCESS in 18m 03s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7c62c90f8ef545808bf1f31d984f49e0 : SUCCESS in 20m 02s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22274e90adfa4ec0853ce6da85b3ce06 : SUCCESS in 17m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/257d48c8bbba4efd87fea8515fb52a65 : SUCCESS in 11m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76ada959ae4f4d579f82be70f9776e1c : SUCCESS in 10m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/04f16f86d17b4b2f959de5ad344bc330 : SUCCESS in 17m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c157b53f68f74e9f8e7ad1327ed0469d : SUCCESS in 12m 41s","accounts_in_message":[],"_revision_number":1},{"id":"6ce7d04732f61a88522ed111195902ceded8829a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 13:14:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5bd0e442bc373436f5df1db9e5c5ad618a9c41e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 15:27:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/39436f3eb62740e7a72147c55dcf58b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeaaa813031e4fcf9f173cd18a66f4a6 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec00a9f55f314633acf8dbe1ccf7dc86 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d240890d66ff4f2a879480fbdb47166a : SUCCESS in 7m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3e32f50155cd4ad584341e9fe234f9ab : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/20921a0db7b548ee89a619afb332e779 : SUCCESS in 6m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b7cf977272b144e7a170d58f5421f6f4 : SUCCESS in 11m 53s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/81d98c4a6dd5447d89f58f77bb1b8786 : SUCCESS in 40m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdaf3659dcc24ec384ebe7a77e5a7762 : SUCCESS in 1h 07m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a682ad803ae04bd5be2e285f93060283 : SUCCESS in 1h 19m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/eaa34ed45b714777a374abb1cb572ffc : SUCCESS in 21m 37s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30d6aba69c4748b1bfd7d939c159ae94 : SUCCESS in 13m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/154625ea3279497db6994e0d0a946f19 : FAILURE in 2h 05m 16s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edf98c53229a45e7af109a9e2e48f1be : SUCCESS in 2h 01m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/00dfbd7574f84fe1ae45f62ecf45fcfe : FAILURE in 17m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f121a77070748af89cb8bb19d8c9375 : SUCCESS in 20m 56s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d2d5da48baf9493babf75463f640e4a7 : SUCCESS in 33m 01s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fae64e57dafe4430a110213515a860d3 : SUCCESS in 30m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/251a254747d143359c4ec88eb9d889d2 : SUCCESS in 15m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32863ec6f9964c19aa4bbafdda1c51a3 : SUCCESS in 12m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ff63b0fd124d475a969b7abb887395d7 : SUCCESS in 18m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a420504b48849a5a81229467ca4c0c8 : SUCCESS in 16m 08s","accounts_in_message":[],"_revision_number":1},{"id":"fbbb9d8b5b2568f9eedebc16ecefd525bed2c785","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-10 15:36:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7a1f3dc8eb5e9b418e9140f7ae6c89e4654f151","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 17:36:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca057999fa1248c2bc13f7f68cdfc6f0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa99f9b555e6403bbc24844337932848 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddc6e2771ec841f8a548038c8e543a88 : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e19716859a2d429ca88d1ec64c50874f : SUCCESS in 6m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9a1208686934eceab847664e5fec20f : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4087308289e84fd88d5ddeab0b550db6 : SUCCESS in 7m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a72ce658042f4fc0b7b1cee756f54980 : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e0a00788f8124c0ba1a7344aa35707a5 : SUCCESS in 10m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9e1f13080b7a490db1f43140b80ff5b2 : SUCCESS in 10m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/523ff158ba9147088ab6c88d96c0b43c : SUCCESS in 1h 10m 07s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a16b43aeb6b442b18925206872f1330b : FAILURE in 10m 04s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c334aa6782b4e72b5411d176d0c9957 : SUCCESS in 1h 01m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/643008710b914c388702aea327a754ef : SUCCESS in 1h 20m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e1392d9d0aeb46beb7036eec0226103d : SUCCESS in 20m 19s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8cb446a6bf404eae9fac4dad58b01f8f : FAILURE in 8m 49s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d3e317ae3804065b024a4aa10d08fd7 : SUCCESS in 18m 30s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1f7d90bb89864cc7946fb4ee7dd6229e : SUCCESS in 1h 54m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/887e26f3e9204221993b62ab9b67a3d6 : SUCCESS in 1h 52m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/65f839f8baf34f7c819e821dd7af36b9 : SUCCESS in 17m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b25f12e49a14418a75e09999f264dae : SUCCESS in 19m 09s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fbaa7a6fdaf547d38625bd36a2caa44c : SUCCESS in 22m 24s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f047cc9d5348482189913d3eb9660479 : SUCCESS in 28m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9157a186ace2487daa2a32bf9507b9f4 : SUCCESS in 17m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/726b0575012c4e2d893bdc39823c680c : SUCCESS in 15m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b40c3e9e08d84fca99f7a8e129aa5043 : SUCCESS in 10m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f43fec7c2b74cb58934b102282b8926 : SUCCESS in 11m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b4a08c2de5b8ae41599d4e496704e68c4c67c16d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 17:36:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c6aa5f3f75f60b44c79dd617e30fea76165a492","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 19:33:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76dc15f000fb49b79f7c48a558e13adf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/403d4d3972d8460c8a328586e74997ef : SUCCESS in 2m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/771fc6a5a5364a1496d599699d21bdce : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a47be15e68f247d38a44af6a51ab8436 : SUCCESS in 7m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/689272488d4147b2bee91954fc30d102 : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/58e67cee512b422aa6216587f41bbb92 : SUCCESS in 9m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5a0696b94633483ab84fe5b77a13c1e9 : SUCCESS in 10m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5c68e4296be3428b905385c72b6da7ec : SUCCESS in 1h 26m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99abdaa31e114015807396c7a689deed : SUCCESS in 1h 01m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/046d8c25a3034043864a6708938aba0a : SUCCESS in 56m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/07b6b6f5b4324bf2999768aff603144d : SUCCESS in 14m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30a69d048a5b48e6b91326ffebef27f1 : SUCCESS in 12m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bea3484df5e04c44adc8e3e259a8a226 : SUCCESS in 1h 52m 55s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12d9238ae3e74e85bb967379ed04863e : SUCCESS in 1h 40m 13s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4ce62f94cb3e4f0889867756fbe4c9e3 : SUCCESS in 16m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4b82533016942039571b439902ae665 : SUCCESS in 12m 45s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7183f6fc433543fe8c51f6fa16fe99c2 : SUCCESS in 35m 03s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/329bf4b8e2024bfabf826617f11033bc : SUCCESS in 20m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/358968e317d340d6befa5fb0b92491b5 : SUCCESS in 14m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/887891aa9d2948158af2f33dcfed0e1e : SUCCESS in 9m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/62cd6a9de69b4a2480965909022cd70a : SUCCESS in 17m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd81a7c589e849d78ce4b4f70b892579 : SUCCESS in 14m 14s","accounts_in_message":[],"_revision_number":1},{"id":"5dea0a58dabbb5aab9f802b8e41c4965e0839ba6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 19:33:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bad978e23508a28dcd4567588b7a6a5be712353e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 19:34:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/337bc3d27bfd40a78ce7858eef843e73\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d855dd8adabf4fffa67856e9114756fa : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"338b5b965a11c1b13e3b282781abe161a5ff0cd9","revisions":{"338b5b965a11c1b13e3b282781abe161a5ff0cd9":{"kind":"REWORK","_number":1,"created":"2025-09-29 15:46:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/962523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/962523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/962523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/962523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/962523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/962523/1"}}},"commit":{"parents":[{"commit":"f6fb05fc66a020edbf531903ec9d029da6b5e356","subject":"CI: Test seed VM with distribution cloud images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6fb05fc66a020edbf531903ec9d029da6b5e356"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-19 13:59:23.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-09-29 15:45:42.000000000","tz":120},"subject":"CI: Fix Ubuntu cloud image tests","message":"CI: Fix Ubuntu cloud image tests\n\nUbuntu VMs use the ens2 interface instead of eth0.\n\nWe also increase the VM provisioning timeout, since 360 seconds is not\nalways enough to reach the point where OpenSSH is ready.\n\nChange-Id: I002cadf4c14325bbb10bbfd2efeeb87084bd4314\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit b491c276e789a1c1a0dd262059f3ef8bc2ba538e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/338b5b965a11c1b13e3b282781abe161a5ff0cd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/338b5b965a11c1b13e3b282781abe161a5ff0cd9"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
