)]}'
{"id":"openstack%2Fkayobe~962859","triplet_id":"openstack%2Fkayobe~master~Ia040773a30e3f479402de6261eaaa62a93a1a29b","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-10-15 09:28:05.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-10-17 10:56:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia040773a30e3f479402de6261eaaa62a93a1a29b","subject":"Switch default Kolla images for centos/rocky 10","status":"MERGED","created":"2025-10-02 17:21:18.000000000","updated":"2025-10-17 10:56:49.000000000","submitted":"2025-10-17 10:56:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962859","meta_rev_id":"1cdf0079654485498550d437faf3a0bb2a68cf35","_number":962859,"virtual_id_number":962859,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-17 10:56:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-10-17 09:16:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-10-15 09:28:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-15 11:59:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-10-17 09:16:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-02 18:16:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-04 09:24:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-10-15 11:59:53.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-10-17 09:16:56.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"bb4b1ca37a347a8a6c43f521830b85f016daa3b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-02 17:21:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76d07f3c52234ad8721b6307970462ab2cf147e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 18:16:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9f9c1b44b6b402394329029f86a90f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aed059127224087af85fdb29b432142 : SUCCESS in 9m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/808d4eb7fb2d49f28b917745b48c992a : SUCCESS in 7m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e2faefe3d254b65b82b9527b4be887f : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c887648ccf741c6b849c16e6de0a308 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3a63340e0584383ab337bde6c683d8e : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/540681bd30274fcbb67a3daa36e1c1b6 : SUCCESS in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/25bb16e28f184d6ea25f1b5e412d027e : SUCCESS in 6m 57s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8a788529b3f44fdf9d1869529e46ee0e : SUCCESS in 2m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f0776b7868124cf685928d5c23bc90a5 : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/981d4b84b50e400bb5f929a523700821 : SUCCESS in 11m 05s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c609f62ea12f4cbabb2ff5237cca304f : SUCCESS in 10m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ea08f55c933246439c01bf68adedfb03 : SUCCESS in 13m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3d17ace403043b58e5ae99e50fd753c : SUCCESS in 5m 47s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c2b5a0abc2e54a13941fcbd2880d3e3c : SUCCESS in 9m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f7907e5b416342c3982232a5aef94642 : SUCCESS in 17m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67e5005e6a384c039cb045a45ccc806d : SUCCESS in 14m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6225d54e99c545cb944df4c639fad9f5 : SUCCESS in 49m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/eb530c0cffa3427c821fc17d5a2efa2b : SUCCESS in 47m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f34f29490d1949ceb5c478a5609d3cfb : SUCCESS in 48m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a14f6fd723664a56b46adc674792dbc0 : SUCCESS in 9m 42s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8c7af369a7c7412ebe4eeae2040eb343 : SUCCESS in 7m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f39e63632d74ab0a0d2b71bb793789a : SUCCESS in 8m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/386bd96509ad450eaabcaf96117a176a : SUCCESS in 14m 17s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/373aa393c83949b0bd84a8ee21cd9f11 : SUCCESS in 16m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e476d520aa5e44f496ea633de1b750fc : SUCCESS in 9m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/803b35c01c7948398bcdbf72e608343b : SUCCESS in 14m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a18178e6947d43de83462b0141cbefac : SUCCESS in 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"460618f0f78c06356f3d5bd571de531b3b19f67b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-03 08:47:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ab9da6670ef5ea587bc72b2d194ed431a5088b6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 09:53:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb1fe4c92ea241eebfc17bba4dd1dadb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e13ae301afa347f1b3770105f3db8558 : SUCCESS in 9m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/80cce27cade84ea493af23727493fefa : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba8ed60ae4ac4d6fb0270d76bc3606da : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c70e4d503727477a8cf372f510ca51cc : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d36ed00be0f42878d875bf8bfba984f : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ef55af89c934735bf364ae11b0ca4e3 : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/387c17bba3d34920971c9fc4fec9cda7 : SUCCESS in 5m 22s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b9fc776b6e82470ab7cc92b1ad758ecc : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9ff77937110e40c6b15e41a29d4b1102 : SUCCESS in 12m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a2d4900f5791423891961022fb597a38 : SUCCESS in 9m 37s\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/c7f90a3884204a0fbf562ad583e3f357 : SUCCESS in 27m 17s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/50e99e57093f46fc8d27f0ccde72d6c0 : SUCCESS in 33m 50s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/54772cd65db24218a490eaccd2434ff5 : SUCCESS in 57m 28s\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/63f2316fade44c0184b9b224148a204a : FAILURE in 14m 22s (non-voting)\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0aa18490ed3d4af99f24cc36809f809d : FAILURE in 6m 01s","accounts_in_message":[],"_revision_number":2},{"id":"262812ba51dbb8faff6b46badc7416b6f2399336","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-04 09:24:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"82a4fbbd0e5d1fbf1e3fd1e1819c6b4878f81b71","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-04 09:25:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5337dad7c3392c05e6b4af0261f5c1fda134bed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-04 10:05:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b5fd770c0894d8a972635c8f1320e24\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcabf9fcac404c008dbc6fb8207e7c22 : SUCCESS in 6m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9e36f0058d7f42e0b8c9c7c90fb1fdf2 : SUCCESS in 6m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d46a96aa56f74713baa4d601ac9939e8 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df4e080c8ce9492182d1b5dc248652cf : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c478eda6cd1e4c96abbeccdc85500996 : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/161c9becc7db41f7a6a818acd4808110 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3732c13ee7c649ebbd1bfd3684efd95b : SUCCESS in 3m 54s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7e637f037dc64fb3a5914006630f1ba4 : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/176b68817d7346219b4f498365c1204d : SUCCESS in 10m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/41e4fefd15d74c05bb8eaee14be43fcf : SUCCESS in 9m 41s\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/1c86123e870c41b0913bc2f911e26f56 : SUCCESS in 27m 00s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/cd5862ca1b19486e9b68fed15ef52973 : SUCCESS in 38m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/731a1a7178754f69aad30f4a69bac6b0 : SUCCESS in 30m 10s\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/e13542d05d0c404a87abf1676e0c8138 : FAILURE in 6m 10s (non-voting)\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/32abe96eb6544535b4e48798ca2036df : FAILURE in 10m 08s","accounts_in_message":[],"_revision_number":2},{"id":"141b66602efe56301fbeb526c28e5a0c0224f35c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-07 11:41:38.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d5d9172453d57550d247c451fce5c2ca85210d75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 12:33:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a353566a2ef54f8481747a28a7f5aaa4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fc971123c1c4ef2937251b68e181efb : SUCCESS in 5m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/96708333c3914d69b7bcbfa23cd7c9b3 : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/125976fb66fa4aaea87d8b2a51a34e44 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b326528861074a56875f4bcb8fa47943 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69ac3c03163a46a6ac89f47f8995f7a0 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/38c25765bbc64a8a9f2b3c2ec6b65b15 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f51311fdbf70472db4fa3f5f896e2946 : SUCCESS in 6m 41s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/df2e854eb85d4adbb69cb1637728a5d9 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/09c0e84e11884411ae99faa97a0125de : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc15747b3dcb4cafb83ae94342117874 : SUCCESS in 11m 54s\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/e44b87d398244c668c68452c3465f8d1 : SUCCESS in 44m 38s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fcf0255a4af948bda9cc8cc7335de265 : SUCCESS in 38m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/90827e4285f14d21991dca81a31043e9 : SUCCESS in 29m 29s\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/b529b5ac22264d488c06bccd08fd567f : FAILURE in 10m 18s (non-voting)\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/08b44105a89c4bc6813826e414ece70d : FAILURE in 11m 38s","accounts_in_message":[],"_revision_number":3},{"id":"0d431ac964b078e7fbb43b06c17fb1031f0e5fdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-10 14:02:21.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"dc770f3d8ad868fd06a93e0d97498a3b6aca73ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 15:00:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6d2128de2ef47b9848e6686202cd3a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1d1351e1dc845c5af96021beefbabfe : SUCCESS in 9m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0fe7a5e89bf24ddaaa89d575d4a45f7a : SUCCESS in 6m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1e5263b913b4154aa26258ae9e82f54 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cd5a699ebfd44f38571165a16fda15d : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20e844648e72456c803cef938e7b06fe : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a28e21539f94a0fb051d5163824d5f6 : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0528e9f142474ea0a380f15853c0a0b7 : SUCCESS in 6m 58s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/da7d0a048bfe4480b74299f7bded5f28 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7a0e773c24a44b82b3190bfbcf4f2e0c : SUCCESS in 13m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dddef12a8dc444cba974d505539d1980 : SUCCESS in 12m 38s\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/33fc658f73d146ac9673ed6633773f55 : SUCCESS in 25m 40s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/304065effe9948baaf81716d3d71b546 : SUCCESS in 27m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f9dd032093a243ae938a53587656a310 : SUCCESS in 28m 35s\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/078374eea72e4cf28a606bf6eb37583d : SUCCESS in 34m 02s (non-voting)\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5c265864cc3d45ad9931205b0466c7ac : SUCCESS in 51m 30s","accounts_in_message":[],"_revision_number":4},{"id":"40ad5ae8a08a861479260ec38d42ad3dac0e65dc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-10 16:14:49.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"377c25cd843fc93fd175f7c5ee9ab28ff6e19086","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-15 07:19:39.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"222370506bef6342b73ac06ff07f58b8dfc83a1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-15 07:36:32.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"be2e1a8835acb2ca3e36d193a4418fdc810f6ada","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-15 08:12:51.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":6},{"id":"042f1438ea371427320dfbf0f26e8858371d6a8f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-15 08:12:53.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":6},{"id":"dfe1f5dcc1566e13cdc33ae153fff049ebd391d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 08:41:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ef80cca959344acbaba8cdf33b7f2e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe74b9fca5b140fab4be84b874d647e9 : SUCCESS in 5m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/531bc4baf6354afd80cc55b726260dd2 : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f61c9bb622f14b6d8ad1db0d30472e3c : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88f3fa4e60d344b496caf2dbf92fef9e : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7858dd6d8ee849e1a0a27feba0340a5b : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c7be984cbed040c6aea9958821550e1b : SUCCESS in 6m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f08e00585f2e425997bb17354e287e68 : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/adfbcf37fd424583bdd8390caa10cc7c : SUCCESS in 8m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/57ed62bbff374be4aef9fd53ab9cc63e : SUCCESS in 9m 44s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0fe9733da5b94a7db96df85b73dbe088 : SUCCESS in 16m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/87322b24ef5647f68349702c6b6d39da : SUCCESS in 6m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ec89165b1ff4299b442023b24fe04e3 : SUCCESS in 13m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7112d9719ad24ef89ddcd0aa61e93c8d : SUCCESS in 9m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0e0592b9561c4ef38d56335917674d2a : SUCCESS in 9m 10s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1cca774694f4658b3006f5351c32c38 : SUCCESS in 6m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48c804c7d1db4d36a91bb758e047915e : SUCCESS in 46m 44s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c242737333274d898ee554b5bdc7a30f : SUCCESS in 1h 01m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b97b7b14c1d4f7993ad4bcd013f3702 : SUCCESS in 54m 34s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a3f2222ca924a74a6d1414549f9c4c9 : SUCCESS in 13m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/46bf68c82e964d87b71012b77e7d8673 : SUCCESS in 7m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/063a562c05b845b187e4bc4a757a5a26 : SUCCESS in 22m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6baec69618be43048a26091ff7b99228 : SUCCESS in 15m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ded2bedf57c445c59ec738db82e54a07 : SUCCESS in 13m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d343161f16344fa876e57fc453856fc : SUCCESS in 13m 56s","accounts_in_message":[],"_revision_number":6},{"id":"e23ef8d4993aaa8d9e3bdb8394c5f4470780c95f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-15 09:28:05.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"3bee52a86c267ab56bf2b4e3b066817e42bb0f59","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-10-15 11:59:53.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"0f8cc521ffa3f9f0f62cad257012509410063f85","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-17 09:16:56.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"556476034eb8e9bfce6f2cc7d266011f6b436500","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 09:17:13.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"15a7e285cb0300b865b733769ff731c3032ec4f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 10:56:00.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/416ce061f8aa401c8fce5510117add6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6217467516994b7c8facc6d61c949fea : SUCCESS in 10m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ed10dd0afe42485fa2d51c276222ffb6 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be27efb26254bbf8dd635bf929fd3d2 : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f9336287f1443098af7c55fa6812319 : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3499cbda01d24a0ab8f035b5fc206507 : SUCCESS in 6m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f56743dee1bd411ca47a0acb807d71f2 : SUCCESS in 3m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c78468d5336242d6b79ddd786d85370e : SUCCESS in 10m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/09c102b6eec8498d8e704c6335f49037 : SUCCESS in 8m 57s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/397002aac95b4083b97ed792780c462a : SUCCESS in 12m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e7e0710bbddf4e1684f4c6398ca35b90 : SUCCESS in 10m 40s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28a4b6be7afb49f8bbbc9e89fb2b475f : SUCCESS in 6m 33s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bbae1c18f1c04032a05626002873ea76 : SUCCESS in 10m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/54c91108b9504b748d1754cfa5634336 : SUCCESS in 10m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6be1e8f018147b7b043b590e2775dbf : SUCCESS in 11m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a1ff02c28ee4d88b6909ae3654d23ac : SUCCESS in 52m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f71786ee2721448f932e33da0d00420c : SUCCESS in 28m 12s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5e20b0664894da4bafaa056b8972ec8 : SUCCESS in 1h 34m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45780d8b95484b90875fb08bf77c8f39 : SUCCESS in 8m 19s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ed4e3319629d4088b8839e185a6ab551 : SUCCESS in 13m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37b6d54f87d1498880ac9585724923a5 : SUCCESS in 16m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/be6371642a5748b2b4ed53d7141d6b40 : SUCCESS in 7m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/489d218e592c4f60a9c4de18a91b7451 : SUCCESS in 15m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0878863f0dc498794f66a5012c837db : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":6},{"id":"724ec255438b233b819f97d72e105d6364271b06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 10:56:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"1cdf0079654485498550d437faf3a0bb2a68cf35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 10:56:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94e545cd8b3d4f4fa4485823a6d462d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01c6d88bc72943a09feaa210879d941b : SUCCESS in 37s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"fd9e45be508d627ae2585af6c1837842874bbe34","revisions":{"657d38fe259db875d4817acfab8f067464cc94b5":{"kind":"REWORK","_number":1,"created":"2025-10-02 17:21:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/962859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/962859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/962859/1"}}},"commit":{"parents":[{"commit":"fc0c01e1b6f5fe45375a60483db9b477fe7f253e","subject":"CI: Add centos/rocky 10 seed jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc0c01e1b6f5fe45375a60483db9b477fe7f253e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-02 17:21:12.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-02 17:21:12.000000000","tz":120},"subject":"Switch default Kolla images to centos/rocky 10","message":"Switch default Kolla images to centos/rocky 10\n\nChange-Id: Ia040773a30e3f479402de6261eaaa62a93a1a29b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/657d38fe259db875d4817acfab8f067464cc94b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/657d38fe259db875d4817acfab8f067464cc94b5"}]},"branch":"refs/heads/master"},"63781f0ff4493e0ecbe449965ad48f983d68791f":{"kind":"REWORK","_number":2,"created":"2025-10-03 08:47:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/962859/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/962859/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/962859/2"}}},"commit":{"parents":[{"commit":"fc0c01e1b6f5fe45375a60483db9b477fe7f253e","subject":"CI: Add centos/rocky 10 seed jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc0c01e1b6f5fe45375a60483db9b477fe7f253e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-03 08:46:45.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-03 08:46:45.000000000","tz":120},"subject":"Switch default Kolla images to centos/rocky 10","message":"Switch default Kolla images to centos/rocky 10\n\nChange-Id: Ia040773a30e3f479402de6261eaaa62a93a1a29b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/63781f0ff4493e0ecbe449965ad48f983d68791f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/63781f0ff4493e0ecbe449965ad48f983d68791f"}]},"branch":"refs/heads/master"},"cae56aa511d45c553e0f0aaea7019ad2eab86004":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-10-07 11:41:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/962859/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/962859/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/962859/3"}}},"commit":{"parents":[{"commit":"2bcc769e49479296ca3667815c6f9b929a693182","subject":"CI: Add centos/rocky 10 seed jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2bcc769e49479296ca3667815c6f9b929a693182"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-03 08:46:45.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-03 08:46:45.000000000","tz":120},"subject":"Switch default Kolla images to centos/rocky 10","message":"Switch default Kolla images to centos/rocky 10\n\nChange-Id: Ia040773a30e3f479402de6261eaaa62a93a1a29b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cae56aa511d45c553e0f0aaea7019ad2eab86004"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cae56aa511d45c553e0f0aaea7019ad2eab86004"}]},"branch":"refs/heads/master"},"fdd5fd0b70854531387950dee22f013595e04865":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-10-10 14:02:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/962859/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/962859/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/962859/4"}}},"commit":{"parents":[{"commit":"0e33803e834b1952c4a0b30e7c04fcbee271e023","subject":"CI: Add centos/rocky 10 seed jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0e33803e834b1952c4a0b30e7c04fcbee271e023"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-03 08:46:45.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-03 08:46:45.000000000","tz":120},"subject":"Switch default Kolla images to centos/rocky 10","message":"Switch default Kolla images to centos/rocky 10\n\nChange-Id: Ia040773a30e3f479402de6261eaaa62a93a1a29b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fdd5fd0b70854531387950dee22f013595e04865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fdd5fd0b70854531387950dee22f013595e04865"}]},"branch":"refs/heads/master"},"3276b0dcd78ffd562ff8dbe7aa3549203f2a143e":{"kind":"REWORK","_number":5,"created":"2025-10-15 07:19:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/962859/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/962859/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/962859/5"}}},"commit":{"parents":[{"commit":"bff673a170130af015ded6fa7a1a87f88c08f9a5","subject":"CI: Add more centos/rocky 10 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bff673a170130af015ded6fa7a1a87f88c08f9a5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-15 07:16:54.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-15 07:19:12.000000000","tz":120},"subject":"Switch default Kolla images for centos/rocky 10","message":"Switch default Kolla images for centos/rocky 10\n\nChange-Id: Ia040773a30e3f479402de6261eaaa62a93a1a29b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3276b0dcd78ffd562ff8dbe7aa3549203f2a143e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3276b0dcd78ffd562ff8dbe7aa3549203f2a143e"}]},"branch":"refs/heads/master"},"fd9e45be508d627ae2585af6c1837842874bbe34":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-10-15 07:36:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/962859/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/962859/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/962859/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/962859/6"}}},"commit":{"parents":[{"commit":"ff1f584246c3f1ce64e0bbf29e6991dc1838ca90","subject":"Merge \"Bump stackhpc.libvirt-host to v1.14.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff1f584246c3f1ce64e0bbf29e6991dc1838ca90"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-15 07:16:54.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-15 07:36:25.000000000","tz":120},"subject":"Switch default Kolla images for centos/rocky 10","message":"Switch default Kolla images for centos/rocky 10\n\nChange-Id: Ia040773a30e3f479402de6261eaaa62a93a1a29b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fd9e45be508d627ae2585af6c1837842874bbe34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fd9e45be508d627ae2585af6c1837842874bbe34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
