)]}'
{"id":"openstack%2Fkayobe~963667","triplet_id":"openstack%2Fkayobe~master~Id9d874024279a8e26561aca0e485c310de38b434","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-10-10 16:55:24.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-10-13 20:58:02.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-10-13 20:58:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9d874024279a8e26561aca0e485c310de38b434","subject":"Bump stackhpc.libvirt-host to v1.14.0","status":"MERGED","created":"2025-10-10 09:09:21.000000000","updated":"2025-10-13 20:58:55.000000000","submitted":"2025-10-13 20:58:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963667","meta_rev_id":"334282d3237a6e57a83c3b05e22839fa7a140390","_number":963667,"virtual_id_number":963667,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-13 20:58:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-10-10 16:55:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2025-10-13 19:20:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-10-10 13:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-10-13 19:20:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-10 10:37:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-10 11:01:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-10-10 11:01:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-10-10 13:07:58.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-10-13 19:20:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"0d333fa812b5e806313349bfcd64b45268a84834","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-10 09:09:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba7596e5d768e04e01216afebe0897712edd6fc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 10:37:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93807f04e7d845fcaae2d756b781e032\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f5705edc6914aa9bb1f2b7105319d14 : SUCCESS in 10m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a57f1f1b499a43b1bd43f7059091df6d : SUCCESS in 5m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a08ede1642454f4ca7d697d3e82146f1 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca919bd704b04e7282b3c8ec2f424b28 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6abc36839678455194a6fe892411c630 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/06e28d9cbce245e3ad79b1f275abb52a : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ccb991431fa4ad989f38d9a39dd6f0a : SUCCESS in 6m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/884c2d2f90b640a3ad9dd9fb90997932 : SUCCESS in 2m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e7741389d3b343739e68bcb4d4b25ee7 : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b643333ac05e4685aee70ff4ed5176ef : SUCCESS in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/41f7c40c3d4046ceb457e9712461c4c3 : SUCCESS in 9m 39s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bce78bca1c814b9fa642c75cf1616656 : SUCCESS in 11m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/87ef3f4c797c43269a87375644dc9e14 : SUCCESS in 13m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb5ac3cdfc594725a13bc7b7cfe2dc58 : SUCCESS in 11m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/31d6fc30a2e34b46b0f25e23a8c7a16e : SUCCESS in 15m 55s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d737164cf4fe48e38c1aa0ae675361e3 : SUCCESS in 22m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2b4b5b824d44ceebd7e1413d3cf8d2b : SUCCESS in 13m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ab4b17ff49d47299642f27900ed58e6 : SUCCESS in 44m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/56dc66175f2c49b0b3e7a051339650c5 : SUCCESS in 49m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49e8f13a0204412cad596ac9241be441 : SUCCESS in 1h 23m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4194822834354c1082b1649ef57ea705 : SUCCESS in 15m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/48ef0b4ada4647a0977562599d58e735 : SUCCESS in 13m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/204ad0c929d347699bd9e566b1fb8bba : SUCCESS in 26m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8563900035904ea283f600ed57b6f92b : SUCCESS in 13m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5ddab57f92474708a237a97a6d804493 : SUCCESS in 13m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6951caf1fec2489aa11f6153f23bd38f : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":1},{"id":"58da60a9c355df7feeb7310d1cbe9232513a3627","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-10 13:07:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8944a37d94ae31c218a47b0541acb2d3802e9b38","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-10 16:55:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3af92f5160d03c1595f9025c5e87e3caebf2ac47","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-13 19:20:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae03cc7532a2d6943003be11c7b331ca8300aadc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 19:21:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1e4120bb6c8a2122129866ad3bd900a1039470e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 20:58:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76cf65b086ba481899aa95505c133584\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/381fe7bae15d49f8b14b0e761b5001eb : SUCCESS in 4m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b63f1c8b8c7940fa830c85ef58be96f2 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c3e31f280ca4f78a7513ad6042454c0 : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e98d15dbbac4f5e86a5b9f208bd422f : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d7e313f3f2e34ece8e03a895759ee732 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62bfefa9c89b4c48be243e71af210a75 : SUCCESS in 2m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e9150f615d424f618f6df60c5d90dec2 : SUCCESS in 3m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/51d7d6c2a0964a3aa7948b187c15f932 : SUCCESS in 11m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/79ca8e757f4b48ada594637c110ea1cc : SUCCESS in 7m 08s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/744c4b4d050a4e9f92ee88dcedff4778 : SUCCESS in 7m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e2b3a71c312143b18e70c0987b2ca14e : SUCCESS in 11m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e90afefff764ad29781c0eec5ea2dd5 : SUCCESS in 12m 26s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0c509541da384e808b73f49f4296019d : SUCCESS in 9m 24s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4baf8f8eb7fc40dcbce7e01d204e8c2e : SUCCESS in 16m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14029caab9994cd9a84d24c7ee2635af : SUCCESS in 7m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/050a536d7ca34922ba6120a3c3c07516 : SUCCESS in 29m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fb8565ded0114cdc9c39c8d2ebc35ef1 : SUCCESS in 50m 08s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8124b74a42447b7a1ec7b18fe0f5226 : SUCCESS in 1h 32m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0be14b6ef50f4f4fa04cf1ec032dcc62 : SUCCESS in 7m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/78495b6b53c54538bf4699ae627624cc : SUCCESS in 13m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4c48ec33e6046b89e12a9a8e52d3adc : SUCCESS in 18m 24s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b5b123a08df54489832bf0227287d1f7 : SUCCESS in 16m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/afe976b0cc6e49bc81c145be7c8677ff : SUCCESS in 13m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b173299b76df495bbb358d5a2eb1e5fd : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":1},{"id":"324624bce88fc0d7265742817dd4c5888334adae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 20:58:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"334282d3237a6e57a83c3b05e22839fa7a140390","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 20:58:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9790bf85401428eaac9f87dfe7df58e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9d95e47ba4b4925aa012c80da382714 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f939bf4132774e72b92d1052096ab0ec : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d48d3b2a7122f02c09707f0bdf4ca30d795c6eae","revisions":{"d48d3b2a7122f02c09707f0bdf4ca30d795c6eae":{"kind":"REWORK","_number":1,"created":"2025-10-10 09:09:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/67/963667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/963667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/963667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/963667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/963667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/963667/1"}}},"commit":{"parents":[{"commit":"5bffc12c085aa6f0468fc3a61e414af44a60df73","subject":"Bump Ceph package repository to Squid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5bffc12c085aa6f0468fc3a61e414af44a60df73"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-10 09:09:06.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-10 09:09:06.000000000","tz":120},"subject":"Bump stackhpc.libvirt-host to v1.14.0","message":"Bump stackhpc.libvirt-host to v1.14.0\n\nThis adds support for centos/rocky 10.\n\nChange-Id: Id9d874024279a8e26561aca0e485c310de38b434\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d48d3b2a7122f02c09707f0bdf4ca30d795c6eae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d48d3b2a7122f02c09707f0bdf4ca30d795c6eae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
