)]}'
{"id":"openstack%2Fkayobe~963744","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~Id9d874024279a8e26561aca0e485c310de38b434","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-10-17 09:17:36.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-10-17 12:12:04.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-10-17 12:12:04.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-10-15 11:59:06.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"Id9d874024279a8e26561aca0e485c310de38b434","subject":"Bump stackhpc.libvirt-host to v1.14.0","status":"MERGED","created":"2025-10-10 20:16:12.000000000","updated":"2025-10-17 12:12:53.000000000","submitted":"2025-10-17 12:12:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963744","cherry_pick_of_change":963667,"cherry_pick_of_patch_set":1,"meta_rev_id":"1505206c9a4f46fccc13fa096317d1e4091a1047","_number":963744,"virtual_id_number":963744,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-17 12:12:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-10-17 09:17:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-10-15 12:42:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-15 11:59:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-10-17 09:17:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-10 22:57:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-14 10:02:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-10-14 10:02:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-10-14 10:02:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-10-15 12:42:54.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"f6cbff5fb1bd8ec66728b67c9657a4d71fc9c931","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-10 20:16:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6588446a80293ee7cb4cf898fd21b8078ad3060a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 22:57:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/304196ef26514ffabcf2d09122b2b1fb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ef413adbb3547d7b1da507df7516c91 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/424d974cb4dc4b05a86e4da98dcd654a : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ee893ac90d5402396703022730fb7dd : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec33a62fe8e44570a3d5b672709a6f59 : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ead15028b8cb44108366c8c1e020cc48 : SUCCESS in 5m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e9c8efe50b0c40a1a28a687c66920d9c : SUCCESS in 3m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/007bca807fa84b2cbf660160cd129145 : SUCCESS in 10m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5d2ab3ce841c46628b2eb59d6d4d0dbe : SUCCESS in 10m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2a41631a2e3a435bbfee8b221692dffb : SUCCESS in 11m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2a8c724feb30418bbcd56ad7fbbeb48c : SUCCESS in 13m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0707e47abcb4ea0b36d420564a968ac : SUCCESS in 11m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c5c6fb2d846d4d7ba39e0d78aebaa56e : SUCCESS in 13m 32s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e0ae3d1d53bb4f0f85f5322db32897b2 : SUCCESS in 19m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afc41c9aefef4ab88771e61ae0ae5ab8 : SUCCESS in 7m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ca1f932aff594331852eab8acbdebe27 : SUCCESS in 1h 06m 29s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/dc633b28d3f1494aad9559e6a4e0fb27 : SUCCESS in 46m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bc10bdff052e4c5286b88afa3b54aa61 : SUCCESS in 54m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d30552c311e4e34bcf492b22eb3282b : SUCCESS in 58m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/95c40c859f864e1c9f68787f9a12f2b1 : SUCCESS in 43m 04s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/710c0fbb60fe4e5ebbbad0ca257426dc : SUCCESS in 2h 23m 02s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0a369cb8e15a4fb3a77dedec9399620e : SUCCESS in 2h 36m 22s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb5d1af2050e4ce1bceb92073c5fe599 : SUCCESS in 1h 40m 36s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4bf62eb09084dfb9aa1bdc5c393c25b : SUCCESS in 2h 12m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/04bed038343a4b71a2cf50bb8e2b58f8 : SUCCESS in 17m 28s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/edc7794d3f7240aab3f27d4953efe27b : SUCCESS in 16m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c86c693bf12748219a0d0c56c12c0db2 : SUCCESS in 14m 58s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a1aa9bd545254d818024f3e955cb43ea : SUCCESS in 14m 30s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/88cb45d2de234cc38a4ced619a7c388c : SUCCESS in 35m 46s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8b3604de6d9f415a8674109e7ffe0521 : SUCCESS in 34m 20s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d1e4eb5f1044c10b64da13d692dd372 : SUCCESS in 34m 18s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dadb246e93d04b708049121b8e12079a : SUCCESS in 16m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6946eb2d113b45f488cdbef3f35656b2 : SUCCESS in 9m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e05eb86a264d4735ad6351d53f7597fa : SUCCESS in 12m 07s","accounts_in_message":[],"_revision_number":1},{"id":"661d6eab3cd1179b0b3380d87ed3ed56de6591ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-14 07:05:42.000000000","message":"Patch Set 2: Cherry Picked from branch master.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d98d695ab3bbdc846018435a6793efd39f17c3b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 09:53:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2dd704436e349c58351cc0d4e8ef837\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee360ebc6114465a9686c506439a1d53 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d86b8763c30049ce8fcbc3848f1779dd : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c8cd9f21abd40d3a71059c20a47fa77 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d416ce47d3164f4fa25e3965b51cd442 : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afa9805b53a346b6ad40df9447c58932 : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7ad45fe1043a4a5cac81001d7c44910c : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6cf4cee11814404cbaa9c727117ba74c : SUCCESS in 12m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6fa0e921eb9346ed9eb1d8ca2539d630 : SUCCESS in 10m 42s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a3ed0a6b269b40348de91aae7efdd29c : SUCCESS in 7m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4bde92ac974842f6a4d85b7889484157 : SUCCESS in 14m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c2549b6a57a4db78cecadee4a04a44d : SUCCESS in 12m 28s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6df2925241f642e289b9e0d1d3d56b04 : SUCCESS in 21m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8587e4cd5ad34e72bcfbd13d6a8f96db : SUCCESS in 18m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d723401471e4c3da392b5dda8716e3d : SUCCESS in 14m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cd54431471da4f9aa138233a519ce65e : SUCCESS in 1h 10m 32s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/a25d8ddfce9b4dc388d644414024c474 : SUCCESS in 34m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/48bd27ed46f847f9971e3b14f69d8385 : SUCCESS in 1h 06m 50s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9474c9bd3e704e799224b0675cba5316 : SUCCESS in 59m 18s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c0526bd74ca64c9991f3ea11ce1e0b9e : SUCCESS in 30m 07s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e0997b704cf8410fa467b8f720b9f8af : SUCCESS in 2h 17m 27s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/aee5f919f00f4533965ec222e2753d9b : SUCCESS in 2h 32m 51s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f59b21bede1c4126a2ddf64181ec3afd : SUCCESS in 2h 03m 36s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b711475d598e456b8d3db8e2879fbef7 : SUCCESS in 2h 07m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/95278f97bb0f45e3944d998eb507f0f0 : SUCCESS in 17m 45s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/bc3c2b3118144e1fa5c4e294cc9dc766 : SUCCESS in 17m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0c29a86cb8e459b81ce7953a587341f : SUCCESS in 16m 51s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2e570f45278c4ab5991ee04960bf3707 : SUCCESS in 14m 08s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8852ea63dce8418db463386ed6fca215 : SUCCESS in 35m 27s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/4207a6d6b5c3441f82273ac413c8ba8d : SUCCESS in 35m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/430fbfad483a432abc06fafcb087e4ce : SUCCESS in 34m 00s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/02575cd057354c568d5476037ec305ba : SUCCESS in 16m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7e083cdb0c9249d4872c2ba6f671ecce : SUCCESS in 15m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0417e47ee86747a2b38bb611584b03a5 : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":2},{"id":"7e192a618f54217637fe9a0a375ba0d3f2e8a083","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-10-15 11:59:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"56c69e779bd38e7e854da7ec3055cfe9dbdd2235","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-15 12:42:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ec2ec5bbe1385c4a947fa827882af732ef849949","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-17 09:17:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9aa893de7ed0a458de76d1c1053f5e0cd307606f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 09:18:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8483465bfea99cd17cce0e0690506682258c4041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 12:12:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4ed9665edb14a7dae1f4bf3e9d60a7a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc12160848d242718ab180da59c05c19 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9775f22cfd44a209f59171ffc1bfa41 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fbc5e51cbcc41cfba32751e0ca1ed51 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bdafab29e1747e294143d4f744efbd9 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e4b8b18df4f1462d8a8e57c3178e54cd : SUCCESS in 3m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e5905a3212f444f395c348d48e1c433a : SUCCESS in 11m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e6c320615c314139b7969c8e40c09a0b : SUCCESS in 9m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f1e483aa71a54e70903f3cb9cea9016e : SUCCESS in 11m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b60b0f1048084819b0626c9b5c9ffba8 : SUCCESS in 12m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df6acf1688d84fea8c610558759e601d : SUCCESS in 11m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7981190fc6b344d7969ee78f74959cc8 : SUCCESS in 20m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/77508aa56d6341b3913d9b4572ec36a0 : SUCCESS in 17m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/805340067f09409fa4b260c5844fcf3f : SUCCESS in 13m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d5854918a2a44b0aac19f38f898997fb : SUCCESS in 1h 04m 08s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/d2ceeef8b8d04b808cb50215008a2b79 : SUCCESS in 52m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1a1fc090a30a43d795e919c3b444fbd3 : SUCCESS in 1h 08m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e1d82b67fae437687cafad12ac5b4d1 : SUCCESS in 56m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/248dae27b044467cbacb07a0c095383c : SUCCESS in 58m 20s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b48a41700f234cd8ad433062ddc5aab2 : SUCCESS in 2h 02m 52s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b95816fe7bf944f9ab28590f72e63a18 : SUCCESS in 1h 53m 16s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71212178094e40b49f6ccc22273f657f : SUCCESS in 2h 08m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32f4f25ecd0346e2a1058827a3ef3d56 : SUCCESS in 2h 30m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/503eb991a83b436cb278b1f834e401d1 : SUCCESS in 19m 21s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/1abdf8cc1576484b9d254facb1d699d8 : SUCCESS in 16m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49ddc69fe601462b8d72b87faabf03df : SUCCESS in 15m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/93dc0c6a2a804cd9bc7c984264ae8806 : SUCCESS in 14m 13s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1a069829088246f1862106d5536d9e85 : SUCCESS in 31m 58s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/9151fad66de9488a923cf322efa7b337 : SUCCESS in 35m 08s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc0f321b85af48bf878060ef51bfc129 : SUCCESS in 24m 59s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/279f2552f4f44b8c8dcbdc4fe4e16eee : SUCCESS in 15m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6dfacd2e66d14c99adabae76c9fdf63b : SUCCESS in 13m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3915cedda07c402c88a878ee4bc4f83c : SUCCESS in 12m 28s","accounts_in_message":[],"_revision_number":2},{"id":"5e07e81209a3a6b9b8b37ea22df726ea1734eeec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 12:12:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1505206c9a4f46fccc13fa096317d1e4091a1047","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 12:12:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98efba6359e94f53bc8403543b9df0ff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74a61c6c3cd14ef388d343dc8e86aa36 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a58837d0afe6406cab5dfc2b31e8fa50 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2ec437bf6beb171ee03bdc6348cefd5f1eaa755d","revisions":{"9ce9110d37b369c0cbb6baf93fe700f5d009a074":{"kind":"REWORK","_number":1,"created":"2025-10-10 20:16:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/963744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/963744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/963744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/963744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/963744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/963744/1"}}},"commit":{"parents":[{"commit":"f1207ecaa6cbea5a8c5a99e3efefe03d150bc999","subject":"CI: Add CentOS Stream/Rocky Linux 10 host configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1207ecaa6cbea5a8c5a99e3efefe03d150bc999"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-10 09:09:06.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-10 20:16:06.000000000","tz":120},"subject":"Bump stackhpc.libvirt-host to v1.14.0","message":"Bump stackhpc.libvirt-host to v1.14.0\n\nThis adds support for centos/rocky 10.\n\nChange-Id: Id9d874024279a8e26561aca0e485c310de38b434\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit d48d3b2a7122f02c09707f0bdf4ca30d795c6eae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ce9110d37b369c0cbb6baf93fe700f5d009a074"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ce9110d37b369c0cbb6baf93fe700f5d009a074"}]},"branch":"refs/heads/stable/2025.1"},"2ec437bf6beb171ee03bdc6348cefd5f1eaa755d":{"kind":"NO_CHANGE","_number":2,"created":"2025-10-14 07:05:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/963744/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/963744/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/963744/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/963744/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/963744/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/963744/2"}}},"commit":{"parents":[{"commit":"f1207ecaa6cbea5a8c5a99e3efefe03d150bc999","subject":"CI: Add CentOS Stream/Rocky Linux 10 host configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1207ecaa6cbea5a8c5a99e3efefe03d150bc999"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-10 09:09:06.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-14 07:05:42.000000000","tz":0},"subject":"Bump stackhpc.libvirt-host to v1.14.0","message":"Bump stackhpc.libvirt-host to v1.14.0\n\nThis adds support for centos/rocky 10.\n\nChange-Id: Id9d874024279a8e26561aca0e485c310de38b434\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit d48d3b2a7122f02c09707f0bdf4ca30d795c6eae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ec437bf6beb171ee03bdc6348cefd5f1eaa755d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ec437bf6beb171ee03bdc6348cefd5f1eaa755d"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
