)]}'
{"id":"openstack%2Fkayobe~964347","triplet_id":"openstack%2Fkayobe~master~I173240e0c7b5559a1ef1eea95067bd9d6c390ab4","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-10-21 17:04:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I173240e0c7b5559a1ef1eea95067bd9d6c390ab4","subject":"CI: Replace deprecated yaml callback plugin","status":"MERGED","created":"2025-10-18 04:27:54.000000000","updated":"2025-10-21 17:05:29.000000000","submitted":"2025-10-21 17:04:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964347","meta_rev_id":"e063ad17278712589961613c4305217d04d5f8c2","_number":964347,"virtual_id_number":964347,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-21 17:04:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-10-20 11:06:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-10-18 09:31:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-21 14:22:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-21 14:22:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-18 06:05:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-18 06:33:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-10-18 06:33:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-10-18 06:33:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-10-18 06:33:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-10-18 06:33:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2025-10-18 06:33:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-10-18 06:33:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2025-10-18 09:31:59.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"6de8ddd3ee250423669125a892161a065ce5933b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-18 04:27:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7fc828d40c7bbc41cc81fb1a9959abc3417b11d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-18 06:05:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3f74c33a331403ba67243c349b4fbbb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/596fe5fd58bb4c469a35e948d1c0c920 : SUCCESS in 7m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e7462c5da2534c599a5dcf64d07f6dd8 : SUCCESS in 5m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b8161b15cab431bbad35bcfea9dff5c : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c0eb0ed3182420ba405cc4020ec5d05 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f278241b56549eda9a41a457d62ac98 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21ac7dbe9da44dfd84ebea5df588cdbe : SUCCESS in 5m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f99244bf371b45a4be9a8e79654dc768 : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/db669969185f4b19a0e87544ca53adc3 : SUCCESS in 12m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a8385816870246138e333b8b7459f2d0 : SUCCESS in 6m 32s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b5b184dd073540d78024b1de08bb4fdb : SUCCESS in 8m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/98eb07fe1f52485ea6a24edd81cf29a1 : SUCCESS in 15m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94103139836d4f8bb058c43802e958c9 : SUCCESS in 11m 18s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e0af27bb97404e639d8ec80654a4ab0c : SUCCESS in 9m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9b4c89d4c5cb475ca5bce1e91e858e0b : SUCCESS in 22m 24s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61f7e91274f24a928b849494f9fdf909 : SUCCESS in 13m 47s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/881e809b1def40d78f31d624d3d3e17f : SUCCESS in 43m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8c82a18ca62d4be0900b76144069ee15 : SUCCESS in 1h 04m 38s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7d878cbd33cc406887c13a84ed81de76 : SUCCESS in 45m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6aa5f164a174b6fb0a7f624a9eb208c : SUCCESS in 37m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/725d8832a62c4946a6daec365aab4c01 : SUCCESS in 42m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bde2db0e4a0401c862b7110d8ceb820 : SUCCESS in 1h 33m 03s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/94ee887e7ffb4c55af58b9dd5347d436 : SUCCESS in 12m 46s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/fb71dfa93ef740c9a1e5d6efde1722ad : SUCCESS in 10m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b0bd7f0b25844e694489f601ddd8882 : SUCCESS in 14m 41s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3431f47a74ee424e9b40dae40e2e72eb : SUCCESS in 13m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21043ebaf79542b5953b39cb5deaa78c : SUCCESS in 27m 40s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f9daca880cc2455ca838a66c06d466d0 : SUCCESS in 8m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6eaba9c7c6ef49278e2b0cee25b61e94 : SUCCESS in 9m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5610b1f5fea747ae92ba7e829f1036dd : SUCCESS in 9m 59s","accounts_in_message":[],"_revision_number":1},{"id":"12084a89ea1ad08eb778ebe9183efcc2371ec852","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-18 09:31:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5e69143fdf48aacfb06306ae12728302f830e370","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-20 11:06:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a9bc161a81bee471045445815191ce8dfbdee484","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-10-21 14:22:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c3048dc7a02a19d68015275b9914c71ba774a800","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 14:22:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02417cbf2b34bcd60c6cda034862ef05357a7a60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 17:04:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdc26d4b159f446e9261ab42072e151c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a39efd65c9e34a208eb28ec843c5e7b4 : SUCCESS in 8m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6c22f32edfd246be82d24392d569ece3 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c6c1a118a4f4f8594f4bb23cda1b54c : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21d53510d6354c179b13f84fc09fbe69 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4393d78029c41fa972522d35e2ce215 : SUCCESS in 6m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/15f13ea4df28472dad72d471a99c38de : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dbcf1aff556b4f15a8cf1e6ba5fc3496 : SUCCESS in 12m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/25f7907d327542a1b934f0611871196d : SUCCESS in 5m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9f38d6d9a3c44ed6b12f913f04a12711 : SUCCESS in 12m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b402637ef8c14c6bb2b8f170436d6285 : SUCCESS in 14m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aabb29b9a4ad4877b0775a81aa1dc0bb : SUCCESS in 10m 55s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/99075ebc8f0849788ac0974438b659cd : SUCCESS in 16m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/987439e9979f461a9ae39b9cbbd26141 : SUCCESS in 11m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1497872550654ffe91dbe6956265fde7 : SUCCESS in 6m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ef337b7a98944c838cbb2e1a9325cc9e : SUCCESS in 25m 45s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/130bdc7651924106bd7eb084a11ae209 : SUCCESS in 28m 41s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/20f7ecf486e847cd92032997f4bf98c7 : SUCCESS in 1h 08m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91798161e41d42f19516618b0b9b75bc : SUCCESS in 49m 29s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5cd5ae536e8342299993a9f2c47c6c6a : SUCCESS in 51m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81280b74763a4f5da3908b8b4a8deb08 : SUCCESS in 1h 52m 33s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5f426c2a74b3424c9a7ad236e9f924c1 : SUCCESS in 8m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/36b83ea44dca428680706a00402e202b : SUCCESS in 9m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffc5fad14e694975af18719d3da61ab3 : SUCCESS in 14m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/deea9f54d6da4ca49fa209d58429e654 : SUCCESS in 14m 37s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfb3ae40614d404f990ed188e3dd46dc : SUCCESS in 24m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4aa1881a421143e68e51ff4090d882d5 : SUCCESS in 12m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8d63a944b88e4c38b15349ae5a92246b : SUCCESS in 7m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94c45ef7f7f647d9a9cc52528efa0d2f : SUCCESS in 8m 23s","accounts_in_message":[],"_revision_number":1},{"id":"43d5ebbe8264ba8f064ac157ab6877052a73b833","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 17:04:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e063ad17278712589961613c4305217d04d5f8c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 17:05:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f9c198582474f499da89733286ae64d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/700c25cf6ea64e0b9748b1906c630fd8 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8882c873055ac74c0916d6d6acc58b6823c496a7","revisions":{"8882c873055ac74c0916d6d6acc58b6823c496a7":{"kind":"REWORK","_number":1,"created":"2025-10-18 04:27:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/47/964347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/964347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/964347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/964347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/964347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/964347/1"}}},"commit":{"parents":[{"commit":"33f39fa6cace1f7b408163f7d5faa901a6491be4","subject":"Merge \"Remove default(omit) from Bifrost inventory template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33f39fa6cace1f7b408163f7d5faa901a6491be4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-18 04:27:18.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-18 04:27:18.000000000","tz":120},"subject":"CI: Replace deprecated yaml callback plugin","message":"CI: Replace deprecated yaml callback plugin\n\n[DEPRECATION WARNING]: community.general.yaml has been deprecated. The\nplugin has been superseded by the option `result_format\u003dyaml` in\ncallback plugin ansible.builtin.default from ansible-core 2.13 onwards.\nThis feature will be removed from community.general in version 12.0.0.\n\nChange-Id: I173240e0c7b5559a1ef1eea95067bd9d6c390ab4\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8882c873055ac74c0916d6d6acc58b6823c496a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8882c873055ac74c0916d6d6acc58b6823c496a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
