)]}'
{"id":"openstack%2Fkayobe~964538","triplet_id":"openstack%2Fkayobe~master~Iea0ea4924ba1faf2a1af4a0f14bb3926c960801a","project":"openstack/kayobe","branch":"master","attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-10-29 13:56:32.000000000","reason":"\u003cGERRIT_ACCOUNT_34076\u003e replied on the change","reason_account":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}},"34076":{"account":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"last_update":"2026-04-09 16:20:47.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-10-24 09:48:34.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-09 16:20:47.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}},"hashtags":[],"change_id":"Iea0ea4924ba1faf2a1af4a0f14bb3926c960801a","subject":"Fix ip-allocation playbook missing network tag","status":"NEW","created":"2025-10-22 12:51:08.000000000","updated":"2026-04-09 16:20:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"e0ae235cc53b2e8610e02b92e3149c0be9df6705","_number":964538,"virtual_id_number":964538,"owner":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-22 14:26:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2025-10-24 12:07:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"date":"2025-10-24 12:07:41.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-22 13:14:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-10-22 14:26:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-23 09:57:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-10-24 09:48:34.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"}],"messages":[{"id":"a02d8c5a48b853daaffa28d3814509c52e7c0a0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2025-10-22 12:51:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8058dad5edf1e5a8c74433c91153d454cc92e914","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-22 13:14:36.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bdfbc885bc541fce45245532400c66030478a93a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 14:26:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/458d898521b74c24bf5869481f340f2c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/755ff2d839fc4f7f86e15c403bde15a3 : SUCCESS in 9m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/80f21d8ddf634fed9f195e43383b16e3 : SUCCESS in 6m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fff9ef72a710449486fc78cb708a2027 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac72a7dcb1074818a5c6bcfc46dff902 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e0a246a8f71437d801284bd42357d0a : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dbfbc839089b4c3492ea4a2bebfe1940 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f69a078d7b34840b635a232310162eb : SUCCESS in 2m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cefc71356e2f4a9ba1d154492ebfdcd9 : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/307cc8ec63ff4693bf56cd5a3494f84e : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e685b7af3ce74d2aaa0b92b8fc356061 : SUCCESS in 10m 17s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/36a3bb6d616b4f398db71b0c1340fd20 : SUCCESS in 13m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3be107ea823040209552a9c839003317 : SUCCESS in 16m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb75f6ac9a2d4e9089498a609790a941 : SUCCESS in 12m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9ab68d16db6b40c199cb69e3ec433af3 : SUCCESS in 21m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/655dd044cbdc4d7a97cc579b313e9e25 : SUCCESS in 19m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3812ef6a227a4641900c3d7964ed39dc : SUCCESS in 14m 06s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c7cbf4a545ff4f3cbf98928f42f869f8 : SUCCESS in 59m 24s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e416abc1da094881b80402c7fb4995b0 : SUCCESS in 37m 27s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/bae0b8e77e404b8591f1c19392367cff : SUCCESS in 46m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86dbce24d50b4122a82140a798f8225d : SUCCESS in 47m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7525b9e4b9644115ba2d738e4357079e : SUCCESS in 50m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2c4128111cc4b9885e514d138e17718 : SUCCESS in 1h 24m 25s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/976d5f39d07942c9844420b149b3e868 : SUCCESS in 17m 40s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c312c5f18de541aaaf32408cf38d8625 : SUCCESS in 14m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6d5e77bb7e8417cbd56c371a6b18041 : SUCCESS in 10m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/33de62ee779b4d1290b2e068a95fb1ad : SUCCESS in 14m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90cb4bb5df2e457bb6de39bbc5be2dda : SUCCESS in 26m 37s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/59cfef52635746fb880d957ae7f326ae : SUCCESS in 8m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a7f7b0fe5ddc4af094f02213083ad3c7 : SUCCESS in 15m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e1153950e7e42489d7f20f2eccc4d64 : SUCCESS in 10m 55s","accounts_in_message":[],"_revision_number":1},{"id":"fd8eabd507e91db2f2ba9a08695ecb5b8852679e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-23 09:57:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58a68bc5bfac22362bd2c825eb1f7d6e61117f5e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-23 13:08:07.000000000","message":"Patch Set 1: -Backport-Candidate -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91693f40be410de3fc4ba1ccc9a3bf16413355f8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-24 09:48:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c6865d96485f30d4e42593f0324383dd557df1f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-24 12:07:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42db37869c6e619383fd29f653ba392cfdfacde5","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2025-10-29 13:56:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0ae235cc53b2e8610e02b92e3149c0be9df6705","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 16:20:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b11f8d18f82a9c907a81327ee2738438bc00c7be","revisions":{"b11f8d18f82a9c907a81327ee2738438bc00c7be":{"kind":"REWORK","_number":1,"created":"2025-10-22 12:51:08.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/38/964538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/964538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/964538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/964538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/964538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/964538/1"}}},"commit":{"parents":[{"commit":"8882c873055ac74c0916d6d6acc58b6823c496a7","subject":"CI: Replace deprecated yaml callback plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8882c873055ac74c0916d6d6acc58b6823c496a7"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2025-10-22 12:45:49.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2025-10-22 12:50:29.000000000","tz":120},"subject":"Fix ip-allocation playbook missing network tag","message":"Fix ip-allocation playbook missing network tag\n\nFixes ip-allocation playbook not being run when host configure\nexecuted with \u0027network\u0027 tag.\n\nCloses-Bug: #2129550\nChange-Id: Iea0ea4924ba1faf2a1af4a0f14bb3926c960801a\nSigned-off-by: Jakub Darmach \u003cjakub@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b11f8d18f82a9c907a81327ee2738438bc00c7be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b11f8d18f82a9c907a81327ee2738438bc00c7be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
