)]}'
{"id":"openstack%2Fkayobe~964693","triplet_id":"openstack%2Fkayobe~master~Ia2d2c824c64fce4779d4e38eec155dc020a4ad7e","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-10-27 20:19:59.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-10-27 16:28:02.000000000","reason":"\u003cGERRIT_ACCOUNT_17669\u003e replied on the change","reason_account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},"38234":{"account":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"last_update":"2025-10-27 22:04:40.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-10-27 22:04:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia2d2c824c64fce4779d4e38eec155dc020a4ad7e","subject":"Account for encoding in _get_direct_url","status":"MERGED","created":"2025-10-23 13:49:54.000000000","updated":"2025-10-27 22:05:35.000000000","submitted":"2025-10-27 22:04:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"964693","meta_rev_id":"5bed8d061537eee5190b40895efa97eab5c232eb","_number":964693,"virtual_id_number":964693,"owner":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-27 22:04:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-10-27 20:19:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2025-10-27 16:28:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-10-27 21:55:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-10-27 20:19:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-10-27 20:19:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-10-27 20:19:59.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-23 13:53:17.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-10-23 15:28:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-24 09:25:40.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-10-27 16:03:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-10-27 17:43:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2025-10-27 21:55:11.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"2c74afa721cdb4fc47eb53e434d5b294fa5f4e07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"date":"2025-10-23 13:49:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c6a8209cc379ff1db48eefc0adab4a70542a3cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 15:28:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/52dc17a49cbc43899f03e418b8226779\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8db3a56f815442a196cbe76b716322cb : SUCCESS in 10m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c8b7ef9514cd41f68343649e6a6dbb08 : SUCCESS in 6m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/208dbb37776c44d1b76d0fb4be63b5fc : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afbc8a8cfc3a4d8bb0a68e77688836ee : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea9b6ff3c5ae459f9524a3faf9296424 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2de47dd6697a4f248be883c97f51627f : SUCCESS in 8m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c3d6b97bf1d4d3a9d91fe8374b9554f : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ab6e6101c7944760b105b983fb1b9e3f : SUCCESS in 7m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dcd75a9804d14f3db7bb2ec06d5bd36b : SUCCESS in 11m 30s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/92f8c171f0554ca2b1135a55a3cd706b : SUCCESS in 8m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e4e9b44146c6439d8b3e8d776a220808 : SUCCESS in 8m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2e543bd57ee4bc5a50a690766514847 : SUCCESS in 6m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/389df45f63314abda445aa48e24f4b3a : SUCCESS in 16m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/93f1e33e6e57413c832e03d4a2bacdc0 : SUCCESS in 17m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e34ec0e88cec428694d4491e0ba18bb1 : SUCCESS in 8m 20s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/23a88c6cb14f4faa992d15d6fd955d02 : SUCCESS in 45m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2eb2e51ac3b74883b4c3fcbcdf521f91 : SUCCESS in 50m 45s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2305e0143590465586d6fcb14d7099d7 : SUCCESS in 51m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d9a5a1fcd02466498bb1220aba1a336 : SUCCESS in 23m 49s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e1424bde3fa94591a5281b1ccef6e4d3 : SUCCESS in 24m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06ed616f6d0344bf87824f67beb2cf1e : FAILURE in 1h 14m 59s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/01563ce06b7d463088d5ad8f0b41a0e2 : SUCCESS in 14m 00s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/12ec659c7b5a4225b23f1a22e6bd8288 : SUCCESS in 9m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/486b1212749844e0b78906abe8a2d744 : SUCCESS in 9m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e70f783e8feb4d508a5d53face86e932 : SUCCESS in 14m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a416375abd54a6a9528fc2fe618811c : SUCCESS in 24m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a0e1755198cd4671b3f7b40821ed56ed : SUCCESS in 9m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/040a545b09754c0ca01db2bedcd856f3 : SUCCESS in 14m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63e806d320b743669df42a144bcccfb3 : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":1},{"id":"243ec2c8eec97596cde2cdb71c6e6ba5606ba902","author":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"date":"2025-10-23 19:23:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f12843843ef42c54e1622c336738473fe6643d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 20:31:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50857d332c7a41858fa34e803a80e22b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/726303944a144d71a15dedab9fac2769 : SUCCESS in 10m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c3cb120434ce4ae18bdac05fd4f8d0ef : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b134363d1494a5688a695d3e76424c1 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50d936e3839840b1a5e7c7ed2ebedb10 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1dc58f66cc5b42a9829487f64e64cc3b : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6116dd8712ae4328841de4389de41838 : SUCCESS in 7m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2969667a58c44892ab81deb3c02bd84c : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aeb75ea77935443d80d000eb57bb2407 : SUCCESS in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b47a7b555ba3435cbba43f010eaae831 : SUCCESS in 8m 46s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e1af9061ec5847faadf1fcddab5045ed : SUCCESS in 7m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2eeb77c3665e4a20bc9a16a19ec4013e : SUCCESS in 16m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22243ab482604fba806c27fd0515d13b : SUCCESS in 10m 41s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/96989d6738bb4568b57d17927a2ea920 : SUCCESS in 16m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c17df07e889840b99ba4c0d1231d115a : SUCCESS in 9m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1466a924ca64f94b68c2ee94334ab56 : SUCCESS in 8m 15s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/09dcd3335990460781a674154b363196 : SUCCESS in 1h 03m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d09cc729ec4d4a7abbaef7974f4c264f : SUCCESS in 46m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c57528cbcf2046b8ac4d649c4de21c33 : SUCCESS in 30m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4be24e58b69e4bab8ea40d8c714f0f0c : SUCCESS in 46m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d426fcf38d694cb28047b6e542f5c1b4 : SUCCESS in 50m 03s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c048b3f77386407584dc5646b2cc3cab : FAILURE in 37m 45s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cd725dfaa2144d469e1cef23f6cabedd : SUCCESS in 13m 00s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a33dc34265384961aaca685791b271c3 : SUCCESS in 14m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4dc1b01dd3b846b98536f4cd6aa65244 : SUCCESS in 13m 42s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/612a5f63c39e4702b38b2a75f5cbbcb7 : SUCCESS in 13m 34s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a10d77ea9fd47609065bf8109f593d0 : SUCCESS in 31m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c042df0f17e141a485cb85992f5eb7cd : SUCCESS in 12m 18s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3a9361af3d474911beda5fe882f0eee7 : SUCCESS in 14m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e90866f1b9a248ce92e6b16ee619be19 : SUCCESS in 13m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b3aa075e0e185182c2f8e7747331c0240c792134","author":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"date":"2025-10-24 08:39:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48147ec29c703357033a3157b34ba499f91aa788","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-24 09:25:40.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6841936cccaa1b5b713586b24a7b9b2b8b16b790","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 09:58:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d86bf732be964cad974bf0a129404613\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98fbfd2ea23a4213ba5abb8820b8bc38 : SUCCESS in 5m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ab0e2518bc7b4e5c8b099e3710a2bca5 : SUCCESS in 3m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76904a2bdc354e0f9fc63476d81142b0 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7270a95b3bd04b6e832cfa253ab2f795 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a11c016346a94022b03de5977a2aae89 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c1224e274b854f15ac46b7c2b2cc4e2d : SUCCESS in 6m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f8e52c7fab8e47b996b1c7e9c38860bf : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c5cde6492954311ba491c3b367b7ef8 : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a41f9cf4c2b34000a57e426931a04d58 : SUCCESS in 10m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bb624a5cdad644a09dd354728397a110 : SUCCESS in 7m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/efd4e778b2ac4e0aa419e15a77245859 : SUCCESS in 11m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5b4318f13e242e9840f7fae107a5109 : SUCCESS in 12m 10s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8c022d153d544595b517184ad5a8b07f : SUCCESS in 18m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8e88325d47ef419f97c7cc6da6a0bf0b : SUCCESS in 17m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/037f2dae878940ecb0ac80ada40891f5 : SUCCESS in 14m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/391771ff7ae746a5b5f16f9511879ff5 : SUCCESS in 25m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/87e78a52a1e046b8938278a020966db7 : SUCCESS in 49m 02s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4ce4d20ed0364f30903cdee47f545e08 : SUCCESS in 1h 13m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2ab44a6a6ac49b99cb60e17888fa841 : SUCCESS in 53m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/32772f11f5604e22a13a8d0075689654 : SUCCESS in 48m 57s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/947bf6264d194037b403946ac287c801 : FAILURE in 45m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/155955b0dfc8415386e7aa1c6bacc6c3 : SUCCESS in 18m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a2210a9c2b5745268bb830ca5f1cbc68 : SUCCESS in 12m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc4b30f01fe644bd832ca088913798bc : SUCCESS in 7m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/71041fa743ab4ea9b55882dc8ac4ce03 : SUCCESS in 13m 07s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de3fa46b057d4c0eaaee36be67f624f2 : FAILURE in 18m 25s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/935c3cc3a0e04d02847527b94f2569ed : SUCCESS in 11m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1049f143b96145f9973692a238b6f52b : SUCCESS in 12m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5325e248ec4418c8615ff0d5559651d : SUCCESS in 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"9b13fc88c45616b7638b7df6d7882c6fc565d75f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"date":"2025-10-24 10:14:04.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"96ba1425cd4685d7484815134203aff587e4f829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 11:33:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/564c009b9bfe453d82a75b8962177ed7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e33cddb78644d96bdb728efe41119c2 : SUCCESS in 9m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/df322311aabf4862aa83119a7931cbdd : SUCCESS in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/431fbbe3a7e441dcae8b9e9b1da03bec : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d96d67b7f3941cdb43567bedd8880b7 : SUCCESS in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bbdd0512e16a4136a96c3aaab5f3309d : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12efca240a884f278ccc60be4276f957 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2737755a1f5c4b2eafd050fd0d46053a : SUCCESS in 2m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/da1f6635e9f64da9bdd8f85112600afb : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c3de323da6c4eb3a10a7b15b8461316 : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6640456c54d8405ab6b75ca3f400f902 : SUCCESS in 10m 26s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c48c9d0ba7c04b21884cabeb5f9a49f2 : SUCCESS in 7m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3f7b5de70caa451ba9cfd45a2c4be8ff : SUCCESS in 14m 27s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1eabddc3d7bc4f719bda9f516bb1c8b5 : SUCCESS in 13m 29s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/019f5b06d2054869bf5c0d0c7eb35d90 : SUCCESS in 13m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bfb440e16dac4cc0b84fc01f36a3b91a : SUCCESS in 18m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a1e951ce34348b0b2f77a14f16c33d0 : SUCCESS in 14m 10s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/34bd63d2894f48ea80298d4ee1247301 : SUCCESS in 59m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f03deeabf99f4efe98ae15deff6d928c : SUCCESS in 1h 02m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ae9f15ebfbad46a0a97cda7d23df4ec5 : SUCCESS in 50m 22s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f22f8fa2062e48e1a18d210095624f67 : SUCCESS in 55m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6a7f4aeeee854a899cc6ada9cca266a4 : SUCCESS in 51m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7c7188221a548b1ba996281a1654af5 : FAILURE in 1h 17m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5ffc269a2ff2420b88c32e94d562a666 : SUCCESS in 18m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6a93d9882a92415d811fe74472883628 : SUCCESS in 18m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd97d41fa6f24c2f86bfd378c50f528c : SUCCESS in 7m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d3b87fc03efd423882e09d7a1408b97e : SUCCESS in 14m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03ab2864ca35475a8d44b7d1eeb72e92 : SUCCESS in 15m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c0a531d3c70b412db2dc9a3ed5edc9d5 : SUCCESS in 17m 07s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f1d05955aafb41eb86b862c7b9b3ddb5 : SUCCESS in 10m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0bb950edb6d4a83a7c9d8f6e5a07cda : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":2},{"id":"7a81b174b372f4ff809f2c710c048a793e2b1963","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-27 11:15:30.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ed0e8fd34ba6ca9807f750e9c965bf816f5fdd34","author":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"date":"2025-10-27 11:24:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b86a0f9edfc1f7990385b7a698afda0a2acaf8f","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-27 11:30:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a86e3d687b0d9c3f9984fa66bb7c3ba7d346f319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 13:05:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4fc4e096a764a75a72a557c410007dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e70f39c0c2424ccb967e871a950b33d5 : SUCCESS in 10m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8b55d5fc6df243e28cc96ad3ca942254 : NODE_FAILURE Node(set) request e19bb61e5e1c4c96bcaac31bea5e4824 failed in 0s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92863d02080441d9879d25c24b8fa8b7 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea75970f218b46a2b5efe6615014bb86 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e4b5c0b8b144444ac185034b1952a00 : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1feabafa7244fba9df59f605e90d5bd : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ba096075dc24bf8bcafc9b8c4a3d98a : SUCCESS in 3m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c9899ca2693f419d97ea8d77ea7942dd : SUCCESS in 3m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9f407c0515e6499781974a6d311cb675 : SUCCESS in 10m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/976043b949d84ee49aa20903f13e7bdc : SUCCESS in 5m 37s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c7e8aaf6e44744e1b100cf474f1a6d84 : SUCCESS in 12m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/373c5a8970b846dfbb4e89781f86c984 : SUCCESS in 14m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ee0f07916fc4658a58f18fecbe1bf03 : SUCCESS in 11m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/47bbe9e6867e4fa19da7ee58da251751 : SUCCESS in 16m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8753ca799fa4414ea20cd6f889570925 : SUCCESS in 17m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f11690c327b24791a7117b804a07a3c4 : SUCCESS in 7m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fc1bc1a1124c4c0a9636dad0a7acbbf1 : SUCCESS in 26m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/682447b015ad4ac99c1a998ec44e4d02 : SUCCESS in 29m 04s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/cc77818406ed4af889b0cd5cf1e6b998 : SUCCESS in 54m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3dd4cfb0472247dfb9a739309edf00a4 : SUCCESS in 49m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/98769a0798184d15960c3f1bc0c17ecf : SUCCESS in 50m 19s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd1be4aa47c14849b11dae2bb46d104f : SUCCESS in 1h 35m 11s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/be846fc3dc644052964457535375e549 : SUCCESS in 14m 33s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/13815aa49f8f45c8bc1e1f95eed9cbd8 : SUCCESS in 13m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d45934b34f914fe2a9b4f9452f044d87 : SUCCESS in 10m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b011c12ce21a4b36b5d6af7e2f9f3cea : SUCCESS in 6m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b7072883c1f4371baab0f4d7c354e83 : SUCCESS in 27m 25s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/282d441315cf4ffba7c462d155d9a1de : SUCCESS in 13m 34s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/99339ce6c7c24e18b095105a0a098b5d : SUCCESS in 12m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27dc5074ea93454ea728086a000dd480 : SUCCESS in 12m 24s","accounts_in_message":[],"_revision_number":2},{"id":"a6ea12e899a5f98e0bc94e00d693e6cd1acbaff6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-27 16:03:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d351946dbe1d4dce155e502d4ebb5d72656841b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"date":"2025-10-27 16:08:44.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c8332e0534e739b02f5d7634e375ecacd00f99b5","author":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"date":"2025-10-27 16:09:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"89f9f7eb441b492a856b1af81cb041c178d2e45e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-27 16:28:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7826ceddd37e733ecb6ca9fc2e03eb09f6da11af","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-27 17:43:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"96eee72b5229ed9b77dca86dede840b3e0c6d7c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 18:04:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16ee6bdcbd94470e8d581929a57cb5a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2bc66ce2f6b43a5bd21dd423cc5e74e : SUCCESS in 10m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3fe6d849334047c4980e40c66b08d4c9 : SUCCESS in 7m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a0b6afeb2424ad8b1ad1891577a9c62 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2005893573c2494c9d4271103641e6b3 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14af6fad2a514a9a95cd9c30a666cff1 : SUCCESS in 5m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a872ea9ecc4648aab566529bad4c08c4 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39b925ae03484da79f63577d561c68dd : SUCCESS in 4m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f69c217dab7449889012229be2601151 : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c0167b7ee35c45daa12a74d9cfcb5ee3 : SUCCESS in 10m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ba59c17c36f34677b84fc2201ed760cb : SUCCESS in 11m 26s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/48723e92d00b40fbbe1b2df52d74e9e3 : SUCCESS in 7m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c0d1216acee44b90ada149a92ea8f019 : SUCCESS in 15m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f119cee47aa46e289db133121de4768 : SUCCESS in 13m 05s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d940b7f1c6974a67a2c0cee3c338ad9c : SUCCESS in 21m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/522395ab53e6482681e24edc3c27e618 : SUCCESS in 21m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2dc96b860d2a4ffc81a7b73db2a68983 : SUCCESS in 13m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/cbb818001c0c4eb1a5ba550e9c9c5acd : SUCCESS in 59m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1800f2f6a35d42b896b0ac306f142547 : SUCCESS in 51m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/485594015eaa4358b9032d81bfacb242 : SUCCESS in 59m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e69e537ffc8f4984857e3a03a04d4f49 : SUCCESS in 55m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d9eacd0f27424377a078617fbc689651 : SUCCESS in 52m 14s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d25350234da4c609f004bda5941b46d : SUCCESS in 1h 51m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b1b3bdba350846879bd1fda2a328c4f0 : SUCCESS in 13m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d8a52cd2c4ce46949f44da42959aca2c : SUCCESS in 13m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/deb7063bcc9f451384dcf7221dc34500 : SUCCESS in 13m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c33337378f0048c893e5dd4bede22e76 : SUCCESS in 13m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee3520f7f7c54bf4a6881aa0cd61df06 : SUCCESS in 29m 59s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/eb6ce03472d3471999f2069780bfb820 : SUCCESS in 12m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9fcd36cf0fc940de97d21fd0eb51b8da : SUCCESS in 14m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da2b76f7e9fb419aacd09e6f81d6f2d4 : SUCCESS in 11m 47s","accounts_in_message":[],"_revision_number":3},{"id":"257710596122933e981ee6a0362160df53ba5067","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-27 20:19:59.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"51696a77a67272777987cae85ffde86e390bc805","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 20:20:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9ebcccf519bd64077dfc043f3783ea0475741e68","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-27 21:55:11.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"734c82e7f6664b2cc6f449b6545fb7c9a149d32b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 22:04:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b8262f15bb44c54a81465b99a0cc1f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b8c55d3416e4f6ab902f5bef2917797 : SUCCESS in 4m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4cef737d4f094d1d9103fa3b792edd58 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cc565a0a5034baab429a75273bacce2 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ad0748962884eb7ab6a1e28deec31d4 : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff1f45e75c6443fd992ae34ad6ee974a : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9df124b70629478faa32da553419877f : SUCCESS in 1m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2e1ab161dd57483f8f44fefaac8bca71 : SUCCESS in 3m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/38d86add8cef442d90eaf6e33f14f5b7 : SUCCESS in 10m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/776482dd485045bfb64a787dbe8f24b5 : SUCCESS in 6m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/74d37121f30049229c6edd7d35bf4a06 : SUCCESS in 15m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/96a7560f5d8d4f3ebd7a1fbe82cf3005 : SUCCESS in 14m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c3859b1e2e04f9ba7a8122119bfe022 : SUCCESS in 11m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/896843ebc3534d94b3ac7aead807ed47 : SUCCESS in 14m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d9988ba42de548bdb85d008110022e99 : SUCCESS in 18m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/952753ce25e14eb8895961b7a603b019 : SUCCESS in 14m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/00c45d4adc2744a282438f0d1b4867bf : SUCCESS in 40m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c467af7b15bc4758bb94c248146b714b : SUCCESS in 29m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3794a336783743bf8f52404a30b7228a : SUCCESS in 53m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1e2d99ecab4497bb882737194cf88fd : SUCCESS in 43m 18s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ed1ba8774f3e4bdda53e4792b54814dd : SUCCESS in 51m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a4054c3ad324b60a15792368bb8a08d : SUCCESS in 1h 39m 41s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2693156634cc4a81a3cfda356e66ce12 : SUCCESS in 8m 28s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/fcbb06a04f5741ceadb456c2b6c8d241 : SUCCESS in 14m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70c6e623bf164d36b09bc76e10af37be : SUCCESS in 14m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b4cd253f48b54a81880cb13ca4b8dc3b : SUCCESS in 12m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/847d7815e37c47629beacdd8dff46ac7 : SUCCESS in 26m 52s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1e62c08a70994b73ba32a826184a0c3a : SUCCESS in 7m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/036e09fc06c7434fb27f31b5a82bc1fa : SUCCESS in 7m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e80cc4b0ee847b6b3cc922ab70ab7f3 : SUCCESS in 11m 20s","accounts_in_message":[],"_revision_number":3},{"id":"1200e90ef8479e6b5588c42857b852a9846466a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 22:04:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5bed8d061537eee5190b40895efa97eab5c232eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 22:05:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d8140835fff4e21998c8e9148f89069\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/341432ac0f634b209e68317c55a997a2 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9510a403e5d4588814671f60e12e8ee : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5eb4cbc6ffd4324501a82325e63c3b35aafe403d","revisions":{"fa3f76239901a1b9221f6e74186723a1029c8dd8":{"kind":"REWORK","_number":1,"created":"2025-10-23 13:49:54.000000000","uploader":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"ref":"refs/changes/93/964693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/964693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/964693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/964693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/964693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/964693/1"}}},"commit":{"parents":[{"commit":"8882c873055ac74c0916d6d6acc58b6823c496a7","subject":"CI: Replace deprecated yaml callback plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8882c873055ac74c0916d6d6acc58b6823c496a7"}]}],"author":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-10-23 13:47:25.000000000","tz":60},"committer":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-10-23 13:49:31.000000000","tz":60},"subject":"Account for encoding in _get_direct_url","message":"Account for encoding in _get_direct_url\n\nModifies the method to strip out url-encoding of the file name as well as a better way to gather the path that avoids double slashes\n\nCloses-Bug: #2129687\nChange-Id: Ia2d2c824c64fce4779d4e38eec155dc020a4ad7e\nSigned-off-by: Ian Watson \u003cianwatson92@hotmail.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa3f76239901a1b9221f6e74186723a1029c8dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa3f76239901a1b9221f6e74186723a1029c8dd8"}]},"branch":"refs/heads/master"},"95c842023844d800b6b03884f8f449673683563a":{"kind":"REWORK","_number":2,"created":"2025-10-24 10:14:04.000000000","uploader":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"ref":"refs/changes/93/964693/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/964693/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/964693/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/964693/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/964693/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/964693/2"}}},"commit":{"parents":[{"commit":"8882c873055ac74c0916d6d6acc58b6823c496a7","subject":"CI: Replace deprecated yaml callback plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8882c873055ac74c0916d6d6acc58b6823c496a7"}]}],"author":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-10-23 13:47:25.000000000","tz":60},"committer":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-10-24 10:13:50.000000000","tz":60},"subject":"Account for encoding in _get_direct_url","message":"Account for encoding in _get_direct_url\n\nModifies the method to strip out url-encoding of the file name as well as a better way to gather the path that avoids double slashes\n\nCloses-Bug: #2129687\nChange-Id: Ia2d2c824c64fce4779d4e38eec155dc020a4ad7e\nSigned-off-by: Ian Watson \u003cianwatson92@hotmail.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95c842023844d800b6b03884f8f449673683563a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95c842023844d800b6b03884f8f449673683563a"}]},"branch":"refs/heads/master"},"5eb4cbc6ffd4324501a82325e63c3b35aafe403d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-10-27 16:08:44.000000000","uploader":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"ref":"refs/changes/93/964693/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/964693/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/964693/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/964693/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/964693/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/964693/3"}}},"commit":{"parents":[{"commit":"8882c873055ac74c0916d6d6acc58b6823c496a7","subject":"CI: Replace deprecated yaml callback plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8882c873055ac74c0916d6d6acc58b6823c496a7"}]}],"author":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-10-23 13:47:25.000000000","tz":60},"committer":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-10-27 16:08:03.000000000","tz":0},"subject":"Account for encoding in _get_direct_url","message":"Account for encoding in _get_direct_url\n\nModifies the method to strip out url-encoding of the file name as\nwell as a better way to gather the path that avoids double\nslashes\n\nCloses-Bug: #2129687\nChange-Id: Ia2d2c824c64fce4779d4e38eec155dc020a4ad7e\nSigned-off-by: Ian Watson \u003cianwatson92@hotmail.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5eb4cbc6ffd4324501a82325e63c3b35aafe403d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5eb4cbc6ffd4324501a82325e63c3b35aafe403d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
